)]}'
{"id":"openstack%2Fopenstack-ansible-os_mistral~671607","triplet_id":"openstack%2Fopenstack-ansible-os_mistral~master~Ia215313f1d6ccdd6d304ecc2888b081cbaf860c6","project":"openstack/openstack-ansible-os_mistral","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"Ia215313f1d6ccdd6d304ecc2888b081cbaf860c6","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-07-18 22:19:06.000000000","updated":"2019-07-19 13:52:43.000000000","submitted":"2019-07-19 13:52:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"671607-1563544363541-ef2d6f53","meta_rev_id":"e65cf0d0084e580a0b06fe54a973017a4023eac8","_number":671607,"virtual_id_number":671607,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-07-19 13:52:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":2,"date":"2019-07-19 12:09:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":1,"date":"2019-07-19 12:09:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-19 12:09:03.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-07-19 13:52:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1ee92c40c9485ad4268e0c6e5123aaf6a908c0ce","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-07-18 22:19:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"de63734ea00c0e3a3b37efb8ecc10849145e04d4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 23:59:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/07/671607/1/check/openstack-ansible-linters/1e6a59b/ : SUCCESS in 6m 22s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/07/671607/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/1a67d99/ : SUCCESS in 1h 05m 03s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/07/671607/1/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/4f92eea/ : FAILURE in 56m 29s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/07/671607/1/check/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/06ee91b/ : SUCCESS in 1h 09m 22s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/07/671607/1/check/openstack-ansible-deploy-aio_metal-debian-stable/a58b121/ : SUCCESS in 1h 07m 47s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/07/671607/1/check/openstack-ansible-deploy-aio_metal-centos-7/52155ec/ : SUCCESS in 1h 32m 07s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/07/671607/1/check/openstack-ansible-deploy-aio_metal-opensuse-15/bfc5944/ : SUCCESS in 1h 13m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/07/671607/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/2c14259/ : SUCCESS in 1h 15m 04s\n- openstack-tox-docs http://logs.openstack.org/07/671607/1/check/openstack-tox-docs/355c8ad/html/ : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":1},{"id":"78fca3f5be1bf9ff38e34b077920ab51c968da56","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-19 12:09:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bb64c3bc1a7fa1d84a7cd41bc8d80a8801fc66a5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 12:09:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"46a1c3248aaa4c7c91d9fa37d7a2c894a5a85bae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 13:52:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"44843d409db73e32a9e66166bb1e64903549fc0e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 13:52:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/07/671607/1/gate/openstack-ansible-linters/8267448/ : SUCCESS in 7m 03s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/07/671607/1/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/f955b41/ : SUCCESS in 1h 03m 34s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/07/671607/1/gate/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/72a191b/ : SUCCESS in 56m 14s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/07/671607/1/gate/openstack-ansible-deploy-aio_metal-debian-stable/adb638e/ : SUCCESS in 1h 13m 14s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/07/671607/1/gate/openstack-ansible-deploy-aio_metal-centos-7/d816bc7/ : SUCCESS in 1h 28m 32s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/07/671607/1/gate/openstack-ansible-deploy-aio_metal-opensuse-15/995d687/ : SUCCESS in 1h 28m 49s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/07/671607/1/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/e702854/ : SUCCESS in 1h 02m 30s\n- openstack-tox-docs http://logs.openstack.org/07/671607/1/gate/openstack-tox-docs/ac27d7a/html/ : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2c3029e5b3dfbd83b54eae62b0af032c52b225d3","revisions":{"2c3029e5b3dfbd83b54eae62b0af032c52b225d3":{"kind":"REWORK","_number":1,"created":"2019-07-18 22:19:06.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/07/671607/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_mistral","ref":"refs/changes/07/671607/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/07/671607/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/07/671607/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/07/671607/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/07/671607/1"}}},"commit":{"parents":[{"commit":"6f31a874c814890d9ebb5caefe8d404ef02f2438","subject":"Merge \"Use systemd-journald instead of log files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/6f31a874c814890d9ebb5caefe8d404ef02f2438"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-18 22:19:04.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-18 22:19:04.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: Ia215313f1d6ccdd6d304ecc2888b081cbaf860c6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/2c3029e5b3dfbd83b54eae62b0af032c52b225d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/2c3029e5b3dfbd83b54eae62b0af032c52b225d3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
