)]}'
{"id":"openstack%2Fopenstack-ansible-os_mistral~671730","triplet_id":"openstack%2Fopenstack-ansible-os_mistral~stable%2Fstein~I5c82e1f570377522f706b5640aaa3fb7d9c879a3","project":"openstack/openstack-ansible-os_mistral","branch":"stable/stein","topic":"static_imports-stable/stein","hashtags":[],"change_id":"I5c82e1f570377522f706b5640aaa3fb7d9c879a3","subject":"Convert dynamic includes to static imports","status":"MERGED","created":"2019-07-19 13:52:50.000000000","updated":"2019-07-19 18:18:39.000000000","submitted":"2019-07-19 18:18:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"671730-1563560319860-02c3ebfc","meta_rev_id":"6e7270a9a665eaba470aacdfed2d1b9c9864aa2b","_number":671730,"virtual_id_number":671730,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-07-19 18:18:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":2,"date":"2019-07-19 16:09:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-19 17:02:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-19 17:03:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-19 16:09:58.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-07-19 17:03:35.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-07-19 18:18:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dccb1fe8776eb6e6412648ed870678f5cbe05f61","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-19 13:52:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d3302d46e9058c3960a8eeb90e77febe902a3716","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-19 13:54:31.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"99788ab0301cf7aed0f6e97b0c60d4824f98d215","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-19 13:56:06.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"087681f1b6ea88b566884f4163ded6f99a54290f","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-19 16:09:58.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"887768d339978f794fee4f370a70f54e911afd73","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 16:46:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/30/671730/3/check/openstack-ansible-linters/1540d8d/ : SUCCESS in 8m 35s\n- openstack-tox-docs http://logs.openstack.org/30/671730/3/check/openstack-tox-docs/98f8df1/html/ : SUCCESS in 5m 29s\n- openstack-ansible-deploy-aio_metal_mistral-centos-7 http://logs.openstack.org/30/671730/3/check/openstack-ansible-deploy-aio_metal_mistral-centos-7/ba9f8be/ : SUCCESS in 1h 30m 00s\n- openstack-ansible-deploy-aio_metal_mistral-debian-stable http://logs.openstack.org/30/671730/3/check/openstack-ansible-deploy-aio_metal_mistral-debian-stable/3e6187f/ : SUCCESS in 59m 32s\n- openstack-ansible-deploy-aio_metal_mistral-ubuntu-bionic http://logs.openstack.org/30/671730/3/check/openstack-ansible-deploy-aio_metal_mistral-ubuntu-bionic/e646b55/ : SUCCESS in 55m 25s\n- openstack-ansible-deploy-aio_distro_metal_mistral-centos-7 http://logs.openstack.org/30/671730/3/check/openstack-ansible-deploy-aio_distro_metal_mistral-centos-7/939bd65/ : SUCCESS in 1h 27m 03s\n- openstack-ansible-deploy-aio_distro_metal_mistral-ubuntu-bionic http://logs.openstack.org/30/671730/3/check/openstack-ansible-deploy-aio_distro_metal_mistral-ubuntu-bionic/724cbdb/ : SUCCESS in 1h 09m 15s","accounts_in_message":[],"_revision_number":3},{"id":"60c923b792351aa15061fd8429d9a1132518e914","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-19 17:02:46.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"0761c0d7ab334de3051ae67e62b30782a720f1ae","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-19 17:03:35.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"60a6bc55f1650bb55958b97343191a0fa3b47946","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 17:03:46.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9c976dcddd590fa4df328585064b76e3844ac66d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 18:18:39.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/30/671730/3/gate/openstack-ansible-linters/fa28999/ : SUCCESS in 6m 28s\n- openstack-tox-docs http://logs.openstack.org/30/671730/3/gate/openstack-tox-docs/0e9b52d/html/ : SUCCESS in 3m 24s\n- openstack-ansible-deploy-aio_metal_mistral-centos-7 http://logs.openstack.org/30/671730/3/gate/openstack-ansible-deploy-aio_metal_mistral-centos-7/26582c5/ : SUCCESS in 1h 08m 03s\n- openstack-ansible-deploy-aio_metal_mistral-debian-stable http://logs.openstack.org/30/671730/3/gate/openstack-ansible-deploy-aio_metal_mistral-debian-stable/96e0958/ : SUCCESS in 56m 07s\n- openstack-ansible-deploy-aio_metal_mistral-ubuntu-bionic http://logs.openstack.org/30/671730/3/gate/openstack-ansible-deploy-aio_metal_mistral-ubuntu-bionic/01a58e7/ : SUCCESS in 55m 21s\n- openstack-ansible-deploy-aio_distro_metal_mistral-centos-7 http://logs.openstack.org/30/671730/3/gate/openstack-ansible-deploy-aio_distro_metal_mistral-centos-7/f6f5558/ : SUCCESS in 1h 10m 07s\n- openstack-ansible-deploy-aio_distro_metal_mistral-ubuntu-bionic http://logs.openstack.org/30/671730/3/gate/openstack-ansible-deploy-aio_distro_metal_mistral-ubuntu-bionic/3cbdf0f/ : SUCCESS in 55m 00s","accounts_in_message":[],"_revision_number":3},{"id":"be440fbbf55b68b197f4c7d90f0b4743211168d6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 18:18:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"565df6e3941330da908373e9c1c757a9ff7565ba","revisions":{"7894830a4b45daaaa997b32ec4c92882dbe9e2dc":{"kind":"REWORK","_number":1,"created":"2019-07-19 13:52:50.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/30/671730/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_mistral","ref":"refs/changes/30/671730/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/30/671730/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/30/671730/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/30/671730/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/30/671730/1"}}},"commit":{"parents":[{"commit":"124004fb847680fa8de4b2b91ac824d9092e25ed","subject":"Merge \"Fix distro installs on Ubuntu\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/124004fb847680fa8de4b2b91ac824d9092e25ed"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-07-17 15:50:05.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-07-19 13:52:38.000000000","tz":180},"subject":"Convert dynamic includes to static imports","message":"Convert dynamic includes to static imports\n\nWhen task/role files are included using include_tasks, tags are not\npassed to the included tasks. As a result, tags like neutron-config\ndo not have the intended effect. This patch changes include_tasks\nto import_tasks for all cases where dynamic vars or loops are not used\nso that tags are properly handled.\n\nReference -\nhttps://docs.ansible.com/ansible/latest/user_guide/playbooks_reuse.html\nhttps://bugs.launchpad.net/openstack-ansible/+bug/1815043\n\nChange-Id: I5c82e1f570377522f706b5640aaa3fb7d9c879a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/7894830a4b45daaaa997b32ec4c92882dbe9e2dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/7894830a4b45daaaa997b32ec4c92882dbe9e2dc"}]},"branch":"refs/heads/stable/stein"},"0aa385089b43d63145e915487d2339d0ea32c9d5":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-07-19 13:54:31.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/30/671730/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_mistral","ref":"refs/changes/30/671730/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/30/671730/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/30/671730/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/30/671730/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/30/671730/2"}}},"commit":{"parents":[{"commit":"124004fb847680fa8de4b2b91ac824d9092e25ed","subject":"Merge \"Fix distro installs on Ubuntu\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/124004fb847680fa8de4b2b91ac824d9092e25ed"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-07-17 15:50:05.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2019-07-19 13:54:25.000000000","tz":0},"subject":"Convert dynamic includes to static imports","message":"Convert dynamic includes to static imports\n\nWhen task/role files are included using include_tasks, tags are not\npassed to the included tasks. As a result, tags like neutron-config\ndo not have the intended effect. This patch changes include_tasks\nto import_tasks for all cases where dynamic vars or loops are not used\nso that tags are properly handled.\n\nReference -\nhttps://docs.ansible.com/ansible/latest/user_guide/playbooks_reuse.html\nhttps://bugs.launchpad.net/openstack-ansible/+bug/1815043\n\nChange-Id: I7e529e9b4d23ffe8658c4dcdf4e98989d0f84f3f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/0aa385089b43d63145e915487d2339d0ea32c9d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/0aa385089b43d63145e915487d2339d0ea32c9d5"}]},"branch":"refs/heads/stable/stein"},"565df6e3941330da908373e9c1c757a9ff7565ba":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-07-19 13:56:06.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/30/671730/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_mistral","ref":"refs/changes/30/671730/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/30/671730/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/30/671730/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/30/671730/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/30/671730/3"}}},"commit":{"parents":[{"commit":"124004fb847680fa8de4b2b91ac824d9092e25ed","subject":"Merge \"Fix distro installs on Ubuntu\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/124004fb847680fa8de4b2b91ac824d9092e25ed"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-07-17 15:50:05.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2019-07-19 13:56:03.000000000","tz":0},"subject":"Convert dynamic includes to static imports","message":"Convert dynamic includes to static imports\n\nWhen task/role files are included using include_tasks, tags are not\npassed to the included tasks. As a result, tags like neutron-config\ndo not have the intended effect. This patch changes include_tasks\nto import_tasks for all cases where dynamic vars or loops are not used\nso that tags are properly handled.\n\nReference -\nhttps://docs.ansible.com/ansible/latest/user_guide/playbooks_reuse.html\nhttps://bugs.launchpad.net/openstack-ansible/+bug/1815043\n\nChange-Id: I5c82e1f570377522f706b5640aaa3fb7d9c879a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/565df6e3941330da908373e9c1c757a9ff7565ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/565df6e3941330da908373e9c1c757a9ff7565ba"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
