)]}'
{"id":"openstack%2Fopenstack-ansible-os_mistral~691983","triplet_id":"openstack%2Fopenstack-ansible-os_mistral~stable%2Fstein~Iab50f8cc6a47abe61e18964dac43e4f4e8ce0753","project":"openstack/openstack-ansible-os_mistral","branch":"stable/stein","topic":"use-opensuse-15","hashtags":[],"change_id":"Iab50f8cc6a47abe61e18964dac43e4f4e8ce0753","subject":"Switch to opensuse-15","status":"MERGED","created":"2019-10-29 21:00:55.000000000","updated":"2019-10-31 03:56:45.000000000","submitted":"2019-10-31 03:51:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"691983-1572493891451-edb63827","meta_rev_id":"199be715d25641385267341063b10f58df645a9b","_number":691983,"virtual_id_number":691983,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-31 03:51:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"all":[{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":2,"date":"2019-10-31 02:31:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-30 21:50:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"all":[{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":1,"date":"2019-10-31 02:31:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-29 21:58:11.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-10-30 21:50:50.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-10-31 02:31:58.000000000","updated_by":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"reviewer":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"state":"REVIEWER"},{"updated":"2019-10-31 03:51:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6a7bd0a6c0ebcc82eb15141a576d5f2883c524c8","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-29 21:00:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8c86b844326ca815d538acc363779b68ff8c5738","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-10-29 21:58:11.000000000","message":"Patch Set 1: Code-Review-1\n\nwhy don\u0027t we switch the nodeset to opensuse-15 instead of removing the  job alltogether?","accounts_in_message":[],"_revision_number":1},{"id":"16b36b9cb148a6830474abcbe1e2543c8815d97f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 22:40:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/6ed6cae2c91a4f0dba6b5604ffcc4e62 : SUCCESS in 6m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f03c4912265f41a799ad9b2a855d1ac5 : SUCCESS in 3m 16s\n- openstack-ansible-deploy-aio_metal_mistral-centos-7 https://zuul.opendev.org/t/openstack/build/9a53cfa5b6df486487bc1cf10884a007 : SUCCESS in 1h 10m 13s\n- openstack-ansible-deploy-aio_metal_mistral-debian-stable https://zuul.opendev.org/t/openstack/build/f849be23197f428fb857e2bf65e3bb98 : SUCCESS in 1h 03m 17s\n- openstack-ansible-deploy-aio_metal_mistral-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f333172692b749e88cc8f3608575f44e : SUCCESS in 58m 41s\n- openstack-ansible-deploy-aio_distro_metal_mistral-centos-7 https://zuul.opendev.org/t/openstack/build/c7037deb9b3946b3b095334f84e44ba2 : SUCCESS in 57m 58s\n- openstack-ansible-deploy-aio_distro_metal_mistral-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9c14fc59e6b44623ace71f3bbfe4e3b7 : SUCCESS in 44m 59s","accounts_in_message":[],"_revision_number":1},{"id":"441e5f2c3cc0dcdb3ab7a0ddc5c28e74ad30c755","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-30 07:19:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8414f54fd9532ccb05fb1d44f5987f6776a5fa27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 08:54:55.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d3d7c7ead8e44191a0c3e534e5979f7d : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0101878e844b4a868bf40e7f46364347 : SUCCESS in 3m 18s\n- openstack-ansible-deploy-aio_metal_mistral-centos-7 https://zuul.opendev.org/t/openstack/build/20e7ace1b9344a66b53d753ae32eb83d : FAILURE in 1h 24m 30s\n- openstack-ansible-deploy-aio_metal_mistral-debian-stable https://zuul.opendev.org/t/openstack/build/0e6fd1ca3d384a3cbcfb2a21f1cd5d24 : FAILURE in 1h 07m 54s\n- openstack-ansible-deploy-aio_metal_mistral-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b3e6c8a10b5c4a5bbe4f9423f5cce935 : FAILURE in 1h 18m 26s\n- openstack-ansible-deploy-aio_distro_metal_mistral-centos-7 https://zuul.opendev.org/t/openstack/build/6f9e2c747c8e45b4aa0c97769ad73a4a : FAILURE in 1h 01m 45s\n- openstack-ansible-deploy-aio_distro_metal_mistral-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/879f75e8bd3f468fb3fb7647424bbb83 : SUCCESS in 51m 59s","accounts_in_message":[],"_revision_number":2},{"id":"98802ac1a13b4468b534bf419fee8920924f9d92","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-30 18:35:43.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"309765d4481beb340a3dd5f307f139172d5d285a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 19:47:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/41bc51744da34c84855b13193837ff20 : SUCCESS in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ca50c395b294c769a8389bf764b8b81 : SUCCESS in 4m 19s\n- openstack-ansible-deploy-aio_metal_mistral-centos-7 https://zuul.opendev.org/t/openstack/build/f57668d448f8428fbd6f2e5a75c396db : SUCCESS in 1h 09m 16s\n- openstack-ansible-deploy-aio_metal_mistral-debian-stable https://zuul.opendev.org/t/openstack/build/2ad01709a40746259a59e170fd965fb5 : SUCCESS in 1h 02m 04s\n- openstack-ansible-deploy-aio_metal_mistral-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b2c0c37cf4f54decb6b68b03b5381513 : SUCCESS in 59m 29s\n- openstack-ansible-deploy-aio_distro_metal_mistral-centos-7 https://zuul.opendev.org/t/openstack/build/02893d6de83246ca8c89b13a8821c440 : SUCCESS in 1h 00m 04s\n- openstack-ansible-deploy-aio_distro_metal_mistral-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e507e77a3bed4607b80792506a0411df : SUCCESS in 44m 19s","accounts_in_message":[],"_revision_number":3},{"id":"54bf3b059728228433ac4a1fb07b398584c175da","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-10-30 21:50:50.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"eff721a3c64026929c74616cdc91e88e150ac0e7","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2019-10-31 02:31:58.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"4d55e36bc932a75903d031e7fe9e0383bdc53218","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-31 02:32:12.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"638dfc10de24b5601f416cc71dc348d3e9091ade","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-31 03:51:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"a1749f540fde49ca152cfbe695e3b64fd78ee17c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-31 03:51:31.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d0b3637a19bb42a18a784a2de9a9512e : SUCCESS in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0184cadc03c44d62b722772c9736ef47 : SUCCESS in 3m 34s\n- openstack-ansible-deploy-aio_metal_mistral-centos-7 https://zuul.opendev.org/t/openstack/build/f1f646960dfe4d468d558db2c99a7661 : SUCCESS in 1h 18m 26s\n- openstack-ansible-deploy-aio_metal_mistral-debian-stable https://zuul.opendev.org/t/openstack/build/5d974034295c4801ac5318e241714699 : SUCCESS in 1h 08m 08s\n- openstack-ansible-deploy-aio_metal_mistral-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9f1d7eea47524d0d918fae9b739fef44 : SUCCESS in 1h 00m 20s\n- openstack-ansible-deploy-aio_distro_metal_mistral-centos-7 https://zuul.opendev.org/t/openstack/build/77f13d7adc9f4d58a92412013c7c819e : SUCCESS in 53m 26s\n- openstack-ansible-deploy-aio_distro_metal_mistral-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1d1d510f0a5942f2be249cd7b3d84add : SUCCESS in 44m 40s","accounts_in_message":[],"_revision_number":3},{"id":"ea79e5d8e10b7c0a6edd70825c8ad57f9b4acac4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-31 03:56:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e21ce91daf34f15b2f568993727c056 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b774f0e62f17d0337cc401ad69664bccc48a8813","revisions":{"4f9cc719dda4662246cc8b2f5cb3c704afb76df2":{"kind":"REWORK","_number":1,"created":"2019-10-29 21:00:55.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/83/691983/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_mistral","ref":"refs/changes/83/691983/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/83/691983/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/83/691983/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/83/691983/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/83/691983/1"}}},"commit":{"parents":[{"commit":"565df6e3941330da908373e9c1c757a9ff7565ba","subject":"Convert dynamic includes to static imports","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/565df6e3941330da908373e9c1c757a9ff7565ba"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-10-29 21:00:27.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-10-29 21:00:51.000000000","tz":60},"subject":"Remove opensuse-150 job","message":"Remove opensuse-150 job\n\nRemove the opensuse-150 job, it is not run anymore on master and was\nremoved there (together with others) since it did not test what it\nshould test.\nWe\u0027re retiring opensuse-150 in OpenStack Infra, so remove the job\ncompletly on this branch as well.\n\nThis is a partial backport of change https://review.opendev.org/658448\n\nChange-Id: Iab50f8cc6a47abe61e18964dac43e4f4e8ce0753\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/4f9cc719dda4662246cc8b2f5cb3c704afb76df2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/4f9cc719dda4662246cc8b2f5cb3c704afb76df2"}]},"branch":"refs/heads/stable/stein"},"818ade1643f1b2380e6a2b8af29ddc697f1dc838":{"kind":"REWORK","_number":2,"created":"2019-10-30 07:19:22.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/83/691983/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_mistral","ref":"refs/changes/83/691983/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/83/691983/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/83/691983/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/83/691983/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/83/691983/2"}}},"commit":{"parents":[{"commit":"565df6e3941330da908373e9c1c757a9ff7565ba","subject":"Convert dynamic includes to static imports","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/565df6e3941330da908373e9c1c757a9ff7565ba"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-10-29 21:00:27.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-10-30 07:19:06.000000000","tz":60},"subject":"Switch to opensuse-15","message":"Switch to opensuse-15\n\nopensuse-150 nodeset is referring to openSUSE 15.0, which is still in\nmaintenance but openSUSE 15.1 has been released already. \"opensuse-15\"\nis going to refer to the \"latest openSUSE 15.x\" build released and\nworking for OpenStack going forward, so add this nodeset and use\nit by default going forward.\n\nNote this is not a backport, master has changed the job setup completly,\ninstead it just updates the jobs.\n\nChange-Id: Iab50f8cc6a47abe61e18964dac43e4f4e8ce0753\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/818ade1643f1b2380e6a2b8af29ddc697f1dc838"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/818ade1643f1b2380e6a2b8af29ddc697f1dc838"}]},"branch":"refs/heads/stable/stein"},"b774f0e62f17d0337cc401ad69664bccc48a8813":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-10-30 18:35:43.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/83/691983/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_mistral","ref":"refs/changes/83/691983/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/83/691983/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/83/691983/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/83/691983/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/83/691983/3"}}},"commit":{"parents":[{"commit":"565df6e3941330da908373e9c1c757a9ff7565ba","subject":"Convert dynamic includes to static imports","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/565df6e3941330da908373e9c1c757a9ff7565ba"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-10-29 21:00:27.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","date":"2019-10-30 18:35:35.000000000","tz":0},"subject":"Switch to opensuse-15","message":"Switch to opensuse-15\n\nopensuse-150 nodeset is referring to openSUSE 15.0, which is still in\nmaintenance but openSUSE 15.1 has been released already. \"opensuse-15\"\nis going to refer to the \"latest openSUSE 15.x\" build released and\nworking for OpenStack going forward, so add this nodeset and use\nit by default going forward.\n\nNote this is not a backport, master has changed the job setup completly,\ninstead it just updates the jobs.\n\nDepends-On: https://review.opendev.org/692137\nChange-Id: Iab50f8cc6a47abe61e18964dac43e4f4e8ce0753\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/b774f0e62f17d0337cc401ad69664bccc48a8813"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/b774f0e62f17d0337cc401ad69664bccc48a8813"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
