)]}'
{"id":"openstack%2Fopenstack-ansible-os_mistral~694320","triplet_id":"openstack%2Fopenstack-ansible-os_mistral~master~I37cc2ee6c2a10c5ac26fd0fe845a1006395423e0","project":"openstack/openstack-ansible-os_mistral","branch":"master","topic":"osa/opendev","hashtags":[],"change_id":"I37cc2ee6c2a10c5ac26fd0fe845a1006395423e0","subject":"Replace git.openstack.org with opendev.org","status":"MERGED","created":"2019-11-14 15:00:10.000000000","updated":"2019-11-19 03:51:19.000000000","submitted":"2019-11-19 03:50:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"694320-1574135417524-c4d0231e","meta_rev_id":"997d357b073144bbeb15e69db4fc1f3ed03d352b","_number":694320,"virtual_id_number":694320,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-11-16 22:11:03.000000000","_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-19 03:50:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-11-18 16:47:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-11-18 12:59:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":1,"date":"2019-11-18 16:47:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-18 12:59:33.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-11-18 16:47:39.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-11-19 03:50:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"92379296d96938d0b54783467845e259c7f1b049","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-11-14 15:00:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a911b15f80c3163f50f1b76eb6682aa1371368d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 18:11:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/086c0372583d43c286592eba52897237 : SUCCESS in 5m 47s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/92e72d5dfcbb45a58c08c220808c47fb : SUCCESS in 1h 31m 43s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/8396a82927ae40b296dc7f44b617cafc : FAILURE in 19m 45s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/737652c3f1a148b59657ad3b8185f53c : SUCCESS in 56m 59s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/401f09d967ab4bdb8bfb0ada474b5de0 : SUCCESS in 1h 19m 43s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a3b309f7fcca4033a7e51600bcef2a85 : SUCCESS in 1h 52m 10s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/8bf12781d12e45f2b08c9f1d6a18c12d : FAILURE in 24m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ad8f5783f4184a119d91cdf94148c237 : SUCCESS in 1h 20m 34s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b36c6dee15394506ac20cf6a458373ea : TIMED_OUT in 3h 04m 28s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f06633b1e3ff42b0b817993a92f4ac73 : SUCCESS in 1h 54m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b199d823403646ed9c4c54bbb7d44534 : SUCCESS in 5m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/54543c977ae44511b1a646d30d92b8ba : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":1},{"id":"85407fb29b52dc80c4b50272a3f3153ced2f29a4","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-11-16 22:11:03.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a90bd07615d21d474abcbabd17d61381d5a2ede7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-17 01:15:16.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/761ebcfb4aa144e7bc21b1aae98270f4 : SUCCESS in 4m 44s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/1742dc15c27c4d8eb8f3bad633096698 : SUCCESS in 1h 03m 16s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/96e6d9aee896459390770e5184f5e45a : SUCCESS in 1h 04m 13s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/944c448945d941268b8f86351f1016f9 : SUCCESS in 58m 30s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/b5e9a21aeac94408aa28aecd043ae02e : SUCCESS in 2h 02m 37s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f10de98eafdc457ba36e60d0024dec16 : SUCCESS in 1h 33m 48s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/85fb57c12f5340969057e7b2a0b2afe9 : SUCCESS in 1h 24m 49s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a59df593abb2435ba30a248baa2d8259 : SUCCESS in 1h 09m 16s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/58fbed8880d147b1b1eda1973d3ef786 : TIMED_OUT in 3h 02m 36s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1828446456d14d25a2193028ff6889bc : SUCCESS in 1h 49m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12e1769a64c44b739ee23e43c53c89f5 : SUCCESS in 5m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e13db355dfe4b5299a9e336bfaf8c3d : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":1},{"id":"9c931b258e01082b646b091fb73ea9f756c9314b","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-11-18 09:55:01.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9a099e8259ff60e0505e43df0014befc73d6cb94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 12:58:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/814c000c95314e60b3b7798b7469512f : SUCCESS in 5m 50s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/491a5c09a2454c9486bfcd0f360f109f : SUCCESS in 1h 10m 37s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/83585ba8e63a4aba8680045386a30a17 : SUCCESS in 1h 05m 04s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a94fb8f0a65843e6b8c05f2b6baa25dd : SUCCESS in 55m 10s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/c5dffe49133f44a8b1aa10fc34edf87f : SUCCESS in 1h 35m 46s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/244a626252e54e4aa3d06d6fdcb8c989 : SUCCESS in 1h 38m 29s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/29f554168ff84f2a91eab10abf294ef9 : SUCCESS in 1h 30m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/993dfffe553c4deea8b784add80f0b4b : SUCCESS in 1h 16m 24s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d33d0f7c1fe5401d857a18706ba4c9d8 : SUCCESS in 3h 00m 57s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b63f52385c994b739c22cdb52eb0497c : SUCCESS in 2h 03m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab6a9682dcba43bd86d85b4b96928a4c : SUCCESS in 6m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/54024e1086a44aa6b1fe8795c73ee220 : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":1},{"id":"662b49738ba9bc7c3bb1f300614061b8e3b71b7f","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-11-18 12:59:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"03465541a79b5941c5908c08fdd25ab610f7acab","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-11-18 16:47:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ba9a8e7a96f86a8aef68df8b1694401668c49661","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 16:47:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fcaca4f87af104fb7821220de48170073d877912","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 19:56:56.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/28a3683a6d02437d84d4098d62f41ee5 : SUCCESS in 5m 53s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/37d588a8aab04a318200fbd1254ca589 : SUCCESS in 1h 26m 36s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/93f0ee277338447f8364b4fc3de03447 : SUCCESS in 1h 00m 19s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/38f5ac2eac5645339ebf17a3286ca97c : SUCCESS in 1h 42m 24s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8e30dc4742ba4f9581ee546e2cba7a10 : SUCCESS in 1h 53m 11s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/24c1da02a431471890c2001d537621da : SUCCESS in 1h 43m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8e7142e233a748d880a7be95d7828681 : SUCCESS in 1h 20m 44s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ba301082884a440f9cd9f9537a3eac61 : TIMED_OUT in 3h 03m 02s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/78234c6ae7444b64b7c6049b64df0a57 : SUCCESS in 1h 58m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3db45bc1fa05429bbb312cab7cad726f : SUCCESS in 5m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/629001ebba104af4ba0de6196ac903c3 : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":1},{"id":"c1dc1c8c6d409c884e4972180d69865cf10a32d2","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-11-18 22:02:06.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"39f131044e0b6ca57808972f40a96165506f4591","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 01:02:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/73837a70dfe54199ac4ebb8edd4c56ab : SUCCESS in 5m 26s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b768bfa7faa647e2911984ad69c7be6d : SUCCESS in 1h 36m 19s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/4435f061f6df4bb3a66cf2fd80583471 : FAILURE in 1h 03m 40s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/52d2d5b888904e7dbd9ffa3cc9ffe481 : SUCCESS in 1h 07m 06s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/2e901c5e27d34cd692a233eaca076bac : SUCCESS in 1h 26m 04s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/62d42191380345b6a854ede54aa384ff : SUCCESS in 1h 47m 21s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/a67afc3a05a4431986ff6c6170bf77e6 : SUCCESS in 1h 26m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/420b0007cc6145469b38549c6ff7c062 : SUCCESS in 1h 20m 01s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/9e962efbb1ad42a6be27863b0f613214 : SUCCESS in 2h 57m 00s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/be1bf0644ae44097b95d26aa09a00f84 : SUCCESS in 2h 01m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc963ad2f65746eab2ba91dec8601186 : SUCCESS in 4m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0edfb12ab6b64030826efd66564852b5 : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":1},{"id":"4033cec36654d9a5d98611f8b20ab874866fe7b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 01:03:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"736288301e7122728aa7b012454f5388e09357e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 03:50:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d98e731b6feb40318b21b4ea0422c5a8 : SUCCESS in 6m 22s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/29f8bf1dd4e84cc48c135c1f28d11145 : SUCCESS in 1h 41m 48s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5ebea0665bd4438180b241b4a4619612 : SUCCESS in 1h 20m 36s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/339858ac91214c968e111fed7610127c : SUCCESS in 1h 40m 18s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0a5cbc0904be44b0b26128cda3d4ef07 : SUCCESS in 1h 50m 30s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/2d0ba42b605c49e9b4e9c4a849eadebd : SUCCESS in 1h 35m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b6a800300ee144aca27e193d07035d2c : SUCCESS in 1h 21m 29s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/4979054768a542a4a66b3fbca3b2eacd : SUCCESS in 2h 46m 21s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f2f109f4a1ef40829fb474cc0bb9d5bf : SUCCESS in 2h 00m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/644d03a2b8144dac94b7599f72b52958 : SUCCESS in 4m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e6191306df704ccb91bd7bd0dff06344 : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":1},{"id":"4e8024720081e237c73639172c7a6025ce6c9074","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 03:50:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"44da44c61ff62f7516e7e2e917c95ef779b2eba8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 03:51:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92b22322985e47fdac0d3dd879fc9519 : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8e1d5ba1d99d4ec3939f1c24cda7568a : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"47330c95933b55448f47601e558256b47a1d7e17","revisions":{"47330c95933b55448f47601e558256b47a1d7e17":{"kind":"REWORK","_number":1,"created":"2019-11-14 15:00:10.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/20/694320/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_mistral","ref":"refs/changes/20/694320/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/20/694320/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/20/694320/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/20/694320/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/20/694320/1"}}},"commit":{"parents":[{"commit":"61bde761c76bdbf84028cedf55bad91f1e3335be","subject":"Update master for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/61bde761c76bdbf84028cedf55bad91f1e3335be"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-11-14 15:00:00.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-11-14 15:00:00.000000000","tz":120},"subject":"Replace git.openstack.org with opendev.org","message":"Replace git.openstack.org with opendev.org\n\nThis patch replaces git.openstack.org with opendev.org as redirection\nfrom old path was enabled.\nAlso we change upper constraints url due to [1]\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-May/006478.html\n\nChange-Id: I37cc2ee6c2a10c5ac26fd0fe845a1006395423e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/47330c95933b55448f47601e558256b47a1d7e17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/47330c95933b55448f47601e558256b47a1d7e17"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
