)]}'
{"id":"openstack%2Fopenstack-ansible-os_mistral~696981","triplet_id":"openstack%2Fopenstack-ansible-os_mistral~stable%2Fstein~Iba8ae5e8ded1398649792ecc5b4d9267551713c4","project":"openstack/openstack-ansible-os_mistral","branch":"stable/stein","topic":"clarify-suse-support-stable/stein","hashtags":[],"change_id":"Iba8ae5e8ded1398649792ecc5b4d9267551713c4","subject":"Add meta to mistral","status":"ABANDONED","created":"2019-12-02 22:12:48.000000000","updated":"2021-08-11 16:10:15.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f59ffc4078b8f83b74b8155a73b744014063f524","_number":696981,"virtual_id_number":696981,"owner":{"_account_id":16197,"name":"Tom Jose Kalapura","email":"tomjosekal@gmail.com","username":"tomjosekal"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-05-21 14:47:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":2,"date":"2020-05-21 16:40:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-02 22:12:48.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2020-05-21 14:47:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-21 16:40:30.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"7ae5b0ac39417e91800f8a135d4c5b26e1c3e138","author":{"_account_id":16197,"name":"Tom Jose Kalapura","email":"tomjosekal@gmail.com","username":"tomjosekal"},"date":"2019-12-02 22:12:48.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"1a6fafdae44fa32b871ffb0b5aa762466073f825","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 23:30:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/172c9160149d44f6abe8b633309df530 : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c4db6e01da948f5a6dd0bb062e818d3 : SUCCESS in 2m 53s\n- openstack-ansible-deploy-aio_metal_mistral-centos-7 https://zuul.opendev.org/t/openstack/build/6f89d97d245348c69a48fe831ea76c87 : SUCCESS in 1h 16m 30s\n- openstack-ansible-deploy-aio_metal_mistral-debian-stable https://zuul.opendev.org/t/openstack/build/919aa2fb44324a039072cc72d56dfe34 : SUCCESS in 1h 01m 31s\n- openstack-ansible-deploy-aio_metal_mistral-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9a268faef0cf43f2b4336b9c8f7f3166 : SUCCESS in 57m 50s\n- openstack-ansible-deploy-aio_distro_metal_mistral-centos-7 https://zuul.opendev.org/t/openstack/build/54e76b9184d44d489abc98e1e6217285 : SUCCESS in 55m 28s\n- openstack-ansible-deploy-aio_distro_metal_mistral-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d3b6c86990144ed9a8721d0eab4bc4ff : SUCCESS in 45m 27s","accounts_in_message":[],"_revision_number":1},{"id":"5930512cd601c00b4db3e949e85c86f7d40fc5f2","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-12-04 08:24:28.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"268d3c88922c598e8b65e527c36ef9da24d541e4","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-21 13:16:43.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"75216a2f47809f20dac273ce74330b52202e927d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 14:47:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/fcfb7bbba6d746fa9d964255c262892b : SUCCESS in 8m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7529261a70e4f179378dfe67c185dbc : SUCCESS in 6m 49s\n- openstack-ansible-deploy-aio_metal_mistral-centos-7 https://zuul.opendev.org/t/openstack/build/86090fab7f0846da890741d193765b88 : SUCCESS in 1h 29m 14s\n- openstack-ansible-deploy-aio_metal_mistral-debian-stable https://zuul.opendev.org/t/openstack/build/3216a5568ea44213992c32a973d13d5b : SUCCESS in 1h 12m 00s\n- openstack-ansible-deploy-aio_metal_mistral-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/36fcc7053bd64b14bfcad9d818c0dd69 : SUCCESS in 1h 07m 19s\n- openstack-ansible-deploy-aio_distro_metal_mistral-centos-7 https://zuul.opendev.org/t/openstack/build/ce1134ac146c4f7482b936017810c079 : SUCCESS in 1h 07m 52s\n- openstack-ansible-deploy-aio_distro_metal_mistral-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/88e5dac293b34207908b78c20df35eb9 : SUCCESS in 44m 33s","accounts_in_message":[],"_revision_number":2},{"id":"76ec1488f047ae88d7b827f33746d80d1d018849","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-21 16:40:30.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f59ffc4078b8f83b74b8155a73b744014063f524","tag":"autogenerated:gerrit:abandon","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-08-11 16:10:15.000000000","message":"Abandoned\n\ncleaning up old patches","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"873a1cfeb5625b81af647a37a46d909ffee793d0","revisions":{"8a2a2004bc40d74bacf38e1481a769fdc9853ffd":{"kind":"REWORK","_number":1,"created":"2019-12-02 22:12:48.000000000","uploader":{"_account_id":16197,"name":"Tom Jose Kalapura","email":"tomjosekal@gmail.com","username":"tomjosekal"},"ref":"refs/changes/81/696981/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_mistral","ref":"refs/changes/81/696981/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/81/696981/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/81/696981/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/81/696981/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/81/696981/1"}}},"commit":{"parents":[{"commit":"b774f0e62f17d0337cc401ad69664bccc48a8813","subject":"Switch to opensuse-15","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/b774f0e62f17d0337cc401ad69664bccc48a8813"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-21 07:18:27.000000000","tz":120},"committer":{"name":"Tom Jose Kalapura","email":"tomjosekal@gmail.com","date":"2019-12-02 22:12:48.000000000","tz":0},"subject":"Add meta to mistral","message":"Add meta to mistral\n\nMistral didn\u0027t get the meta/ information that is usually present\non all our roles.\n\nThis fixes it. I added mnaser and his company in meta, as he is\nthe original author of this role.\nAt the opposite of other roles, I didn\u0027t add the apt dependency.\n\nChange-Id: Iba8ae5e8ded1398649792ecc5b4d9267551713c4\n(cherry picked from commit 432a90e4e409c1137246979905b59fc4639a1014)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/8a2a2004bc40d74bacf38e1481a769fdc9853ffd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/8a2a2004bc40d74bacf38e1481a769fdc9853ffd"}]},"branch":"refs/heads/stable/stein"},"873a1cfeb5625b81af647a37a46d909ffee793d0":{"kind":"REWORK","_number":2,"created":"2020-05-21 13:16:43.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/81/696981/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_mistral","ref":"refs/changes/81/696981/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/81/696981/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/81/696981/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/81/696981/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/81/696981/2"}}},"commit":{"parents":[{"commit":"b774f0e62f17d0337cc401ad69664bccc48a8813","subject":"Switch to opensuse-15","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/b774f0e62f17d0337cc401ad69664bccc48a8813"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-21 07:18:27.000000000","tz":120},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-05-21 13:16:32.000000000","tz":0},"subject":"Add meta to mistral","message":"Add meta to mistral\n\nMistral didn\u0027t get the meta/ information that is usually present\non all our roles.\n\nThis fixes it. I added mnaser and his company in meta, as he is\nthe original author of this role.\nAt the opposite of other roles, I didn\u0027t add the apt dependency.\n\nChange-Id: Iba8ae5e8ded1398649792ecc5b4d9267551713c4\n(cherry picked from commit 432a90e4e409c1137246979905b59fc4639a1014)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/873a1cfeb5625b81af647a37a46d909ffee793d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/873a1cfeb5625b81af647a37a46d909ffee793d0"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
