)]}'
{"id":"openstack%2Fopenstack-ansible-os_mistral~709498","triplet_id":"openstack%2Fopenstack-ansible-os_mistral~master~I1b00ccf2a16a5bc97fb70fa75e9e58809299f6cf","project":"openstack/openstack-ansible-os_mistral","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I1b00ccf2a16a5bc97fb70fa75e9e58809299f6cf","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2020-02-24 13:11:58.000000000","updated":"2020-03-07 23:37:17.000000000","submitted":"2020-03-07 23:36:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"709498-1583624170209-c6fdaf72","meta_rev_id":"363fc79a46821c4f5e1b40c8d4eccc3a1ed4b506","_number":709498,"virtual_id_number":709498,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-07 23:36:10.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-07 22:03:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-03-07 22:03:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-07 22:03:20.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-03-07 23:36:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2c4e036a8b4b0142fa727efbeb91f0defec673a3","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-02-24 13:11:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cb8527e311fa995763cbc17684f8a4e14486d6fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 15:16:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/73fac7fb7654470c945c336e42611bab : SUCCESS in 6m 47s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/6ebd4d04700b4c4b875b040da431ceef : SUCCESS in 1h 25m 27s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/877a633c0f9c4debb8d82bb3c683b1eb : FAILURE in 30m 06s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5fa781c35e2540bb9c4dbdbcf63cf561 : SUCCESS in 59m 29s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/ec98dd2fe6654e0487adc5e79a4fd558 : SUCCESS in 1h 24m 08s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/2a68fc4933f64c74ae80019feeeb8f33 : SUCCESS in 1h 47m 03s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/e871513bdfca4db7909c937bb8e3aa36 : SUCCESS in 1h 30m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8cdc02cb621d46638119e0a3901c7adf : SUCCESS in 1h 17m 57s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/71a529f238b54681b45b07c792f37d4a : FAILURE in 1h 25m 06s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2e165d2758cd49529697ff7a3a125ac2 : FAILURE in 1h 23m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfff0771ca0d422e9850e068e164e484 : SUCCESS in 7m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ff6ad77f78df49fca6d8193dc4ac6ef1 : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":1},{"id":"75685e0b47b9c8c6a999dc62ee4687efbaae261d","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-02-25 08:50:00.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c983ee8a65a3733f0b65dcc17bad09374f4272ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 10:45:41.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/bbcc901fc4154276b55fc21e6061a953 : SUCCESS in 9m 31s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/23058ee86c8b4cd489a120907e4d581c : SUCCESS in 1h 18m 46s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/ecf26142856d4e689080a8f1d76977a9 : FAILURE in 32m 22s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0389b64bd1114da08b0c6ddaf96b4ac9 : SUCCESS in 1h 18m 53s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/8e9430a479f14a7198435c32ac6c3c15 : FAILURE in 37m 49s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c66eb8af14b64368989d7580428312f8 : SUCCESS in 1h 43m 16s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/e5beae88d14e470b9b6615277da3fe4e : FAILURE in 1h 39m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5cd54f9eaad24c90b48f79b07a480c75 : SUCCESS in 1h 24m 06s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f5c7e8120e8b4b81bdec703c51b3cca7 : FAILURE in 1h 30m 01s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e7ea80e4943d402d8ce921021eb2b689 : FAILURE in 1h 21m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/819659bf1e7b4876a163693ead51e864 : SUCCESS in 10m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/23caf1ed7ba040f187422b90fe323c6b : SUCCESS in 6m 45s","accounts_in_message":[],"_revision_number":1},{"id":"4716995725932a5a2f6c963e357891d4a6234ac0","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-02-25 21:06:47.000000000","message":"Patch Set 1:\n\nreheck","accounts_in_message":[],"_revision_number":1},{"id":"81563cb475d65ab97fdb5a3c3641ce324c45bb59","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-03-06 17:57:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8509acb11f348cac09435ef510c54cdda68a7669","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 20:37:31.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/e77b97eca3b24eaab3e0a57fdd5c88f3 : SUCCESS in 8m 27s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/78c5d561c663448aaa102bfc6d2b52e4 : SUCCESS in 1h 13m 01s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/f2c06841c7dc43af84a1ddd0bd21f7ac : FAILURE in 35m 16s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/73a34da759df427abc8826c3794cb386 : SUCCESS in 58m 40s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/6eab1b8a62ae4956bc9f053dc18d0e40 : SUCCESS in 1h 12m 46s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/64d8ef32b6d443d29cfa89ee6200fae8 : FAILURE in 1h 32m 40s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/47a1196613f343d5966e4bd4d830f9d5 : SUCCESS in 1h 30m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/832f0017812f4832887d9a7706bd7eb4 : SUCCESS in 2h 13m 22s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8f1498fb2d744c5fa9fcfea6640d8291 : FAILURE in 1h 37m 51s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/47352917b04442688b6f134ed0ff7728 : FAILURE in 1h 12m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17cba3c635574ef193106be6b8036c9b : SUCCESS in 8m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/93d8b09331d84fc1b8e1b147f0928c57 : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":2},{"id":"1c225138fe0c1c9f773a8870199c67cc2623cda4","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-03-07 19:50:24.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"8e73193784b52cb857d38294b30cfe27c38284a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-07 21:51:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/bfd8231e2c824a129fa07e2f86d765d2 : SUCCESS in 7m 03s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/757dd720e881485ab9cd5b698e140503 : SUCCESS in 1h 12m 39s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/9fb10b27ab4a4e16a5a62f29354db137 : FAILURE in 29m 28s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c612990c895a422c9fada0136f3ba80c : SUCCESS in 56m 47s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/1b78289ecd514c92a00a087460c2825c : SUCCESS in 1h 05m 07s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3ce27ea8b04e4cfda2e96b7201ab80a2 : SUCCESS in 1h 54m 39s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/9eda2dba5ba5459eb920263bd7e392ce : SUCCESS in 1h 39m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1aa33fa41e674ab3bc43db8e51cd7bc3 : SUCCESS in 1h 27m 21s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8ce2af3251b64f0d962b07cdd729fbb0 : FAILURE in 1h 28m 30s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d73f044acad845c3b3f1604ddd93898c : FAILURE in 1h 03m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42de1b4a356c448aa00c7c5126bc3242 : SUCCESS in 5m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c5212122e23e49e7bae254b462f5a3f6 : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":2},{"id":"282e7f0cd0d3f573fbf2ddd10df753fd89150077","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-03-07 22:03:20.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"690a819507216b0ce2211b359fc84d81876db943","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-07 22:03:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4622f9756f087d028c3f5b421ef6fcdfacd2471f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-07 23:36:10.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/78ec1c637d844b0b8fad13517f772b69 : SUCCESS in 6m 18s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/340d6eec0a804d80982d02b935e42af0 : SUCCESS in 1h 15m 37s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b244cbf1c19e4f45b29689923fb89b69 : SUCCESS in 58m 18s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/c50daf4869ee4cd2a413dcfa864a4f5c : SUCCESS in 1h 05m 21s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e0f54e5b83d94d94b3e6f9c67e824fb7 : SUCCESS in 1h 29m 13s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/23005fe9fe7a472babb3fc4b99c5aa60 : SUCCESS in 1h 27m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4c98c8d84df244bca12b1df02f6265c5 : SUCCESS in 1h 15m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/196f22037b2a41bd8ca5b96c5dbcbcb8 : SUCCESS in 4m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a496eed1a59843f69d4b1294b206c1c4 : SUCCESS in 3m 23s","accounts_in_message":[],"_revision_number":2},{"id":"39b2f1cfb330cee4668a2d7922e8b48bbbb5f459","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-07 23:36:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"63a1ca91befa95ec31d81fe5e1c000a31516a117","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-07 23:37:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72e72f0221084dfb96559e743b5d981a : SUCCESS in 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e6b7f8be3324404bac3fab05fad45eed : SUCCESS in 54s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ee4c54e5104332413c1de64421e7173fea8303aa","revisions":{"4cb9dcbe9d41459381aa066fc06daa65c59f5402":{"kind":"REWORK","_number":1,"created":"2020-02-24 13:11:58.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/98/709498/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_mistral","ref":"refs/changes/98/709498/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/98/709498/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/98/709498/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/98/709498/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/98/709498/1"}}},"commit":{"parents":[{"commit":"b9f815d6d3df5b620e4bc23e456f8cdfed3279d5","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/b9f815d6d3df5b620e4bc23e456f8cdfed3279d5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-24 13:11:54.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-24 13:11:54.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I1b00ccf2a16a5bc97fb70fa75e9e58809299f6cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/4cb9dcbe9d41459381aa066fc06daa65c59f5402"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/4cb9dcbe9d41459381aa066fc06daa65c59f5402"}]},"branch":"refs/heads/master"},"ee4c54e5104332413c1de64421e7173fea8303aa":{"kind":"REWORK","_number":2,"created":"2020-03-06 17:57:52.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/98/709498/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_mistral","ref":"refs/changes/98/709498/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/98/709498/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/98/709498/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/98/709498/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/98/709498/2"}}},"commit":{"parents":[{"commit":"132ab7b140db396f780b451bcb892aec075501a4","subject":"[ussuri][goal] Updates for python 2.7 drop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/132ab7b140db396f780b451bcb892aec075501a4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-06 17:57:50.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-06 17:57:50.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I1b00ccf2a16a5bc97fb70fa75e9e58809299f6cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/ee4c54e5104332413c1de64421e7173fea8303aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/ee4c54e5104332413c1de64421e7173fea8303aa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
