)]}'
{"id":"openstack%2Fopenstack-ansible-os_mistral~742940","triplet_id":"openstack%2Fopenstack-ansible-os_mistral~stable%2Fussuri~Idedaa4729e01cb787bd06af3dc3427113fc0bfe3","project":"openstack/openstack-ansible-os_mistral","branch":"stable/ussuri","topic":"osa/el8-stable/ussuri","hashtags":[],"change_id":"Idedaa4729e01cb787bd06af3dc3427113fc0bfe3","subject":"Add centos-8 support","status":"MERGED","created":"2020-07-24 16:31:13.000000000","updated":"2020-07-25 11:33:30.000000000","submitted":"2020-07-25 11:32:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"742940-1595676746753-1d1e633c","meta_rev_id":"2e15ea606602d172f88421890161faab87e4ee33","_number":742940,"virtual_id_number":742940,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-25 11:32:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},{"value":2,"date":"2020-07-24 16:32:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-24 20:08:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"date":"2020-07-24 16:31:13.000000000","_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-07-24 20:08:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-24 16:31:13.000000000","updated_by":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"reviewer":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"state":"REVIEWER"},{"updated":"2020-07-24 20:08:34.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-07-25 11:32:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"53640f017979477545b2a78038352a496d1e4622","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-07-24 16:31:13.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"4b81cd6d7affa88a6a8764fd3d3d07bdf641e3bf","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-07-24 16:32:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ed39b87bc4f296d7a0ff975022a2c9fc001e2c76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 18:43:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/c284e93deb9f42bba7e7c9923e9121e8 : SUCCESS in 9m 02s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/97f72c2eefc24220ac1e3b70a57fa6b0 : SUCCESS in 1h 09m 22s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/b37b50fc287e46d886542226c628afc6 : SUCCESS in 1h 05m 39s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1fb160491edb4ba5b380df44aeee26af : SUCCESS in 1h 02m 01s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/d11a0fef5aa84709ad676af6575e2fbf : SUCCESS in 1h 10m 27s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/064f0cb5ea104271a809bbc4c55e6f41 : SUCCESS in 1h 31m 25s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/8a14fb9bf01843699cf1e7efba548cf6 : SUCCESS in 1h 25m 43s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b6302282fcd14f5086c626e2abf10ba8 : SUCCESS in 1h 18m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c701b43d6f274a2c998f1a1f071eabb4 : SUCCESS in 1h 05m 53s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/21dd5f8cdbe1489083e6e9b4f9569146 : FAILURE in 1h 33m 11s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f12f83c975d941ceb14562fc78947d05 : SUCCESS in 2h 03m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bfa2d55cad249a6a0c34c4ec49d48f1 : SUCCESS in 7m 04s","accounts_in_message":[],"_revision_number":1},{"id":"25b89a71ac9020a32dd78100709113e7d1cc223c","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-07-24 20:08:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a3b40c317756f603b438bd6988a275ba4d603851","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 20:08:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"629d688d6b863124b0f92e2e617405e7b7154f1e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 23:18:39.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/a60364998ad74202a660f795fbd96f55 : SUCCESS in 8m 16s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/6227f3df095a4d3aac91475706c7de84 : SUCCESS in 1h 30m 04s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/63fcdc05b31341d2a927f06d29e9aa3a : SUCCESS in 1h 23m 33s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4747d20258d7488fb2e82e15b09bf83e : SUCCESS in 1h 19m 06s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/22d5a07aeb7b43229cea5b1f08351972 : TIMED_OUT in 3h 03m 14s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/9531e835f7b942d2ad9e27480724febb : SUCCESS in 1h 45m 41s\n- openstack-ansible-deploy-aio_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/a6c4d306987c4b238fc09829ea99e9a4 : SUCCESS in 2h 30m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/05d048dc4f44473eab7690dbaf36be18 : SUCCESS in 1h 54m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ebcac55e790498cbddd61ac4f41efa7 : SUCCESS in 1h 16m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23d7469c6a2d4fe38ba3fd174ffa76fe : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":1},{"id":"d3e0135050609385c1958cd24b167db85f0e57f7","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-07-25 07:12:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f24d34440dccb56acf2b585ef65c1bf62813a82c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-25 09:18:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/749bcb7929624f1f93b8bb5fa1db8307 : SUCCESS in 9m 09s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d03a9feff808461fb41da73584b09937 : SUCCESS in 59m 49s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/852af8500dd94146bfbb67278617fc97 : SUCCESS in 1h 04m 29s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ff6f922e0b9946a2a24145b156af4a3f : SUCCESS in 57m 28s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/fa4a5080e9b347cdb889354a2c9cb60b : SUCCESS in 1h 09m 48s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3253c9929b7d4d8ca3f50f2998da72ee : SUCCESS in 1h 01m 46s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/713f5736aafe4e4da6c53426369edaf8 : SUCCESS in 1h 20m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9ad3c5ed00d74480a35024390ef1e049 : SUCCESS in 1h 13m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5a333a5766cc4597987d1a08c59b108a : SUCCESS in 1h 07m 13s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/79246e8c141345c5a7f3ce3c4b6ea30f : FAILURE in 1h 33m 37s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ebfc25b0b74946e8999fd681c5932b52 : SUCCESS in 1h 56m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88dbeaf96d5b426e8767682475cb582b : SUCCESS in 5m 32s","accounts_in_message":[],"_revision_number":1},{"id":"f023c9a3ef1b02f3334d28cbf5ad6bda075b47e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-25 09:18:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e165796bff84c5490a925f2b0081f1976a49c0d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-25 11:32:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/4f969e6cda724c77b5a4c661a8c20bb2 : SUCCESS in 7m 14s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0965b0a965cf48f39709b4db0847f8aa : SUCCESS in 1h 08m 18s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/94e73ed36c8741d59f891d065cf744ab : SUCCESS in 44m 57s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c61615e3fe914d9b80307824d760f5a0 : SUCCESS in 54m 04s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/a1fbacea1ef748238e3d2621e35cd392 : SUCCESS in 1h 11m 56s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f796e25ac7b94daabb2c76c64b4f00d9 : SUCCESS in 1h 23m 58s\n- openstack-ansible-deploy-aio_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/08e35313c89b4d09a121127cd093242d : SUCCESS in 2h 11m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b97a502e8d0d488d90fa792760b50997 : SUCCESS in 1h 14m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2ebb521d881e4f329a09486c5cf878f2 : SUCCESS in 1h 10m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0a4604354714099a31c218b46d39335 : SUCCESS in 5m 21s","accounts_in_message":[],"_revision_number":1},{"id":"5a15fb5492b48ceabdf23f8571849201efde94fe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-25 11:32:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"eb10ddfe4b2e9a7ce7b403c25c34e172845ea7bf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-25 11:33:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/845a5c2f22174b9ca52dc3347111443a : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1e7caea88de4d85a16e71eb58ffea4d5eb533440","revisions":{"1e7caea88de4d85a16e71eb58ffea4d5eb533440":{"kind":"REWORK","_number":1,"created":"2020-07-24 16:31:13.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/40/742940/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_mistral","ref":"refs/changes/40/742940/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/40/742940/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/40/742940/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/40/742940/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/40/742940/1"}}},"commit":{"parents":[{"commit":"d958df78e573c35960ee2ba03375cd1a2eb91797","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/d958df78e573c35960ee2ba03375cd1a2eb91797"}]}],"author":{"name":"root","email":"satish.txt@gmail.com","date":"2020-07-07 03:49:40.000000000","tz":-240},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-07-24 16:31:13.000000000","tz":0},"subject":"Add centos-8 support","message":"Add centos-8 support\n\nRenaming redhat-7.yml to redhat.yml for centos-8 support.\n\nChange-Id: Idedaa4729e01cb787bd06af3dc3427113fc0bfe3\n(cherry picked from commit 22701d0aebb467c6d81e91512463e0484d21cca2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/1e7caea88de4d85a16e71eb58ffea4d5eb533440"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/1e7caea88de4d85a16e71eb58ffea4d5eb533440"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
