)]}'
{"id":"openstack%2Fopenstack-ansible-os_mistral~760636","triplet_id":"openstack%2Fopenstack-ansible-os_mistral~master~Ifc6a975055559919f8e4234f315f1748ca4f5de0","project":"openstack/openstack-ansible-os_mistral","branch":"master","topic":"api_threads","hashtags":[],"change_id":"Ifc6a975055559919f8e4234f315f1748ca4f5de0","subject":"Reduce number of processes on small systems","status":"MERGED","created":"2020-10-30 16:35:39.000000000","updated":"2020-11-02 10:34:54.000000000","submitted":"2020-11-02 10:33:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"760636-1604313185403-5b3c2ee3","meta_rev_id":"3c5cbea902839851fd8b70ffcba88b422aa71916","_number":760636,"virtual_id_number":760636,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-02 10:33:05.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2020-11-02 08:41:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-11-01 17:14:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2020-11-02 08:41:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-01 17:14:43.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-11-02 08:41:37.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2020-11-02 10:33:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"952ec867b5bf31af0bca2b783662d28426f647d2","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-30 16:35:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0b93e479ec1eb99c58a3ba3b922101c06d0b8ff2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 19:01:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/3a19e1e2447f43f2bedf1eb0aa7aa6cf : SUCCESS in 11m 44s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/fea9b233a836452e9a51fb248ed5c030 : SUCCESS in 1h 05m 16s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0bf36010b6be4cce89f8e50150f25a10 : SUCCESS in 1h 02m 11s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/7b584fc34fc245e0a8754545ae5ffdb9 : SUCCESS in 1h 08m 15s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/69b4ebc079db4ae6bafd30fc1ed2fc1a : SUCCESS in 1h 37m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c8c3bae7547f4b16af6365cf0c3dc8d9 : SUCCESS in 1h 26m 44s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/82034509e1f54801bb1cc6f8d406f7c8 : SUCCESS in 1h 18m 33s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/8b86e84e66eb48998f5fe57d10fa7a07 : SUCCESS in 2h 10m 11s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8a2234ef8a7747eeb8cf4bd5e90b4da1 : SUCCESS in 1h 50m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/638cf5cc6cbf442898f923bac382f8fc : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":1},{"id":"395c0a7b9f1205fed9525240cb533703c4345d3f","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-11-01 17:14:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ecf1c78a5be95228ab4d4e883b2450fc3df1b4db","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2020-11-02 08:41:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"eac00381a4bb57229b5d47f3c0c8aca37e475240","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-02 08:41:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"040aeb9d633e1a00f66155c709c5c9e89dfd5de8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-02 10:33:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8378a713793d946748f854069790e93c22d4c3cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-02 10:33:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/243900242e5747c9a04073edcd3aaed1 : SUCCESS in 12m 25s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/c41c0a62a6e44bb8bc320d3b86c3deed : SUCCESS in 1h 13m 55s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/039987a304884c01882d2504d7b6e868 : SUCCESS in 1h 00m 38s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/92575fdc819542edb4c96f97350de36e : SUCCESS in 1h 17m 17s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/179eee1c453b455890f9a84a1e9e899c : SUCCESS in 1h 35m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b5e3378722a145c8b7b06f9dc349cd9c : SUCCESS in 1h 22m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ca1e432e8a1344ad9311991ef60a5537 : SUCCESS in 1h 12m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45c144a6c12d427ca2b1fca64d00f739 : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":1},{"id":"61736920f6dfcfb5e14d55ac0cf77424338f1f59","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-02 10:34:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba2498c84bed4a26989eebd19b1a63d3 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fc31f37f696ed27d6b2bd0c3cb3c9220d460f917","revisions":{"fc31f37f696ed27d6b2bd0c3cb3c9220d460f917":{"kind":"REWORK","_number":1,"created":"2020-10-30 16:35:39.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/36/760636/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_mistral","ref":"refs/changes/36/760636/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/36/760636/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/36/760636/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/36/760636/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/36/760636/1"}}},"commit":{"parents":[{"commit":"053982c22a99ea168f6b8720fd9a15bfcc44d868","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/053982c22a99ea168f6b8720fd9a15bfcc44d868"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-10-30 16:35:30.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-10-30 16:35:30.000000000","tz":120},"subject":"Reduce number of processes on small systems","message":"Reduce number of processes on small systems\n\nEven the most modest 4C/8T system would run with the maximum 16 processes\ndue to the calculation being VCPU*2.\n\nWe devide amount of CPUs to number of threads for hyperthreaded CPUs\n\nChange-Id: Ifc6a975055559919f8e4234f315f1748ca4f5de0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/fc31f37f696ed27d6b2bd0c3cb3c9220d460f917"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/fc31f37f696ed27d6b2bd0c3cb3c9220d460f917"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
