)]}'
{"id":"openstack%2Fopenstack-ansible-os_mistral~792613","triplet_id":"openstack%2Fopenstack-ansible-os_mistral~master~Id2c4f09b15bfa184113a83f1eb3e825d7fe43129","project":"openstack/openstack-ansible-os_mistral","branch":"master","topic":"integrated_linters","hashtags":[],"change_id":"Id2c4f09b15bfa184113a83f1eb3e825d7fe43129","subject":"Replace linters test with integarted one","status":"MERGED","created":"2021-05-21 12:53:06.000000000","updated":"2021-05-26 14:48:06.000000000","submitted":"2021-05-26 14:43:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"792613-integrated_linters","meta_rev_id":"251bb940b4274efe44333fbb56f670ad156f673c","_number":792613,"virtual_id_number":792613,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-26 14:43:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2021-05-25 08:54:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-05-25 14:20:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-05-25 14:20:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-21 15:42:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-25 08:54:51.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2021-05-25 14:20:08.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"c80ff5ec55c81ab675729b7bd068503af4b94e28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-05-21 12:53:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"16eb57f0d8239ca0765618a7657f8f7f9fa4a5fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 15:42:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0ca2011c133b4f169289fcc8d1df82b0 : SUCCESS in 24m 54s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/91842efea5184aa1a54edfd86dc7d023 : SUCCESS in 1h 04m 17s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2eb71d59c9154f5490a4fcfb0ff895f6 : SUCCESS in 1h 03m 41s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/4f7faad532f14047b0e50096b295088d : SUCCESS in 1h 16m 57s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/c6ebe86c0d4f4a6196174cc8d51048c5 : SUCCESS in 1h 34m 06s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/9143d4ad794f426cb0ad775ec82cb802 : SUCCESS in 1h 25m 49s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4c92f5beff46455da9b5be09a28a0ec5 : SUCCESS in 1h 31m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/367f2da914d24ea793b4b2ede4e17743 : SUCCESS in 1h 23m 49s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/efc5419966a94bfaa9818c340f995747 : SUCCESS in 2h 01m 10s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f4d89480100e4992b75faea56a69d92b : SUCCESS in 1h 50m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5de42cd648334681ac72cec899dc03d4 : SUCCESS in 7m 22s","accounts_in_message":[],"_revision_number":1},{"id":"65898f76ff2a9c838cb8b02436cfd4c440eef9d7","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-05-25 08:54:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b705e509eb5290a8bcd0981b6d2bc2ccc53321e1","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-05-25 14:20:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8477d23f6f5e550e0108935cbeddf3459fc3a468","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 14:21:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"828a574a17ec7ce9b35de21c27b4abf25807619a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 16:03:57.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ff43ce6414814ed7913a977548772ff1 : SUCCESS in 20m 31s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/c7811d667f2749089743fd402f57946b : FAILURE in 1h 00m 02s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/10a515b3edcb4adaaa05fac177c90f18 : SUCCESS in 50m 43s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/3a7edf7c43334b3f97d8a88a2b8c84e8 : SUCCESS in 1h 08m 18s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/1d729b76db894c6cbe71bd4650008338 : SUCCESS in 1h 19m 34s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/a4f9151300484fe99391a4b029b40309 : FAILURE in 1h 29m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f7b860e7653648fe9e944b97a5279498 : SUCCESS in 1h 32m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bf917d9cc061406a9af8f42e748db1fb : SUCCESS in 1h 17m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9724d430b0f64eb08af51a08199f47d5 : SUCCESS in 6m 35s","accounts_in_message":[],"_revision_number":1},{"id":"8bf6d848940ac40d3e3f68e839430bb40117f5d6","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-05-25 23:29:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a5003df33de966026dea0318fc4dbbdd77e24a21","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-05-25 23:29:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0cabb8f68ee7a70cd29d573d3464bc5bc2b973a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 01:33:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/63d90245d655407692283f1ef4d417d2 : SUCCESS in 21m 35s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/a244053f132342a094af3007523960ff : SUCCESS in 1h 11m 55s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/97ab8f400c1b41d59e55c1ad8e598bb0 : SUCCESS in 1h 01m 30s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/96f6df1be4a048d3974cf86bebcd6354 : SUCCESS in 1h 16m 30s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/1eb9bbb4999740728e4cbc1821fc842f : SUCCESS in 1h 44m 30s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/1fff72b50758416b8d2ca3c729e558cb : SUCCESS in 1h 25m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b65aaa0e523b4c06a9f2e2cc81c9993b : SUCCESS in 1h 29m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5607811899de428897a66034cec54e2b : SUCCESS in 1h 09m 40s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/dc2b884c475742eca0c3672f70ffc831 : SUCCESS in 1h 41m 29s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7303afa684854d2e8aacf5fcacd9e285 : SUCCESS in 1h 54m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62e15d37c2da4e6d9ad9de898bcd4d0c : SUCCESS in 6m 50s","accounts_in_message":[],"_revision_number":1},{"id":"fc9b96fd711bf9cb4452b2e4f9d34686fdd798e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 01:33:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"418041a564d403e1485ed2a2c8d06e047f66eb06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 02:53:30.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/69abe6564bdf41e3bee3671b537fa03c : SUCCESS in 16m 37s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/19ec2f5db46040f680a7833ff7d2b0f9 : SUCCESS in 54m 00s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e53f2d0f1ead4173b02944cb2b9da68a : SUCCESS in 56m 27s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/6fad1058da754253948d54a6ccd320b3 : SUCCESS in 1h 09m 04s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/0ed25c9982944fd7a24cc2a5dd04a821 : FAILURE in 36m 30s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/f9fac46af965458a8f2096ba4d774a75 : SUCCESS in 1h 17m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e589e7e90496499b95f6f742002ec6f8 : SUCCESS in 1h 00m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/72c9d3f401144d0cbedb2dac25842842 : SUCCESS in 1h 01m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1dc50bcc40a4113b4513989837fbc1f : SUCCESS in 6m 07s","accounts_in_message":[],"_revision_number":1},{"id":"e5112e5fe2847df294feb1068918de10e01a8da5","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-05-26 06:30:44.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2b4963e935e4ef1add11d931b76cc13119b7b4b1","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-05-26 06:32:03.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a419b6c1edb3a7b7975f154e2de961c34c554b5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 08:48:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6200e228f88045f980aa0cf9071ea8f2 : SUCCESS in 29m 46s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/c3a3a951ccbe4ae2b2f9043e619fa526 : SUCCESS in 1h 07m 28s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/69af502e1bcd46619f09980baeccf43b : SUCCESS in 1h 05m 59s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/7ec04ccfb19a446685330f1c5992fe7f : SUCCESS in 1h 19m 14s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/80404aba8f264e88b6b0c599c39e8e4e : SUCCESS in 1h 35m 08s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/13e697daeac946809696c2674ec0fab7 : SUCCESS in 1h 33m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/337136a8a03a4faab8425736900f982d : SUCCESS in 1h 36m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1272400dec97405fae4d7d8c15eca520 : SUCCESS in 1h 26m 34s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/6263882b46d040ad83d90aeab3a2954f : SUCCESS in 2h 12m 54s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/22fc81a253d546dfad338436d44d960b : SUCCESS in 1h 57m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe76b65a16cf448ca483e4d13f476e22 : SUCCESS in 8m 28s","accounts_in_message":[],"_revision_number":1},{"id":"08450e9dedeaa623ba4b8c58435721c453647d1a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 08:49:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5c18e44b623a810b810380f682260ab0ef5a3fab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 10:24:50.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/af0d0a6127c74ccdb0eabe7a1ed96820 : SUCCESS in 19m 29s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/c6258fbc2c614653af17ee32a22c1ad0 : SUCCESS in 1h 02m 38s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/13d1a466d6ae405b81753fcd5aa13b1b : SUCCESS in 1h 00m 58s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/6f36464ecc9f4bb78c8f9e3f9d7d533d : SUCCESS in 1h 13m 21s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/c64cbc0ff616414aa71dcc805cbec75b : FAILURE in 45m 54s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/8e71dd36b00b4650bcdc1465f34cdf0a : SUCCESS in 1h 30m 39s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/641ef9bfcf394344ac056930dffd4a40 : SUCCESS in 1h 10m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f1be08c2785240709accd02983f7388d : SUCCESS in 1h 12m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ed3862d17ba4348937eb787f1126a73 : SUCCESS in 6m 03s","accounts_in_message":[],"_revision_number":1},{"id":"970e421134310a9a18619f8ec183ccb9e3022b90","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-05-26 10:50:23.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d8632db289b2a45423cad0cd9b5448ef69bbc48d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 13:00:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e9969b9da13e415bbba3a821214df683 : SUCCESS in 18m 53s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/0f54014f441a4d799ad18f0f943009bf : SUCCESS in 1h 10m 02s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0eae9dbeb3974a09aea4cdf24c5d9c3b : SUCCESS in 58m 44s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/57c45797393949a3a6dd536b6179b794 : SUCCESS in 1h 22m 46s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/056d485dbf3c4bff907b1261720d12dc : SUCCESS in 1h 33m 00s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/d0df2aba4c344e61aea389934c60ff5c : SUCCESS in 1h 31m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f3fff9720bc04a09b3f7a3a1efbb62b3 : SUCCESS in 1h 29m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/023b2e26217e4d7a876c97842681f78b : SUCCESS in 1h 15m 08s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/13fa4fffec8643a6908a193c213120ec : SUCCESS in 2h 04m 41s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/21d3b48a14e8471585f6e499f5b10bfd : SUCCESS in 1h 48m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4caee75fe2446259ae94b4d82f2581b : SUCCESS in 6m 53s","accounts_in_message":[],"_revision_number":1},{"id":"12648ba1fbbc774d609193853b6eb2d5ab30e0a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 13:01:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9fe4125a0cac91f25547636d8dae683d9c33a25d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 14:43:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f2696b2a890e4c5c85ce68c89ec9f502 : SUCCESS in 19m 13s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/0da5008f6fb54ef8b898341482a0a7fb : SUCCESS in 1h 13m 53s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/14512ca887634824bb22b8382d6a9d8e : SUCCESS in 1h 08m 32s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/2fb9628af704475fb75ed8f440545692 : SUCCESS in 1h 24m 15s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/4ac87169b4f44054a625527f69c2f4c6 : SUCCESS in 1h 29m 00s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/fb013d3bf37d435f97dc105faa00c9b2 : SUCCESS in 1h 37m 01s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/deac25ac4789471d95d1856d1fb7e90f : SUCCESS in 1h 25m 21s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7bae4186ca8b462b9b263d4e16cead86 : SUCCESS in 1h 16m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1549437ebb0e45b8897b33a582581a96 : SUCCESS in 6m 30s","accounts_in_message":[],"_revision_number":1},{"id":"e07b17ba83e68a9ce0a714b8d06a5562239e014e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 14:43:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"251bb940b4274efe44333fbb56f670ad156f673c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 14:48:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/810016b3de464ebababf3cae0fe04eb6 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"efd06dfaadb8ddafc66aa48fd615617917d6d897","revisions":{"efd06dfaadb8ddafc66aa48fd615617917d6d897":{"kind":"REWORK","_number":1,"created":"2021-05-21 12:53:06.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/13/792613/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_mistral","ref":"refs/changes/13/792613/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/13/792613/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/13/792613/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/13/792613/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/13/792613/1"}}},"commit":{"parents":[{"commit":"858375bd58d86a84c57fa1fc88c6d6ce327a3b7d","subject":"Merge \"Add variables for rabbitmq ssl configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/858375bd58d86a84c57fa1fc88c6d6ce327a3b7d"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-05-21 12:53:03.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-05-21 12:53:03.000000000","tz":180},"subject":"Replace linters test with integarted one","message":"Replace linters test with integarted one\n\nWe\u0027ve created integrated linters check job a while back and it\u0027s successfully\nworking for several releases. At the moment we experience difficulties\nwith future maintenance of the linters check from the openstack-ansible-tests\nrepo. So instead of fixing current one, we replace it with modern version of\nthe test.\n\n\nChange-Id: Id2c4f09b15bfa184113a83f1eb3e825d7fe43129","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/efd06dfaadb8ddafc66aa48fd615617917d6d897"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/efd06dfaadb8ddafc66aa48fd615617917d6d897"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
