)]}'
{"id":"openstack%2Fopenstack-ansible-os_mistral~798550","triplet_id":"openstack%2Fopenstack-ansible-os_mistral~master~I0386a5754e78d9c608afe0051ae8ff64b11659a3","project":"openstack/openstack-ansible-os_mistral","branch":"master","topic":"reno-wallaby","hashtags":[],"change_id":"I0386a5754e78d9c608afe0051ae8ff64b11659a3","subject":"Update master for stable/wallaby","status":"ABANDONED","created":"2021-06-29 08:47:47.000000000","updated":"2021-07-01 10:08:18.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"589003e0c9739ecc21e8d0f71613919265bd332d","_number":798550,"virtual_id_number":798550,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-06-29 09:23:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-29 09:23:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"51b9be5ffb654bee4fb9ce7caf3c5d9add0c0908","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2021-06-29 08:47:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a8f37a5b399319c55cb4f5f9d7bd901aaf8670e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 09:23:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/edfbe74612544e11ab1ff8cbc64d9071 : SUCCESS in 16m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c39fc110f94478b8d9d4a6880beb957 : SUCCESS in 7m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d5577b364e72472aaf6e5b47debf006f : SUCCESS in 6m 14s","accounts_in_message":[],"_revision_number":1},{"id":"589003e0c9739ecc21e8d0f71613919265bd332d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-07-01 10:08:18.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f0b5c1e649e2d0d1b3d1b951c1786262f18cd117","revisions":{"f0b5c1e649e2d0d1b3d1b951c1786262f18cd117":{"kind":"REWORK","_number":1,"created":"2021-06-29 08:47:47.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/50/798550/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_mistral","ref":"refs/changes/50/798550/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/50/798550/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/50/798550/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/50/798550/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/50/798550/1"}}},"commit":{"parents":[{"commit":"797835ffa67b97646c3be0d2a420101fbb15f974","subject":"Merge \"Use ansible_facts[] instead of fact variables\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/797835ffa67b97646c3be0d2a420101fbb15f974"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-06-29 08:47:46.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-06-29 08:47:46.000000000","tz":0},"subject":"Update master for stable/wallaby","message":"Update master for stable/wallaby\n\nAdd file to the reno documentation build to show release notes for\nstable/wallaby.\n\nUse pbr instruction to increment the minor version number\nautomatically so that master versions are higher than the versions on\nstable/wallaby.\n\nSem-Ver: feature\nChange-Id: I0386a5754e78d9c608afe0051ae8ff64b11659a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/f0b5c1e649e2d0d1b3d1b951c1786262f18cd117"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/f0b5c1e649e2d0d1b3d1b951c1786262f18cd117"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
