)]}'
{"id":"openstack%2Fopenstack-ansible-os_mistral~810234","triplet_id":"openstack%2Fopenstack-ansible-os_mistral~master~Ic050f71a2e9a48e12e2549ffdf60223d2b0c601d","project":"openstack/openstack-ansible-os_mistral","branch":"master","topic":"osa/galera_pki","hashtags":[],"change_id":"Ic050f71a2e9a48e12e2549ffdf60223d2b0c601d","subject":"Refactor galera_use_ssl behaviour","status":"MERGED","created":"2021-09-21 13:20:12.000000000","updated":"2021-10-07 13:03:29.000000000","submitted":"2021-10-07 13:01:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"810234-osa/galera_pki","meta_rev_id":"e8352bd48ef52e2ce5c0eb5d5b8dcdb0bf4f7fc2","_number":810234,"virtual_id_number":810234,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-07 13:01:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2021-10-06 09:15:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2021-10-05 17:39:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2021-10-06 09:15:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-21 15:44:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-05 17:39:04.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"},{"updated":"2021-10-06 09:15:54.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"77540b88aad4e7792b24a90352023f5dd6b46891","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-09-21 13:20:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"29dd7652bda3f3fcc9fee280b50273c439354cf7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-21 15:44:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a655be86ebe445d9d48edc28d63e4c7 : SUCCESS in 16m 58s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/dd5201f80d92458d9fa751a4719bc35e : FAILURE in 53m 02s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4e4f05a026bd4bac8a5d40bce98a29a4 : SUCCESS in 1h 00m 44s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/087bf939f6554286b21e2c079c00903d : SUCCESS in 1h 17m 30s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/71efd214e6994314894f36401e3e5fec : SUCCESS in 1h 20m 26s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/5061c2569c414ff1b00925f300984333 : SUCCESS in 1h 30m 13s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/924edfd41a6b4c2fa0b660e0270847a5 : SUCCESS in 1h 25m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7d131cf8a40843ba8a248021d0fbdb3b : SUCCESS in 1h 24m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f42c0a926c5a49b88469eaef41eabbee : SUCCESS in 1h 27m 24s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/ca6c15a6d5544fbd850803e7f84e0f1e : SUCCESS in 2h 12m 17s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4d59e62000db41a79cbda6d3bcfda0ae : SUCCESS in 1h 46m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b42aaf90e9b34a88b37f6105d28b1642 : SUCCESS in 6m 18s","accounts_in_message":[],"_revision_number":1},{"id":"854cc222423499cc1f41f7f40fd4f5af6150771c","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2021-10-05 17:39:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4be018d394f200ec0f7ff6149e83a0945c7b95a3","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-10-06 09:15:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bf3c9a6e927fc2d53ad7e7c15df2aa7dbde5a9ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-06 09:34:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4793d0dbbef748c9357b467ed769de441807884c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-06 11:43:16.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/252d2cd8924947f0993d5bd10ec35fe5 : SUCCESS in 17m 26s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2a3ac755ea674d2cae2b444e755f5eca : SUCCESS in 57m 18s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/9740238bf4d6462f8a707bd09b0cf4c9 : SUCCESS in 1h 21m 34s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/1d2ce7c7210b429896c1cc6084919437 : SUCCESS in 1h 22m 28s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/b394a26c3f1846f58c754416af8ed868 : SUCCESS in 1h 31m 18s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/35d06b2a60644375b329bc83ce9dae28 : FAILURE in 1h 29m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b5978d09d38f499faf43cd2cf58eb732 : SUCCESS in 1h 37m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7c7e0669d40545239c981387147d61d7 : SUCCESS in 1h 21m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a295b6da189d4e5e95f4d5d3ae9aaa6b : SUCCESS in 6m 31s","accounts_in_message":[],"_revision_number":1},{"id":"671e536b6aebed4e244b76b2b4a0b29db5b42c2c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-10-06 11:59:50.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6323025f6be9fe8f0624b7b274d30a9cfe5399e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-06 14:22:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/123748538cad441dad3e6efe06634b1a : SUCCESS in 21m 46s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/c4dcd0965b3d4d1b9a4e280162b6fba9 : FAILURE in 46m 35s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e77187e10c85425288293e107f81435d : SUCCESS in 1h 09m 29s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/0cdb28106e0249f7a63f57937d745854 : SUCCESS in 1h 29m 58s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/4272512164a34913811d723b131855e7 : SUCCESS in 1h 16m 13s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/ab3b59d9eaf547cd89b7c9f7e77b0924 : SUCCESS in 1h 32m 28s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/589fd5d001b1451c8cae481429b64376 : SUCCESS in 1h 29m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/da9b0bae80cb4a499bc3ec800bb21fec : SUCCESS in 1h 30m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/96dca4e76d424691b8f8a82981f7455b : SUCCESS in 1h 17m 47s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/4e8cb627214c4e868037177acb8ccabe : SUCCESS in 2h 07m 49s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c19c4db2c9504da7850c643c625272de : SUCCESS in 1h 59m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83c1c099d51c4930878acc5b40ceec89 : SUCCESS in 8m 05s","accounts_in_message":[],"_revision_number":1},{"id":"69d1d86f9f8a27c4d1f7cba32545d3e66a89d1cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-06 14:31:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4afd09d097767c7027a895b6d6b2c8cafff53e5a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-06 16:07:19.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5750f4ca61b648a6aa3d321a4817c24b : SUCCESS in 14m 57s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1cfff671c4834d73aaf25ff1d261a759 : SUCCESS in 57m 34s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/08f7af7407504497b9c790c31ee87597 : SUCCESS in 1h 22m 49s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/9c42013c8b874b85a01448d89f5096ba : SUCCESS in 1h 15m 29s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/fd1fe54ef35f48d9927422d518c517a4 : SUCCESS in 1h 24m 22s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/ba88169567504f9cb6f35028b0f04b3b : FAILURE in 1h 28m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bc28a08a9ed847f3bcbc9869633d1b53 : SUCCESS in 1h 15m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b1fc5ef4ed0c418a973a35870f9ce2f5 : SUCCESS in 1h 19m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45c9c3074b8b422f82b286fad52461cd : SUCCESS in 5m 55s","accounts_in_message":[],"_revision_number":1},{"id":"53ef7a3e73e2d3389dbcd010957af26bff13bdc5","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-10-07 09:00:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"76d4bfb60a05e678c7edf6d5b52c45666c538016","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 11:17:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7d26d22abe564ee8b781cef9761fe6da : SUCCESS in 22m 44s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/cac8697f1e0b4937a54788ebb7aaecc0 : FAILURE in 49m 01s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3713d777d9584672a3ed85ebc8e3e851 : SUCCESS in 1h 02m 38s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/6c36743da3eb40cfbe8a03321cc23449 : SUCCESS in 1h 17m 10s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/e55210a3c7e9449b8c8a334fe8869cda : SUCCESS in 1h 16m 22s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/5a038ca034c34cd1a96a62c31e1336ea : SUCCESS in 1h 24m 29s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/7b87ab0257df4387a0ad04e60749305c : SUCCESS in 1h 30m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9005e8fddefa44b0aeb43f53d410e183 : SUCCESS in 1h 32m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b83bb9a551c64c658624985ca5882901 : SUCCESS in 1h 26m 03s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/8b9740c6c81c4190b5d132c65fa9a25d : SUCCESS in 2h 10m 04s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/98975ac5f8d74c1eb2a92c1d8e701305 : SUCCESS in 1h 54m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72b93a5b77d14f97a258a2376a9931a1 : SUCCESS in 7m 58s","accounts_in_message":[],"_revision_number":1},{"id":"508f73508ba9d0b2c644e3d9905a6c15e4c88dd3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 11:20:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b2fc3ee477d78e13a37ec009bd520ba0922b4ae2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 13:01:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/90b77f71ebe94b21b892a443e4354c3f : SUCCESS in 14m 01s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1b98de4168bd48718e34ed8ca6cc8c17 : SUCCESS in 1h 12m 09s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/d6575fecd7424fe68838f687b6fd23d1 : SUCCESS in 1h 26m 16s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/511c44355ddd40dd9503ac09673c054c : SUCCESS in 1h 23m 22s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/508ba692acd94cf1b2eab8287a26da7c : SUCCESS in 1h 37m 38s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/f189c2fa66b649e89d819c7fd65ea22d : SUCCESS in 1h 33m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5926249e0795464a8ed77371ca1752e8 : SUCCESS in 1h 37m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d6077c808cc746d287b6ff3fd1525864 : SUCCESS in 1h 16m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70dbec2d1bcd4e8db62a8d16345a07bd : SUCCESS in 6m 27s","accounts_in_message":[],"_revision_number":1},{"id":"a8414aaa1ef503f294b0ff0003353714e3afb004","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 13:01:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e8352bd48ef52e2ce5c0eb5d5b8dcdb0bf4f7fc2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 13:03:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03ba5b9297fb4875b50d8958bba16908 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ce27e5ec3904d493a6ab58f9ce850b6765105251","revisions":{"ce27e5ec3904d493a6ab58f9ce850b6765105251":{"kind":"REWORK","_number":1,"created":"2021-09-21 13:20:12.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/34/810234/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_mistral","ref":"refs/changes/34/810234/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/34/810234/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/34/810234/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/34/810234/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/34/810234/1"}}},"commit":{"parents":[{"commit":"b351843e017dd8df0bd75b917dbb7b08cfd39c35","subject":"Changed minversion in tox to 3.18.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/b351843e017dd8df0bd75b917dbb7b08cfd39c35"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-09-21 13:20:03.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-09-21 13:20:03.000000000","tz":180},"subject":"Refactor galera_use_ssl behaviour","message":"Refactor galera_use_ssl behaviour\n\nWith PKI role in place in most cases you don\u0027t need to explicitly\nprovide path to the CA file because PKI role ensures that CA is trusted\nby the system overall. In the meanwhile in PyMySQL [1] you must either\nprovide CA file or cert/key or enable verify.\n\nSince current behaviour is to provide path to the custom CA we expect\ncertificate being trusted overall. Thus we enable cert verification when\ngalera_use_ssl is True.\n\n[1] https://github.com/PyMySQL/PyMySQL/blob/78f0cf99e5d5351df0821442e4dc35c49a6390c6/pymysql/connections.py#L267\n\nChange-Id: Ic050f71a2e9a48e12e2549ffdf60223d2b0c601d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/ce27e5ec3904d493a6ab58f9ce850b6765105251"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/ce27e5ec3904d493a6ab58f9ce850b6765105251"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
