)]}'
{"id":"openstack%2Fopenstack-ansible-os_mistral~820237","triplet_id":"openstack%2Fopenstack-ansible-os_mistral~master~I6f71c588ccec56e37ec707fc7f9c29037fcd668f","project":"openstack/openstack-ansible-os_mistral","branch":"master","topic":"db-pooling","attention_set":{},"removed_from_attention_set":{"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2021-12-03 20:25:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6f71c588ccec56e37ec707fc7f9c29037fcd668f","subject":"Database connection pooling improvements","status":"MERGED","created":"2021-12-02 18:20:48.000000000","updated":"2021-12-03 20:27:09.000000000","submitted":"2021-12-03 20:25:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"820237-db-pooling","meta_rev_id":"879a17e62f6280dd8ef9a81f6384e25ad6bc22bb","_number":820237,"virtual_id_number":820237,"owner":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-03 20:25:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2021-12-03 13:52:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-12-03 13:55:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-12-03 13:55:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-02 20:46:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-03 13:52:50.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2021-12-03 13:55:20.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"9f3fd0560d2d642c276bd5ffac356f5a8f0fdbe4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2021-12-02 18:20:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d069b323404944f91ea77f9e9680db3952d8a06d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 20:46:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2fca345dcbd34c69881949855b25d4d5 : SUCCESS in 19m 11s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/142dac4ceb3b4d1aa5f1eee0d02783d0 : FAILURE in 36m 34s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/81a7342e10c749a188ea2abbec818ec6 : FAILURE in 42m 23s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/cf6755dc3d39423a8d6efe04ef89bbce : FAILURE in 1h 19m 14s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/7914f50b52b542399f99284643637f12 : FAILURE in 1h 05m 35s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/cb0b53efa6b64f8c8514aa28723118b7 : FAILURE in 57m 47s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/908ce0c231d643a6a340c604567accf4 : FAILURE in 1h 00m 33s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b839f1ef8754447bbc78ba67ac163d16 : FAILURE in 1h 32m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/876db735caca459289d799c00d362688 : FAILURE in 1h 18m 10s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/aee5cb37d82f48f5bdde5b747201bba0 : FAILURE in 2h 03m 57s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/60e59a87dc8b49fd84792d3b449eea57 : FAILURE in 1h 56m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26da3cb869e6473cb279211250c76cb6 : SUCCESS in 6m 24s","accounts_in_message":[],"_revision_number":1},{"id":"de959bd80141945cfdd42c8f3d1eaa15c1d4aabf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2021-12-03 10:41:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8ff580c6bf1b2619974041a482d02136e1874363","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-03 13:05:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/015f22a97e2341d8b7c09ab8ffa276be : SUCCESS in 14m 03s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/2a0dbfbba6214705a81e86ac1b745e09 : FAILURE in 39m 54s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c38c7634bcc746f29a24024b687dc3df : SUCCESS in 1h 05m 18s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/b59cd5d8fa3b47a1a9bec23795ea6a97 : SUCCESS in 1h 34m 18s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/4ee115732fa340bfa6470f7d28b063d3 : SUCCESS in 1h 20m 14s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/c13d67d62c0b40549189c7077efc4e2d : SUCCESS in 1h 29m 04s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/a7f93ab8b3a145e390c10801349dc060 : SUCCESS in 1h 42m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b0c512daec934cb499cadd6becbd2a48 : SUCCESS in 1h 33m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f1646aa54b40476881b9b0f467b3ab54 : SUCCESS in 1h 37m 07s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/830e752c958d49658741dc19e537716d : SUCCESS in 2h 12m 26s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e4ad458f521446e0813248efed930f3e : SUCCESS in 2h 05m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad1dfc78ca7448aab307464d2676d623 : SUCCESS in 5m 21s","accounts_in_message":[],"_revision_number":2},{"id":"b64481aa5efbfe8e335a70f3322bea05258b482d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-03 13:52:50.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"7668f6dbf128aac12ada5389fa3339e2284e7964","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-12-03 13:55:20.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8b6103c6ebceb25c1c2d9f83c170fc681ef6d3d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-03 13:57:25.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"dcc7b18db90a20a466de0d05ceddb5dd4eb211c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-03 15:40:32.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/820a50c47cfd417786730e7bd51c993d : SUCCESS in 19m 19s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f38891c2efb14896a12e439152ec5e40 : SUCCESS in 1h 07m 01s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/198890c8d452489f9d4bc8a2ae6fd550 : SUCCESS in 1h 35m 00s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/57b492bf7f634d47b60c99999bed5caa : SUCCESS in 1h 07m 24s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/c5e41b659eeb48ea9c2f04c9ed4fba59 : FAILURE in 1h 24m 34s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/5aa461d84e5f4c47b9df0a2bb662c08c : FAILURE in 1h 21m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/706fccd7d3594ccfb91fa1f9135a9d1d : SUCCESS in 1h 42m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d55bf5deae8e481597cb8957d64f5ec1 : SUCCESS in 1h 24m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3ed0db0788d4080a887b66cf17b3e5d : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":2},{"id":"6903bed92aa0d8fd2a5f730b8fbe325eda438fb4","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-03 16:37:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6cea84adbcea5235e681586d5bac0d55c9d7b042","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-03 18:47:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0da53eaec49c4bc79c5dd4edcbc612ea : SUCCESS in 15m 09s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/b14c785900f746f793e9944665dd0cc1 : FAILURE in 43m 20s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d3070bc861ae44f4afda90ee767fdeec : SUCCESS in 1h 03m 55s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/5170efb0999e482d9abce17465456c6d : SUCCESS in 1h 35m 53s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/3108539a21d34164a8adab26f7a208de : SUCCESS in 1h 18m 31s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/2440f522e2c8473e840b22529b3fe34b : SUCCESS in 1h 30m 07s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/565cb7facec645eda75979e51964075c : SUCCESS in 1h 27m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/57f92a481c9b43a591c7e93eca05e91e : SUCCESS in 1h 35m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2313a6cabf03413086eff0995c71d2e1 : SUCCESS in 1h 31m 23s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/3a97ec34923842c19b1408542a8e8dec : SUCCESS in 2h 06m 08s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/02fdea53cec240a4a8d8478ae6f9ede5 : SUCCESS in 2h 03m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/355696735cb94fb18fd193972ee76074 : SUCCESS in 6m 25s","accounts_in_message":[],"_revision_number":2},{"id":"e38f1ae1801842a921d4141a91cb854b4368a6d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-03 18:49:59.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b3b4beb8addcd88074dc5c35c303844980958ed2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-03 20:25:39.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f9ef797e5b504911a26b96e1f11bafb2 : SUCCESS in 12m 16s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d83423d2be77431ba7c5afaafc45b8ab : SUCCESS in 1h 03m 58s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/c00550a93cec489dbf6b648e88ef36a2 : SUCCESS in 1h 34m 23s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/d482d41db026437da66809b4747c6669 : SUCCESS in 1h 15m 11s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/0a4146d437f948b4aae8aae14b4652fd : SUCCESS in 1h 29m 03s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/57236c53cd144a92bf5c89b6b99cd578 : SUCCESS in 1h 27m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c79dba1f7741452b9b003171e5644ea6 : SUCCESS in 1h 33m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cecf7ca8dfcc43fbaf0a817463f2d09d : SUCCESS in 1h 28m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84c1c4a5f8c448d1af42bba6ee36fae0 : SUCCESS in 5m 23s","accounts_in_message":[],"_revision_number":2},{"id":"b89353daec5f36cacb9eb764944db4fc3b37c9be","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-03 20:25:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"879a17e62f6280dd8ef9a81f6384e25ad6bc22bb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-03 20:27:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0a3485337184146ab4141b3c8725d55 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fc7cc65df2689c987f7d8d5a9cacd10ef1debfc5","revisions":{"6111c1943235e14950c2b15120120129f2099b58":{"kind":"REWORK","_number":1,"created":"2021-12-02 18:20:48.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/37/820237/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_mistral","ref":"refs/changes/37/820237/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/37/820237/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/37/820237/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/37/820237/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/37/820237/1"}}},"commit":{"parents":[{"commit":"ce27e5ec3904d493a6ab58f9ce850b6765105251","subject":"Refactor galera_use_ssl behaviour","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/ce27e5ec3904d493a6ab58f9ce850b6765105251"}]}],"author":{"name":"Damian Dabrowski","email":"damian@dabrowski.cloud","date":"2021-12-02 18:20:38.000000000","tz":60},"committer":{"name":"Damian Dabrowski","email":"damian@dabrowski.cloud","date":"2021-12-02 18:20:46.000000000","tz":60},"subject":"Database connection pooling improvements","message":"Database connection pooling improvements\n\n- Implemented new variable ``connection_recycle_time`` responsible for SQLAlchemy\u0027s connection recycling\n- Set new default values for db pooling variables which are inherited from the global ones.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/819424\nChange-Id: I6f71c588ccec56e37ec707fc7f9c29037fcd668f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/6111c1943235e14950c2b15120120129f2099b58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/6111c1943235e14950c2b15120120129f2099b58"}]},"branch":"refs/heads/master"},"fc7cc65df2689c987f7d8d5a9cacd10ef1debfc5":{"kind":"REWORK","_number":2,"created":"2021-12-03 10:41:19.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/37/820237/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_mistral","ref":"refs/changes/37/820237/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/37/820237/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/37/820237/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/37/820237/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/37/820237/2"}}},"commit":{"parents":[{"commit":"ce27e5ec3904d493a6ab58f9ce850b6765105251","subject":"Refactor galera_use_ssl behaviour","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/ce27e5ec3904d493a6ab58f9ce850b6765105251"}]}],"author":{"name":"Damian Dabrowski","email":"damian@dabrowski.cloud","date":"2021-12-02 18:20:38.000000000","tz":60},"committer":{"name":"Damian Dabrowski","email":"damian@dabrowski.cloud","date":"2021-12-03 10:41:15.000000000","tz":60},"subject":"Database connection pooling improvements","message":"Database connection pooling improvements\n\n- Implemented new variable ``connection_recycle_time`` responsible for SQLAlchemy\u0027s connection recycling\n- Set new default values for db pooling variables which are inherited from the global ones.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/819424\nChange-Id: I6f71c588ccec56e37ec707fc7f9c29037fcd668f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/fc7cc65df2689c987f7d8d5a9cacd10ef1debfc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/fc7cc65df2689c987f7d8d5a9cacd10ef1debfc5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
