)]}'
{"id":"openstack%2Fopenstack-ansible-os_mistral~820435","triplet_id":"openstack%2Fopenstack-ansible-os_mistral~master~Id34047927b5295af22218525efffa4053c2e3cf2","project":"openstack/openstack-ansible-os_mistral","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"Id34047927b5295af22218525efffa4053c2e3cf2","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2021-12-04 17:40:31.000000000","updated":"2021-12-08 18:12:41.000000000","submitted":"2021-12-08 18:09:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"820435-openstack/openstack-ansible-tests/sync-tests","meta_rev_id":"141ac62a09a583335a48e5b2d3903c4e56e325e5","_number":820435,"virtual_id_number":820435,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-08 18:09:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2021-12-04 18:43:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2021-12-04 18:43:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-04 18:43:50.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2021-12-04 20:00:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ebac8759dfbcbc165b2753afe576bcd065cd2482","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-12-04 17:40:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eed442776e67b6b474d78257ed52a4e9e28224aa","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-04 18:43:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d12874a21db5602ea3c8273972ea915511ba069d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-04 20:00:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/67c08088985a42849f1153104a785549 : SUCCESS in 22m 36s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/9911cad301644dbca80b70d0fb115e45 : FAILURE in 38m 05s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/96d096833ef149b49c3262989dc4b246 : SUCCESS in 1h 06m 41s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/c9d25b9742bc47d68c7e7ba9e75c59ce : SUCCESS in 1h 24m 09s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/6c1541f7e54447d7939dbca2a0bc920c : SUCCESS in 1h 24m 33s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/8e93d2bfe64e4a888754c874f4b1551e : SUCCESS in 1h 25m 53s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/e33c97c799a44021a1260e549f0a833d : SUCCESS in 1h 36m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8179997a991e4c09870cc0451f0c81e9 : SUCCESS in 1h 34m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f25db21c710a4a54a9a48a812d40fe0c : SUCCESS in 1h 21m 37s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/cf568f4185fc4525b286809052d1a698 : SUCCESS in 2h 16m 08s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/df381f58ca5145a1bbf97d12c242c27a : SUCCESS in 1h 57m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a38e56ae9404852b44b26408acafa6f : SUCCESS in 7m 58s","accounts_in_message":[],"_revision_number":1},{"id":"4e315208f55fad720bc3b8fafb84ed40043c47ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-04 20:02:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a4601c8d241d5a9ccd7a12cbb8a23eaeeb7c55a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-04 22:20:47.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/03db49d3fccd438e882d8e96affb90b7 : SUCCESS in 18m 23s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e8a43d4d9e8d4b9f870a081286ea619c : SUCCESS in 1h 07m 59s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/62d9c204cf0e4fa9bb8206f1dc4f6511 : SUCCESS in 2h 13m 57s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/999cad09b42c45b68d408009652fee65 : SUCCESS in 1h 16m 24s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/449d6e923e8a4a209df9a9760cbfbb8c : SUCCESS in 1h 38m 58s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/0799e3b05c5945dcb4e44b15a2b1c3c5 : FAILURE in 1h 30m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/590e8061195343a7bf3088d60713aa46 : SUCCESS in 1h 36m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a5d75ee15a7d4750ab55158ae348fa64 : SUCCESS in 1h 31m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d37ab8ae10e45ff83d7c1d9e7de49c9 : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":1},{"id":"9d3bc95d06e2c521068ddf3caaa1aef7f057d9e0","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-06 07:54:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"46b2fe850b0ad08f2e146a327eee382b02348f77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-06 11:01:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d51af3f5fc1340739667b6452cfaf6b1 : SUCCESS in 11m 14s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/30164cf3030b4f9a927c58c02148edd0 : FAILURE in 40m 14s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/21249e5e0e314a1a94612145ef89b723 : FAILURE in 36m 40s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/d846d7bd5be145a0964ae11a3b06a618 : SUCCESS in 1h 50m 54s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/e6587da6bfa746dfacb1b97e21659652 : SUCCESS in 1h 01m 19s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f5331ad57df741ea8835a154226950bc : SUCCESS in 1h 19m 48s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/1a98a93ca1be4e90a4e41335052c2ec6 : SUCCESS in 1h 19m 33s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/99f8fbcc5a8c4f768ac792fb8b56a941 : SUCCESS in 2h 05m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0963cc2e10f0429c89fffb3bead04169 : SUCCESS in 1h 19m 59s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/6e5fc618353842cab2ae65fc4c32b884 : SUCCESS in 1h 56m 51s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/05a7c93067384247a98d8966e0e99014 : TIMED_OUT in 3h 04m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e82f072b8b143a6a1d862cc3f9089d6 : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":1},{"id":"f44912e6d31111553750fdb90e91616dac89fb56","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-08 08:10:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a2c4b67998c1ae682d7c2d6df549231aa4460a64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 10:22:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/924c0427ada8439b873f50df95a87282 : SUCCESS in 12m 08s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/29765f02bea442ab8bb2eb3118597f87 : FAILURE in 39m 52s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dcc5d3d685864363be718079fcd83bc1 : SUCCESS in 58m 59s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/bde97cf39dc44c6ead3e46cfff953c45 : SUCCESS in 1h 38m 48s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/554c61767cc749e78cd19a4443131556 : SUCCESS in 1h 16m 50s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/21d802331d7148559cc960b10a86ad9b : SUCCESS in 1h 25m 14s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/69cfa2eaaa9446d6bde5e5386b319d32 : SUCCESS in 1h 26m 01s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/20098849090141fa89bef1d7795afda4 : SUCCESS in 1h 35m 24s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/22352d5866e34c1089c364d06e2963f4 : SUCCESS in 1h 26m 24s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/9fd2a1e5d2c546a489744403a30b3123 : SUCCESS in 2h 06m 04s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2abea46de4964ad0850420ca4863909e : SUCCESS in 1h 57m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2ee26696ca74995bbf055ea7c7db2f5 : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":1},{"id":"c15fa0e2bade30df9e0edbc026b223823a68d8ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 10:24:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f41776242ff1f51497d2fe256320675875dc2ff5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 12:07:53.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/74af24c6218645988101e9da5b21fa11 : SUCCESS in 15m 13s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/99d0043acb474bf5ad9da3f4cb87013f : SUCCESS in 1h 08m 38s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/3ad22430333d4b338e3d57b155d63a2d : SUCCESS in 1h 39m 41s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/c50798f2ba644b348324af995a1aa978 : SUCCESS in 1h 15m 44s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/38404b66fb3f4f6ab3a228652c0112b7 : FAILURE in 1h 31m 12s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/1a61f1b5b49e48c097d482c9752520a2 : FAILURE in 1h 27m 16s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a217dfd2dfcb46148cfc3343675df45d : SUCCESS in 1h 36m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6008094cbc684b7ebc2eba037593f1bc : SUCCESS in 1h 27m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5972dd3e4125441ca644a05d74ec70b7 : SUCCESS in 6m 37s","accounts_in_message":[],"_revision_number":1},{"id":"33d362995e80e85df45e9d689d77a59ca1ab79ba","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-08 13:55:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6baed4498878917a08b89ac3c7b7d1b152823c24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 16:27:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/96557bcaf30a4002a45c092bf1950bea : SUCCESS in 18m 28s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/488d5f9113794820b437f8b915eebbdf : FAILURE in 35m 31s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3d246cefeab94d3da82a9952dd074334 : SUCCESS in 58m 11s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/d7d47f203e954b26a76b180bca986cbd : SUCCESS in 1h 43m 44s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/6ac071e6e0054d11ace63cabc924955a : SUCCESS in 1h 27m 36s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f3754a2981a44e459a1f8dcf92839ba4 : SUCCESS in 1h 27m 17s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/514f0ac19a6a48e1b0eed145321928c3 : SUCCESS in 1h 28m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/88dc869567ee447faae68be799215bae : SUCCESS in 1h 41m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f9c8dd73523470085cbb441b5d4fd2d : SUCCESS in 1h 33m 18s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/d138da63a7574387a9f6775fc38c70f7 : SUCCESS in 2h 13m 04s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e7edb9d3494d4691bbb6b63203613b74 : SUCCESS in 2h 14m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f3600fa7d8c4218adaf46a4e27ac638 : SUCCESS in 5m 58s","accounts_in_message":[],"_revision_number":1},{"id":"243881dadc4394c55937d405ea98ca578af0e255","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 16:31:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3b1c55b33f1b19c4b1ec5c5e7564e74a089db39f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 18:09:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e2100e5cdc54481db6e4d94b68fe3fa7 : SUCCESS in 13m 43s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ecdf94cb78174be583f4e019a5fc14a8 : SUCCESS in 59m 44s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/7013da84ae60438cab40c112f2f4db9f : SUCCESS in 1h 34m 21s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/cbf952e09d8345dc95130cfb3be25718 : SUCCESS in 1h 09m 18s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/6925ed478fd544e0b747908a029b061f : SUCCESS in 1h 23m 43s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/17870cc1d4fe4b828b6eb27e32310c05 : SUCCESS in 1h 26m 29s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d38b4e0ee387441f8090bf9732ac8bf1 : SUCCESS in 1h 35m 08s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8f842d3f95384945ad6d8c53157f964f : SUCCESS in 1h 25m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85161ebf26364269b0dc82fc97b15b1d : SUCCESS in 6m 52s","accounts_in_message":[],"_revision_number":1},{"id":"9cc84ea534b42a5cb564676fabd0433e8528048f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 18:09:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"141ac62a09a583335a48e5b2d3903c4e56e325e5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 18:12:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07d47f34ea0a4b36ae2444e1a94ae830 : SUCCESS in 1m 26s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dc98b6ee024f3831cf56ddd52589a6550b4ceea2","revisions":{"dc98b6ee024f3831cf56ddd52589a6550b4ceea2":{"kind":"REWORK","_number":1,"created":"2021-12-04 17:40:31.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/35/820435/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_mistral","ref":"refs/changes/35/820435/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/35/820435/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/35/820435/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/35/820435/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/35/820435/1"}}},"commit":{"parents":[{"commit":"1a22dc47de2eaa62d79d391e2f185c9a24bd3cc8","subject":"Merge \"Database connection pooling improvements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/1a22dc47de2eaa62d79d391e2f185c9a24bd3cc8"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-12-04 17:40:30.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-12-04 17:40:30.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: Id34047927b5295af22218525efffa4053c2e3cf2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/dc98b6ee024f3831cf56ddd52589a6550b4ceea2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/dc98b6ee024f3831cf56ddd52589a6550b4ceea2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
