)]}'
{"id":"openstack%2Fopenstack-ansible-os_mistral~849519","triplet_id":"openstack%2Fopenstack-ansible-os_mistral~stable%2Fwallaby~I03aaecd4286535ae80aca249778ea9b209b87129","project":"openstack/openstack-ansible-os_mistral","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"13095":{"account":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"last_update":"2022-07-13 20:11:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I03aaecd4286535ae80aca249778ea9b209b87129","subject":"Add mistral-extra in the mistral venv","status":"MERGED","created":"2022-07-12 18:37:39.000000000","updated":"2022-07-13 20:13:02.000000000","submitted":"2022-07-13 20:11:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"849519","cherry_pick_of_change":849575,"cherry_pick_of_patch_set":1,"meta_rev_id":"9fa899296683ddc743c05ae3badb0ff947691ae5","_number":849519,"virtual_id_number":849519,"owner":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-13 20:11:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2022-07-13 12:34:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-07-13 10:46:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2022-07-13 12:34:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-12 18:40:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-13 07:40:53.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"},{"updated":"2022-07-13 10:46:14.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2022-07-13 12:34:41.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"e9d3f9b19d57ffcea458b62721d094ebb8a0bbfb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-07-12 18:37:39.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"a1f23ec5ee387d62eee66589edc967d92d0dce32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 18:40:13.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"59302f9eedb81920311a7060e5f7a3603c25c1ff","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-07-13 07:40:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f591a943b1d55979194f6efb939e8a1831714cdc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 09:31:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7fc7a479632b47eeb3a1f6c9074ed55e\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/02e6fe1b40b84ba7a9f4873f9f240a0f : SUCCESS in 19m 34s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/0ea74a444db84e55bbc2fe4d34eee6ee : FAILURE in 45m 47s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6469a7d06e6b4cd3a505e82fd0bdfd2f : SUCCESS in 1h 05m 05s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/05da7ac93af0444b903a048d1e3ba28f : SUCCESS in 1h 25m 17s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/109adbd6c5844b93814d655da5186b39 : SUCCESS in 1h 24m 54s (non-voting)\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/a40a349344bf4eb895051e1f9897c131 : SUCCESS in 1h 37m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d433470f1f344042bb5e884fe1039565 : SUCCESS in 1h 26m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/788c52ac455745bba27f5f3dce00a339 : SUCCESS in 1h 19m 39s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/21ac3d51bde64e1da8a0c40627f22af4 : SUCCESS in 1h 44m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16a6e9f2c634478da7f79d5247d4284d : SUCCESS in 5m 56s","accounts_in_message":[],"_revision_number":1},{"id":"6f024c487c631e965764b839114cb54434f1d3de","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-07-13 10:46:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2477d995a8983529ed9d28cd2430d5417e56eda6","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-07-13 12:34:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"befed1fd7f37a9ff037d52045f1e7c5bc71ff751","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 12:34:58.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ffc1fa979b6b4955a5da48d37aae185a\n\n\nWarning:\n  Change 849518 in project openstack/openstack-ansible does not share a change queue with 849519 in project openstack/openstack-ansible-os_mistral","accounts_in_message":[],"_revision_number":1},{"id":"cf15c9f77efa9fa435ccf05a910b9c380d54e4e6","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-07-13 13:55:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5fbc854c7f5210ddc22c2923229eac5f1600c096","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 16:16:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4b89e151bb764da3b89acae877138946\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/da78de7239e34998a3d715f15bad5d51 : SUCCESS in 22m 29s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/e5da034fc6bc4e558a3e1f7102f39bf6 : FAILURE in 46m 26s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9a7b1fb8f2324ea0a5cbbed2f8b6c5ab : FAILURE in 54m 55s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/c79d0970860c4db49a8dacac09d860f9 : FAILURE in 1h 19m 44s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/7abfff7beffd484aba81a27021b4b3e0 : FAILURE in 1h 15m 25s (non-voting)\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/621bcd84daa44c159d783e160dfec07f : SUCCESS in 1h 45m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/61effbe5f53b49ccaffb9ed107294a19 : RETRY_LIMIT in 17m 44s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f0e6b7b01fb943fca4cb752d1491d5a6 : SUCCESS in 1h 35m 17s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/98d406d17ae244dcb81186b04d92d12a : RETRY_LIMIT in 17m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/200536c7c5a941229fc966734a361d55 : SUCCESS in 6m 58s","accounts_in_message":[],"_revision_number":1},{"id":"8ac91260ef4ba5627c5baf18311ce2dceb5840de","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-07-13 16:17:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8bbdd6de8168c0ee5a5a1384f82c02648e2197ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 18:21:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd029578ea6e491daa56b1f48acd93f5\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bc5d176e94a845afbb6e35e23fafb904 : SUCCESS in 21m 46s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/5f5106bead58431995989c48a93f8c27 : FAILURE in 44m 07s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/de26db63bfb24295987d0feb488139bf : SUCCESS in 1h 06m 03s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/7eb3311657574e1d9a8bfc158a581631 : SUCCESS in 1h 15m 33s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/86866dc872ee49d180815a05b065151c : SUCCESS in 1h 19m 38s (non-voting)\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/542067a720324f9bada39b02eae07ae8 : SUCCESS in 1h 35m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f3d9b9dfbb6c467fa85c439fc9c8c8f6 : SUCCESS in 1h 32m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/71ee0f0120294502b6427f9f624e96c1 : SUCCESS in 1h 04m 45s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bacce8888ad64452ac180e0cd31d6b8b : SUCCESS in 1h 56m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f71058b9084404a871298ebaa1e57e1 : SUCCESS in 7m 01s","accounts_in_message":[],"_revision_number":1},{"id":"1c27dce4122ea527a06af92776da014293cfe43e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 18:23:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"74c824ff37d5d1f8fb6834d20d3f2d0991bf4b73","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 20:11:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ef530b9b4744069aae95ab2910893b7\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8da1d61c5eb74fb89119826df7e259ea : SUCCESS in 36m 36s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/79e6b2942e8c4ef0949c0aeaa9890c33 : SUCCESS in 1h 15m 56s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/916d7936ca0c44d3864cc95d772b31de : SUCCESS in 1h 18m 18s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/b00b11b141824435ab69e30f6b337df2 : SUCCESS in 1h 48m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f0ef4a4cce064676b0b054a87779fc78 : SUCCESS in 1h 26m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b7879f528fdc41dcbe5475b42b7b3742 : SUCCESS in 1h 22m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0bf9d9726bf41d396d722ad939fa8f2 : SUCCESS in 7m 52s","accounts_in_message":[],"_revision_number":1},{"id":"ad9904739d50b7fb09a491dc2c36cf7f1fcea182","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 20:11:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9fa899296683ddc743c05ae3badb0ff947691ae5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 20:13:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf9a2c0118004985a0977f301852d887\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/270dfec83284491c8892f90ccead5e70 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"34343349d86991feed1bfbbfcf735a9cea1300ef","revisions":{"34343349d86991feed1bfbbfcf735a9cea1300ef":{"kind":"REWORK","_number":1,"created":"2022-07-12 18:37:39.000000000","uploader":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"ref":"refs/changes/19/849519/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_mistral","ref":"refs/changes/19/849519/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/19/849519/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/19/849519/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/19/849519/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/19/849519/1"}}},"commit":{"parents":[{"commit":"fce6170874c229e1f03bda4c9435c72aca63333e","subject":"Update TOX_CONSTRAINTS_FILE for stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/fce6170874c229e1f03bda4c9435c72aca63333e"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2022-05-31 12:22:36.000000000","tz":-240},"committer":{"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","date":"2022-07-12 18:37:39.000000000","tz":0},"subject":"Add mistral-extra in the mistral venv","message":"Add mistral-extra in the mistral venv\n\nMistral has split the openstack action from the main repo, this patch\ninstall the extra actions into the venv so we can have openstack-action\nin it.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/849518\nChange-Id: I03aaecd4286535ae80aca249778ea9b209b87129\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/34343349d86991feed1bfbbfcf735a9cea1300ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/34343349d86991feed1bfbbfcf735a9cea1300ef"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
