)]}'
{"id":"openstack%2Fopenstack-ansible-os_mistral~849522","triplet_id":"openstack%2Fopenstack-ansible-os_mistral~stable%2Fussuri~I03aaecd4286535ae80aca249778ea9b209b87129","project":"openstack/openstack-ansible-os_mistral","branch":"stable/ussuri","attention_set":{},"removed_from_attention_set":{"13095":{"account":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"last_update":"2024-01-03 16:02:30.000000000","reason":"Change was abandoned"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2024-01-03 16:02:30.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I03aaecd4286535ae80aca249778ea9b209b87129","subject":"Add mistral-extra in the mistral venv","status":"ABANDONED","created":"2022-07-12 18:58:11.000000000","updated":"2024-01-03 16:02:30.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":849578,"cherry_pick_of_patch_set":1,"meta_rev_id":"dcf2b8fb4d9efff509012cd40127b8014b5b01d3","_number":849522,"virtual_id_number":849522,"owner":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-01-03 10:09:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"}],"CC":[{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-12 21:51:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-12 17:55:12.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"CC"},{"updated":"2022-10-21 07:55:57.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"},{"updated":"2022-10-24 08:01:22.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2022-10-24 10:24:14.000000000","updated_by":{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},"reviewer":{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},"state":"REVIEWER"}],"messages":[{"id":"ef57a22175ab1e3fb1674fd8d2850f135a4c3878","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-07-12 18:58:11.000000000","message":"Patch Set 1: Cherry Picked from branch stable/victoria.","accounts_in_message":[],"_revision_number":1},{"id":"560196aa04829392e5988a0e4f36f2cbc79eb81d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 21:51:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d228cd52fe8849b0b9d92511ef1a3f44\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/ad49eaac366b427fb190754c324de535 : FAILURE in 12m 05s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/9f43f4ef948b445eb1c03af379eb3c51 : FAILURE in 1h 03m 48s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/098bf48e15aa45fca631d2716fbfa673 : FAILURE in 27m 57s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c51fce69b5924576904d87205a437d96 : SUCCESS in 1h 08m 33s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/3de6ce52ee5b49f0bbe181c70a081a4f : FAILURE in 1h 20m 44s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/bad8714f8e1a42a0badc0609433135a3 : FAILURE in 1h 43m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1c7b3ab38eb1460eb8558d684ab1b5f3 : FAILURE in 33m 26s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/765c1f7be11a4c94a02670caf75b2cf7 : SUCCESS in 1h 19m 59s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a1b85e20ddf04a70a4aac3d18868e9ff : FAILURE in 2h 51m 02s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1432a4550a6240fc8d5e26b2c2b700b8 : FAILURE in 32m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ace41347d99493394f9c0db30957d3b : SUCCESS in 7m 11s","accounts_in_message":[],"_revision_number":1},{"id":"7ef45df1ff95285777bab028011cf1a25ca92a54","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-08-12 17:55:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ceaf7a9422628eba01c74fa90ea859a5c4427dba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-12 21:17:45.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/30870d6cf0954e418645d95ef86bc51a\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/120e0a2ff5c04360b07082ad9653a4e8 : FAILURE in 21m 11s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/4462ab91d8d84db1a438b1363a6c5b45 : SUCCESS in 1h 07m 07s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/db313ee166514f9c9961a0121159c681 : SUCCESS in 1h 17m 48s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5d6a0b7a0579476797e14e569a95f6e3 : SUCCESS in 58m 14s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/50a7f48d9cce400991fdccdadd58d5e0 : SUCCESS in 1h 11m 10s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c0207778853e4d26bf009240a06d1889 : SUCCESS in 1h 25m 26s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6c0744922e0946aabbcec0acfaf05c18 : SUCCESS in 1h 27m 16s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c16af48fb4364fa9add8abaf087e434c : SUCCESS in 1h 04m 45s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/340d7bdcf26d4ff7acb39acbf2d06116 : SUCCESS in 3h 00m 48s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9740737b951d4dbaa17b98e518c73ceb : FAILURE in 23m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca2257cc097b43e3adcb8936923a59b8 : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":1},{"id":"a10fd61bd6d2b5fe45e18d816a21c2a53053b4a6","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-10-21 07:55:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"997c2587bb06b07f5e2a1f773caa80c463a2dc83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-21 10:08:29.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/212159c7d8d44f9a83ce5393571945fa\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/8053b7de0ef24dd0ab5ec1b4bf5aa13e : FAILURE in 7m 28s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a19a115d5ed445d38b0e72152fc06fc3 : SUCCESS in 1h 02m 15s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c04576d3507841af8403b71271987091 : SUCCESS in 52m 13s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/92273a02df5340399abf3978217793c2 : SUCCESS in 55m 07s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/93db9a403bf045dc9b04005beb2aa1f2 : SUCCESS in 1h 10m 01s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3deb3e8f04d340978e2d9e2f48d5aa09 : SUCCESS in 1h 22m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e4b76a32229041cdad5b441546665bc5 : SUCCESS in 1h 22m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa578d13208d4b3d846a693456f8c075 : SUCCESS in 1h 08m 52s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8cbf2684571e4e7fb580b73f181eea3b : SUCCESS in 1h 58m 37s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ab8c6405edd147edb2efc39199c9998f : SUCCESS in 1h 32m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88a90a3d5eec44e2ae32af3947658b48 : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":1},{"id":"38aa7b9e442392176abd3cd2140a0a961d099de3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-10-21 15:57:14.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"9780d89591dce23701f523c0b0c3ffed501c45cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-21 18:25:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6675f9cb790e4bd7bb203dcead32a000\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/42c1fd958a7b4700bbc855b3d902397f : SUCCESS in 8m 02s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e788fda795e34ff9ae4980c3049f1e72 : SUCCESS in 1h 09m 35s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f87af608f01648babd7e2b4483fb5c86 : SUCCESS in 57m 01s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1e918db2a1014b269a83e5e2170e3add : SUCCESS in 55m 30s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/cfa4b3167a464cc9a80279cce9bb9d93 : SUCCESS in 1h 05m 37s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/2afc98910e3c4ce29d5ebe7d20f9642c : SUCCESS in 1h 26m 25s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/10bdaf496371432b9d8e58da7004f879 : SUCCESS in 1h 09m 43s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8fc299af624a4089a27aa63e99ee9115 : SUCCESS in 1h 06m 53s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/61a906bff5e54e9cb826e1b960c70633 : SUCCESS in 2h 24m 03s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6adf4645e07d47d08282e69629ded966 : SUCCESS in 1h 56m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6907d41659d4e29b91b79136f014788 : SUCCESS in 5m 28s","accounts_in_message":[],"_revision_number":2},{"id":"18aa0aa574e1122b4b42602ea1fea613ce5da844","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-10-24 08:01:22.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1e745466b7f380c07476138c8cf55ac0a9a2afce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-24 08:01:41.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5bae2da9d3d54010828e4ac3cd9fd927\n\n\nWarning:\n  Change 849521 in project openstack/openstack-ansible does not share a change queue with 849522 in project openstack/openstack-ansible-os_mistral","accounts_in_message":[],"_revision_number":2},{"id":"75ac7547df26d413491fd1bbbb8636d584a838df","author":{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},"date":"2022-10-24 10:24:14.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"9be53fab78bbb938325ba11a8656908b6abb338a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-10-24 10:34:17.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fbf64bb69bc629d5732a8faf9e0f6a1fe2af3dc7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-24 13:12:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c42fabcb15142ad9d21baf5290ecfbf\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/71c0488d7a6b4bc2821496e61abbc442 : SUCCESS in 6m 45s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a55cfe857fcd4884b26f660ac692a6a2 : SUCCESS in 1h 04m 42s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/15d9b9a918c14316ba0c93472041481e : SUCCESS in 58m 19s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/211fe5b0a34e4b2d82cb8ad5ae773ab1 : SUCCESS in 58m 11s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/a2858aa297664caa905036bb222cb3e3 : SUCCESS in 1h 10m 09s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3415a82c411b4098b99ad09518ff67b1 : SUCCESS in 1h 19m 08s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f0ffc00c52ac4c9e926b1822f9929baa : SUCCESS in 1h 10m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6a44c0e02e1a484b9ea462a684411e51 : SUCCESS in 1h 10m 03s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/78857fdb7b2940fc87dfb84158606518 : SUCCESS in 2h 28m 07s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a44a37c19299462f919dc60e5d1b8c8d : SUCCESS in 1h 58m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efcae3ee750648dab5e98880b9c1d09c : SUCCESS in 5m 54s","accounts_in_message":[],"_revision_number":2},{"id":"0155cf3e4b94774b59cfb208ccd377f60eb9e8b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-24 13:14:37.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ebf31fd1efd5ea3a023df2dcc6b527168c988692","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-24 14:50:15.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/59e74072ae5a447d9c4dbcb511f727ac\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/353731bb5ec44e43867ec45477328119 : SUCCESS in 7m 59s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/4e2808e0ea5b4a439172e7f532e16220 : SUCCESS in 1h 04m 46s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/247820f980214980825bc68a2734597b : SUCCESS in 59m 59s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4370d081209641539e429cfd45b55453 : SUCCESS in 53m 05s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/ee6bf0e0f4ad4459a8c388078f8833bb : FAILURE in 20m 28s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7555773fe64c466b91c238dcdc9d2576 : SUCCESS in 1h 30m 26s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/08a77d552bbc4b1da30881755a843f58 : SUCCESS in 1h 10m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ecc206f444a4411ea832d2ab5c87cf9f : SUCCESS in 1h 05m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/004d8033f6b84cf9bdd56deb34007542 : SUCCESS in 6m 10s","accounts_in_message":[],"_revision_number":2},{"id":"55d6d673b9f1da5469e2ba129a96651ed2ee642e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-10-24 15:14:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ef7356cf4b18cc0d471d0ef281df068daf9dc364","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-24 17:46:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/df831926bd2a498e9001d32102f79dc3\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f4e289d8b58b4ccba5f3147b93fa7f65 : SUCCESS in 7m 35s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3cb175f7fe8a44ca8a6098a6f8d4a9f6 : SUCCESS in 1h 05m 52s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4f849105b7a646df866b16287c90b32d : FAILURE in 20m 30s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5bfc1d9ed4804727ac9fb7397d80d48e : FAILURE in 21m 42s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/474026fd8d214641955fc286f49fe352 : FAILURE in 21m 45s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8d6c84fb5bf54288964ba6a1bcb62007 : SUCCESS in 1h 39m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cc04fb2058bc442ca1792041636169dd : FAILURE in 21m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ac08d0d8cfe84ef6806408598af6c7d3 : FAILURE in 22m 12s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/57fecf168ba84005978410ad28cd30a9 : SUCCESS in 2h 26m 25s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e66eb37819d64132bf4a08cf6e920b45 : SUCCESS in 2h 01m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ded0f45adbe54febb8eca86c2d5dce65 : SUCCESS in 5m 52s","accounts_in_message":[],"_revision_number":2},{"id":"5e906f2991f4e792dacea384c239fa219a317e09","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-10-25 08:52:44.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"b1f4d4d3f08f227bf33e12428fc9fd9f42ae6a4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-25 11:09:23.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cb123466afb640689c0cd49584dff596\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/110ca6bc37f4424b83eee229c250c4fa : SUCCESS in 9m 08s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e68b585f720442dcb21a3a75030641da : FAILURE in 28m 06s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e6bf56e7a3b640f091b483aa8f35b0d8 : SUCCESS in 56m 39s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/10a3fa9b384d4740b0d9d430f9c7b14d : SUCCESS in 1h 04m 15s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/1459f07a4e38496db4fe5b4debdc11ad : SUCCESS in 1h 12m 34s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d6c56850b64c4869ba2b34bcbb7b5dcb : FAILURE in 21m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/74238e8cc2134e72a071c7161eab7c51 : SUCCESS in 1h 21m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8bf1a2485350438c96f328dcb74da48d : SUCCESS in 1h 09m 24s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b78c160bbf8e45f8b1610ce4ff84105a : FAILURE in 20m 52s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5c38f300e290490dab75e2e875b210a3 : SUCCESS in 2h 14m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90a7a02cde524d6cbf7678159044ad02 : SUCCESS in 5m 41s","accounts_in_message":[],"_revision_number":3},{"id":"bca630c24dc4416df3fca293ef87ba368ba4dee2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-01-03 07:59:33.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"5d05386f10ca5a94dfbdd2b857ff1bc8133237ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-03 10:09:54.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/00fae9a9e36f4f9c80fa960a179aeb06\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f4e68562419744e79d937ae3fd0f6848 : SUCCESS in 6m 42s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e31452b120584abf9c394122dda4a323 : FAILURE in 19m 28s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5fd5b40069304c809c1f89f01d34e48a : SUCCESS in 1h 03m 37s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7c9b0d83842c43ccbd1f34b9903d075b : SUCCESS in 49m 10s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/adb32f61c38a48f69ae31ce9290be27c : SUCCESS in 1h 03m 19s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b73c68cbe5b0496380256c671a32881b : FAILURE in 29m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fec4b75df2ff4ec0a3d333ef292ad6f2 : SUCCESS in 1h 08m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3bef22195b2042a190138dce06464823 : SUCCESS in 59m 22s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/37ab6be5e4464b91808ea6c1c28af229 : FAILURE in 26m 11s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cfe2ff4a93dd49fa81f14f5fe70afcb4 : SUCCESS in 1h 57m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c78cb365b1ef453eaa5b8a6811661604 : SUCCESS in 5m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/90e9537fcba6452bb86f786bd0c3aa77 : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":4},{"id":"dcf2b8fb4d9efff509012cd40127b8014b5b01d3","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-01-03 16:02:30.000000000","message":"Abandoned\n\nU has been EOL\u0027ed, so patch can\u0027t be merged.","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"5d333b8ac81b5eb651bd3fd6adb713d476d32b24","revisions":{"82ca4aa8da2790c85911f5b054527d7fc161c6b5":{"kind":"REWORK","_number":1,"created":"2022-07-12 18:58:11.000000000","uploader":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"ref":"refs/changes/22/849522/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_mistral","ref":"refs/changes/22/849522/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/22/849522/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/22/849522/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/22/849522/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/22/849522/1"}}},"commit":{"parents":[{"commit":"00f80b9fa09cbb455ed139a927accc1fe86802cb","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/00f80b9fa09cbb455ed139a927accc1fe86802cb"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2022-05-31 12:22:36.000000000","tz":-240},"committer":{"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","date":"2022-07-12 18:58:11.000000000","tz":0},"subject":"Add mistral-extra in the mistral venv","message":"Add mistral-extra in the mistral venv\n\nMistral has split the openstack action from the main repo, this patch\ninstall the extra actions into the venv so we can have openstack-action\nin it.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/849521\nChange-Id: I03aaecd4286535ae80aca249778ea9b209b87129\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/82ca4aa8da2790c85911f5b054527d7fc161c6b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/82ca4aa8da2790c85911f5b054527d7fc161c6b5"}]},"branch":"refs/heads/stable/ussuri"},"8aa3048170e3d98d1d1a13e034f2148e0e56889b":{"kind":"REWORK","_number":2,"created":"2022-10-21 15:57:14.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/22/849522/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_mistral","ref":"refs/changes/22/849522/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/22/849522/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/22/849522/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/22/849522/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/22/849522/2"}}},"commit":{"parents":[{"commit":"00f80b9fa09cbb455ed139a927accc1fe86802cb","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/00f80b9fa09cbb455ed139a927accc1fe86802cb"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2022-05-31 12:22:36.000000000","tz":-240},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-10-21 15:57:13.000000000","tz":0},"subject":"Add mistral-extra in the mistral venv","message":"Add mistral-extra in the mistral venv\n\nMistral has split the openstack action from the main repo, this patch\ninstall the extra actions into the venv so we can have openstack-action\nin it.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/849521\nChange-Id: I03aaecd4286535ae80aca249778ea9b209b87129\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/8aa3048170e3d98d1d1a13e034f2148e0e56889b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/8aa3048170e3d98d1d1a13e034f2148e0e56889b"}]},"branch":"refs/heads/stable/ussuri"},"1afc9d65fb181d132c448783ec5d4e6c00242acd":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-10-25 08:52:44.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/22/849522/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_mistral","ref":"refs/changes/22/849522/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/22/849522/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/22/849522/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/22/849522/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/22/849522/3"}}},"commit":{"parents":[{"commit":"00f80b9fa09cbb455ed139a927accc1fe86802cb","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/00f80b9fa09cbb455ed139a927accc1fe86802cb"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2022-05-31 12:22:36.000000000","tz":-240},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-10-25 08:52:44.000000000","tz":0},"subject":"Add mistral-extra in the mistral venv","message":"Add mistral-extra in the mistral venv\n\nMistral has split the openstack action from the main repo, this patch\ninstall the extra actions into the venv so we can have openstack-action\nin it.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-rabbitmq_server/+/862564\nChange-Id: I03aaecd4286535ae80aca249778ea9b209b87129\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/1afc9d65fb181d132c448783ec5d4e6c00242acd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/1afc9d65fb181d132c448783ec5d4e6c00242acd"}]},"branch":"refs/heads/stable/ussuri","description":"Edit commit message"},"5d333b8ac81b5eb651bd3fd6adb713d476d32b24":{"kind":"REWORK","_number":4,"created":"2023-01-03 07:59:33.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/22/849522/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_mistral","ref":"refs/changes/22/849522/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/22/849522/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/22/849522/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/22/849522/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/22/849522/4"}}},"commit":{"parents":[{"commit":"00f80b9fa09cbb455ed139a927accc1fe86802cb","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/00f80b9fa09cbb455ed139a927accc1fe86802cb"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2022-05-31 12:22:36.000000000","tz":-240},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-01-03 07:59:33.000000000","tz":0},"subject":"Add mistral-extra in the mistral venv","message":"Add mistral-extra in the mistral venv\n\nMistral has split the openstack action from the main repo, this patch\ninstall the extra actions into the venv so we can have openstack-action\nin it.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-rabbitmq_server/+/862564\nChange-Id: I03aaecd4286535ae80aca249778ea9b209b87129\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/5d333b8ac81b5eb651bd3fd6adb713d476d32b24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/5d333b8ac81b5eb651bd3fd6adb713d476d32b24"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
