)]}'
{"id":"openstack%2Fopenstack-ansible-os_mistral~849575","triplet_id":"openstack%2Fopenstack-ansible-os_mistral~stable%2Fxena~I03aaecd4286535ae80aca249778ea9b209b87129","project":"openstack/openstack-ansible-os_mistral","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"13095":{"account":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"last_update":"2022-07-21 02:47:52.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2022-07-20 22:37:42.000000000","reason":"\u003cGERRIT_ACCOUNT_25023\u003e replied on the change","reason_account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}},"hashtags":[],"change_id":"I03aaecd4286535ae80aca249778ea9b209b87129","subject":"Add mistral-extra in the mistral venv","status":"MERGED","created":"2022-07-12 18:36:40.000000000","updated":"2022-07-21 02:49:14.000000000","submitted":"2022-07-21 02:47:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"849575","meta_rev_id":"39275b6065fa848bd1e86ddd667e466240c566cd","_number":849575,"virtual_id_number":849575,"owner":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-21 02:47:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2022-07-19 15:28:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-07-19 16:13:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-07-19 16:13:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-12 20:31:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-18 11:05:22.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"CC"},{"updated":"2022-07-19 15:28:26.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2022-07-19 16:13:36.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"e0b6821fb2a16469af55d07ac5da32037c07fddb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-07-12 18:36:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"43ec89ddc78b3f369bc5c633a4b98e971ab635d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 20:31:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0ac9fc83c4bc458a8734d9d6d97f2ba9\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bc92b2cf2abe4238bea10832ffa73bd7 : SUCCESS in 19m 03s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/c4f856dd518d45feb15e189f811e5262 : FAILURE in 34m 00s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/83d9b6b2dff944a8a1828deda0f7e9ce : FAILURE in 1h 04m 35s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/4b580e1bcd8b4fa59ab5fef618b7b4ca : FAILURE in 1h 47m 29s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/694f53d14ebc4a94bd1288e0489d6dcf : FAILURE in 1h 15m 20s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/ab0cd53124104f9f9ea67cdbf67740ec : FAILURE in 1h 35m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aee6a14548734259bd9b4ef09a2e9183 : FAILURE in 1h 44m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/93cc9bea2e5f4fefabbcc24cc1d11cd5 : FAILURE in 1h 32m 15s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/f5c27713cfc6486c8bde2efc27c1b01c : FAILURE in 1h 28m 59s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0ad98f09848e4e4d97e4e34a45a10afb : FAILURE in 1h 01m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fe19a54c2c849629bad13b90c715321 : SUCCESS in 6m 50s","accounts_in_message":[],"_revision_number":1},{"id":"0aa14f849eb4bf2b0f4aeb8b29ca4851e4ca68c8","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-07-18 11:05:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3066d82de8ac484006923a525d79799b26b37d32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 13:20:18.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/307f40a45885402a9cd058801deb39c0\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/792c7e418bb44bee9b284ff2208d81fb : SUCCESS in 13m 59s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/5c9d731cf7484cd2998d287325b02d0f : FAILURE in 30m 05s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c4f78346d8f94e9481147532a3a72648 : SUCCESS in 1h 06m 54s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/8a8f172457324fee8cb22417858fe517 : SUCCESS in 1h 40m 07s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/3c9d5a9aa3ac434291ca13dd1dacb586 : SUCCESS in 1h 19m 30s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/5ae7ae233ae84cba983c1e9db7cb2510 : FAILURE in 34m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/91e872cf15d54ac4a08f619bc7561539 : SUCCESS in 1h 43m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0878f1e9d0fc4d56832bf2ed963cf1d6 : SUCCESS in 1h 33m 29s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/b0cf9e111ca3424db56bf6842b7f1f04 : FAILURE in 1h 31m 24s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3ca64828714e46bf91d7560a30ae56ab : SUCCESS in 2h 03m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/202a32f4320a4c2f860cdf65756c3d4a : SUCCESS in 9m 20s","accounts_in_message":[],"_revision_number":1},{"id":"86d0608d49aac86a2316d6e4692847fd1042759a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-07-19 08:56:20.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"c6df10b4b055fbb985350267f9045f95c17ae2c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 11:18:07.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/68c8862cb483410b919261575a363c44\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/518adb4e59a145948d7df0ae1fd37318 : SUCCESS in 19m 36s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/1fdab257f1a1491c8a452d34de2a86f6 : FAILURE in 35m 50s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6900609694f943998c9ba3f7c671d959 : FAILURE in 29m 51s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/990076ca6212498fa0198e80d99f296d : SUCCESS in 1h 33m 34s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/bd91d36b660241c9a777206e0af55af6 : SUCCESS in 1h 23m 16s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/1858108e461549e3bce48bb49ff43aa5 : SUCCESS in 1h 35m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/79318fd5a6d749aaae2608537f2bcda5 : SUCCESS in 1h 31m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c18a91b2a409453ca1a1215004b301d7 : SUCCESS in 1h 26m 25s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/0dd5dca4a1ef491cb7c0d624190255b4 : SUCCESS in 2h 06m 27s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6fa0f65024024ced8b987a2b49a045b6 : SUCCESS in 1h 53m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12ac3625f2a44617b9c67b567c8a4b9d : SUCCESS in 7m 33s","accounts_in_message":[],"_revision_number":2},{"id":"2999aefc57548f09c5c4896d1123673d6ebab76d","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-07-19 11:31:38.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"03930628fea988cb0941bab3d0daffa03984afca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 14:20:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41b1d717a04042dd9e7f295fc3ae4dd2\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6d067bf54cff4a52aa0045fadc59e5a1 : SUCCESS in 19m 36s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/5eef92be9b854a1898d22658ce32e64c : FAILURE in 34m 27s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ef0d4e9dbee94381811caa7744d9890e : SUCCESS in 1h 11m 10s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/0722b701c3d24d439d95e4a4a185fd41 : SUCCESS in 1h 41m 26s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/169b645bb7dd4624a3b2b2c8f8eec9f8 : SUCCESS in 1h 14m 25s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/4ddd3b69f889444d802a0cd9b54f1d8b : SUCCESS in 1h 33m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c60de97aaacb4af2b99e00525df549bf : SUCCESS in 1h 37m 29s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ebcd1f1db0d7497b9c18f72e340275bc : SUCCESS in 1h 46m 05s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/4bf8106c79e1476f80a99a350aef72c2 : SUCCESS in 2h 39m 40s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d62f96fc4fbb4681b1e53de70004191f : SUCCESS in 2h 08m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98cc31bf1f234db194a19d9e167e5c46 : SUCCESS in 6m 45s","accounts_in_message":[],"_revision_number":2},{"id":"83468b6fd2d4c73a02d54c7f7089fa2f7cb2ed7f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-07-19 15:28:26.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3f5441bb307b716a0c72cfee9a35fb839c9e5c13","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-07-19 16:13:36.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8ecbe656461b5d8d877666fc65d9970e56ab5e9b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 16:13:50.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/05b30136272249b1a702010c96aa83bb\n\n\nWarning:\n  Change 849574 in project openstack/openstack-ansible does not share a change queue with 849575 in project openstack/openstack-ansible-os_mistral","accounts_in_message":[],"_revision_number":2},{"id":"b20b4a78a4236bda17825e104b1b9ff9e3d993b5","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-07-20 08:02:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1056520e2e1900a75ae154bbad8293e57ec01963","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 10:05:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82f9d9947583476da97b06ad37190b27\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/01cd17af6eb3446d893883e053ae430a : SUCCESS in 18m 47s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/9c9334d594034382878fdfdec8a8d746 : FAILURE in 30m 31s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6fa8bdceb13c4fe58571f78d88a8e58c : SUCCESS in 51m 23s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/1bfe9026ee8047019c79628059c1a221 : SUCCESS in 1h 23m 59s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/a11de82f2b314699a3f4b0d106018bc6 : SUCCESS in 1h 05m 43s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/0019d34d128c48399709b31498870210 : SUCCESS in 1h 17m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b8468cc97ac34b7499d865523fe929c6 : SUCCESS in 1h 22m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d06ebb45ad1144d287b62f3f7a70032f : SUCCESS in 1h 15m 49s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/a4bcb5b2aea9400089c611a5e4107932 : SUCCESS in 1h 58m 50s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bc613226849d4f58aa3ecf05ac2ea9df : SUCCESS in 1h 48m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ee1f30e797d4591bb74b8faefe2d6c0 : SUCCESS in 5m 23s","accounts_in_message":[],"_revision_number":2},{"id":"d62236e4c2ec016a631ac74046f575df7459ab51","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 10:07:37.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b8561d26d186c76290be77fc024fbabe4b508186","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 12:00:29.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9eac9a77f78d4282b690116381594796\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2686829d78054bc296f35f8f2b21e18c : SUCCESS in 38m 19s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cc09b2b4846a4072ad897d701572191e : FAILURE in 40m 31s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/48d4d20a791247c4a461047c61efacdf : SUCCESS in 1h 45m 35s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/6501bd3430504d2abcb872de6866bfea : SUCCESS in 1h 30m 26s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/3ec7486d1d0847b38c4942add6b9d6f2 : SUCCESS in 1h 41m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b3260cd4e9a547388e15c0ea42133b7a : SUCCESS in 1h 52m 01s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4639b2c1fc264149b8eaf1fb39531ecc : SUCCESS in 1h 41m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6e48f94bd0746659119ed237160911e : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":2},{"id":"e87e3df3e801f88d2b4254ea5866de5c6013296c","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-07-20 22:37:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9cc47f4d891e7fcc2879b3491b1d4135b0e661f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 01:14:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/189a9525c243495e95ec7765e8146c00\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7e5773f6c822483cbd128db7674cd732 : SUCCESS in 11m 03s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/128f3b65fdec4106b496945e21da8726 : FAILURE in 29m 56s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/308378dd1d424904b87d1ea84740d55c : SUCCESS in 52m 02s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/a5e61923fd0648dc8c30fe0d3dc717db : SUCCESS in 1h 33m 56s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/9faf5ffb588d45968daa11ac3bd5daf9 : SUCCESS in 1h 14m 31s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/69b73089149947d98daae798de97c660 : SUCCESS in 1h 36m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f46e7f72c5db4b1c8c086ef8ad2901d3 : SUCCESS in 1h 38m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/58a3cc1bdc4a4f91a45c35947bf9f31c : SUCCESS in 1h 24m 07s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/242f80f35df548588f4b94962c7d4c89 : SUCCESS in 2h 28m 53s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8ebcb3e537224315ba545a094d506908 : SUCCESS in 2h 06m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2eaa20ec625e4908b41df52c6200af4b : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":2},{"id":"a479bdcdaa7aca0381c0882f788f1bc1b6112100","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 01:15:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b93829e019807b3bb3821c7f72d616e8fba46c96","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 02:47:52.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d255ed89f3e4e01b639cd506a960856\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/43d9b71b0d1d427aba7870daa2c28449 : SUCCESS in 35m 20s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2eecf955329d43c497d833a6a47f71ab : SUCCESS in 54m 57s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/b063a8c3188941229d94e6cd1b23ebee : SUCCESS in 1h 26m 21s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/4b6c0a2a90044edb80017a8f4e844332 : SUCCESS in 1h 05m 55s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/a353e94713544c4fa53cfb644e8e0b87 : SUCCESS in 1h 27m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/314ca68cd20c4b8e86960c183af0215d : SUCCESS in 1h 31m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/03fcfcb6086143408427a82a2f3a0d41 : SUCCESS in 1h 25m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0995babccdeb4f5b8853ddd144a3ba1a : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":2},{"id":"4d75e1d826d71ac58292baf576bb57aace2485c0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 02:47:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"39275b6065fa848bd1e86ddd667e466240c566cd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 02:49:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3eab9ccc70e4d969d0168e80266372c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/333c53af98484b25bf9d8eb63ae13267 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"58c7abd44e8eca98dc30915de06a67a255d35ab3","revisions":{"e5eb5050115fb9fb7e9db1e01476b0798b52d8aa":{"kind":"REWORK","_number":1,"created":"2022-07-12 18:36:40.000000000","uploader":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"ref":"refs/changes/75/849575/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_mistral","ref":"refs/changes/75/849575/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/75/849575/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/75/849575/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/75/849575/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/75/849575/1"}}},"commit":{"parents":[{"commit":"f3fbe6f52fe204a5cbd62ee64b7ada5ce2279d98","subject":"Merge \"Updated from OpenStack Ansible Tests\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/f3fbe6f52fe204a5cbd62ee64b7ada5ce2279d98"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2022-05-31 12:22:36.000000000","tz":-240},"committer":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2022-07-12 18:36:40.000000000","tz":-240},"subject":"Add mistral-extra in the mistral venv","message":"Add mistral-extra in the mistral venv\n\nMistral has split the openstack action from the main repo, this patch\ninstall the extra actions into the venv so we can have openstack-action\nin it.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/849574\nChange-Id: I03aaecd4286535ae80aca249778ea9b209b87129\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/e5eb5050115fb9fb7e9db1e01476b0798b52d8aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/e5eb5050115fb9fb7e9db1e01476b0798b52d8aa"}]},"branch":"refs/heads/stable/xena"},"58c7abd44e8eca98dc30915de06a67a255d35ab3":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-07-19 08:56:20.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/75/849575/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_mistral","ref":"refs/changes/75/849575/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/75/849575/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/75/849575/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/75/849575/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/75/849575/2"}}},"commit":{"parents":[{"commit":"f3fbe6f52fe204a5cbd62ee64b7ada5ce2279d98","subject":"Merge \"Updated from OpenStack Ansible Tests\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/f3fbe6f52fe204a5cbd62ee64b7ada5ce2279d98"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2022-05-31 12:22:36.000000000","tz":-240},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-07-19 08:56:20.000000000","tz":0},"subject":"Add mistral-extra in the mistral venv","message":"Add mistral-extra in the mistral venv\n\nMistral has split the openstack action from the main repo, this patch\ninstall the extra actions into the venv so we can have openstack-action\nin it.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/849574\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-rabbitmq_server/+/850233\nChange-Id: I03aaecd4286535ae80aca249778ea9b209b87129\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/58c7abd44e8eca98dc30915de06a67a255d35ab3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/58c7abd44e8eca98dc30915de06a67a255d35ab3"}]},"branch":"refs/heads/stable/xena","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
