)]}'
{"id":"openstack%2Fopenstack-ansible-os_mistral~849716","triplet_id":"openstack%2Fopenstack-ansible-os_mistral~master~I25b622de40c02f3cc410070e04e2d5d0fd67abc9","project":"openstack/openstack-ansible-os_mistral","branch":"master","topic":"mistral_wsgi_or_cron_triggers","attention_set":{},"removed_from_attention_set":{"34653":{"account":{"_account_id":34653,"name":"Danila Balagansky","email":"dbalagansky@me.com","username":"dbalagansky"},"last_update":"2022-07-13 22:04:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I25b622de40c02f3cc410070e04e2d5d0fd67abc9","subject":"Allow running Mistral API service without uWSGI","status":"MERGED","created":"2022-07-13 15:56:20.000000000","updated":"2022-07-13 22:05:48.000000000","submitted":"2022-07-13 22:04:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"849716-mistral_wsgi_or_cron_triggers","meta_rev_id":"17222a5e4eb4a0bdbfbf03684d45da6d9b50fea5","_number":849716,"virtual_id_number":849716,"owner":{"_account_id":34653,"name":"Danila Balagansky","email":"dbalagansky@me.com","username":"dbalagansky"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-13 22:04:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2022-07-13 18:11:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-07-13 20:32:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-07-13 20:32:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2022-07-13 18:12:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-13 18:11:06.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2022-07-13 18:19:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-13 20:32:05.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"c3e36ffc0bf5e8af2e91fa60269b6af098d734b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34653,"name":"Danila Balagansky","email":"dbalagansky@me.com","username":"dbalagansky"},"date":"2022-07-13 15:56:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9e88953a22b98c84e559fdc4f6a739be1456020a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-07-13 18:11:06.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b3e7b121c8ed328426f7212bc5c58a7a0014d5a3","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-07-13 18:12:19.000000000","message":"Patch Set 1: Backport-Candidate+1","accounts_in_message":[],"_revision_number":1},{"id":"a37564118a0d7d8513db9036b177b34787389e58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 18:19:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20c7add4e4c9439c923be053d01dab23\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8842aac219b948539ae6d6c24e13d23b : SUCCESS in 18m 22s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/6b8cdf22ae5045d78abb0f9556b2f12a : FAILURE in 34m 25s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ff910a41a5a64a5583b2f64004afd8c3 : FAILURE in 33m 23s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/d6a4bcb766c24d9093e723e61580dacf : SUCCESS in 1h 21m 18s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/d651849e4e4d41c0bfe1c11af16dbb77 : SUCCESS in 1h 30m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e7972c58baf74e188f1e373bf6f2d045 : SUCCESS in 1h 31m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1baaecc63ea24cb88b5e6af414d7171c : SUCCESS in 1h 29m 19s (non-voting)\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/cf3bcb78f892403abb5bfa8c088814bd : SUCCESS in 2h 03m 05s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/820cfeb540df4601b1e1d0d5d7675a23 : SUCCESS in 2h 17m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95e1bb7852af47acacffa9916b7eeafc : SUCCESS in 5m 48s","accounts_in_message":[],"_revision_number":1},{"id":"dd6df64af7f5ef1908c6284aba67cfcb233437f3","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-07-13 20:32:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"078f27847a39e6279a40e9b9d235b2418e643cf8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 20:34:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6efc7fd9fbf27377594c1706566b939b37256aad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 22:04:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7bd626fc7d9e4bbf8d1c2886fa804ac4\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d83519cd6ff547b088a8974ec6b68c4f : SUCCESS in 23m 24s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/f87f444beb5b4328a96cc632e1949163 : SUCCESS in 1h 18m 07s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/c035fe60322b4f7e92e737c7258bcf20 : SUCCESS in 1h 21m 21s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/be88b54fca6a4a149e68ca20b2fc4f1e : SUCCESS in 1h 29m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4edd97f6cdc476c8a47e5993ae0dffc : SUCCESS in 5m 31s","accounts_in_message":[],"_revision_number":1},{"id":"8f2423273023404a44d14225594f44b48d4b4bfd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 22:04:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"17222a5e4eb4a0bdbfbf03684d45da6d9b50fea5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 22:05:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9865f9a01e44489492d4c4ef4af92ff6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2ef8900dcb84ea4a774118a4c5b104d : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"05b37e554247bf4b128fd017bf7792e705a7802f","revisions":{"05b37e554247bf4b128fd017bf7792e705a7802f":{"kind":"REWORK","_number":1,"created":"2022-07-13 15:56:20.000000000","uploader":{"_account_id":34653,"name":"Danila Balagansky","email":"dbalagansky@me.com","username":"dbalagansky"},"ref":"refs/changes/16/849716/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_mistral","ref":"refs/changes/16/849716/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/16/849716/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/16/849716/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/16/849716/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/16/849716/1"}}},"commit":{"parents":[{"commit":"fd223ea205128f2e59da77b757414e513e1e5a50","subject":"Merge \"Add mistral-extra in the mistral venv\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/fd223ea205128f2e59da77b757414e513e1e5a50"}]}],"author":{"name":"Danila Balagansky","email":"dbalagansky@me.com","date":"2022-07-13 15:48:04.000000000","tz":180},"committer":{"name":"Danila Balagansky","email":"dbalagansky@me.com","date":"2022-07-13 15:48:04.000000000","tz":180},"subject":"Allow running Mistral API service without uWSGI","message":"Allow running Mistral API service without uWSGI\n\nWhile running Mistral API service within uWSGI, Mistral\u0027s Cron\nTriggers do not create Workflow Exectuions. This change would disable\nCron Triggers for all existing installations as running API service\nwithin uWSGI stays enabled by default.\n\nChange-Id: I25b622de40c02f3cc410070e04e2d5d0fd67abc9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/05b37e554247bf4b128fd017bf7792e705a7802f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/05b37e554247bf4b128fd017bf7792e705a7802f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[]}
