)]}'
{"id":"openstack%2Fopenstack-ansible-os_mistral~849804","triplet_id":"openstack%2Fopenstack-ansible-os_mistral~master~Ic6829cf816a3924fbaaa5b86422bfbc3851afc95","project":"openstack/openstack-ansible-os_mistral","branch":"master","topic":"mistral_api_use_uwsgi_relnotes","attention_set":{},"removed_from_attention_set":{"34653":{"account":{"_account_id":34653,"name":"Danila Balagansky","email":"dbalagansky@me.com","username":"dbalagansky"},"last_update":"2022-07-25 16:13:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic6829cf816a3924fbaaa5b86422bfbc3851afc95","subject":"Add release notes for mistral_api_use_uwsgi","status":"MERGED","created":"2022-07-14 10:49:25.000000000","updated":"2022-07-25 16:14:38.000000000","submitted":"2022-07-25 16:13:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"849804-mistral_api_use_uwsgi_relnotes","meta_rev_id":"9c752dcb70e5d922d9b79574ccc02ac5143686ae","_number":849804,"virtual_id_number":849804,"owner":{"_account_id":34653,"name":"Danila Balagansky","email":"dbalagansky@me.com","username":"dbalagansky"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-25 16:13:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2022-07-15 13:03:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-07-25 15:48:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-07-25 15:48:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-14 11:03:01.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2022-07-14 11:12:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-25 15:48:26.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"260046959869a5f950edd1dd1c81ab03a79a39af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34653,"name":"Danila Balagansky","email":"dbalagansky@me.com","username":"dbalagansky"},"date":"2022-07-14 10:49:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6a638d11be22266bed0f49895c15e6e9c4e5bab7","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-07-14 11:03:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"24e07be12016adf2e30e75fe39dbc1be57fb08a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 11:12:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a99c4daceae94f6a90d2981d7e1a4ce1\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fea1b76cc02947b09ac28d0ccc71a6da : SUCCESS in 20m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f45cd9eb3b0544358ca6334e4347b6a2 : SUCCESS in 5m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/898416e6424c4b2c80f0ad29348f2742 : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":1},{"id":"da15806be4fbeb0a73cc6e919b5d312d5f989c35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34653,"name":"Danila Balagansky","email":"dbalagansky@me.com","username":"dbalagansky"},"date":"2022-07-14 13:52:17.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"02bf4175e9281d36198c27851deba8e245b0ca5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 14:08:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31ef40e4c99341df8a8cf5126305ece6\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/277538da29be4e0f9edd488c5c14a9a4 : SUCCESS in 10m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60395eb4fdd444bf952821ec25e6d692 : SUCCESS in 4m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/53bb4500a04848c29ab740dc9adc853d : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":2},{"id":"a6b4bc932cca922d2cb683452b7ea7adf5bd5041","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-07-15 13:03:53.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c1dfab4a767c817c3d0e403b16a27b53937c2829","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-07-25 15:48:26.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"cd7e74e9972a4d7dcefad5dde4909c152180a922","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 15:51:05.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a17b8b07ced5f0341442c9eee813d68017e625d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 16:13:20.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/942b0dcd23934107ab24494513734cd2\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/297106ef36db4de6bfb8a2b9c5a0b6cc : SUCCESS in 21m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc36f55d108e469fa82f3a253b3015f8 : SUCCESS in 6m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c3aa7f2380a44a7ea2070ac65b555ee8 : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":2},{"id":"4282785a2595f98d8abf0cc7bed6ff7fb1b58eac","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 16:13:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"9c752dcb70e5d922d9b79574ccc02ac5143686ae","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 16:14:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d76ad3f1c62147f1bcfc515e1b9e5ade\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94124f6cd6624aa3a45d26a331b3cfd3 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"cc9b8443a61137e126acf8ba12aedd28885a0d1f","revisions":{"f0c152b883ba5a206a1b686782849f34698c8489":{"kind":"REWORK","_number":1,"created":"2022-07-14 10:49:25.000000000","uploader":{"_account_id":34653,"name":"Danila Balagansky","email":"dbalagansky@me.com","username":"dbalagansky"},"ref":"refs/changes/04/849804/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_mistral","ref":"refs/changes/04/849804/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/04/849804/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/04/849804/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/04/849804/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/04/849804/1"}}},"commit":{"parents":[{"commit":"05b37e554247bf4b128fd017bf7792e705a7802f","subject":"Allow running Mistral API service without uWSGI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/05b37e554247bf4b128fd017bf7792e705a7802f"}]}],"author":{"name":"Danila Balagansky","email":"dbalagansky@me.com","date":"2022-07-14 10:38:42.000000000","tz":180},"committer":{"name":"Danila Balagansky","email":"dbalagansky@me.com","date":"2022-07-14 10:48:07.000000000","tz":180},"subject":"Add release notes for mistral_api_use_uwsgi","message":"Add release notes for mistral_api_use_uwsgi\n\nDepends-on: https://review.opendev.org/c/openstack/openstack-ansible-os_mistral/+/849716\nChange-Id: Ic6829cf816a3924fbaaa5b86422bfbc3851afc95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/f0c152b883ba5a206a1b686782849f34698c8489"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/f0c152b883ba5a206a1b686782849f34698c8489"}]},"branch":"refs/heads/master"},"cc9b8443a61137e126acf8ba12aedd28885a0d1f":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-07-14 13:52:17.000000000","uploader":{"_account_id":34653,"name":"Danila Balagansky","email":"dbalagansky@me.com","username":"dbalagansky"},"ref":"refs/changes/04/849804/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_mistral","ref":"refs/changes/04/849804/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/04/849804/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/04/849804/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/04/849804/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/04/849804/2"}}},"commit":{"parents":[{"commit":"05b37e554247bf4b128fd017bf7792e705a7802f","subject":"Allow running Mistral API service without uWSGI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/05b37e554247bf4b128fd017bf7792e705a7802f"}]}],"author":{"name":"Danila Balagansky","email":"dbalagansky@me.com","date":"2022-07-14 10:38:42.000000000","tz":180},"committer":{"name":"Danila Balagansky","email":"dbalagansky@me.com","date":"2022-07-14 13:52:04.000000000","tz":180},"subject":"Add release notes for mistral_api_use_uwsgi","message":"Add release notes for mistral_api_use_uwsgi\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_mistral/+/849716\nChange-Id: Ic6829cf816a3924fbaaa5b86422bfbc3851afc95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/cc9b8443a61137e126acf8ba12aedd28885a0d1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/cc9b8443a61137e126acf8ba12aedd28885a0d1f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
