)]}'
{"id":"openstack%2Fopenstack-ansible-os_mistral~918128","triplet_id":"openstack%2Fopenstack-ansible-os_mistral~master~Ic5714dc3d6cd42ce1745c2e2044caf812dc94022","project":"openstack/openstack-ansible-os_mistral","branch":"master","topic":"osa/messaging_improvements","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2024-05-13 10:54:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic5714dc3d6cd42ce1745c2e2044caf812dc94022","subject":"Implement variables to address oslo.messaging improvements","status":"MERGED","created":"2024-05-03 18:28:58.000000000","updated":"2024-05-13 10:55:39.000000000","submitted":"2024-05-13 10:54:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"918128-osa/messaging_improvements","meta_rev_id":"d885095b449be416e608f731f25fd83cfd7c53cf","_number":918128,"virtual_id_number":918128,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-13 10:54:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2024-05-13 09:06:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-05-13 09:04:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2024-05-13 09:06:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-03 21:14:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-09 07:14:54.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"CC"},{"updated":"2024-05-13 09:04:20.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2024-05-13 09:06:34.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"87d2e4bd6c779fe965e1bf9fd30f23384197b55e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-05-03 18:28:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6850ac64cad314d78265c02781563e954f47f657","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-03 21:14:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a7a4e86180744e78a36e423dec871f97\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/73c3ead04c714d3eb6fbaf96e4cba960 : SUCCESS in 11m 18s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7e390192df314a199380d445b27c618b : SUCCESS in 1h 15m 03s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/21799e074f594903acd0d6fd312641fb : SUCCESS in 1h 08m 54s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/51431f3ea9714ff58b818a655171e17d : SUCCESS in 1h 07m 24s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/a6d08e5f7fe44d92a9dfa44f22837f1a : SUCCESS in 1h 24m 48s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/aa17fdefde7d4b37a9d9dbbe73ae25be : SUCCESS in 1h 25m 16s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/385b64a943fd46d9b68941b5b33108bd : SUCCESS in 1h 28m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eeedc747b52943de8f8c34fb4f1a2fd3 : SUCCESS in 1h 27m 57s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/556503a522a542fca072aef1b6c276e2 : SUCCESS in 1h 35m 36s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a801aebb9e954191bb6e6570d644a826 : SUCCESS in 1h 21m 16s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a41e97d5b6634e12a09bb5897e4fc9f4 : SUCCESS in 1h 24m 23s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c269d30991cd475c9cd51828b317176a : SUCCESS in 2h 13m 04s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bcaa853eda2e456684b404bb7fa1b6e5 : FAILURE in 43m 58s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c437bb012d6f4048a56e6fa83f6e4325 : SUCCESS in 2h 07m 16s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6506c83c078d432fa19fbb2bfa8b10e9 : SUCCESS in 2h 28m 32s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a54333a60c054efca244dd5598974817 : SUCCESS in 2h 17m 19s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/70991f845e294f10bff89f7680c8f149 : SUCCESS in 1h 53m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a514876f198411fa9dbbc801a7e0884 : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":1},{"id":"2b7209a477b6b8d7135717d13ca1a63f47f3620f","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-05-09 07:14:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c59d0a0fe9193f974ad193ed1d15f5f528053bfc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 10:14:50.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/efc964e2c9f54152bd1ee1763cda1e07\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c8c36f29e127485a97fcdb8567d4ac48 : SUCCESS in 13m 07s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3a4e4a4724334a1ea919ccbf6b08b6eb : FAILURE in 26m 39s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5709cc558eb542faa8e869cafb9907a5 : SUCCESS in 1h 07m 32s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/e5bf1bdd4de749a39d5c09cdc8480e50 : SUCCESS in 1h 20m 08s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/0cf49f460e1b43118517b02bc07c0562 : SUCCESS in 1h 22m 01s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/c9d139e944284bdfb4c9dc652909a591 : SUCCESS in 1h 11m 36s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/be82ace236d24d24a08b817eedddf532 : SUCCESS in 1h 28m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/86710f14c654431a8387ca35ebf54bf0 : SUCCESS in 1h 17m 26s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d292a1af1bfb4a5b8c9245289029d539 : SUCCESS in 1h 38m 17s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ac2325ef2bb74d4fbcac2de0e0fee088 : SUCCESS in 1h 21m 06s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7dc456e94b9f49058dd8d3916dd4661f : SUCCESS in 1h 18m 41s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e34a1d0023644af2b7a260520ad550aa : SUCCESS in 1h 58m 27s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/90f348ab12f940de98317e9cbb338092 : SUCCESS in 2h 39m 29s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/126bdf89cea141ba805432648dddc3fa : FAILURE in 1h 53m 52s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/de918fab1490499fb883d5888683bc83 : SUCCESS in 2h 17m 26s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/88613ea518694514b5909923fce85286 : SUCCESS in 2h 21m 24s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/03ba4fa0b0a84ef485135f9a4ea8ab05 : SUCCESS in 1h 55m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21337e80e13c4a9a82facc849217c470 : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":1},{"id":"d7e8829a996b8bd1dddc742f0b5c9087e33aae5b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-05-10 17:26:37.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"899a708ba27d849018626cce2517c4ef3157b8cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-10 20:18:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/829bd278b7fa467e90ba9dd1d75cb86a\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/98f8127c9cbf4c5cbeec43415ee2bc00 : SUCCESS in 19m 41s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c096ab16986d452192ae7c6478de7976 : FAILURE in 26m 17s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e370082f92d5425195fb1a60be41b08f : SUCCESS in 57m 28s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/1c52d2064ead4f84b3149d34dea9e731 : SUCCESS in 1h 33m 00s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/5ee6334d017e468c8b6dd40d30b205df : SUCCESS in 1h 37m 26s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/ba009e927abf425bba3a90d56fd5a7d2 : SUCCESS in 1h 37m 40s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b95cd83164d24440a28fa3f9dcfa9bd7 : SUCCESS in 1h 27m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5a3d6dde1a464dcb9b35542225d8507c : SUCCESS in 1h 24m 12s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f9290684a13642b89c0bd250fdaf79ae : SUCCESS in 1h 32m 30s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/454fdc08b5bd49a0991a42e0d1685885 : SUCCESS in 1h 36m 44s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1275f29bd6db450bb84e806377d8009e : SUCCESS in 1h 24m 19s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/483c0242f4bf4f10ba1d794371741d46 : SUCCESS in 2h 07m 11s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/80871c3d85b24270990d75e2e4d76162 : SUCCESS in 2h 44m 36s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/245480170fd64a2d8fdff108ed40bdba : SUCCESS in 1h 55m 10s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2268d2a163094949b7f07084fcc65024 : SUCCESS in 2h 42m 43s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/74703720e5df4cfd96f1e65023b375d2 : SUCCESS in 2h 17m 49s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dc8f58bcd0bc44c4a712102c019e39ba : SUCCESS in 1h 43m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3797e727a4b84417b30ffc09d4a57610 : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":2},{"id":"33ebbdfccca325ecaba5826da62f7a0b26839a59","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-05-13 09:04:20.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"76fa378b46856ceec5a644a962fc0c9fb9d0e38f","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2024-05-13 09:06:34.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a358b3da80857e70a8113f5dbe31d778951a95eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 09:08:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"30a913f297ba918c0ee5cf2f3bf47216d43db99c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 10:54:36.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b450244fd5748649efd94771c7a0900\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/148bf6b96dc541cab8e2fede84b828fb : SUCCESS in 13m 14s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/18b7a35ac69c4dffa1548e28259a4ba0 : SUCCESS in 1h 12m 10s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/677e9f9c303749aaa14408a3c8d399d1 : SUCCESS in 1h 20m 22s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/289ae056ac1e4e06822615f258a31cf5 : SUCCESS in 1h 40m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1643a94c55854af3937b39435a99a4fc : SUCCESS in 1h 24m 13s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7fa6d27d27ed4d479d4da0955a463ff8 : SUCCESS in 1h 29m 28s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ef21ba75992d40cca2e20102e93f9f1e : SUCCESS in 1h 38m 20s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/224aa3a9e6864978a67b67ac3bc00d95 : SUCCESS in 1h 40m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/737918acb453437eaff9e073f608095a : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":2},{"id":"48add9388aa6a3c1d4e23a44f4d9fcfd5c259d48","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 10:54:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"d885095b449be416e608f731f25fd83cfd7c53cf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 10:55:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03519a71036e4971b3c668d8e5e7eebf\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bda70059e0574c13806a70e4e5b90515 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7b89942398defc65835d6a9579c88cd3d87a40d2","revisions":{"d5382d4d046bb4b1c17dc935dcba4faa0fdf1300":{"kind":"REWORK","_number":1,"created":"2024-05-03 18:28:58.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/28/918128/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_mistral","ref":"refs/changes/28/918128/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/28/918128/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/28/918128/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/28/918128/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/28/918128/1"}}},"commit":{"parents":[{"commit":"d770855d414f173d864baccead6d55a8422e0459","subject":"Add variable to globally control notifications enablement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/d770855d414f173d864baccead6d55a8422e0459"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-05-03 18:28:48.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-05-03 18:28:48.000000000","tz":120},"subject":"Implement variables to address oslo.messaging improvements","message":"Implement variables to address oslo.messaging improvements\n\nDuring last release cycle oslo.messaging has landed [1] series of extremely\nuseful changes that are designed to implement modern messaging\ntechniques for rabbitmq quorum queues.\n\nSince these changes are breaking and require queues being re-created,\nit makes total sense to align these with migration to quorum queues by default.\n\n[1] https://review.opendev.org/q/topic:%22bug-2031497%22\n\nChange-Id: Ic5714dc3d6cd42ce1745c2e2044caf812dc94022\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/d5382d4d046bb4b1c17dc935dcba4faa0fdf1300"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/d5382d4d046bb4b1c17dc935dcba4faa0fdf1300"}]},"branch":"refs/heads/master"},"7b89942398defc65835d6a9579c88cd3d87a40d2":{"kind":"REWORK","_number":2,"created":"2024-05-10 17:26:37.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/28/918128/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_mistral","ref":"refs/changes/28/918128/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/28/918128/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/28/918128/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/28/918128/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/28/918128/2"}}},"commit":{"parents":[{"commit":"d770855d414f173d864baccead6d55a8422e0459","subject":"Add variable to globally control notifications enablement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/d770855d414f173d864baccead6d55a8422e0459"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-05-03 18:28:48.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-05-10 17:26:32.000000000","tz":120},"subject":"Implement variables to address oslo.messaging improvements","message":"Implement variables to address oslo.messaging improvements\n\nDuring last release cycle oslo.messaging has landed [1] series of extremely\nuseful changes that are designed to implement modern messaging\ntechniques for rabbitmq quorum queues.\n\nSince these changes are breaking and require queues being re-created,\nit makes total sense to align these with migration to quorum queues by default.\n\n[1] https://review.opendev.org/q/topic:%22bug-2031497%22\n\nChange-Id: Ic5714dc3d6cd42ce1745c2e2044caf812dc94022\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/7b89942398defc65835d6a9579c88cd3d87a40d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/7b89942398defc65835d6a9579c88cd3d87a40d2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
