)]}'
{"id":"openstack%2Fopenstack-ansible-os_mistral~950513","triplet_id":"openstack%2Fopenstack-ansible-os_mistral~master~I8e576e17ed76f71a8d8bc0cc69193935966083a7","project":"openstack/openstack-ansible-os_mistral","branch":"master","topic":"osa-wsgi-module","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-06-02 17:09:34.000000000","reason":"Change was submitted"},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-06-02 17:09:34.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2026-06-02 17:09:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8e576e17ed76f71a8d8bc0cc69193935966083a7","subject":"Switch from wsgi script to wsgi module","status":"MERGED","created":"2025-05-21 10:14:46.000000000","updated":"2026-06-02 17:12:16.000000000","submitted":"2026-06-02 17:09:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"950513-osa-wsgi-module","meta_rev_id":"9544d83dd92e647c93a91e91ceadc5b52a4fd9b6","_number":950513,"virtual_id_number":950513,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-02 17:09:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":2,"date":"2026-06-02 15:12:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":2,"date":"2026-06-02 15:07:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":1,"date":"2026-06-02 15:12:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-21 12:37:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-22 11:59:28.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2025-06-07 22:49:56.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2026-05-31 10:38:38.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-06-02 15:12:36.000000000","updated_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"}],"messages":[{"id":"549878ef5cffde28007cfdd152392b693efb417b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-05-21 10:14:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"97eacedafac97bd52847d06a116d51f9057f6a63","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-05-21 10:46:40.000000000","message":"Topic set to osa-wsgi-module","accounts_in_message":[],"_revision_number":1},{"id":"8374103654534f2308436826f8c3e51fff317290","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 12:37:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/060126326ea84f96b8e8c866bf53fe6f\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4c838cde7f3843e5ac1244fc5acee761 : SUCCESS in 7m 09s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f2cb13de587743e6892eff9ba6df3820 : SUCCESS in 1h 03m 18s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/298e9f575ceb465e8861a96c901b5b49 : SUCCESS in 1h 05m 55s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/cd904a3b2cf84186805a57995b7a33a7 : SUCCESS in 1h 19m 42s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/0a51086ca13c4a2d8d9a584deeec6aa3 : SUCCESS in 1h 23m 22s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/56046d10087747b3b06d6dc2b81a9d76 : SUCCESS in 1h 26m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/16bfed14a25e4efb992b46d5c4d4312f : SUCCESS in 1h 23m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/00cb97033f08443c9a0b590835d24234 : SUCCESS in 39m 28s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/561277f251604fff8089e8cbdb13b749 : SUCCESS in 42m 06s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e1143f9e8f634396b3edfe896ee5d6d9 : SUCCESS in 1h 19m 59s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/584c0598be784d2283aeaaf51c9ebc59 : SUCCESS in 1h 29m 17s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5eb3586ca9164f539727c8f7055eba0d : SUCCESS in 2h 12m 46s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/929e5ab3e44246fc97e78495b5599cae : SUCCESS in 2h 17m 24s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7a27ee9bb9a5436ca251e6d0f6853a41 : SUCCESS in 1h 51m 59s\n- openstack-ansible-upgrade_2024.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2e1c2ce444134e248b24db3c4a9f8502 : SUCCESS in 1h 54m 00s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/531bb7caac104db7ad26601afd504ff3 : SUCCESS in 2h 02m 39s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a0d56981b4824d009c02bc52cd9d5d3f : SUCCESS in 1h 55m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/775da3897c0e48468529fe6e9d424528 : SUCCESS in 5m 21s","accounts_in_message":[],"_revision_number":1},{"id":"92811f656ca0235bf82904496ce4281631ad1dd6","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-22 11:59:28.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e57b7909c83d532fd0a1a6d8689f76eb05043253","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-05-22 18:38:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5445c65422131dfd21621e29ad3112e98612a39c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 20:47:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84f47328248b40b9b95b4a340a1fac4f\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eaa0e68ee0ab4a36980dcb091f96ecb8 : SUCCESS in 5m 45s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/16daf48e2c1f4fbbb37a6519c708b3fb : FAILURE in 55m 01s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/3a8524063c9147c3b13d8ccf87fe024e : SUCCESS in 1h 27m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/628778028a504162b45cfc087c190549 : SUCCESS in 1h 23m 29s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ef46c5a5b5874443a8c60a7e3bf3f069 : SUCCESS in 1h 13m 09s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9bf45f96e480421081ae693b148de470 : SUCCESS in 1h 53m 25s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d76c1c51cec54132aed0064392b7ab8d : SUCCESS in 2h 06m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d20b996ddb85425199a68560bbae3706 : SUCCESS in 7m 02s","accounts_in_message":[],"_revision_number":1},{"id":"8f2e7f319bf241104d4e1c226b4608c6b4157303","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-06-05 21:34:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0a5cbca6dc83c03f54c477e45d48462975ae99da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"real_author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-06-10 15:28:09.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_25023\u003e\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"_revision_number":2},{"id":"54ddf13f3c1d4114ae52d9742a1712fbb4103098","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-10 17:59:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5bd039b22b494b26b12a1a9c4f47286e\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/83dfe57c16dd4593bc16b21666a25021 : SUCCESS in 7m 42s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/86b457279c8e403398301e63481966ea : FAILURE in 1h 01m 40s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/2f93dbb5ad094624b726c835963a0993 : SUCCESS in 1h 27m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1e917455870b4ae2952e3015fbd18a60 : SUCCESS in 1h 23m 15s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/115c9aacd9fb4c689306af6fca746c91 : SUCCESS in 1h 34m 03s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c02655accfb7443f9d5418eec2fd1296 : SUCCESS in 2h 09m 41s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/33b5dc97ce7f4fcb8d016e3ee328eb5b : SUCCESS in 2h 02m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcfcf2f0742449d4815ebc051def5ae1 : SUCCESS in 5m 28s","accounts_in_message":[],"_revision_number":2},{"id":"46a375668711ab0236c29de1301eec2b72b5ba40","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-06-27 08:23:26.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fe67cb9561c471181b8330a40aa0219be6d4ea37","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-07-17 07:58:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f87050d5e776677c65ab4ec9dbe8187d7fec6084","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 09:05:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d511fe202dd45d6995fe01e462ca68e\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/955bdeab73c249718235649e24ff1d1c : SUCCESS in 6m 06s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/786d70f2e29e4677bb6cf99784f59a8f : FAILURE in 59m 33s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/9270235efbbb4f209ae8dfd7d2a66a7e : SUCCESS in 40m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/be0269bdf7e64cd6be96be08aaf67cf3 : SUCCESS in 44m 13s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d9eff3eb89d948c8964a5d50be738354 : SUCCESS in 42m 44s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a3c36bc6f03641ad82343c7616d20b50 : SUCCESS in 55m 08s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/55be38191f8b402f8df65041f46f14d7 : SUCCESS in 1h 01m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ff3259c779845398993a90adb3ad79c : SUCCESS in 5m 57s","accounts_in_message":[],"_revision_number":2},{"id":"58b3ac26daf5723575e8f250b7f85b3ffa28e37e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-11-10 09:58:38.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5a4095e43e523218cdde95894380408d63af0c74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-10 12:18:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42275e00cca04b4886f2f3872da2ab18\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3a3f22727ec44f91a21a49ae52fc58f8 : SUCCESS in 5m 32s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ed438d1c8a8e48009080532890adb457 : FAILURE in 7m 59s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/75d048ac2a244e7d890af354055146c8 : FAILURE in 55m 01s (non-voting)\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/cbab58848da74a1cbabaa2d07129a490 : SUCCESS in 1h 05m 24s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/01ef1fbae1c540bdbe1e25093a829325 : SUCCESS in 1h 21m 30s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/b63483bfa39e49e39d313bd3abe000c0 : SUCCESS in 1h 21m 30s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/2d00e31d3b1a4bf48377cb236a62a211 : SUCCESS in 1h 08m 47s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d44830d4ab1347d8b12a8aa5789c81cd : SUCCESS in 1h 32m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1e712724afcc41fb97f92e7f609a74f4 : SUCCESS in 1h 23m 12s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a3ca54c1605d4629b6dab5fee2bd2c15 : SUCCESS in 1h 27m 21s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/df0dba33b73b4dce955ba9b2231b4a3a : SUCCESS in 2h 04m 19s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d4b22ccb4d7942feb606995645dba5b0 : SUCCESS in 2h 06m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22dabda873b54bacbe6bf00443e87d9f : SUCCESS in 6m 54s","accounts_in_message":[],"_revision_number":2},{"id":"38c0576b8c680eb663728c0bd345bedb5ddbbc38","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-11-10 13:20:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7c0bd02db62627213ef50431cdbc1b51fe74eebf","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-29 23:16:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b2d88432c7a2546d5cd16964bafa10d6b79c8569","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-29 23:17:31.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"0bed2f1d677c5eb67c8b0978acfc13eacb72fe6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-29 23:17:37.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"c48505107d07a7e9f6d7e2d0868d2380adef2e3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 01:55:14.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9e3b86141ef34cb5a42ed6a8438be6bc\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a0183ac1dc814574854d107ecc91d359 : SUCCESS in 12m 46s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6ed8e71a13f041c288af8b69ccfcf0ff : FAILURE in 8m 50s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/78b340b7dddd4ebd98843923af3b12b0 : FAILURE in 54m 17s (non-voting)\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/0935d92e3b2a466188c70cd1b54eaafe : SUCCESS in 59m 35s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/13f7b7630c754ac49a5ee50ba0dd2f75 : SUCCESS in 59m 21s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/624da237b22d4442b7a4b026a9682fa8 : SUCCESS in 1h 31m 36s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ed54b3ac13834a9baab775a194729ebc : SUCCESS in 51m 40s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/fd964d695c614804bc107a67bfdbbc3b : SUCCESS in 53m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c951d439d7434e1d80066978d1dadd0f : SUCCESS in 1h 41m 14s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9b6bb6d7874540f7b8f19d62b36e9d1e : SUCCESS in 1h 46m 40s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a38d4907ebd248349459a1552d8cab39 : SUCCESS in 1h 22m 25s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/88306412fe6a4bddbabffacdff68db23 : SUCCESS in 2h 04m 15s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6d1926b4eae44a5c87fbc1c210df9c13 : FAILURE in 34m 41s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9b4c850ce07b40048493918f12ed55fb : SUCCESS in 2h 28m 48s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c7d0fc29c52345efa892d30608122596 : SUCCESS in 1h 27m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42d8fd1ea91544df98df77741181be2b : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":4},{"id":"a254b5f567b29427b47b1de4500bc9cd67ac7b8a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-30 20:23:30.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"f797c38177a951a71b08d773ad28ba51196a313e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-30 22:55:04.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dca023b937754467a74fc47ed0b040ef\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/491433cdab00492dab809e7118622820 : SUCCESS in 14m 53s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8aeff047963a46d1a2305558a320a143 : FAILURE in 1h 06m 06s (non-voting)\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/3ec84890e8ed434a80682be266687c73 : SUCCESS in 51m 38s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/0fd89cd279854dc8977672d4c94243ae : SUCCESS in 1h 33m 39s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/a9e33ad3fa4848499410e07ba3f167ba : SUCCESS in 1h 32m 12s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/cdbf949b555f4fb294cb902541dbabfb : SUCCESS in 1h 27m 00s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/5efbeeacebd847e689c05faba91d6dc4 : SUCCESS in 1h 20m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/73ffb0662422454a9390e3e804ce0630 : SUCCESS in 1h 35m 06s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0e07312fc70142c78dd9082653d6e75a : SUCCESS in 1h 35m 42s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e7c195d245a645459d9ba1d67e72ffef : SUCCESS in 1h 49m 04s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/64c9730c03b741af97c5e7b5b5d50f73 : SUCCESS in 1h 08m 52s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/20d124436e8c4012bf16fb25a46c162b : SUCCESS in 2h 19m 36s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cd367c9e77864063857c5097db3c67bd : SUCCESS in 2h 20m 37s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a826e465cda0424980eedf759ac1a43f : SUCCESS in 2h 21m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2b263f442b64899ac0a070c25a7f314 : SUCCESS in 6m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/84a86594301a45feaa320558581e1a52 : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":5},{"id":"f335ee52ee8089d026fdbcfe39ae4043478cac23","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-31 08:04:05.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"faa6650afae924a8c4cb80f0f48297a31cd58766","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-31 10:38:38.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"9ca8ef947df476c772549cbedb74d92fc2f3c0f8","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-06-01 12:07:10.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"71c5954081495e7e3f2d857f1f77f74a2ee80f80","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-06-01 12:15:35.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"b0ccf071ee805833e1e9427bcf9d1eeea6ea9c99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 15:06:43.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f291b5fb900c482ab2102fd032700fab\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9fb95ad277a94fdeaa466b37bb69a06a : SUCCESS in 11m 12s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e98395c61e7c4e9baf0044d4d02f1332 : FAILURE in 1h 02m 41s (non-voting)\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/f4f900060c144634823628bc8cbdf04a : SUCCESS in 1h 34m 33s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/06a9177d019d4abcb2a545ae2c16dc86 : SUCCESS in 1h 36m 49s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/08f6179d9d2349a99eb35f178581d745 : SUCCESS in 1h 32m 55s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/db9645011c06450fb1c3508e8ec3b054 : SUCCESS in 1h 26m 56s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/42a297d4608f4e1ba540e06e1c9c4801 : SUCCESS in 1h 23m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ef2dcb8dab41404786bf8beb011d7f49 : SUCCESS in 1h 44m 02s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/54ec27bfb7e246dcae159766207766d7 : SUCCESS in 1h 45m 15s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/fa166255fec44406b43f61517f87704d : SUCCESS in 1h 15m 18s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/da142a922a75445ba8a0bcdeff32e7dc : SUCCESS in 2h 01m 48s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6978b9859e48479582466f889012643b : SUCCESS in 2h 31m 18s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6b690b4cfe344161bbfc2ec94ee19fd1 : SUCCESS in 2h 35m 54s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/31fa86979ab745c6984e791e80339688 : SUCCESS in 2h 28m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7b8f323d4f343208e80f2b541096f8a : SUCCESS in 3m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fd539501c0954176bb9fe4d6c5eee9d1 : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":6},{"id":"79bd210d75bfd15373e5fd545b384f777befe2df","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-06-02 15:07:12.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"e3caf25fdafbd1995cd73d70a3c526a840612d61","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2026-06-02 15:12:36.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"957eb991ae58f33470719279acf6f08da8bef32c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 15:15:43.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"6fc269cbf09958c0ab735dfae467888fc6406d01","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 17:09:34.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2856633f7af480a8309b27054ff82a8\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6b97b3330b794ac69906fb46a1d89a1a : SUCCESS in 15m 17s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/584eb5abc59c4f56b6ccc6731ed1ff39 : SUCCESS in 1h 37m 25s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/86bf764391204023a18858703e583c5f : SUCCESS in 1h 39m 09s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/f92f7e59ade7423d8993161d038b814c : SUCCESS in 1h 34m 14s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b0a70fec854d441dbc74b0502ee0ac69 : SUCCESS in 1h 30m 31s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ece0c9fb4bc043a5b56af343a1373c5b : SUCCESS in 1h 51m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/24c56394fbe046abb3d46ff590bc11df : SUCCESS in 1h 41m 03s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/92a850a306a34af7a439f349eee41692 : SUCCESS in 1h 41m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a2d4b5c4ad64367a33557c62e99f16b : SUCCESS in 9m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dfb2afb5fa8842be8da3733117f5b9b3 : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":6},{"id":"7aa16243052d9b85b5e07d12bccf3c4ba04c6723","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 17:09:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"9544d83dd92e647c93a91e91ceadc5b52a4fd9b6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 17:12:16.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ced306e9b9bf419f96f9dfa1871bd380\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf789a2ce4074008a0a70f77c5e32118 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"913e670b307065d3089e2b0d0e35f0d7c11fc890","revisions":{"486a8d677298fa7e141ab123a92d57ea160d0d89":{"kind":"REWORK","_number":1,"created":"2025-05-21 10:14:46.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/13/950513/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_mistral","ref":"refs/changes/13/950513/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/13/950513/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/13/950513/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/13/950513/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/13/950513/1"}}},"commit":{"parents":[{"commit":"fa40c1b6ffe777e04ba0997a7917072c8c97fd6c","subject":"Remove tags from README and small fix with markings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/fa40c1b6ffe777e04ba0997a7917072c8c97fd6c"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2025-05-21 10:14:27.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2025-05-21 10:14:27.000000000","tz":60},"subject":"Switch from wsgi script to wsgi module","message":"Switch from wsgi script to wsgi module\n\nChange-Id: I8e576e17ed76f71a8d8bc0cc69193935966083a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/486a8d677298fa7e141ab123a92d57ea160d0d89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/486a8d677298fa7e141ab123a92d57ea160d0d89"}]},"branch":"refs/heads/master"},"2e46afa50325473ffbc2cb107ce61ff8f593c73b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-06-10 15:28:09.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"real_uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/13/950513/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_mistral","ref":"refs/changes/13/950513/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/13/950513/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/13/950513/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/13/950513/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/13/950513/2"}}},"commit":{"parents":[{"commit":"efbceec3a5ba0babfe853defc607ea648578509d","subject":"Replace deprecated UPPER_CONSTRAINTS_FILE variable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/efbceec3a5ba0babfe853defc607ea648578509d"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2025-05-21 10:14:27.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2025-06-10 15:28:09.000000000","tz":0},"subject":"Switch from wsgi script to wsgi module","message":"Switch from wsgi script to wsgi module\n\nChange-Id: I8e576e17ed76f71a8d8bc0cc69193935966083a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/2e46afa50325473ffbc2cb107ce61ff8f593c73b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/2e46afa50325473ffbc2cb107ce61ff8f593c73b"}]},"branch":"refs/heads/master","description":"Rebase"},"52858c85d70f56d6690bc1445a1c1077904657cd":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-03-29 23:17:31.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/13/950513/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_mistral","ref":"refs/changes/13/950513/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/13/950513/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/13/950513/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/13/950513/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/13/950513/3"}}},"commit":{"parents":[{"commit":"efbceec3a5ba0babfe853defc607ea648578509d","subject":"Replace deprecated UPPER_CONSTRAINTS_FILE variable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/efbceec3a5ba0babfe853defc607ea648578509d"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2025-05-21 10:14:27.000000000","tz":60},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-03-29 23:17:31.000000000","tz":0},"subject":"Switch from wsgi script to wsgi module","message":"Switch from wsgi script to wsgi module\n\nChange-Id: I8e576e17ed76f71a8d8bc0cc69193935966083a7\nSigned-off-by: Jonathan Rosser \u003cjonathan.rosser@rd.bbc.co.uk\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/52858c85d70f56d6690bc1445a1c1077904657cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/52858c85d70f56d6690bc1445a1c1077904657cd"}]},"branch":"refs/heads/master","description":"Edit commit message"},"d65be9bbccfe7c6062737efadc3c0a23f6f4eb8e":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-03-29 23:17:37.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/13/950513/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_mistral","ref":"refs/changes/13/950513/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/13/950513/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/13/950513/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/13/950513/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/13/950513/4"}}},"commit":{"parents":[{"commit":"024652289e2c339911a142896c3989181f45fa46","subject":"Add hashi_vault pki backend support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/024652289e2c339911a142896c3989181f45fa46"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2025-05-21 10:14:27.000000000","tz":60},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-03-29 23:17:37.000000000","tz":0},"subject":"Switch from wsgi script to wsgi module","message":"Switch from wsgi script to wsgi module\n\nChange-Id: I8e576e17ed76f71a8d8bc0cc69193935966083a7\nSigned-off-by: Jonathan Rosser \u003cjonathan.rosser@rd.bbc.co.uk\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/d65be9bbccfe7c6062737efadc3c0a23f6f4eb8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/d65be9bbccfe7c6062737efadc3c0a23f6f4eb8e"}]},"branch":"refs/heads/master","description":"Rebase"},"4751d5e7c94e4c13bb2d45ff1fca13e59c192672":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-05-30 20:23:30.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/13/950513/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_mistral","ref":"refs/changes/13/950513/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/13/950513/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/13/950513/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/13/950513/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/13/950513/5"}}},"commit":{"parents":[{"commit":"c03617599a092d8b5cd6cf7a6cb3c405532498a4","subject":"Use correct libxml package name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/c03617599a092d8b5cd6cf7a6cb3c405532498a4"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2025-05-21 10:14:27.000000000","tz":60},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-30 20:23:30.000000000","tz":0},"subject":"Switch from wsgi script to wsgi module","message":"Switch from wsgi script to wsgi module\n\nChange-Id: I8e576e17ed76f71a8d8bc0cc69193935966083a7\nSigned-off-by: Jonathan Rosser \u003cjonathan.rosser@rd.bbc.co.uk\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/4751d5e7c94e4c13bb2d45ff1fca13e59c192672"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/4751d5e7c94e4c13bb2d45ff1fca13e59c192672"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"d65be9bbccfe7c6062737efadc3c0a23f6f4eb8e","theirs":"c03617599a092d8b5cd6cf7a6cb3c405532498a4","contains_conflicts":false}},"913e670b307065d3089e2b0d0e35f0d7c11fc890":{"kind":"REWORK","_number":6,"created":"2026-06-01 12:15:35.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/13/950513/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_mistral","ref":"refs/changes/13/950513/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/13/950513/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/13/950513/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/13/950513/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_mistral refs/changes/13/950513/6"}}},"commit":{"parents":[{"commit":"c03617599a092d8b5cd6cf7a6cb3c405532498a4","subject":"Use correct libxml package name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/c03617599a092d8b5cd6cf7a6cb3c405532498a4"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2025-05-21 10:14:27.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2026-06-01 12:15:35.000000000","tz":0},"subject":"Switch from wsgi script to wsgi module","message":"Switch from wsgi script to wsgi module\n\nChange-Id: I8e576e17ed76f71a8d8bc0cc69193935966083a7\nSigned-off-by: Jonathan Rosser \u003cjonathan.rosser@rd.bbc.co.uk\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/913e670b307065d3089e2b0d0e35f0d7c11fc890"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_mistral/commit/913e670b307065d3089e2b0d0e35f0d7c11fc890"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
