)]}'
{"id":"openstack%2Fopenstack-ansible-os_monasca-agent~660272","triplet_id":"openstack%2Fopenstack-ansible-os_monasca-agent~master~I08abf1796b0c4dc7b68dfe89795f59086113cfbf","project":"openstack/openstack-ansible-os_monasca-agent","branch":"master","topic":"fix_leap15_monasca_agent","hashtags":[],"change_id":"I08abf1796b0c4dc7b68dfe89795f59086113cfbf","subject":"Fix monasca agent for leap 15","status":"MERGED","created":"2019-05-21 07:56:46.000000000","updated":"2019-05-31 03:53:38.000000000","submitted":"2019-05-31 03:53:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660272-1559274818354-f81bca3b","meta_rev_id":"2329447c83c36d54566b377bcb4de9d9beafbec9","_number":660272,"virtual_id_number":660272,"owner":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":2,"date":"2019-05-31 03:53:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":2,"date":"2019-05-31 02:27:05.000000000","_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-21 11:37:03.000000000","_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":1,"date":"2019-05-31 02:27:05.000000000","_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-21 11:37:03.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-05-31 02:27:05.000000000","updated_by":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"reviewer":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"state":"REVIEWER"},{"updated":"2019-05-31 03:53:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4cea37b8b33a947f379af56319c034f610d247d5","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-05-21 07:56:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e73d1338e077ebda7df674b3ae068ea1f93a9450","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-05-21 08:25:18.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"a54d58cf5bd2b485b7fb756b0943dfd8acdf11b9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 09:47:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- noop noop : SUCCESS in 0s\n- openstack-ansible-linters http://logs.openstack.org/72/660272/2/check/openstack-ansible-linters/41b0277/ : SUCCESS in 5m 10s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/72/660272/2/check/openstack-ansible-deploy-aio_metal-debian-stable/90ac8c1/ : SUCCESS in 59m 44s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/72/660272/2/check/openstack-ansible-deploy-aio_metal-centos-7/c312994/ : SUCCESS in 1h 15m 26s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/72/660272/2/check/openstack-ansible-deploy-aio_metal-opensuse-150/00e11ef/ : SUCCESS in 1h 03m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/72/660272/2/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/6d68087/ : SUCCESS in 54m 21s\n- openstack-tox-docs http://logs.openstack.org/72/660272/2/check/openstack-tox-docs/f8685c7/html/ : SUCCESS in 2m 54s","accounts_in_message":[],"_revision_number":2},{"id":"eb00a9936607ef1c5d73ae6124cd2c1e3350576a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-05-21 11:37:03.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"876ba7995b70e156efdc08d22793ed0ff0e0047f","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-31 02:27:05.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9cc4b537a82217330a9a92603ae52aba7762a94e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 02:27:19.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"106d1b77439e79a72fcb4198e8a2beeb28b8eb64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 03:53:38.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- noop noop : SUCCESS in 0s\n- openstack-ansible-linters http://logs.openstack.org/72/660272/2/gate/openstack-ansible-linters/8d2731e/ : SUCCESS in 4m 02s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/72/660272/2/gate/openstack-ansible-deploy-aio_metal-debian-stable/823736a/ : SUCCESS in 54m 41s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/72/660272/2/gate/openstack-ansible-deploy-aio_metal-centos-7/2b51c51/ : SUCCESS in 1h 20m 40s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/72/660272/2/gate/openstack-ansible-deploy-aio_metal-opensuse-150/baa09a9/ : SUCCESS in 1h 03m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/72/660272/2/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/30dbf77/ : SUCCESS in 47m 33s\n- openstack-tox-docs http://logs.openstack.org/72/660272/2/gate/openstack-tox-docs/17746c4/html/ : SUCCESS in 2m 29s","accounts_in_message":[],"_revision_number":2},{"id":"75cd8e39fbd5325f8c9f1c69d93770e72fbaeeb0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 03:53:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"af52c060bddf77a0df0b6d71eeeb7171c1fc2c70","revisions":{"168feea67a389fbec8694b0877344196dc6f577e":{"kind":"REWORK","_number":1,"created":"2019-05-21 07:56:46.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/72/660272/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_monasca-agent","ref":"refs/changes/72/660272/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_monasca-agent refs/changes/72/660272/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_monasca-agent refs/changes/72/660272/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_monasca-agent refs/changes/72/660272/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_monasca-agent refs/changes/72/660272/1"}}},"commit":{"parents":[{"commit":"0258241f99003c03cc7a0f461783d2e5dca2cb6b","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_monasca-agent/commit/0258241f99003c03cc7a0f461783d2e5dca2cb6b"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-21 07:54:16.000000000","tz":120},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-21 07:56:26.000000000","tz":120},"subject":"Fix monasca agent for leap 15","message":"Fix monasca agent for leap 15\n\nThis ensures vars are properly loaded for Leap 15.\nIn order to test this actively, we\u0027ll rely on the integrated\njobs.\n\nChange-Id: I08abf1796b0c4dc7b68dfe89795f59086113cfbf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_monasca-agent/commit/168feea67a389fbec8694b0877344196dc6f577e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_monasca-agent/commit/168feea67a389fbec8694b0877344196dc6f577e"}]},"branch":"refs/heads/master"},"af52c060bddf77a0df0b6d71eeeb7171c1fc2c70":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-05-21 08:25:18.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/72/660272/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_monasca-agent","ref":"refs/changes/72/660272/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_monasca-agent refs/changes/72/660272/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_monasca-agent refs/changes/72/660272/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_monasca-agent refs/changes/72/660272/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_monasca-agent refs/changes/72/660272/2"}}},"commit":{"parents":[{"commit":"cd6016cf6f83398361614636c1b8d82801aeb98d","subject":"Drop private argument for include/import role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_monasca-agent/commit/cd6016cf6f83398361614636c1b8d82801aeb98d"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-21 07:54:16.000000000","tz":120},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-21 08:25:18.000000000","tz":0},"subject":"Fix monasca agent for leap 15","message":"Fix monasca agent for leap 15\n\nThis ensures vars are properly loaded for Leap 15.\nIn order to test this actively, we\u0027ll rely on the integrated\njobs.\n\nChange-Id: I08abf1796b0c4dc7b68dfe89795f59086113cfbf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_monasca-agent/commit/af52c060bddf77a0df0b6d71eeeb7171c1fc2c70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_monasca-agent/commit/af52c060bddf77a0df0b6d71eeeb7171c1fc2c70"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
