)]}'
{"id":"openstack%2Fopenstack-ansible-os_monasca-agent~663817","triplet_id":"openstack%2Fopenstack-ansible-os_monasca-agent~master~I73398d6e5ea2f2b4dde058f0cbb5855f36793fc1","project":"openstack/openstack-ansible-os_monasca-agent","branch":"master","hashtags":[],"change_id":"I73398d6e5ea2f2b4dde058f0cbb5855f36793fc1","subject":"Cap sphinx for py2 to match global requirements","status":"ABANDONED","created":"2019-06-07 02:04:20.000000000","updated":"2019-06-19 01:28:05.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f722b50e33a1964c9109e9717a8b7d620152aecf","_number":663817,"virtual_id_number":663817,"owner":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2019-06-07 02:21:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-07 02:21:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3bf8c3901bae7026934e9807581efbc11d84e7a0","author":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"date":"2019-06-07 02:04:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0375f09978bf89a21be49a31da53e7ff365a4bf4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 02:21:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- noop noop : SUCCESS in 0s\n- openstack-ansible-linters http://logs.openstack.org/17/663817/1/check/openstack-ansible-linters/c434b40/ : SUCCESS in 7m 00s\n- requirements-check http://logs.openstack.org/17/663817/1/check/requirements-check/fd335ad/ : SUCCESS in 4m 00s\n- openstack-tox-docs http://logs.openstack.org/17/663817/1/check/openstack-tox-docs/ecef8d0/html/ : SUCCESS in 2m 37s\n- build-openstack-releasenotes http://logs.openstack.org/17/663817/1/check/build-openstack-releasenotes/987f123/html/ : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":1},{"id":"f722b50e33a1964c9109e9717a8b7d620152aecf","author":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"date":"2019-06-19 01:28:05.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"43f3d36abde4b79c305ed2446501d9f5d76919b9","revisions":{"43f3d36abde4b79c305ed2446501d9f5d76919b9":{"kind":"REWORK","_number":1,"created":"2019-06-07 02:04:20.000000000","uploader":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"ref":"refs/changes/17/663817/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_monasca-agent","ref":"refs/changes/17/663817/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_monasca-agent refs/changes/17/663817/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_monasca-agent refs/changes/17/663817/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_monasca-agent refs/changes/17/663817/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_monasca-agent refs/changes/17/663817/1"}}},"commit":{"parents":[{"commit":"9b5ff6c85a7b20620b8e8f53028f8df7b9a92799","subject":"Merge \"Fix monasca agent for leap 15\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_monasca-agent/commit/9b5ff6c85a7b20620b8e8f53028f8df7b9a92799"}]}],"author":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2019-06-07 02:03:57.000000000","tz":480},"committer":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2019-06-07 02:04:14.000000000","tz":480},"subject":"Cap sphinx for py2 to match global requirements","message":"Cap sphinx for py2 to match global requirements\n\nChange-Id: I73398d6e5ea2f2b4dde058f0cbb5855f36793fc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_monasca-agent/commit/43f3d36abde4b79c305ed2446501d9f5d76919b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_monasca-agent/commit/43f3d36abde4b79c305ed2446501d9f5d76919b9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
