)]}'
{"id":"openstack%2Fopenstack-ansible-os_monasca-agent~796620","triplet_id":"openstack%2Fopenstack-ansible-os_monasca-agent~master~I449d3127d9a7f2b0ff541e46f6eb61f05fc4a917","project":"openstack/openstack-ansible-os_monasca-agent","branch":"master","hashtags":[],"change_id":"I449d3127d9a7f2b0ff541e46f6eb61f05fc4a917","subject":"Re-activate Monasca Agent role for Openstack-Ansible","status":"NEW","created":"2021-06-16 09:51:57.000000000","updated":"2021-06-16 11:31:20.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ae3953ffe28322a884c183e726f469f077aabb6b","_number":796620,"virtual_id_number":796620,"owner":{"_account_id":32836,"name":"Furkan Çelik","email":"furkancelik.yalova@gmail.com","username":"furkance"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-06-16 11:31:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-16 10:48:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"530f68412d1a26fe3dd8d436ef687bc88302fa27","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32836,"name":"Furkan Çelik","email":"furkancelik.yalova@gmail.com","username":"furkance"},"date":"2021-06-16 09:51:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"07bf6b04a3b95324b52eb152991a781a298bd9a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 10:48:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/5e3d78289bf142f4a26940ba5babc843 : FAILURE in 11m 50s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5388c6e894a844818e49408c74f81c7e : SUCCESS in 3m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c98e6386b6d347b1bc5eb6fa735f105e : FAILURE in 4m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ebc85d3f0d8649b28ed9c524bea8dea2 : FAILURE in 5m 31s","accounts_in_message":[],"_revision_number":1},{"id":"1b0a71aec09a97b081fe3012cb683fd3c73b745b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32836,"name":"Furkan Çelik","email":"furkancelik.yalova@gmail.com","username":"furkance"},"date":"2021-06-16 11:22:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ae3953ffe28322a884c183e726f469f077aabb6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 11:31:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/1ee1cd09eed0421e8fa3ee8de345b4b2 : FAILURE in 7m 18s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ce6024c1724e437e9f63948fc5302855 : SUCCESS in 3m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c8a30a8a9b445e5871855acb5df8964 : FAILURE in 4m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6a587a0c6637407b9dd7e80683fa65d5 : FAILURE in 3m 59s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8e0d380e05677b18aac587e1415edbbcbf04a4a2","revisions":{"03884d9aa5d331f7c9099115cb42d2fa62158c8d":{"kind":"REWORK","_number":1,"created":"2021-06-16 09:51:57.000000000","uploader":{"_account_id":32836,"name":"Furkan Çelik","email":"furkancelik.yalova@gmail.com","username":"furkance"},"ref":"refs/changes/20/796620/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_monasca-agent","ref":"refs/changes/20/796620/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_monasca-agent refs/changes/20/796620/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_monasca-agent refs/changes/20/796620/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_monasca-agent refs/changes/20/796620/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_monasca-agent refs/changes/20/796620/1"}}},"commit":{"parents":[{"commit":"56e16a7ba6ef96196417df53ee0462d9f29afd8f","subject":"Retire openstack-ansible-os_monasca-agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_monasca-agent/commit/56e16a7ba6ef96196417df53ee0462d9f29afd8f"}]}],"author":{"name":"Furkan Çelik","email":"furkancelik.yalova@gmail.com","date":"2021-06-16 09:50:49.000000000","tz":0},"committer":{"name":"Furkan Çelik","email":"furkancelik.yalova@gmail.com","date":"2021-06-16 09:51:54.000000000","tz":0},"subject":"Re-activate Monasca Agent role for Openstack-Ansible","message":"Re-activate Monasca Agent role for Openstack-Ansible\n\nI want to activate and maintain this role. Now that Monasca Agent can be installed by OSA.\n\nChange-Id: I449d3127d9a7f2b0ff541e46f6eb61f05fc4a917\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_monasca-agent/commit/03884d9aa5d331f7c9099115cb42d2fa62158c8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_monasca-agent/commit/03884d9aa5d331f7c9099115cb42d2fa62158c8d"}]},"branch":"refs/heads/master"},"8e0d380e05677b18aac587e1415edbbcbf04a4a2":{"kind":"REWORK","_number":2,"created":"2021-06-16 11:22:37.000000000","uploader":{"_account_id":32836,"name":"Furkan Çelik","email":"furkancelik.yalova@gmail.com","username":"furkance"},"ref":"refs/changes/20/796620/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_monasca-agent","ref":"refs/changes/20/796620/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_monasca-agent refs/changes/20/796620/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_monasca-agent refs/changes/20/796620/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_monasca-agent refs/changes/20/796620/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_monasca-agent refs/changes/20/796620/2"}}},"commit":{"parents":[{"commit":"56e16a7ba6ef96196417df53ee0462d9f29afd8f","subject":"Retire openstack-ansible-os_monasca-agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_monasca-agent/commit/56e16a7ba6ef96196417df53ee0462d9f29afd8f"}]}],"author":{"name":"Furkan Çelik","email":"furkancelik.yalova@gmail.com","date":"2021-06-16 09:50:49.000000000","tz":0},"committer":{"name":"Furkan Çelik","email":"furkancelik.yalova@gmail.com","date":"2021-06-16 11:22:26.000000000","tz":0},"subject":"Re-activate Monasca Agent role for Openstack-Ansible","message":"Re-activate Monasca Agent role for Openstack-Ansible\n\nI want to activate and maintain this role. Now that Monasca Agent can be installed by OSA.\n\nChange-Id: I449d3127d9a7f2b0ff541e46f6eb61f05fc4a917\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_monasca-agent/commit/8e0d380e05677b18aac587e1415edbbcbf04a4a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_monasca-agent/commit/8e0d380e05677b18aac587e1415edbbcbf04a4a2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
