)]}'
{"id":"openstack%2Fopenstack-ansible-os_monasca~654523","triplet_id":"openstack%2Fopenstack-ansible-os_monasca~stable%2Fstein~I05c395a1847edc9666676f95fa7c61d4c134dd12","project":"openstack/openstack-ansible-os_monasca","branch":"stable/stein","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I05c395a1847edc9666676f95fa7c61d4c134dd12","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-04-22 15:39:06.000000000","updated":"2019-04-22 16:25:03.000000000","submitted":"2019-04-22 16:25:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654523-1555950303309-11ac1cc1","meta_rev_id":"83a1aec46816af4d0c4be550ee0a27f8e7e02205","_number":654523,"virtual_id_number":654523,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-04-22 16:25:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2019-04-22 16:13:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2019-04-22 16:13:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-22 16:13:18.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-04-22 16:25:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6a11ecb966a0af588a4cba57f116079951eca8dd","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-04-22 15:39:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d0246b51f6eda5c18292ae56b9be5f0f15dc2b49","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 16:05:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- noop noop : SUCCESS in 0s\n- openstack-ansible-linters http://logs.openstack.org/23/654523/1/check/openstack-ansible-linters/a8c9dcc/ : SUCCESS in 4m 50s\n- openstack-tox-docs http://logs.openstack.org/23/654523/1/check/openstack-tox-docs/88d6d3a/html/ : SUCCESS in 3m 05s","accounts_in_message":[],"_revision_number":1},{"id":"7623aec0041e1b1e70c6dce41eff87de8c11707c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-04-22 16:13:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"63be594b399b438f3ff9741eff5cad7040105ed6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 16:13:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"572eed1be2b6166f0413a92254c5c1ebf0aa4d25","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 16:25:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8c036e9845bccb08d097b7721cf071b2df1e0437","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 16:25:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- noop noop : SUCCESS in 0s\n- openstack-ansible-linters http://logs.openstack.org/23/654523/1/gate/openstack-ansible-linters/3505d89/ : SUCCESS in 5m 34s\n- openstack-tox-docs http://logs.openstack.org/23/654523/1/gate/openstack-tox-docs/fc01d80/html/ : SUCCESS in 2m 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"70c902c0e8b3b038b17204ca720eb73bf06935aa","revisions":{"70c902c0e8b3b038b17204ca720eb73bf06935aa":{"kind":"REWORK","_number":1,"created":"2019-04-22 15:39:06.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/23/654523/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_monasca","ref":"refs/changes/23/654523/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_monasca refs/changes/23/654523/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_monasca refs/changes/23/654523/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_monasca refs/changes/23/654523/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_monasca refs/changes/23/654523/1"}}},"commit":{"parents":[{"commit":"aeba47431600ece77d36aa9df6758b14623dbf06","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_monasca/commit/aeba47431600ece77d36aa9df6758b14623dbf06"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-22 15:39:04.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-22 15:39:04.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I05c395a1847edc9666676f95fa7c61d4c134dd12\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_monasca/commit/70c902c0e8b3b038b17204ca720eb73bf06935aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_monasca/commit/70c902c0e8b3b038b17204ca720eb73bf06935aa"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
