)]}'
{"id":"openstack%2Fopenstack-ansible-os_monasca~663822","triplet_id":"openstack%2Fopenstack-ansible-os_monasca~master~I7f8ec629291952e0f0c5d4afbca3ce4bf92e2a41","project":"openstack/openstack-ansible-os_monasca","branch":"master","hashtags":[],"change_id":"I7f8ec629291952e0f0c5d4afbca3ce4bf92e2a41","subject":"Cap sphinx for py2 to match global requirements","status":"ABANDONED","created":"2019-06-07 02:15:18.000000000","updated":"2019-06-18 17:15:41.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6db54425902c85de17a39dc3f4e95f4cfd3e87d5","_number":663822,"virtual_id_number":663822,"owner":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2019-06-07 02:40:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-07 02:40:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"955bfe10600b671a10cd3a49ec47a2574a9d189b","author":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"date":"2019-06-07 02:15:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b85a0cc1219deaf04fa1eca3202ed08c35171b50","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 02:40:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- noop noop : SUCCESS in 0s\n- requirements-check http://logs.openstack.org/22/663822/1/check/requirements-check/edfc205/ : SUCCESS in 2m 07s\n- openstack-ansible-linters http://logs.openstack.org/22/663822/1/check/openstack-ansible-linters/24822d6/ : SUCCESS in 5m 33s\n- openstack-tox-docs http://logs.openstack.org/22/663822/1/check/openstack-tox-docs/9c034a5/html/ : SUCCESS in 2m 51s\n- build-openstack-releasenotes http://logs.openstack.org/22/663822/1/check/build-openstack-releasenotes/a56bea1/html/ : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":1},{"id":"6db54425902c85de17a39dc3f4e95f4cfd3e87d5","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2019-06-18 17:15:41.000000000","message":"Abandoned\n\nmonasca has been abandonned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ddf161ffae96f66bba131f7a6f585ab0645b05fa","revisions":{"ddf161ffae96f66bba131f7a6f585ab0645b05fa":{"kind":"REWORK","_number":1,"created":"2019-06-07 02:15:18.000000000","uploader":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"ref":"refs/changes/22/663822/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_monasca","ref":"refs/changes/22/663822/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_monasca refs/changes/22/663822/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_monasca refs/changes/22/663822/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_monasca refs/changes/22/663822/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_monasca refs/changes/22/663822/1"}}},"commit":{"parents":[{"commit":"754c6ab1c2239f3e8fe6fa3ed178d7bad328d8f7","subject":"Merge \"Change the tests source to opendev\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_monasca/commit/754c6ab1c2239f3e8fe6fa3ed178d7bad328d8f7"}]}],"author":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2019-06-07 02:14:49.000000000","tz":480},"committer":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2019-06-07 02:15:12.000000000","tz":480},"subject":"Cap sphinx for py2 to match global requirements","message":"Cap sphinx for py2 to match global requirements\n\nChange-Id: I7f8ec629291952e0f0c5d4afbca3ce4bf92e2a41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_monasca/commit/ddf161ffae96f66bba131f7a6f585ab0645b05fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_monasca/commit/ddf161ffae96f66bba131f7a6f585ab0645b05fa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
