)]}'
{"id":"openstack%2Fopenstack-ansible-os_monasca~664096","triplet_id":"openstack%2Fopenstack-ansible-os_monasca~stable%2Fqueens~Idb8ece742acb66b12d47290b557f9dc8ad5a8baf","project":"openstack/openstack-ansible-os_monasca","branch":"stable/queens","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"Idb8ece742acb66b12d47290b557f9dc8ad5a8baf","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-06-07 23:26:20.000000000","updated":"2019-06-08 17:29:14.000000000","submitted":"2019-06-08 17:29:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664096-1560014954862-2d8656ec","meta_rev_id":"35929d3137d8a4b6db2ba95403c41f25716bd16c","_number":664096,"virtual_id_number":664096,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-06-08 17:29:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":2,"date":"2019-06-08 17:18:36.000000000","_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":1,"date":"2019-06-08 17:18:36.000000000","_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-08 17:18:36.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-06-08 17:29:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4346e4fd821fd7437e58b79c8696919ce0b6eaef","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-07 23:26:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6319dbeaa62621be679dc498edf699c048fe6c0d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-08 00:00:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- noop noop : SUCCESS in 0s\n- build-openstack-sphinx-docs http://logs.openstack.org/96/664096/1/check/build-openstack-sphinx-docs/90a19d2/html/ : SUCCESS in 3m 45s\n- openstack-ansible-linters http://logs.openstack.org/96/664096/1/check/openstack-ansible-linters/98cfe43/ : SUCCESS in 5m 48s\n- openstack-ansible-functional-ubuntu-xenial-nv http://logs.openstack.org/96/664096/1/check/openstack-ansible-functional-ubuntu-xenial-nv/31bc629/ : FAILURE in 21m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"65664d4a6a03edfb0e72f2c3b4cb667265b339f7","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-06-08 17:18:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e99f07c1e2e60a11b37a944eeee05210cc2dce8e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-08 17:18:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b23ca62ddcf518afc04b8195f8a712da2ec9d0c6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-08 17:29:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"fc51d0c7d190977e1cadeaf7bbcb545bd1802c59","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-08 17:29:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- noop noop : SUCCESS in 0s\n- build-openstack-sphinx-docs http://logs.openstack.org/96/664096/1/gate/build-openstack-sphinx-docs/aac9661/html/ : SUCCESS in 3m 20s\n- openstack-ansible-linters http://logs.openstack.org/96/664096/1/gate/openstack-ansible-linters/46bbf98/ : SUCCESS in 5m 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4bf560064bf496ba7e57815f50b3126f59c87b43","revisions":{"4bf560064bf496ba7e57815f50b3126f59c87b43":{"kind":"REWORK","_number":1,"created":"2019-06-07 23:26:20.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/96/664096/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_monasca","ref":"refs/changes/96/664096/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_monasca refs/changes/96/664096/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_monasca refs/changes/96/664096/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_monasca refs/changes/96/664096/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_monasca refs/changes/96/664096/1"}}},"commit":{"parents":[{"commit":"b91009186150acf6d3532e7fee6499f7d0738ea7","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_monasca/commit/b91009186150acf6d3532e7fee6499f7d0738ea7"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-07 23:26:16.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-07 23:26:16.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: Idb8ece742acb66b12d47290b557f9dc8ad5a8baf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_monasca/commit/4bf560064bf496ba7e57815f50b3126f59c87b43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_monasca/commit/4bf560064bf496ba7e57815f50b3126f59c87b43"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
