)]}'
{"id":"openstack%2Fopenstack-ansible-os_murano~692772","triplet_id":"openstack%2Fopenstack-ansible-os_murano~master~I66f2d65bb1a69ee45bb2077fc3901c74748a9ae2","project":"openstack/openstack-ansible-os_murano","branch":"master","topic":"fix","hashtags":[],"change_id":"I66f2d65bb1a69ee45bb2077fc3901c74748a9ae2","subject":"Fix a typo in main.yml","status":"MERGED","created":"2019-11-04 08:31:29.000000000","updated":"2019-11-08 05:20:47.000000000","submitted":"2019-11-08 05:19:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"692772-1573190384112-c91a571c","meta_rev_id":"ae51bad1adc5dd98faf634196594cf26b0bc4d8b","_number":692772,"virtual_id_number":692772,"owner":{"_account_id":23317,"name":"ShangXiao","email":"shangxiaobj@inspur.com","username":"ShangXiao"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-11-08 00:20:39.000000000","_account_id":23317,"name":"ShangXiao","email":"shangxiaobj@inspur.com","username":"ShangXiao"},{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-08 05:19:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"all":[{"value":0,"_account_id":23317,"name":"ShangXiao","email":"shangxiaobj@inspur.com","username":"ShangXiao"},{"value":2,"date":"2019-11-06 14:18:46.000000000","_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-11-06 14:34:46.000000000","_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":23317,"name":"ShangXiao","email":"shangxiaobj@inspur.com","username":"ShangXiao"},{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-11-06 14:34:46.000000000","_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23317,"name":"ShangXiao","email":"shangxiaobj@inspur.com","username":"ShangXiao"},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-06 14:18:46.000000000","updated_by":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"reviewer":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"state":"REVIEWER"},{"updated":"2019-11-06 14:34:46.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-11-08 05:19:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ade634a765c0841434632d95ea571431926a9c27","author":{"_account_id":23317,"name":"ShangXiao","email":"shangxiaobj@inspur.com","username":"ShangXiao"},"date":"2019-11-04 08:31:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"34e073623e87bf05d25730cc86880b747b37b668","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-04 10:59:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/22e58088335d4586bcd13da29b4466f3 : FAILURE in 1h 22m 02s (non-voting)\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/abfb8080f716424b95bb6f75e65205fc : SUCCESS in 2h 05m 11s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/20748e7c38eb43e59c3ce01c7d362e16 : SUCCESS in 1h 35m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f3799e4b47cb48f6bcd07f776e331ccf : SUCCESS in 1h 22m 18s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/068e1638009b47a1912021ab3fac6658 : SUCCESS in 2h 27m 31s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2f0ed1bbd9b74ccf9c2083e60955e9ba : SUCCESS in 1h 42m 25s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e8e81b48279c4c43bbe5b1c1c464bf73 : SUCCESS in 1h 26m 49s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/6d744a46962c42bcb956798ef45d6811 : FAILURE in 52m 36s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1ce791bdc1ec4317a9a9adc0d9fd61c0 : SUCCESS in 56m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09b1388b61d7463398007ffa5a5598d8 : SUCCESS in 2m 30s","accounts_in_message":[],"_revision_number":1},{"id":"2df53645dad18aa1e02f41d88c186a1d523b89a5","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2019-11-06 14:18:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8c66473f5c91e7872296b0bc7125fa64704cdb85","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-11-06 14:34:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9f8d33b2a513729ac53dfc15f671ac108b331e9b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-06 14:34:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a01e88db7f6c9218c71ac1a94f612cdd268d38b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-06 17:04:45.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/6a316475e6d241dd8f818a9d3eae6374 : FAILURE in 1h 36m 04s (non-voting)\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/6cab49f9a76348e3812ece22d9329302 : SUCCESS in 1h 49m 57s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/a5b360447c004a2298a6b13a6fc6c8ba : FAILURE in 24m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/be45b682e4d44dfd999ef0b0624f309c : SUCCESS in 1h 11m 28s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7f7c83edb56141d7b0d56e1e624762e3 : SUCCESS in 2h 24m 48s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4ca4e211be1940a598620c8edab9d352 : SUCCESS in 1h 43m 15s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c41c4e6d5f964306a823da4f842ae1d0 : SUCCESS in 1h 18m 27s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/48ffc54372c84be7baaf6e287f4fd006 : SUCCESS in 57m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38dd88f0a95246b89df29089374de16e : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":1},{"id":"5a9e804b03aa9997f8e31462711700a37b8657a4","author":{"_account_id":23317,"name":"ShangXiao","email":"shangxiaobj@inspur.com","username":"ShangXiao"},"date":"2019-11-07 02:52:44.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e0878aa9b88fa706b1382aec9e027def542891ef","author":{"_account_id":23317,"name":"ShangXiao","email":"shangxiaobj@inspur.com","username":"ShangXiao"},"date":"2019-11-07 02:59:23.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"491de0c9ab1166102fd5310a50ee9cc8fbb70101","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 05:37:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/fa706d3213e14787b39730bbc452aa45 : FAILURE in 1h 29m 20s (non-voting)\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5330e7d6de934b098104498274e3f50b : SUCCESS in 2h 01m 14s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/45a0719ff02e4711a6921e98ab4ed845 : SUCCESS in 1h 47m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9d4230f919ec418eb195c2baac434ace : FAILURE in 19m 37s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/86bcddab36224a7fa6c01aac7c233e6f : SUCCESS in 2h 41m 26s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/efc176ec75434c64a8351a2dc1c84b24 : FAILURE in 1h 08m 54s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ce3ab93370974d159d1a7f0c2e1f54f2 : SUCCESS in 1h 31m 25s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/55475f60b96d4843bdb28c1277ed4009 : FAILURE in 1h 02m 20s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b83a800cc0b2469e8d6cd326497da134 : FAILURE in 18m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/218d0251b30a4d77938a1d340a27d81f : SUCCESS in 2m 37s","accounts_in_message":[],"_revision_number":1},{"id":"1552a47512b0a09f4e1a56d20838f18bbe738e2f","author":{"_account_id":23317,"name":"ShangXiao","email":"shangxiaobj@inspur.com","username":"ShangXiao"},"date":"2019-11-08 00:20:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1bd55f04c36578bf0ff6ec8b10a6d19ce69d33ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-08 02:38:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/d6edcfde86ec445fb14505e1fdf40f71 : FAILURE in 1h 15m 54s (non-voting)\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/62e9ca31f731492bbb7d3b37598ed5b6 : SUCCESS in 1h 47m 13s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/00b1d9be28b4403194e64ead62b9f9ee : SUCCESS in 1h 33m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f40eb47dc58b4a9a8bcc52eb608cf54d : SUCCESS in 1h 15m 03s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/76d70cafb7f64615a1264950fbf0e075 : SUCCESS in 2h 16m 59s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/db2f08f602264ad3a464ff41480f26ae : SUCCESS in 1h 37m 13s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b7127800834048c381e3cfd70401172b : SUCCESS in 1h 13m 18s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/b1d60ed8fd0e48db94f9cbdc0aa9e4a8 : FAILURE in 56m 21s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f8f41571bdf8467fae91b24a0addb489 : SUCCESS in 58m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1d879b0c2d84c7bb496f0e864fae75b : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":1},{"id":"e75a74e0aee278a6800ce85bd071a2f2b96985ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-08 02:39:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b1cb157eee6695b197803974746edfc0662500bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-08 05:19:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/f1cb7e2836c6459184d366cce76988bc : FAILURE in 1h 21m 21s (non-voting)\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/2495f07ab6af470b950e71b7199598f5 : SUCCESS in 1h 56m 20s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/d815eb8616db4e5d9e334d5d1c44ecac : SUCCESS in 1h 35m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/daf4a58f771444cb995a67001934251f : SUCCESS in 1h 28m 10s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/164b5511b462438e96cd883edc6edcf7 : SUCCESS in 2h 38m 51s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/660d834380134dff8272258bbe1d16ad : SUCCESS in 2h 02m 27s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/81427711502d4997ae2aba749da64959 : SUCCESS in 1h 26m 46s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1fab29953ef3497a82e1dfd5223c25c1 : SUCCESS in 55m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bccce43da88849029bf1f3b0d1454e37 : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":1},{"id":"6d1bbd7241e76ec63aed8be3835e898ecedb91d5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-08 05:19:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9dc9e9e324cb4ad42846b1cf0688a8dd286d69c1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-08 05:20:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49a70d5a8efc41d2a9baea1283e0ca37 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bb18ed2e2285ea7723fa64439a9e81e00195b188","revisions":{"bb18ed2e2285ea7723fa64439a9e81e00195b188":{"kind":"REWORK","_number":1,"created":"2019-11-04 08:31:29.000000000","uploader":{"_account_id":23317,"name":"ShangXiao","email":"shangxiaobj@inspur.com","username":"ShangXiao"},"ref":"refs/changes/72/692772/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_murano","ref":"refs/changes/72/692772/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/72/692772/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/72/692772/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/72/692772/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/72/692772/1"}}},"commit":{"parents":[{"commit":"9f5a5eac6911935ab7662f7b66723620d312f0bf","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/9f5a5eac6911935ab7662f7b66723620d312f0bf"}]}],"author":{"name":"ShangXiao","email":"shangxiaobj@inspur.com","date":"2019-11-04 08:28:48.000000000","tz":480},"committer":{"name":"ShangXiao","email":"shangxiaobj@inspur.com","date":"2019-11-04 08:31:25.000000000","tz":480},"subject":"Fix a typo in main.yml","message":"Fix a typo in main.yml\n\nFix the error word \"successfull\" to \"successful\".\n\nChange-Id: I66f2d65bb1a69ee45bb2077fc3901c74748a9ae2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/bb18ed2e2285ea7723fa64439a9e81e00195b188"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/bb18ed2e2285ea7723fa64439a9e81e00195b188"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
