)]}'
{"id":"openstack%2Fopenstack-ansible-os_murano~694324","triplet_id":"openstack%2Fopenstack-ansible-os_murano~master~Iaf81d0c02738a65f1f1d54e77aabdb7b419c2360","project":"openstack/openstack-ansible-os_murano","branch":"master","topic":"osa/opendev","hashtags":[],"change_id":"Iaf81d0c02738a65f1f1d54e77aabdb7b419c2360","subject":"Change upper constraints url","status":"MERGED","created":"2019-11-14 15:07:45.000000000","updated":"2019-12-03 12:31:28.000000000","submitted":"2019-12-03 12:29:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"694324-1575376173273-232123cc","meta_rev_id":"0786da5f872ff3602175596c190a08ce3c678d23","_number":694324,"virtual_id_number":694324,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"date":"2019-12-03 06:34:42.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"date":"2019-11-28 06:09:56.000000000","_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-03 12:29:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"all":[{"value":2,"date":"2019-11-26 12:41:00.000000000","_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-11-18 09:28:42.000000000","_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"all":[{"value":1,"date":"2019-11-26 12:41:00.000000000","_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-18 09:28:42.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-11-26 12:41:00.000000000","updated_by":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"reviewer":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"state":"REVIEWER"},{"updated":"2019-11-28 06:09:56.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-12-03 12:29:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"62f3b37dcbefa55c0e4e7499752f08e40b73397d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-11-14 15:07:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a132aaef5a0b9afd67aa7df495e858704b51ec7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 18:26:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/ce12d17ad6ef4b38bdc84b67996a91b1 : FAILURE in 1h 26m 27s (non-voting)\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8da5c322ef65407bb60deb9ed734f923 : SUCCESS in 1h 51m 04s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/6c833cf1270e4dbf88832d5657d94ad7 : FAILURE in 26m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a5ca160217e84cdcb003d1f91296d839 : SUCCESS in 1h 17m 13s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0ead4c2d48414847912edddf34f0e721 : TIMED_OUT in 3h 04m 46s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c13e95584c4349ee81806fd662f7decd : SUCCESS in 2h 25m 24s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0dec8104a2f44e0182d962ba515698d5 : SUCCESS in 1h 16m 01s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/8b0d5a0432fc4edab38eeae01f114ea6 : FAILURE in 22m 29s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/67e32b353029428e802af54ca96725e7 : SUCCESS in 1h 00m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d23ac1248d154bb79c4696668e8aede5 : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":1},{"id":"572dd5a822b005a1deaafb986758d3ca7c4deba6","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-11-17 12:11:00.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2c853f591e8bd767d8346f3f196000e4e4991d2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-17 15:07:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/e54eb542dc3746a0b5c2f8531b7d544a : FAILURE in 1h 37m 09s (non-voting)\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/08cf36d01c00447da79bbe0d8f7f2274 : SUCCESS in 1h 42m 46s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/33cbab77319f412fa586ab0e16ce202b : SUCCESS in 1h 45m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/819dea6f41bc41629b472bd897fb396e : SUCCESS in 1h 34m 35s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/27e75ccc2630475db159ff2f1a99cb52 : SUCCESS in 2h 52m 35s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/23edd8ad669146879837ced031a5eacb : SUCCESS in 1h 51m 10s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e710925049c54af88db33e578e58faec : SUCCESS in 1h 34m 43s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/cecd0e07bee3408da4caa0acd4f7abcd : FAILURE in 1h 06m 04s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2862c54e6c1b48d8bb69fc63c5375996 : SUCCESS in 1h 17m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92e0892612214efd934912a779393994 : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":1},{"id":"d93c7f32e7e8db0082971fbdd9ee9183d5367693","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-11-18 09:28:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f7d5abce3bbd2d4cf1e941b9effaee3259a3ad3f","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2019-11-26 12:41:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5e8f6ffbdd80e0f04ee83f6a39173ec5890cfa72","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 12:41:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a689b8ae17c7087daaeefd81d2c054eca10d4f6b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 15:02:19.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/9b18408ab9504ba983ad43d27a80ffb3 : FAILURE in 1h 21m 30s (non-voting)\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b14c6acc35754498bb62c326a4a8df8e : SUCCESS in 2h 08m 41s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/25b09d4b0a5045e1b482dff4a918bfa8 : SUCCESS in 1h 42m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b16ffe563914455aa2a11547e65be7a6 : SUCCESS in 1h 20m 07s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7398e0b464d54aca81822c05916b9bcc : FAILURE in 6m 13s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e5f6486e9946465ea16df77fc3c4c4ea : FAILURE in 5m 30s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/cd629d70332e4fd6a9f5d96a2879494f : SUCCESS in 1h 22m 31s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ed591ab2d7bf4cbaadd4cc05eaa58d78 : SUCCESS in 1h 13m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b646d0cb853411d9bed313283426953 : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":1},{"id":"9b99ec8ea83f792f394937abe537363cfc52398b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-11-26 16:44:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c0c1c40d7053d5363370298f2718786da76c117e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 19:58:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/88e1fd3fbd3f4afea7cfcacc7883c7cc : FAILURE in 1h 18m 18s (non-voting)\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/4b500eea26eb48b8a62fba4c08eafda0 : SUCCESS in 1h 50m 14s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/2d7c416f78104c6198063e4e006f1199 : SUCCESS in 1h 35m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/99b60adcd14249148312e69d29780404 : SUCCESS in 1h 17m 41s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/efe9575ec0a54b848512eb9b4838a24d : TIMED_OUT in 3h 02m 40s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e5061d84fcc9463ca9d46b43b418748d : SUCCESS in 1h 49m 21s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/504e6fbb6cd5495a80896ea039593f1c : SUCCESS in 1h 12m 02s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/8454bd1635754b8f8f3a8d4087adb2a6 : FAILURE in 1h 00m 37s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/88aa4f9c3f8f4ea7a0925474690303b6 : SUCCESS in 1h 09m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef6b8868be1e45daa58acf0a8db56023 : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":1},{"id":"0f88aa57f30ae25ea0dfda5f0a5c213b8fb57204","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-11-27 13:07:17.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"603f924ec33080273d56cec9175772f92e1691d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 16:15:10.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/960e2818c6ab49dc89b3f2dbfe2e48e4 : FAILURE in 1h 28m 09s (non-voting)\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a81376fe33a948b4b5f772c5c71c6e39 : SUCCESS in 1h 52m 13s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/e4a261868eaa4f638d2f657775d0755c : SUCCESS in 1h 49m 26s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d41419b6b7c742bb8c2fac232d97acec : SUCCESS in 1h 15m 06s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/abb2f24bd60b495497f1bc766545216e : TIMED_OUT in 3h 02m 31s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1efccd5e12ec4e2a81c4b7dd5a81be7d : SUCCESS in 2h 16m 43s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e662b617e4744f3b83806a833ccdb80d : SUCCESS in 1h 22m 51s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/ca84d89b48bd46e9bfe217912cf8e98c : FAILURE in 1h 02m 53s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/87ec91832e874ed1bed03572c4c1ce6b : SUCCESS in 1h 06m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1a615e65c104c7bb6f1e3539423994c : SUCCESS in 5m 04s","accounts_in_message":[],"_revision_number":1},{"id":"2f0fbd28ab08c9fb190812532e77fcb148eb9e89","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-11-28 06:09:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"30943b1057002cf789d6cda40e59cb04f9e34b85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 09:21:14.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/1a15b14614d54bc8ac1ed7b70e886d83 : FAILURE in 1h 40m 57s (non-voting)\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/89fb99a42c9845fea01b8482b73a4e4a : SUCCESS in 1h 46m 36s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/12e646912bbc4d66abcb13c5d96837ec : SUCCESS in 1h 43m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b43092051800471dab748995cf6c2286 : SUCCESS in 1h 17m 10s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7be52ee9a0bd41709d11cef633277088 : TIMED_OUT in 3h 02m 37s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/51e610aa491547a1855283ac28f03b46 : SUCCESS in 2h 08m 08s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/59612cbbc7c043dd932d202e694a8861 : SUCCESS in 1h 28m 48s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/4319fadef30047b99e737e3db2638d5d : FAILURE in 53m 58s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cb7d89e88fdc4d789c4bbb74a8931983 : SUCCESS in 57m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88ca3ae6c09e411e94eee95f99ee91b5 : SUCCESS in 2m 55s","accounts_in_message":[],"_revision_number":1},{"id":"d222e953ba54f9cfceabee0b4bdffbb9ccfb3179","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-12-03 06:34:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5a0b7a03404e5fcb67060d2adf65bf66d3edb3cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 09:31:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/e2e2748d4dcb454e99460fa629b7f8d4 : FAILURE in 1h 18m 15s (non-voting)\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/02cb1b675ed641b5b1e40ec38acffd7f : SUCCESS in 1h 52m 38s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/e73745f7504b4a8894ba0dc510c43ee5 : SUCCESS in 1h 31m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/908ca9bc67aa444ab4fdea7842ed519f : SUCCESS in 1h 15m 33s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/81b2e2c1b5a7472bbe32445857959349 : SUCCESS in 2h 54m 55s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aed85b4ad8794a2ab1301201f7be5c40 : SUCCESS in 1h 59m 50s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d18ceff4a88e4cc0aac1e8b72f68f471 : SUCCESS in 1h 17m 15s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/9c3d929e06664613a244caf979b86a97 : FAILURE in 57m 56s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d8366c4a0cad4970864c070129a0ea5b : SUCCESS in 56m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2200454a901a4adb88a52804dbdc3da3 : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":1},{"id":"91fec8e5d149a5c59abbbee5453739fa25712058","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 09:32:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b90737f0881cc9be4b1a6c37bf083125cd638aa5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 12:29:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"54968af9740d63ee13422fb7136e2baed8a7d408","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 12:29:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/c77f7aa450424e9d9ba7bccd64863f14 : FAILURE in 1h 44m 04s (non-voting)\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d58f244d7e2947b7a88ea382b0215b62 : SUCCESS in 2h 09m 36s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/92dd63f28f9f4f0181e51d522280c391 : SUCCESS in 1h 31m 25s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f5bbfad05e1a492dbf8e38b28dd68ff4 : SUCCESS in 1h 16m 36s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0354392244ec4b38ade02d6c383b75fd : SUCCESS in 2h 55m 54s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/601e8721494b4b779d606e2bc944be62 : SUCCESS in 2h 05m 47s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/9b40cf57a3724528b91935063430164d : SUCCESS in 1h 16m 39s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/431403128dbd4dd8aa3c0737fd59570d : SUCCESS in 53m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/017093513c40437a8da6f0a9d2cd82d2 : SUCCESS in 3m 16s","accounts_in_message":[],"_revision_number":1},{"id":"8a3b0fb57adc028241fcfd038386133894062634","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 12:31:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0691f47b0d6645e2b5e980a09cf84fa1 : SUCCESS in 1m 29s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fa9f24a0f87025c9cf38d57dfa9f5185032daa1c","revisions":{"fa9f24a0f87025c9cf38d57dfa9f5185032daa1c":{"kind":"REWORK","_number":1,"created":"2019-11-14 15:07:45.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/24/694324/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_murano","ref":"refs/changes/24/694324/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/24/694324/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/24/694324/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/24/694324/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/24/694324/1"}}},"commit":{"parents":[{"commit":"bb18ed2e2285ea7723fa64439a9e81e00195b188","subject":"Fix a typo in main.yml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/bb18ed2e2285ea7723fa64439a9e81e00195b188"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-11-14 15:07:33.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-11-14 15:07:33.000000000","tz":120},"subject":"Change upper constraints url","message":"Change upper constraints url\n\nTo be consistent with other projects, we are changing upper constraints\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-May/006478.html\n\nChange-Id: Iaf81d0c02738a65f1f1d54e77aabdb7b419c2360\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/fa9f24a0f87025c9cf38d57dfa9f5185032daa1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/fa9f24a0f87025c9cf38d57dfa9f5185032daa1c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
