)]}'
{"id":"openstack%2Fopenstack-ansible-os_murano~743538","triplet_id":"openstack%2Fopenstack-ansible-os_murano~stable%2Fussuri~Icac9d84f317ed0ba5a61adde4e441bc85a434add","project":"openstack/openstack-ansible-os_murano","branch":"stable/ussuri","topic":"osa/el8-stable/ussuri","hashtags":[],"change_id":"Icac9d84f317ed0ba5a61adde4e441bc85a434add","subject":"Add centos-8 support","status":"ABANDONED","created":"2020-07-28 13:43:44.000000000","updated":"2021-11-22 11:58:40.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4d2501a9db918a70824a3bcdf384c96f90b7ee93","_number":743538,"virtual_id_number":743538,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},{"value":0,"date":"2020-10-13 08:31:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-07-28 17:19:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"date":"2020-07-28 13:43:44.000000000","_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-28 13:43:44.000000000","updated_by":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"reviewer":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"state":"REVIEWER"},{"updated":"2020-07-28 17:19:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"65d3efbc0f2f3d5e2c477056d338c24fb25f1119","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-07-28 13:43:44.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"f801a3d6d9c1d0ef6f7d79255a3aa2df5f105ebc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 17:19:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/92face8afacf4083b2f44fd6fa330b04 : FAILURE in 2h 14m 15s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/06c8adf01f5e42b09b723c7ac7f04779 : TIMED_OUT in 3h 08m 38s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/98e9c1fb4cfa4a7e906712adbe5e3ed6 : TIMED_OUT in 3h 05m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d5f3f81d8ca746a8a5d08cf7d61d2bda : TIMED_OUT in 3h 06m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/efb1af20202c44668ee42058446ad68c : TIMED_OUT in 3h 05m 53s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/4c208e5ccbd94dbaadb0362b45161cee : FAILURE in 1h 29m 01s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f32cfcfcb6d04cb2bd5d4ceb6b9d7ca9 : TIMED_OUT in 3h 04m 59s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/69a40fdc877b47ec9af56b2fc72bd4f6 : SUCCESS in 1h 17m 15s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/3b9253d9176d45c3873143d83f9a99cc : FAILURE in 2h 37m 50s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a50efed620264f789fef49fd225e46cf : SUCCESS in 58m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8462499d518749108216841b6883bea1 : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":1},{"id":"8323bcd6a2cb0cecfd66c11572ce94607f6c42ce","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-08-25 06:31:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"72e3b9fcd2b3f1c347afc53eb38c2695bde971ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 09:56:15.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/5ea5f359ee1d41e5921a857ec10a2763 : SUCCESS in 1h 52m 23s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/9d01212a65434df093a68fcd756195a6 : FAILURE in 2h 42m 40s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/bb424703fc8b4ded90a068bd9ce5bb54 : TIMED_OUT in 3h 07m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/460f2e6fed6e43dc8156c6748c37a4df : TIMED_OUT in 3h 07m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cd7bb828a310437db658498461c30c07 : TIMED_OUT in 3h 05m 45s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5ab9b4f9271a4f2cb1e806894c19a045 : FAILURE in 1h 25m 40s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8cc367eb06e146469fb77eb36ab096cf : TIMED_OUT in 3h 03m 42s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7ec4f56e658f490283956796889a9616 : SUCCESS in 1h 22m 29s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/4e716e4b34a14a52b89d44c74bc0b40c : FAILURE in 2h 30m 35s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fd7514b3e7344238bd808866dc81bf31 : SUCCESS in 1h 01m 36s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/18c5d04b828845cc92ae3f8d1d39682a : SUCCESS in 57m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57ce555e07aa423386bddae774688cb5 : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":1},{"id":"179bcf0eeb3571f14057fc17f5d6324a157f0b66","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-13 08:31:08.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4d2501a9db918a70824a3bcdf384c96f90b7ee93","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-11-22 11:58:40.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5ecde9a22f487ec5cdd8ee2228912121c68f95a5","revisions":{"5ecde9a22f487ec5cdd8ee2228912121c68f95a5":{"kind":"REWORK","_number":1,"created":"2020-07-28 13:43:44.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/38/743538/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_murano","ref":"refs/changes/38/743538/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/38/743538/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/38/743538/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/38/743538/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/38/743538/1"}}},"commit":{"parents":[{"commit":"d106eabcb897f678ecc1c0501614abd7fdee2f2d","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/d106eabcb897f678ecc1c0501614abd7fdee2f2d"}]}],"author":{"name":"root","email":"satish.txt@gmail.com","date":"2020-07-07 02:31:10.000000000","tz":-240},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-07-28 13:43:44.000000000","tz":0},"subject":"Add centos-8 support","message":"Add centos-8 support\n\nRenaming redhat-7.yml to redhat.yml to support centos-8.\n\nChange-Id: Icac9d84f317ed0ba5a61adde4e441bc85a434add\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/5ecde9a22f487ec5cdd8ee2228912121c68f95a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/5ecde9a22f487ec5cdd8ee2228912121c68f95a5"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
