)]}'
{"id":"openstack%2Fopenstack-ansible-os_murano~747236","triplet_id":"openstack%2Fopenstack-ansible-os_murano~master~Idaaf3d2262cfa42985ae3ee3aa48726d1b6f4a64","project":"openstack/openstack-ansible-os_murano","branch":"master","topic":"bind-to-mgmt","hashtags":[],"change_id":"Idaaf3d2262cfa42985ae3ee3aa48726d1b6f4a64","subject":"Use the utility host for db setup tasks","status":"ABANDONED","created":"2020-08-20 16:05:21.000000000","updated":"2021-11-22 11:50:40.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8274c307439b1ba5fd2e32725a284797151c9a28","_number":747236,"virtual_id_number":747236,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-02-11 11:33:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-20 20:34:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-21 11:25:23.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-08-21 14:08:51.000000000","updated_by":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"reviewer":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"state":"REVIEWER"},{"updated":"2020-10-05 19:41:08.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-10-27 08:12:39.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"0c01276f83a9f0e966300ece3b7f0c2ff6e46664","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-08-20 16:05:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4ef1a9d6e701593d4fa9ee5b18dba4877999be4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 20:34:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/ad81bd5757d349dd8986a9be1d71ce12 : TIMED_OUT in 3h 06m 48s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/86d9dbb1f32b4922a09c9f911d18cf83 : FAILURE in 2h 45m 39s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/483c1638fac74f9ab350e8d7f3067424 : FAILURE in 1h 10m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6842daab232342b6838bf28c30667b77 : FAILURE in 2h 53m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2b3a9138d1624107974944043af4406f : FAILURE in 2h 05m 23s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ad2cdce95034468a8f727bc188751416 : TIMED_OUT in 3h 02m 51s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/71170604be914dd9afdb3727cf4a469a : TIMED_OUT in 3h 02m 58s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/85ac638ec2bf4026a1d2b17fdd0bcb15 : SUCCESS in 1h 16m 40s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/531c41db3fe94d7d98c9602e9d720a11 : FAILURE in 49m 02s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/718f18404997476fb15e58a48be84a65 : SUCCESS in 47m 14s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b3603e74d1434661b2a2a3a0a655f324 : SUCCESS in 59m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1631250831504793921981ff69d4aec2 : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":1},{"id":"9e2d8524dac38a16838b5f1c7954a2ff27ec823d","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-08-21 07:39:10.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"43fd77e170e50a24ea2fa09631efb03f0182b0e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 11:15:19.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/670cdfb813c4469dbf466cbad40122d9 : FAILURE in 2h 17m 36s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/641eaf21d80a49a191f91ae570199135 : TIMED_OUT in 3h 06m 35s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/a4b1850464214249b390fa691ade6f14 : FAILURE in 1h 47m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/63bff4c340c449f69614f4dc2faaff24 : TIMED_OUT in 3h 07m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/034a6711087b436a9befab3a6ee7bcf1 : TIMED_OUT in 3h 09m 12s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/289f4c67ec824808aafbeb5e7cad684f : TIMED_OUT in 3h 03m 27s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e1963f7b8dfa4812903fd494b0aa2901 : TIMED_OUT in 3h 03m 02s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/40000464ac4e41cfa764e4e5b627db7f : SUCCESS in 1h 27m 41s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e0225a701aa244cbba5e906489d388ec : FAILURE in 1h 01m 12s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f19f53dcb1944572a25895489b9034df : SUCCESS in 1h 07m 19s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ef1db1d815846fa906981f912cfe20a : SUCCESS in 59m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75fd8f2c3d454be284817ec1aae195d9 : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":1},{"id":"6bbe77a5f4c3901c5c2ab8362f1a2705d20f857c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-08-21 11:25:23.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3ef594381cc1a42a510d79f7e5ebc9e154ad194c","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2020-08-21 14:08:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c895af83b32dc1cf11ffa81a66746b3bde398915","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 14:54:41.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/a8b30b3962fd40a0bb4231ba92aa1847 : FAILURE in 2h 24m 48s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e84f25715a4944e180d22911e76c0c2c : TIMED_OUT in 3h 06m 35s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/71f002e1e1a4457a981110558ebe410b : FAILURE in 1h 12m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c62d92d1cd71442c9ac79e060dd2b3fd : FAILURE in 2h 42m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/770aff8f24564db48d9df0d0f3a5a7e6 : FAILURE in 2h 46m 01s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d51a82e4cae848529b26f84afc6588e1 : TIMED_OUT in 3h 03m 29s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e3d73134ddfb4d48a5588dc1a5c1535f : TIMED_OUT in 3h 03m 03s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/429e2ceb8b1a49e9ab074be325819b70 : SUCCESS in 1h 11m 10s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/6f95082e80c84e698aefb703d8a5a918 : FAILURE in 53m 01s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/45c0198bf88f4e41b45d7d829a3c63ee : SUCCESS in 1h 05m 08s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd43f53de922476aa588de173cab52c7 : SUCCESS in 56m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ea3ac2c8f8c42e5ab0721cd25875a9f : SUCCESS in 5m 54s","accounts_in_message":[],"_revision_number":1},{"id":"0e9aabc1a2ca6f2e54f13d0f2c2772cc52ade595","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-05 19:41:08.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0ac542aa7e61ca0f57a181b47e2316f928cec38f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 23:14:12.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/8afd96f418f8403b8d68674f5f836ece : FAILURE in 2h 47m 18s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/12fd74817c0d4850a756433549f4d98e : FAILURE in 1h 14m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7bfc2c9eb0ed484d9746b686040c4cc3 : SUCCESS in 2h 09m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/70422552c46844cc8d38593bd0478080 : FAILURE in 2h 35m 39s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/1533bdc631b348f097123ac998a38eee : FAILURE in 33m 38s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b9cfb2e6ffc043e9a4a4c09bb4da8b45 : TIMED_OUT in 3h 02m 33s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/117a466675e14f2d8e12578ed85794cc : FAILURE in 1h 14m 06s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5e9f307682244cc29b7cf0950289de1d : SUCCESS in 1h 06m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c166095aa66248aa8f27fee355b2d74a : SUCCESS in 3m 28s","accounts_in_message":[],"_revision_number":1},{"id":"6ccb402cb1dcd25f96b3cf936588de0a2430f9b2","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2020-10-27 08:12:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ceb2f5397f8406853c0088b46b9e761ebf7ad458","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-27 11:23:28.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/07eb7427e0654b5c964ef6ffcba8663f : FAILURE in 1h 10m 06s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/8ef6b4944e6c4f919c6cdbd7d675a862 : FAILURE in 1h 24m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2ef93327fa344b28ad5bab13058d0752 : FAILURE in 1h 20m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cbcdabd1423e49b6b9f3a06a8904bb01 : FAILURE in 1h 06m 36s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/9d73e7e9e0e843a581c0543f87be01a1 : FAILURE in 1h 14m 40s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d6c5c24afdbf4a119452f3812350daa7 : TIMED_OUT in 3h 02m 54s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/71f43eda6e6e424aaaa2aeb22acd6d19 : FAILURE in 54m 51s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/565b892bea8f48e99f1e4b8955ce996f : FAILURE in 50m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/354f0068fc6d422fa5da1f072daa4a54 : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":1},{"id":"69ac4b76fc001ca58bb88a4c250cec2a193ac63d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-11-18 13:15:18.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9508b9951a5d43f0786b9466b66334254fb3d8f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-18 16:18:55.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/8bc6195551904d3188c21826f5319cf1 : FAILURE in 1h 11m 21s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/2d7bd734580b40de82873db516d4c743 : FAILURE in 1h 33m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/51224ba7f1d247d4ab9ca7f0618a49f5 : FAILURE in 1h 18m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c767d4dcbdb043c9b33287da3f61e8da : FAILURE in 1h 05m 20s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/1dd282ad16954b2d8ae72749df555ffb : FAILURE in 1h 05m 08s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/079936431cc14eb1a78610a2589b8928 : FAILURE in 2h 52m 38s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/90445d1f0d1c44e98feb2495bbf653a8 : FAILURE in 1h 03m 26s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1e44894620eb40c3986162becdd5803c : FAILURE in 46m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f81ad86aa5b5409e8cc34b287951b22d : SUCCESS in 3m 14s","accounts_in_message":[],"_revision_number":1},{"id":"733602ac9f590b6ad2f5c592bd90b4f75345c1e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-01-26 09:05:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cdfe8ac75981d188f16e70d9b64f70e0c01be6c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 10:38:29.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/67827275509a45d294a3eb134bcaa4bf : FAILURE in 1h 10m 38s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/fcabd5b4c2484eb0928ad10797dfaf4a : FAILURE in 1h 18m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bf2da0b79a1b4bee80e2a7b550680d14 : FAILURE in 1h 13m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9663292c9d9a4acdacc9b0f3c2fa1a43 : FAILURE in 58m 44s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/9d4e009e08b04f739e691ea5979598ed : FAILURE in 1h 19m 16s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/34ffaf13a7534239831a450b8c957c7a : FAILURE in 55m 43s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/125d20492b8940cdb689fcf5f78bd6b3 : FAILURE in 1h 05m 45s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c860b176caa4b0bb0ee64f9f479119a : FAILURE in 47m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe5aea93fea849bf93c92545a00c25d3 : SUCCESS in 4m 53s","accounts_in_message":[],"_revision_number":2},{"id":"631119faa3a93459190d63278712125ce90312a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-11 08:16:00.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"83c6f999f94912d0ec8abd049c5a71f119a6007d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-11 08:18:53.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"25aea277a2fa542179257e6956a4e2ef31a79863","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 11:33:53.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/45efde55b4234573b46f36c7d4db73aa : FAILURE in 2h 38m 41s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/2f258cd7e26942bfae386be8886a4ef7 : TIMED_OUT in 3h 06m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/187b6e0883db462d80e8afe77343a8a7 : TIMED_OUT in 3h 06m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a0273de191e543cc8204a2b3619e5e66 : FAILURE in 2h 58m 30s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/4da358ea3a9d4735acc2e4bb08f9a272 : FAILURE in 1h 08m 49s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5e938a78ad024dc0a2b3727a24a0a4fe : FAILURE in 1h 00m 26s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/183f5380ffe0498a9f7ca384506c1ad1 : FAILURE in 53m 33s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9cb49f4f44cb450595a8ad575f463e2e : FAILURE in 48m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bd1e3fea9194801b18f00c18d29dec0 : SUCCESS in 3m 11s","accounts_in_message":[],"_revision_number":4},{"id":"8274c307439b1ba5fd2e32725a284797151c9a28","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-11-22 11:50:40.000000000","message":"Abandoned\n\nimplemented with https://review.opendev.org/c/openstack/openstack-ansible-os_murano/+/781239","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"c2d76764de2a8f5e362bffa195cd2ef5ad1fef54","revisions":{"18d2db4ed9a6bc5a213da30d2aceb2d268d1eba4":{"kind":"REWORK","_number":1,"created":"2020-08-20 16:05:21.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/36/747236/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_murano","ref":"refs/changes/36/747236/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/36/747236/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/36/747236/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/36/747236/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/36/747236/1"}}},"commit":{"parents":[{"commit":"5e0a408fdb89ff2d5a0fe890230e083113e89451","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/5e0a408fdb89ff2d5a0fe890230e083113e89451"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-08-20 16:05:11.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-08-20 16:05:11.000000000","tz":180},"subject":"Use the utility host for db setup tasks","message":"Use the utility host for db setup tasks\n\nMove it to the service setup host (defaults to utility[0]) instead\nof the galera[0] host, and use galera_address (defaults to internal VIP)\nas the endpoint instead of a local connection on the db host.\n\nChange-Id: Idaaf3d2262cfa42985ae3ee3aa48726d1b6f4a64\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/18d2db4ed9a6bc5a213da30d2aceb2d268d1eba4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/18d2db4ed9a6bc5a213da30d2aceb2d268d1eba4"}]},"branch":"refs/heads/master"},"f08bb42efd1ebd7e60fb4f2fcae5390b69cc927b":{"kind":"REWORK","_number":2,"created":"2021-01-26 09:05:33.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/36/747236/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_murano","ref":"refs/changes/36/747236/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/36/747236/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/36/747236/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/36/747236/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/36/747236/2"}}},"commit":{"parents":[{"commit":"5e0a408fdb89ff2d5a0fe890230e083113e89451","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/5e0a408fdb89ff2d5a0fe890230e083113e89451"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-08-20 16:05:11.000000000","tz":180},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-01-26 09:05:08.000000000","tz":0},"subject":"Use the utility host for db setup tasks","message":"Use the utility host for db setup tasks\n\nMove it to the service setup host (defaults to utility[0]) instead\nof the galera[0] host, and use galera_address (defaults to internal VIP)\nas the endpoint instead of a local connection on the db host.\n\nThis patch has I731716385fe5499f36ab43bfd2ce94a4f83a7075 squashed into it.\n\nChange-Id: Idaaf3d2262cfa42985ae3ee3aa48726d1b6f4a64\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/f08bb42efd1ebd7e60fb4f2fcae5390b69cc927b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/f08bb42efd1ebd7e60fb4f2fcae5390b69cc927b"}]},"branch":"refs/heads/master"},"334467a170d1f146f20e6076e28eff68f14cdb04":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-02-11 08:16:00.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/36/747236/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_murano","ref":"refs/changes/36/747236/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/36/747236/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/36/747236/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/36/747236/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/36/747236/3"}}},"commit":{"parents":[{"commit":"ba219e03ab95b8901df36e1d18f36c330279c54c","subject":"Add global override for service bind address","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/ba219e03ab95b8901df36e1d18f36c330279c54c"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-08-20 16:05:11.000000000","tz":180},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-11 08:16:00.000000000","tz":0},"subject":"Use the utility host for db setup tasks","message":"Use the utility host for db setup tasks\n\nMove it to the service setup host (defaults to utility[0]) instead\nof the galera[0] host, and use galera_address (defaults to internal VIP)\nas the endpoint instead of a local connection on the db host.\n\nThis patch has I731716385fe5499f36ab43bfd2ce94a4f83a7075 squashed into it.\n\nChange-Id: Idaaf3d2262cfa42985ae3ee3aa48726d1b6f4a64\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/334467a170d1f146f20e6076e28eff68f14cdb04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/334467a170d1f146f20e6076e28eff68f14cdb04"}]},"branch":"refs/heads/master","description":"Rebase"},"c2d76764de2a8f5e362bffa195cd2ef5ad1fef54":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-02-11 08:18:53.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/36/747236/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_murano","ref":"refs/changes/36/747236/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/36/747236/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/36/747236/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/36/747236/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/36/747236/4"}}},"commit":{"parents":[{"commit":"d615d9504eb254274e001d767308dbdb4940e0d7","subject":"Add global override for service bind address","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/d615d9504eb254274e001d767308dbdb4940e0d7"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-08-20 16:05:11.000000000","tz":180},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-11 08:18:53.000000000","tz":0},"subject":"Use the utility host for db setup tasks","message":"Use the utility host for db setup tasks\n\nMove it to the service setup host (defaults to utility[0]) instead\nof the galera[0] host, and use galera_address (defaults to internal VIP)\nas the endpoint instead of a local connection on the db host.\n\nThis patch has I731716385fe5499f36ab43bfd2ce94a4f83a7075 squashed into it.\n\nChange-Id: Idaaf3d2262cfa42985ae3ee3aa48726d1b6f4a64\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/c2d76764de2a8f5e362bffa195cd2ef5ad1fef54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/c2d76764de2a8f5e362bffa195cd2ef5ad1fef54"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
