)]}'
{"id":"openstack%2Fopenstack-ansible-os_murano~772292","triplet_id":"openstack%2Fopenstack-ansible-os_murano~master~I06e158880863eca1ffa266a911d939256af2409e","project":"openstack/openstack-ansible-os_murano","branch":"master","topic":"osa-new-pip","hashtags":[],"change_id":"I06e158880863eca1ffa266a911d939256af2409e","subject":"Move murano pip packages from constraints to requirements","status":"ABANDONED","created":"2021-01-25 10:16:09.000000000","updated":"2021-11-22 11:51:05.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1563d97441ada2881a524edf108d2ddd6acce470","_number":772292,"virtual_id_number":772292,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-01-26 10:51:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-25 14:57:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3b24b68abe1eebbc81fb15728c0ecd915c8ad7bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-01-25 10:16:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1b6d9eb66579389b33a25bac8f10bcfcbbbdf6a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 14:57:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/07a369ad07d044fcb78b35a28bd4dbd5 : FAILURE in 1h 05m 57s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/6a7440c2bfad47bab1fbcc74348aa30e : FAILURE in 1h 14m 34s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6a7f642946944ae6bc66d666a0a397bf : FAILURE in 1h 09m 01s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/513f9d8573e6440dbd7d6895d98c7b98 : FAILURE in 1h 08m 12s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/310e6ff4220643008bc3850360b4139b : FAILURE in 1h 15m 37s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/efe7671fc1d2478b974da34d2a31a755 : FAILURE in 1h 03m 03s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/dbb48da2df6d40faaeb471db6655eab9 : FAILURE in 53m 33s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e490e3ca80264869bda738cbca5f57ae : FAILURE in 54m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95310390e02249448542f21b3ba0faa6 : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":1},{"id":"c8dcfe4b624e9dd39866eb946b4117819e341032","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-01-26 09:07:41.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"7614fbc66f4a800ac654c0766328a9c457ade617","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 10:51:24.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/63d2c31b812948d9b42c2d6d17c7c412 : FAILURE in 1h 18m 13s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/040e1ad0e4ee46f1be90bc9e06f7642b : FAILURE in 1h 13m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e073037e69ca4b3796e0c36b98731f2f : FAILURE in 1h 15m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/94e8a771035d43b6896db3aa8f360d39 : FAILURE in 1h 06m 48s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/2e0db912a7404f2db2b62991455d40c4 : FAILURE in 1h 26m 12s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d457cbf4b7e2487d8d896d6b4ca03369 : FAILURE in 51m 29s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e0969962efd84fa3a250533b8db4df4c : FAILURE in 57m 32s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/60d3c01794d24ee6af59b17e4868afec : FAILURE in 51m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e00299a579a84f23891845d9ed4bfd5a : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":2},{"id":"1563d97441ada2881a524edf108d2ddd6acce470","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-11-22 11:51:05.000000000","message":"Abandoned\n\nimplemented with https://review.opendev.org/c/openstack/openstack-ansible-os_murano/+/781239","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7355174b466b36043cf1453d58f6812c2fa5c400","revisions":{"aec407c03110b2efd8ad949bb412be334f4c249b":{"kind":"REWORK","_number":1,"created":"2021-01-25 10:16:09.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/92/772292/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_murano","ref":"refs/changes/92/772292/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/92/772292/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/92/772292/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/92/772292/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/92/772292/1"}}},"commit":{"parents":[{"commit":"5e0a408fdb89ff2d5a0fe890230e083113e89451","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/5e0a408fdb89ff2d5a0fe890230e083113e89451"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-01-25 10:15:01.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-01-25 10:16:04.000000000","tz":0},"subject":"Move murano pip packages from constraints to requirements","message":"Move murano pip packages from constraints to requirements\n\nThis is necessary to use the new pip resolver\n\nChange-Id: I06e158880863eca1ffa266a911d939256af2409e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/aec407c03110b2efd8ad949bb412be334f4c249b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/aec407c03110b2efd8ad949bb412be334f4c249b"}]},"branch":"refs/heads/master"},"7355174b466b36043cf1453d58f6812c2fa5c400":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-01-26 09:07:41.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/92/772292/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_murano","ref":"refs/changes/92/772292/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/92/772292/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/92/772292/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/92/772292/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/92/772292/2"}}},"commit":{"parents":[{"commit":"5e0a408fdb89ff2d5a0fe890230e083113e89451","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/5e0a408fdb89ff2d5a0fe890230e083113e89451"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-01-25 10:15:01.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-01-26 09:07:41.000000000","tz":0},"subject":"Move murano pip packages from constraints to requirements","message":"Move murano pip packages from constraints to requirements\n\nThis is necessary to use the new pip resolver\n\nDepends-On: Idaaf3d2262cfa42985ae3ee3aa48726d1b6f4a64\nChange-Id: I06e158880863eca1ffa266a911d939256af2409e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/7355174b466b36043cf1453d58f6812c2fa5c400"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/7355174b466b36043cf1453d58f6812c2fa5c400"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
