)]}'
{"id":"openstack%2Fopenstack-ansible-os_murano~775077","triplet_id":"openstack%2Fopenstack-ansible-os_murano~master~Ic99c64e52875355998f7c2af8b2b325dfa31cddc","project":"openstack/openstack-ansible-os_murano","branch":"master","topic":"bind-to-mgmt","hashtags":[],"change_id":"Ic99c64e52875355998f7c2af8b2b325dfa31cddc","subject":"Add global override for service bind address","status":"ABANDONED","created":"2021-02-11 08:15:18.000000000","updated":"2021-11-22 11:50:53.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"22a70cb9d927659237ae1965c27a1f59d0d3271a","_number":775077,"virtual_id_number":775077,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-02-11 09:57:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-11 09:57:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"09441694d6d91a0a4e4fe56d6ae5c49f63171eaf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-11 08:15:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8f31fd4de6a442a2a997eb02c6e57faefa460e2c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-11 08:18:39.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"939a9ce2d9b9104701b9e0cd39b36a658920362f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 09:57:37.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/843593d431ff4baf82a9787a0d9af25c : FAILURE in 1h 09m 14s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/1dc30d6899c248fa90fa9846129cb87c : FAILURE in 1h 19m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/55699728d0a44d288e0ab0bb0a975085 : FAILURE in 1h 15m 16s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e97ac8d074ce4b5b8969bbb85247831e : FAILURE in 58m 24s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e0923b76b1b9488eae32ee3560cb903a : FAILURE in 1h 15m 44s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c8ff9ca424cf4917bbdf9d8ab6e0e9aa : FAILURE in 1h 13m 26s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/bd7952c973194bdaa8080a8aa6e4f589 : FAILURE in 55m 03s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fc3b953ea9bb45d7a9e6cbfc3b66be6a : FAILURE in 42m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ffcced3cc194d1d85dfa09d8ebe7214 : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":2},{"id":"22a70cb9d927659237ae1965c27a1f59d0d3271a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-11-22 11:50:53.000000000","message":"Abandoned\n\nimplemented with https://review.opendev.org/c/openstack/openstack-ansible-os_murano/+/781239","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d615d9504eb254274e001d767308dbdb4940e0d7","revisions":{"ba219e03ab95b8901df36e1d18f36c330279c54c":{"kind":"REWORK","_number":1,"created":"2021-02-11 08:15:18.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/77/775077/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_murano","ref":"refs/changes/77/775077/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/77/775077/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/77/775077/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/77/775077/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/77/775077/1"}}},"commit":{"parents":[{"commit":"5e0a408fdb89ff2d5a0fe890230e083113e89451","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/5e0a408fdb89ff2d5a0fe890230e083113e89451"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-11 08:14:16.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-11 08:15:06.000000000","tz":0},"subject":"Add global override for service bind address","message":"Add global override for service bind address\n\nChange-Id: Ic99c64e52875355998f7c2af8b2b325dfa31cddc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/ba219e03ab95b8901df36e1d18f36c330279c54c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/ba219e03ab95b8901df36e1d18f36c330279c54c"}]},"branch":"refs/heads/master"},"d615d9504eb254274e001d767308dbdb4940e0d7":{"kind":"REWORK","_number":2,"created":"2021-02-11 08:18:39.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/77/775077/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_murano","ref":"refs/changes/77/775077/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/77/775077/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/77/775077/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/77/775077/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/77/775077/2"}}},"commit":{"parents":[{"commit":"5e0a408fdb89ff2d5a0fe890230e083113e89451","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/5e0a408fdb89ff2d5a0fe890230e083113e89451"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-11 08:14:16.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-11 08:18:34.000000000","tz":0},"subject":"Add global override for service bind address","message":"Add global override for service bind address\n\nAlso rename centos vars file to work with centos-8\n\nChange-Id: Ic99c64e52875355998f7c2af8b2b325dfa31cddc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/d615d9504eb254274e001d767308dbdb4940e0d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/d615d9504eb254274e001d767308dbdb4940e0d7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
