)]}'
{"id":"openstack%2Fopenstack-ansible-os_murano~803996","triplet_id":"openstack%2Fopenstack-ansible-os_murano~stable%2Fussuri~I35da8bfb3d714d086fdbd376c2f3cf9eba4e6b21","project":"openstack/openstack-ansible-os_murano","branch":"stable/ussuri","topic":"debian-stretch-rm","hashtags":[],"change_id":"I35da8bfb3d714d086fdbd376c2f3cf9eba4e6b21","subject":"Remove Debian jobs","status":"MERGED","created":"2021-08-09 22:57:28.000000000","updated":"2021-08-10 23:06:32.000000000","submitted":"2021-08-10 23:00:54.000000000","submitter":{"_account_id":32627,"name":"ianw.admin","username":"ianw.admin"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"803996-debian-stretch-rm","meta_rev_id":"a5bc542e2ee6579b44e02d1963f1192f46c3be06","_number":803996,"virtual_id_number":803996,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":32627,"name":"ianw.admin","username":"ianw.admin"},"all":[{"value":2,"date":"2021-08-10 23:00:54.000000000","_account_id":32627,"name":"ianw.admin","username":"ianw.admin"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-08-10 00:52:07.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32627,"name":"ianw.admin","username":"ianw.admin"},"all":[{"value":2,"date":"2021-08-10 23:00:54.000000000","_account_id":32627,"name":"ianw.admin","username":"ianw.admin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32627,"name":"ianw.admin","username":"ianw.admin"},"all":[{"value":1,"date":"2021-08-10 23:00:54.000000000","_account_id":32627,"name":"ianw.admin","username":"ianw.admin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32627,"name":"ianw.admin","username":"ianw.admin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-10 00:52:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-10 23:00:54.000000000","updated_by":{"_account_id":32627,"name":"ianw.admin","username":"ianw.admin"},"reviewer":{"_account_id":32627,"name":"ianw.admin","username":"ianw.admin"},"state":"REVIEWER"}],"messages":[{"id":"874e912201a0fcd934b9d63725e817bf20128888","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-08-09 22:57:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e46e3ba1043430c74e7ade63fb4cf3e2bbc87d5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 00:52:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/760a186f6dda46a586b35b6b52887f50 : FAILURE in 1h 18m 35s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a4c68b34a6cc44dfb48e466643084798 : FAILURE in 1h 40m 34s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/257c37c03f0e4f078649ac4701b4cef5 : FAILURE in 1h 02m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/75e5a5668e4545ca8b963dd61750a237 : FAILURE in 1h 23m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a9c36338e574a67a37efc7d6d421cff : FAILURE in 1h 48m 19s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ca58e9050a4d410486c9848cee4c8589 : FAILURE in 1h 15m 25s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9dadf647ee524bfa8634642e6b9830dd : FAILURE in 1h 13m 49s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7d43c1e46543435e8f4da8dc41d58977 : SUCCESS in 1h 11m 26s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/c70f2e64e0ed46758c0328b6fa19c3e3 : FAILURE in 50m 49s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/51065a049e484cf9baf4eecb20769234 : SUCCESS in 1h 03m 07s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/de7270467f9e47a2b7e9062d39f12426 : SUCCESS in 1h 20m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbc52162d1a048d9b5c73941ba1869b5 : SUCCESS in 4m 40s","accounts_in_message":[],"_revision_number":1},{"id":"93d247db477b0b68d2db2f9d569fabe81be35fd6","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-08-10 01:30:32.000000000","message":"Topic set to debian-stretch-rm","accounts_in_message":[],"_revision_number":1},{"id":"15bbd9974301ceea9c8c2fbff171b7b887350b5a","author":{"_account_id":32627,"name":"ianw.admin","username":"ianw.admin"},"date":"2021-08-10 23:00:54.000000000","message":"Patch Set 1: Verified+2 Workflow+1 Code-Review+2\n\nBypassing Zuul to merge this.","accounts_in_message":[],"_revision_number":1},{"id":"73f39c6e0110ef663626b2ae12aa740f1484b1fd","tag":"autogenerated:gerrit:merged","author":{"_account_id":32627,"name":"ianw.admin","username":"ianw.admin"},"date":"2021-08-10 23:00:54.000000000","message":"Change has been successfully merged by ianw.admin","accounts_in_message":[],"_revision_number":1},{"id":"a5bc542e2ee6579b44e02d1963f1192f46c3be06","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 23:06:32.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b6dab515694450f890eb64f0b28b939 : FAILURE in 34s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c512282bb4f4c7db625d18256055f2e24de119e7","revisions":{"c512282bb4f4c7db625d18256055f2e24de119e7":{"kind":"REWORK","_number":1,"created":"2021-08-09 22:57:28.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/96/803996/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_murano","ref":"refs/changes/96/803996/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/96/803996/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/96/803996/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/96/803996/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/96/803996/1"}}},"commit":{"parents":[{"commit":"d106eabcb897f678ecc1c0501614abd7fdee2f2d","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/d106eabcb897f678ecc1c0501614abd7fdee2f2d"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2021-08-09 22:54:36.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2021-08-09 22:57:06.000000000","tz":600},"subject":"Remove Debian jobs","message":"Remove Debian jobs\n\nDebian Stretch (the current target of \"debian-stable\") is no longer\nsupported by infra.  Remove these jobs.\n\nCherry-Picked-From: cf9c3c510a0c374eabc80667da3004ddbd955246\nChange-Id: I35da8bfb3d714d086fdbd376c2f3cf9eba4e6b21\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/c512282bb4f4c7db625d18256055f2e24de119e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/c512282bb4f4c7db625d18256055f2e24de119e7"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":32627,"name":"ianw.admin","username":"ianw.admin"}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":32627,"name":"ianw.admin","username":"ianw.admin"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":32627,"name":"ianw.admin","username":"ianw.admin"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
