)]}'
{"id":"openstack%2Fopenstack-ansible-os_murano~803997","triplet_id":"openstack%2Fopenstack-ansible-os_murano~stable%2Fvictoria~I35da8bfb3d714d086fdbd376c2f3cf9eba4e6b21","project":"openstack/openstack-ansible-os_murano","branch":"stable/victoria","topic":"debian-stretch-rm","hashtags":[],"change_id":"I35da8bfb3d714d086fdbd376c2f3cf9eba4e6b21","subject":"Remove Debian jobs","status":"MERGED","created":"2021-08-09 22:58:08.000000000","updated":"2021-08-10 23:06:34.000000000","submitted":"2021-08-10 23:01:07.000000000","submitter":{"_account_id":32627,"name":"ianw.admin","username":"ianw.admin"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"803997-debian-stretch-rm","meta_rev_id":"d53cdee776d28aeadb6074bddaf44a5dc7ca7ee8","_number":803997,"virtual_id_number":803997,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":32627,"name":"ianw.admin","username":"ianw.admin"},"all":[{"value":2,"date":"2021-08-10 23:01:07.000000000","_account_id":32627,"name":"ianw.admin","username":"ianw.admin"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-08-10 00:33:09.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32627,"name":"ianw.admin","username":"ianw.admin"},"all":[{"value":2,"date":"2021-08-10 23:01:07.000000000","_account_id":32627,"name":"ianw.admin","username":"ianw.admin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32627,"name":"ianw.admin","username":"ianw.admin"},"all":[{"value":1,"date":"2021-08-10 23:01:07.000000000","_account_id":32627,"name":"ianw.admin","username":"ianw.admin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32627,"name":"ianw.admin","username":"ianw.admin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-10 00:33:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-10 23:01:07.000000000","updated_by":{"_account_id":32627,"name":"ianw.admin","username":"ianw.admin"},"reviewer":{"_account_id":32627,"name":"ianw.admin","username":"ianw.admin"},"state":"REVIEWER"}],"messages":[{"id":"f86d2c70afee9f442ddf72a696451c75478304dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-08-09 22:58:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"02666f8bcd41d134f6b7a15d149ea59a2996e0cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 00:33:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/74de65edd6204c95942bb53ec2f7f8da : FAILURE in 1h 02m 07s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/6983298a81e84ad7b11fda0e7e390a22 : FAILURE in 1h 17m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3962779f6ac5463fa5528f1bd3cf8b58 : FAILURE in 1h 09m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1d539290f3c047888431e5d861226be4 : FAILURE in 1h 02m 13s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/1464dacecac345b6a7620e226a791f68 : FAILURE in 1h 06m 05s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/23e62215553a4b9c971c80f6a587029e : FAILURE in 1h 28m 40s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/2289f36ae17e4186aea473b51f57151f : FAILURE in 1h 02m 08s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/15403b7c490742b491c1dfdf58802665 : FAILURE in 50m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2338eb100ab34c6cb10d33a287eef53b : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":1},{"id":"97a7b5af672701648bee418fb52e1d7a698f0470","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-08-10 01:31:15.000000000","message":"Topic set to debian-stretch-rm","accounts_in_message":[],"_revision_number":1},{"id":"c0c84eeb21928d54229023c0e3538b764aeb1a47","author":{"_account_id":32627,"name":"ianw.admin","username":"ianw.admin"},"date":"2021-08-10 23:01:07.000000000","message":"Patch Set 1: Verified+2 Workflow+1 Code-Review+2\n\nBypassing Zuul to merge this.","accounts_in_message":[],"_revision_number":1},{"id":"6e16c2d05935c272876c2f8e11a153e31edf0359","tag":"autogenerated:gerrit:merged","author":{"_account_id":32627,"name":"ianw.admin","username":"ianw.admin"},"date":"2021-08-10 23:01:07.000000000","message":"Change has been successfully merged by ianw.admin","accounts_in_message":[],"_revision_number":1},{"id":"d53cdee776d28aeadb6074bddaf44a5dc7ca7ee8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 23:06:34.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23b5eb8b41894773a1c0b4f1e133ce38 : FAILURE in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2e084ae35b4ac541cab9417623d31599bcaff863","revisions":{"2e084ae35b4ac541cab9417623d31599bcaff863":{"kind":"REWORK","_number":1,"created":"2021-08-09 22:58:08.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/97/803997/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_murano","ref":"refs/changes/97/803997/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/97/803997/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/97/803997/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/97/803997/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/97/803997/1"}}},"commit":{"parents":[{"commit":"31fbbb2d8952aaf88337d307b68f9119aa1a725a","subject":"Merge \"Updated from OpenStack Ansible Tests\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/31fbbb2d8952aaf88337d307b68f9119aa1a725a"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2021-08-09 22:54:36.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2021-08-09 22:57:44.000000000","tz":600},"subject":"Remove Debian jobs","message":"Remove Debian jobs\n\nDebian Stretch (the current target of \"debian-stable\") is no longer\nsupported by infra.  Remove these jobs.\n\nCherry-Picked-From: cf9c3c510a0c374eabc80667da3004ddbd955246\nChange-Id: I35da8bfb3d714d086fdbd376c2f3cf9eba4e6b21\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/2e084ae35b4ac541cab9417623d31599bcaff863"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/2e084ae35b4ac541cab9417623d31599bcaff863"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":32627,"name":"ianw.admin","username":"ianw.admin"}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":32627,"name":"ianw.admin","username":"ianw.admin"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":32627,"name":"ianw.admin","username":"ianw.admin"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
