)]}'
{"id":"openstack%2Fopenstack-ansible-os_murano~846147","triplet_id":"openstack%2Fopenstack-ansible-os_murano~master~I9c432b4a24127fad3bfe0f24d02fa750170f483e","project":"openstack/openstack-ansible-os_murano","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2022-06-17 17:58:05.000000000","reason":"Dmitriy Rabotyagov replied on the change"}},"hashtags":[],"change_id":"I9c432b4a24127fad3bfe0f24d02fa750170f483e","subject":"Delegate package import to setup host","status":"MERGED","created":"2022-06-16 11:30:28.000000000","updated":"2022-06-17 22:48:36.000000000","submitted":"2022-06-17 22:47:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"846147","meta_rev_id":"cd20bb5a38f89bf787fb632fd326c569fa9fb7ad","_number":846147,"virtual_id_number":846147,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-17 22:47:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":2,"date":"2022-06-16 15:21:14.000000000","_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-06-16 18:21:54.000000000","_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-06-16 18:21:54.000000000","_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-16 15:21:14.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"},{"updated":"2022-06-16 15:23:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-16 18:21:54.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"4289b4957e82a8bf63c24fc08eb96b3b0400debf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-06-16 11:30:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"31ab2c678ec91a45604eaebf6ac5fe8f763a1447","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-06-16 12:55:44.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"3c86c8031f98bf48444987bab8102133c704c382","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-06-16 15:21:14.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"7fdf708cbc828019271a8b6b4bff0f9876b7dc26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 15:23:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/fc0e0bb5c26443d7ad947e71344a5b26 : SUCCESS in 1h 11m 34s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/6d62e1942d4148a29dd996dc40169faf : SUCCESS in 1h 38m 58s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/c673c9ddf7ba49a29a158530a2a5a054 : SUCCESS in 1h 20m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/27e0b989c9934ad4a4d20aee0b5144f2 : SUCCESS in 1h 28m 34s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c7b0d80e06014516be23ac02f2092a63 : SUCCESS in 1h 30m 15s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/4e79d01dd7524bf9a7284853da682031 : SUCCESS in 1h 40m 30s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/8e4c6928626f49a5b87e1914f2ac1bb8 : FAILURE in 2h 21m 25s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/179f54ab0f364134b78332226f71988e : FAILURE in 1h 41m 03s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/1d1c16e26c85411897dee2f411e318f5 : SUCCESS in 1h 01m 52s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7d326cb1bd7549e69e435793029b199b : SUCCESS in 55m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e34c13ce593417e96b397367fa7de55 : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":2},{"id":"6ac852162c43a24b8bd8577dfcd22392aba3e04a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-06-16 18:21:54.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d8954f0bda6978aceb63c3e9232b5e5f41175ed9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 18:22:09.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n\nWarning:\n  Change 846162 in project openstack/openstack-ansible does not share a change queue with 846147 in project openstack/openstack-ansible-os_murano","accounts_in_message":[],"_revision_number":2},{"id":"f961526ddee241dd4383e05440cf5f426a820dbc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 20:55:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/9ed2fcdd9afc440abd4d2b32123c2cf1 : SUCCESS in 1h 27m 40s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/ca8f1c5b15834f47ab7e8d49ea46089a : SUCCESS in 1h 46m 07s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/24e5bdad07f645cc81eca4aff27c9c50 : SUCCESS in 1h 29m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2a5865f56d1e400384173e8c10caf75a : SUCCESS in 1h 17m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2b64ea385beb4b37b48ff9d5c1c700ca : SUCCESS in 1h 30m 48s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/6965f11a79174b219728d1809201df14 : SUCCESS in 1h 37m 33s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/e9fd2250667c43efae24190b5a7bb519 : SUCCESS in 2h 15m 20s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e130765347fe436dafa4e80b41bf07ab : SUCCESS in 2h 12m 49s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/616e702aceac45a29fe7329c8c2b5085 : SUCCESS in 1h 22m 36s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f61723ae4514457bba54472e1ab39f93 : SUCCESS in 1h 02m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d15b343612a94a3aab4f3c782636c571 : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":2},{"id":"940e7667d1c2953cbbb5e1afa722848665a6420d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 20:55:37.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n\nWarning:\n  Change 846162 in project openstack/openstack-ansible does not share a change queue with 846147 in project openstack/openstack-ansible-os_murano","accounts_in_message":[],"_revision_number":2},{"id":"05802e5667497006a846a70b23ff5c4180844813","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-06-16 21:20:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"634738e1bffef3df850913ff7aed5e15932759ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 23:49:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/246546f2306643e79a7be7d87c534f24 : SUCCESS in 1h 25m 43s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/eba28ca0c0ea46e7b2edb49c5ff45b4a : SUCCESS in 1h 41m 55s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/a36aac290d7841598c2904d0917cedbb : SUCCESS in 1h 28m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/56f9f1597ae548e0a00928ddc596848b : SUCCESS in 1h 40m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/94ce33ac3e8d4a52aac3131d85d979c9 : SUCCESS in 1h 38m 17s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/049f9ac45c1140379a15368b839d8f25 : SUCCESS in 1h 36m 37s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/964dc692808642569dc010eb244409f4 : SUCCESS in 2h 20m 46s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/35e41703e47e4036a49dcc62e0746428 : SUCCESS in 1h 59m 31s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/a96f68ffde5444c9a04bddbdc0578a40 : SUCCESS in 1h 14m 12s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0b63e8bea34a4732964f721dce81de1c : SUCCESS in 1h 02m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a3d7713179247d0905f2b37373dea26 : SUCCESS in 3m 23s","accounts_in_message":[],"_revision_number":2},{"id":"97cebce1ab6edc4f68754c45cc693fe8efb1d4f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 23:49:58.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n\nWarning:\n  Change 846162 in project openstack/openstack-ansible does not share a change queue with 846147 in project openstack/openstack-ansible-os_murano","accounts_in_message":[],"_revision_number":2},{"id":"676baa5d78adc89237e3dc739f2306b0770d341a","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-06-17 13:53:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ca9c2f722e5573abdf99ab71da30a48bc37be5bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 16:13:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/6091fc09076f484fa86beb286c545027 : SUCCESS in 1h 29m 34s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/612c02c8905e4dd8b1e8ab652be76f6c : SUCCESS in 1h 33m 53s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/36db63beabb546149127ef3ffed02833 : SUCCESS in 1h 33m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/58fca4d3bf634faaa877c49f633ac2b6 : SUCCESS in 1h 29m 49s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ea89ec8f67cb43aaa350e5464b39ac37 : SUCCESS in 1h 35m 02s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/757430255ae446618a9d900cdd47dcb3 : SUCCESS in 1h 40m 17s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/ae5e7f3a48d94a048cfadb85320ec492 : SUCCESS in 2h 01m 43s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b8a2b9c2ec6e4b8888574ce9f2f557bd : SUCCESS in 2h 16m 32s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/a0ff0fdf20ba44638abc0dd7feb1833c : SUCCESS in 1h 15m 09s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0bfe4e7e5ddb41d98184e104f2eac8f7 : SUCCESS in 1h 07m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6147a8570efb47cc82d23968562a3b59 : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":2},{"id":"4f12e9dc60834e2684599fcedb30c53f184fb434","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 16:14:05.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n\nWarning:\n  Change 846162 in project openstack/openstack-ansible does not share a change queue with 846147 in project openstack/openstack-ansible-os_murano","accounts_in_message":[],"_revision_number":2},{"id":"0d9e48133725e0fa72cd473b207efc54345567d4","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-06-17 17:58:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2235808771a926d7faecb0eaf964c8bf707314b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 20:27:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/28f8a89595014814847cbf268c571967 : SUCCESS in 1h 24m 11s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/07d599d86de741b8baa5bf0684675647 : SUCCESS in 1h 32m 29s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/04b65115afd44d3bada2f477d42963d8 : SUCCESS in 1h 34m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/426e499668004251a134dface5d416d1 : SUCCESS in 1h 42m 26s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/97658f70259b4f5ba5975c8d7250f7ef : SUCCESS in 1h 32m 14s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/10a215bc9e004966aa58623ec2c1208b : SUCCESS in 1h 41m 25s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/de587802535348c78564dca17c6bcd1b : SUCCESS in 2h 14m 23s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fead55ad66e444aa8eeef0fb7aea6b13 : SUCCESS in 2h 07m 42s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/52185f1bb79f4332be823a27fb7391ae : SUCCESS in 1h 14m 36s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8492957a735b4398897a41f330a5baa9 : SUCCESS in 1h 01m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a20a57b983a1479d98f8ef5ff6229f35 : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":2},{"id":"81369b9013ebbd3bd2b55245ced41f7448006c10","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 20:29:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"81a7a349ff1a921445dcb9277d724e97d8ccda50","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 22:47:26.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/9b9f6dbaba624da49e1e47e8297c8431 : SUCCESS in 1h 28m 35s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/a07fc8193c564d1089f530f5c2d8bce0 : SUCCESS in 1h 36m 10s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/a424483e3ab745b0964d30264729bee8 : SUCCESS in 1h 25m 24s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0b74e3ad1da940d58968d6032b9c7dbf : SUCCESS in 1h 31m 46s\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/636ee6b83c4a46dd97e04a0f552a42b8 : SUCCESS in 1h 39m 31s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3dfd73a40a004143a3074e4cacc6696c : SUCCESS in 1h 08m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bd24c2d986b4121b0aaa22d5e358e4b : SUCCESS in 3m 37s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/92c2f001bdf64a1daedc895ae267c580 : SUCCESS in 2h 12m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1ef7d98959a768b8dee4dd9b580ee1f83869ba5d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 22:47:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"cd20bb5a38f89bf787fb632fd326c569fa9fb7ad","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 22:48:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d183c090ff9e4a96bf27ebb8ee2b8389 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c85b8def8b66625b421a945a581fe80c4ce888a9","revisions":{"0643030626780eca0e35c0bbd4ae5ba0550edbe4":{"kind":"REWORK","_number":1,"created":"2022-06-16 11:30:28.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/47/846147/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_murano","ref":"refs/changes/47/846147/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/47/846147/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/47/846147/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/47/846147/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/47/846147/1"}}},"commit":{"parents":[{"commit":"6960333d5d5f27eeed4532f795237408696b0563","subject":"Switch sphinx language to en","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/6960333d5d5f27eeed4532f795237408696b0563"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-06-16 11:30:20.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-06-16 11:30:20.000000000","tz":120},"subject":"Delegate package import to setup host","message":"Delegate package import to setup host\n\nInstead of running muranoclient on murano host, we\u0027re delegating\nexecution to service_setup_host. This reduce\ncomplexity and remove some requirements for murano hosts.\n\nWith that we also replace usage of muranoclient\nwith openstackclient.\n\nChange-Id: I9c432b4a24127fad3bfe0f24d02fa750170f483e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/0643030626780eca0e35c0bbd4ae5ba0550edbe4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/0643030626780eca0e35c0bbd4ae5ba0550edbe4"}]},"branch":"refs/heads/master"},"c85b8def8b66625b421a945a581fe80c4ce888a9":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-06-16 12:55:44.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/47/846147/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_murano","ref":"refs/changes/47/846147/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/47/846147/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/47/846147/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/47/846147/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/47/846147/2"}}},"commit":{"parents":[{"commit":"6960333d5d5f27eeed4532f795237408696b0563","subject":"Switch sphinx language to en","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/6960333d5d5f27eeed4532f795237408696b0563"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-06-16 11:30:20.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-06-16 12:55:44.000000000","tz":0},"subject":"Delegate package import to setup host","message":"Delegate package import to setup host\n\nInstead of running muranoclient on murano host, we\u0027re delegating\nexecution to service_setup_host. This reduce\ncomplexity and remove some requirements for murano hosts.\n\nWith that we also replace usage of muranoclient\nwith openstackclient.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/846162\nChange-Id: I9c432b4a24127fad3bfe0f24d02fa750170f483e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/c85b8def8b66625b421a945a581fe80c4ce888a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/c85b8def8b66625b421a945a581fe80c4ce888a9"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
