)]}'
{"id":"openstack%2Fopenstack-ansible-os_murano~881972","triplet_id":"openstack%2Fopenstack-ansible-os_murano~stable%2Fzed~I3ecb1d9892145906d2f384d53ec47e722ab2649b","project":"openstack/openstack-ansible-os_murano","branch":"stable/zed","topic":"cherrypick-osa/systemd_restart_on_unit_change-hbqfsn1p95","hashtags":[],"change_id":"I3ecb1d9892145906d2f384d53ec47e722ab2649b","subject":"Ensure service is restarted on unit file changes","status":"MERGED","created":"2023-05-02 11:01:31.000000000","updated":"2023-05-02 20:17:50.000000000","submitted":"2023-05-02 20:16:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"881972-cherrypick-osa/systemd_restart_on_unit_change-hbqfsn1p95","cherry_pick_of_change":879974,"cherry_pick_of_patch_set":2,"meta_rev_id":"8e4dfb7218f3fa91628870572d12061896389932","_number":881972,"virtual_id_number":881972,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-02 20:16:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2023-05-02 17:48:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2023-05-02 17:48:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-02 13:51:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f99bfcd17af634ee8ee547f02fcc60033416fceb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-05-02 11:01:31.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"1ae33cade3123d56d2dff7eb19bdd4c355608fb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 13:51:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c6d0ba1d1e1455c9721671248fb5b9f\n\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/3435ab403c2e46f7bcbeca3efbeea3eb : SUCCESS in 1h 20m 08s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/d80904848ef64c1abea328e2321425af : FAILURE in 1h 24m 37s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/79c95b49ba964cc8b06d470c149d61ab : SUCCESS in 1h 21m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c43f260b6927471ab1276d9b734d1f84 : SUCCESS in 1h 22m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/748798ede9bc478b8c14133f2e64c560 : SUCCESS in 1h 13m 39s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f30518fb075244d19cf3280450ad26d1 : SUCCESS in 2h 29m 44s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e193444917464cfbac4baa36b70b10b8 : SUCCESS in 1h 58m 44s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/afa05a02591344e1b45c3bb87d6edd25 : FAILURE in 41m 00s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0afe6316174444b8be735e29170dd31a : FAILURE in 1h 06m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fb300eefe2b4c7d938aeda162523ef9 : SUCCESS in 2m 57s","accounts_in_message":[],"_revision_number":1},{"id":"b44f8e0cba843162f54a94fedb0c92255cf5f969","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-05-02 17:48:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b78a183061d6f6638d9ba1e568779639e32e5c02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 17:49:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"da1d18681e5139f0f9784e303625d27b6489fbcb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 20:16:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eea0367072a84d68b8c65e8bac365b34\n\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/846e5d7f22aa46c6816317d9e7af9ec2 : SUCCESS in 1h 12m 07s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/db8a3a4da707490dab6ba9b105eddbc5 : SUCCESS in 1h 27m 21s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bfd7fe65a6bc4a439a8d12cc70b978fd : SUCCESS in 1h 15m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e908f4cb6c44cc69c3038620c4ba066 : SUCCESS in 3m 00s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/55de091813a548629d46decb91a61217 : SUCCESS in 2h 08m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9bb538c74c3415f30586e0b58a6b259de5cefc79","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 20:16:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8e4dfb7218f3fa91628870572d12061896389932","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 20:17:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96cebbf6d4d64d419f0d0adf3f406aa6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0525c4df74340fe876c4f4ef8d4dea2 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e22f175dea9a5fc3264f985d252a64595a7a9a73","revisions":{"e22f175dea9a5fc3264f985d252a64595a7a9a73":{"kind":"REWORK","_number":1,"created":"2023-05-02 11:01:31.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/72/881972/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_murano","ref":"refs/changes/72/881972/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/72/881972/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/72/881972/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/72/881972/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/72/881972/1"}}},"commit":{"parents":[{"commit":"3dc5d4a0cfe7c896b5d0570f201daf90876c6a16","subject":"Update tox.ini to work with 4.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/3dc5d4a0cfe7c896b5d0570f201daf90876c6a16"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-04-10 14:37:51.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-05-02 11:01:31.000000000","tz":0},"subject":"Ensure service is restarted on unit file changes","message":"Ensure service is restarted on unit file changes\n\nAt the moment we don\u0027t restart services if systemd unit file is changed.\n\nWe knowingly prevent systemd_service role handlers to execute\nby providing `state: started` as otherwise service will be restarted twice.\nWith that now  we ensure that role handlers will also listen for systemd\nunit changes.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_heat/+/879963\nChange-Id: I3ecb1d9892145906d2f384d53ec47e722ab2649b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/e22f175dea9a5fc3264f985d252a64595a7a9a73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/e22f175dea9a5fc3264f985d252a64595a7a9a73"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
