)]}'
{"id":"openstack%2Fopenstack-ansible-os_murano~890127","triplet_id":"openstack%2Fopenstack-ansible-os_murano~master~I3b0657b8d3b062eb6c8af44f0905066dba150a88","project":"openstack/openstack-ansible-os_murano","branch":"master","topic":"osa/db_port","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-08-21 09:11:48.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2023-08-21 09:11:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3b0657b8d3b062eb6c8af44f0905066dba150a88","subject":"Use proper galera port in configuration","status":"MERGED","created":"2023-07-31 13:19:59.000000000","updated":"2023-08-21 09:12:56.000000000","submitted":"2023-08-21 09:11:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"890127-osa/db_port","meta_rev_id":"24c3a44e9547ca3279f1ba1435456c354f4a1cb6","_number":890127,"virtual_id_number":890127,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-21 09:11:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2023-08-21 07:11:05.000000000","_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-08-17 14:58:38.000000000","_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2023-08-21 07:11:05.000000000","_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-31 16:27:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-17 07:34:11.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2023-08-17 11:02:54.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"c113462d3342f259999828f138f19ab1f1a4cd2b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-07-31 13:19:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c700fbdd4fdf8b90231f92e9423363b287d77dc0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-31 16:27:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/951ddd15c4b8449ab8a48bd9efc5d4af\n\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/2718d5d9aa8f4381b0baefecdf308ed2 : SUCCESS in 1h 44m 16s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/73f15c72595e4246bf6d106c360475cb : SUCCESS in 1h 56m 47s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/48b302bcc883458cb355af9ecf671ca7 : SUCCESS in 2h 02m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/db4102b2dafa49168f07840e538fe243 : FAILURE in 30m 40s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9a3c9582fe3f4379960f66d6d3859d92 : SUCCESS in 1h 57m 27s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ab53a8478ac547b78052819dd77b564c : FAILURE in 27m 33s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5a592458866d498789a18c9a89ca9abb : SUCCESS in 2h 58m 16s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/28c89d0eef92460dbe52c698d97e81c6 : FAILURE in 23m 51s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5062046ecfb0446ebbfc225d2d855e95 : SUCCESS in 1h 17m 09s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a389307734f04cd883bdc7fa879efb0e : FAILURE in 21m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fd443539cbb444da9cc90eb5e0dba53 : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":1},{"id":"664f81ae818e4ca69819b52a58290e55225a18ee","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-07 07:00:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"78067d47fceb5c20eb3b8c4d3c2764bbfefdaed2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-07 09:57:58.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/757384ed08054549a57fa5d174ca472b\n\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/77a5448debed4a8baf0eaf0d8ee1a850 : SUCCESS in 1h 17m 50s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/0a06c44146e34308b3751ddd517a1c7e : SUCCESS in 1h 45m 06s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a7ebd8eb843043ec920245348cbaba1a : SUCCESS in 1h 53m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/da8c08ec20f246d1bce6e92cee4cb54a : SUCCESS in 1h 37m 32s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2f20316534374bdaa52aab80b7e0251b : SUCCESS in 1h 42m 29s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0c48eb1465fa4390a7a78098eeca0abb : FAILURE in 27m 15s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c6a64fb4a25e4951ba401f4841f9c348 : SUCCESS in 2h 49m 45s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/73e945ad2666418a9668680ec1239566 : FAILURE in 22m 32s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/cdac579efb2c4b39bced4f7ba08e704d : SUCCESS in 1h 04m 09s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e65a86ab14ab4307bab7421d1b52a805 : FAILURE in 20m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2aaf67f50e94706b2b7b5c8a57d901e : SUCCESS in 5m 51s","accounts_in_message":[],"_revision_number":1},{"id":"5d2e9abfdf8d2a0dd58786a770e130c8db4cabba","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-08 15:35:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f3d08b4801afe5e33b6993144e253e00608416c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 18:40:07.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5a812087b75745dcaab84fa817fb8b74\n\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/d3bffd032c5644e68a0f0cfd03ccc555 : SUCCESS in 1h 32m 38s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/f00359249c354f5aa21d542135e55b02 : SUCCESS in 1h 38m 54s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3e956c7de2914b2a90fc7e19d52f9ea8 : SUCCESS in 1h 32m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3582ec6ac18d43a6a073a24e3190c502 : FAILURE in 25m 02s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a7e2d526c7a3423faa03b892271108d7 : SUCCESS in 1h 50m 51s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c2d9017907df446e9f5670e4d13e6b22 : SUCCESS in 1h 46m 23s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1b773eb85e2c40dc9d848199de78bdb1 : SUCCESS in 2h 46m 09s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/64b116f12863428da502878db9b43052 : SUCCESS in 2h 26m 04s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c8b5994d575b48eca064c7f836906696 : SUCCESS in 1h 10m 17s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/158c5ec44b70414ea41af852036baa79 : SUCCESS in 1h 00m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fa2d9581b604663b8f67cd0324dee0f : SUCCESS in 3m 07s","accounts_in_message":[],"_revision_number":1},{"id":"e9fd5d27b72f2cc74b379b9352299707cb67fe56","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-09 18:51:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8e68bf77b062fc2d1eed500dba009babab597b51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 21:09:44.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/15da184e2b464df18d35f1a4430f5328\n\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/93dcaf304858432cb83665e5351f7ba8 : SUCCESS in 1h 23m 09s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/10e19f0ab92d4dc4ba6007c915986331 : SUCCESS in 1h 40m 22s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7a40d732c0ec4afe9011e0664faa93d6 : SUCCESS in 1h 52m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2081e0772efa4820b254dbfd141e9faa : SUCCESS in 1h 40m 41s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4f01d8e5f8c74f9db10d67c82d6acb27 : SUCCESS in 1h 54m 58s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/427e2b7c9efb4aaab1ef2edd272a3251 : SUCCESS in 1h 39m 57s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0356b4ebf7874769ae14e030353a5576 : SUCCESS in 2h 10m 59s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2804e8c771c34e4d9959d7f1eeb8db3c : FAILURE in 22m 12s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/504857dd097e47029a4a8482326afef6 : SUCCESS in 1h 03m 12s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9a5556493e3c419091d9562f65ab48e2 : FAILURE in 18m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52a2b008f51f47eb879a94780353be9a : SUCCESS in 3m 00s","accounts_in_message":[],"_revision_number":1},{"id":"486f8960011735f5527dd5c789be191fab1a7934","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-16 13:36:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"79e20914637a16aae8e256a54cc080857d2897a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 16:40:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4134b24b7662416c8518ff9a7c1f0d84\n\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/72b32b7984db4a419bbc34c21d2418d3 : SUCCESS in 1h 45m 51s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/a7469741f4ce41649723c2b8d076f345 : SUCCESS in 1h 57m 57s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7687c97efe8f48499cf06957b969c979 : SUCCESS in 1h 50m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d6ab9b324b3841918b9022ba796be4c7 : SUCCESS in 1h 39m 53s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e7fa61072f0f49698d42ebee86678ff4 : SUCCESS in 1h 48m 58s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/66ed5da65bb8445eb946f448863314a6 : SUCCESS in 2h 02m 28s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7a38cf02d75d4ea5b843e03943c1aadd : SUCCESS in 2h 54m 17s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/95460c32eaff4000a2a09c0651b8270f : SUCCESS in 2h 34m 09s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0c1da56f21d64ec9874c7a2ec7a7bab8 : SUCCESS in 1h 21m 14s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c5145b5519c546b9a5c08b7153ce81d2 : SUCCESS in 1h 04m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3433990fdb441fb84f3ac760ddd4fe3 : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":1},{"id":"cb731ccf58776c4bdd13896a80ceac09d837f54f","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2023-08-17 07:34:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"00ce6fc3fe051f7ce9fda092adde46b11d1c5316","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-08-17 11:02:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2010ffb5418b9d768889044ea426d47a590c1d09","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 11:04:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5cbb06617f154a144168e71dbc3139c33dc778f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 13:10:39.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c52fda193b6c4b1c80c5dffcd535e2c0\n\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/ca34e2194fbc42fe9b55eff6bde0d319 : SUCCESS in 1h 30m 48s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d9c641b3c4064b43af0a3323b54b0cbb : SUCCESS in 1h 58m 39s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/05a41aba46c646d8a078c2a4c15f215f : SUCCESS in 1h 40m 37s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f434573d533d47a49867959140566de2 : SUCCESS in 1h 56m 39s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/482ac476171d4b64805414e9409bcd4e : SUCCESS in 1h 40m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/179ecd1d872643cda50e1c04077af23b : FAILURE in 3m 22s","accounts_in_message":[],"_revision_number":1},{"id":"04023a94bcc0bb71914192226eba7c977683765f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-17 14:58:38.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"77660f75be260029b73db6cafcd93dbacc37bb5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 18:32:46.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/190f402def1441859fd16e50b22acbc8\n\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/4fe99bf6c60a43749d7c0f7f3288fd6e : SUCCESS in 2h 02m 41s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/2a73911b9ba745c0b3ec3bac8fff1aa4 : SUCCESS in 2h 07m 02s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1fcda11d1b5b4f29a5b6c53d85133139 : SUCCESS in 1h 49m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b96f148dcfcb49d9bb19bd63b4072579 : SUCCESS in 1h 59m 22s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/722dea1df01648b6be9b7278c917a8f3 : SUCCESS in 2h 02m 01s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/04db087ca5864267935eae699ee67312 : SUCCESS in 1h 45m 16s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/dd1429521f9949439fb30e4f3522bc01 : TIMED_OUT in 3h 05m 48s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1d10f428c04a4b47b05c093c40613e8d : SUCCESS in 2h 43m 22s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bc256f97bac34112ac5c03b5fb9da344 : SUCCESS in 1h 07m 34s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/074a245fe93f4991875a53ded3b1019d : SUCCESS in 1h 12m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6224968753243b5910d5f2c6b102d8d : SUCCESS in 6m 02s","accounts_in_message":[],"_revision_number":2},{"id":"e41694030f4fd213bef769739ee339733bcd8677","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-18 13:48:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f7268af6a869543f689617c126df930012b0f91d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 16:31:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a3295f1b3ed46a4abe7271de7d10753\n\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/2b82239988f84755a23e97e71a616571 : SUCCESS in 1h 32m 34s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/1c44b0357d2d4a199267ca2ed83ca8f2 : SUCCESS in 1h 42m 21s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c4f31738d6a04f6ba2b865747c7fa3e4 : SUCCESS in 1h 31m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e226a95105444a4ab816f4ec6fef3779 : SUCCESS in 1h 31m 16s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/cf25bdb4a19445278caf623582b1866f : SUCCESS in 1h 56m 54s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/16c3bf7107204b7f8552f34abf37e132 : SUCCESS in 1h 39m 23s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f8175669121f4aca8cc04af4b09eb3b6 : SUCCESS in 2h 39m 09s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/463d7d34dd134d5a807764e0fd3e8688 : SUCCESS in 2h 00m 52s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0aa19adada6d4860b5e5cd17e5f61bb0 : SUCCESS in 56m 08s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5844c33888a1439db69fe270a78a87e3 : SUCCESS in 1h 05m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c12a445c5bf543bc91e1248134fc95ad : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":2},{"id":"eada27402fe14f07564d40df1bf905007fbd7703","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2023-08-21 07:11:05.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"919f3839bedd14e9643b5e524938c380fecbdf1d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 07:13:11.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a4dbbd86add91478544b2e521a5f40d07fb43d9d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 09:11:47.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ac9b06efd4b4e6382f5fe86af99bfad\n\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/404895267bb34b419b44926589383c1e : SUCCESS in 1h 32m 02s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b66e8200b4c247f9b87e0b01a51b51a5 : SUCCESS in 1h 47m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/def82f1aea664c208e6cd88da095b680 : SUCCESS in 1h 42m 17s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/61f29bc7eca94fa6a58a1189d0be5c8e : SUCCESS in 1h 54m 02s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2e7e85c50e564343ad771e89f041f9c8 : SUCCESS in 1h 51m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b505259d00f42d4acf38b6d07a2185b : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":2},{"id":"1b96de1e5b2067b533eb1b476948254147db162d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 09:11:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"24c3a44e9547ca3279f1ba1435456c354f4a1cb6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 09:12:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be0724e93bb94721b25679a82ba611f8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca3e50ac0f9a4c5baf6f22d29ba37a73 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"27726e7643bb785da40279b60059364ec9f22935","revisions":{"f01ce7f24bcafce575f17a77871444de76d66489":{"kind":"REWORK","_number":1,"created":"2023-07-31 13:19:59.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/27/890127/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_murano","ref":"refs/changes/27/890127/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/27/890127/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/27/890127/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/27/890127/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/27/890127/1"}}},"commit":{"parents":[{"commit":"ef685574cbe902cabe23f7f75aa48a94ec79dd04","subject":"Add TLS support to murano backends","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/ef685574cbe902cabe23f7f75aa48a94ec79dd04"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-07-31 13:19:50.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-07-31 13:19:50.000000000","tz":120},"subject":"Use proper galera port in configuration","message":"Use proper galera port in configuration\n\nWhile \u003cservice\u003e_galera_port is defined and used for db_setup\nrole, it\u0027s not in fact used in a connection string for oslo.db.\n\nChange-Id: I3b0657b8d3b062eb6c8af44f0905066dba150a88\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/f01ce7f24bcafce575f17a77871444de76d66489"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/f01ce7f24bcafce575f17a77871444de76d66489"}]},"branch":"refs/heads/master"},"27726e7643bb785da40279b60059364ec9f22935":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-08-17 14:58:38.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/27/890127/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_murano","ref":"refs/changes/27/890127/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/27/890127/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/27/890127/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/27/890127/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/27/890127/2"}}},"commit":{"parents":[{"commit":"2049ef6fd4f95fc1a9b5c6287f2f04ab6dcc96d9","subject":"Define constraints file for docs and renos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/2049ef6fd4f95fc1a9b5c6287f2f04ab6dcc96d9"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-07-31 13:19:50.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-08-17 14:58:38.000000000","tz":0},"subject":"Use proper galera port in configuration","message":"Use proper galera port in configuration\n\nWhile \u003cservice\u003e_galera_port is defined and used for db_setup\nrole, it\u0027s not in fact used in a connection string for oslo.db.\n\nChange-Id: I3b0657b8d3b062eb6c8af44f0905066dba150a88\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/27726e7643bb785da40279b60059364ec9f22935"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/27726e7643bb785da40279b60059364ec9f22935"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
