)]}'
{"id":"openstack%2Fopenstack-ansible-os_murano~898922","triplet_id":"openstack%2Fopenstack-ansible-os_murano~master~Ic9cf5ffae328b5dff3fb357a27628ffab7698b22","project":"openstack/openstack-ansible-os_murano","branch":"master","topic":"osa/quorum_queues","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-10-30 13:11:56.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"Ic9cf5ffae328b5dff3fb357a27628ffab7698b22","subject":"Add quorum queues support for service","status":"MERGED","created":"2023-10-20 10:42:50.000000000","updated":"2023-10-30 18:07:09.000000000","submitted":"2023-10-30 18:06:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"898922-osa/quorum_queues","meta_rev_id":"0bb31c3c82b65b5a1a09f4efc188c48bbf036b61","_number":898922,"virtual_id_number":898922,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-30 18:06:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2023-10-27 10:56:56.000000000","_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-10-26 20:38:57.000000000","_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2023-10-27 10:56:56.000000000","_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-20 15:40:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-26 20:38:57.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2023-10-27 10:56:56.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"05f53c27ee9728226914ebe32a23877732006a74","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-10-20 10:42:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4d4749776b04ff2343ac21175f474d0138188c69","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-10-20 12:35:22.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"d2a6ee34a21e13e804ba1a298d7cff745e4b7472","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-20 15:40:24.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9fc6e6c12af443cbb773794f7c5ab715\n\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/16333b6a78b7484ca3a68bb4be296416 : FAILURE in 1h 29m 16s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/77821f7b1073430383e39440377d4f63 : FAILURE in 1h 44m 12s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1865cd9df0d04b91ac72343e7c0b094c : FAILURE in 1h 38m 49s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/90f0f82f511a4f8b8aa7e37f8e1f7987 : FAILURE in 1h 10m 53s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e92cbc62ab224c35a43b7d0cf733540e : FAILURE in 1h 29m 05s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fdb25a6a631747d8ab810feb8b21008c : FAILURE in 1h 26m 19s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4c7963da7bae42238f5b5dfa2a0f408f : FAILURE in 2h 40m 10s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c7dd97c0f59744728220b8a47b498b10 : FAILURE in 2h 29m 37s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ba8a608ccefa40a8ba1615f512309f72 : FAILURE in 2h 58m 20s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a5203a442c154d10b7dc4aeddfe64653 : FAILURE in 2h 18m 20s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5783c62fa5aa49d087f74f896e485abf : FAILURE in 46m 47s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/61718320f94b4874afa9b232e6b7b93e : FAILURE in 47m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62df07d61f934f4b8759944d1762332e : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":2},{"id":"d49fdd8bf82ac8161ffb172cc96612595f90f3b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-10-25 10:26:00.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"7621b36d8695972fca2cfc59bec5739d370e433a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-25 13:39:54.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/644d21c71d7643e5a98d4052e7b926c7\n\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/3b09179fa1ce4a32a4de1afb2e7066b0 : SUCCESS in 1h 53m 57s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/f8c6ec431e424ce5ac7949f48e2a253b : SUCCESS in 1h 59m 06s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ef904398565a470694e31bcfac90bdd7 : SUCCESS in 2h 00m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e880387e83f4498c9e8c3bf98de524f2 : SUCCESS in 1h 51m 06s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/40637d475a3c4ae48a483b9e99d4edf8 : SUCCESS in 1h 59m 58s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3ee71b264420499da150aa028bae2fcb : SUCCESS in 1h 54m 31s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fe6042bee9ec41bab1044d5e0b19c0a8 : SUCCESS in 3h 01m 48s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f0e173f5e11f406796c5bdc3bfd85185 : TIMED_OUT in 3h 05m 24s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1994f03b4e1d46aa9ae7debb4bfda65c : TIMED_OUT in 3h 04m 43s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/afb717a669b14caeaae44a7bb2a25e58 : TIMED_OUT in 3h 04m 37s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/73b5528fd00a4443a044758c8a964e8f : SUCCESS in 1h 09m 13s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7ddf75742882472e940ffdb091f9a04e : SUCCESS in 1h 02m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1710b9d306ce4351a4d2e0eab98b4cae : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":3},{"id":"429075139b58dc189cb5e15c5615b8e78c3e00c5","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-10-25 14:50:37.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d4ecc91058d9a5aea0330351bb08651d46ec8379","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-25 18:04:33.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e00a37f8656c48ac89fbeba8b0bd6e1f\n\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/527d7f087e844b5ea0252a43f86a8c04 : SUCCESS in 1h 42m 30s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/e9e5ba095f42458ea142f256233262e9 : SUCCESS in 1h 44m 29s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/05df97d8603b41019e70969ee079dfbd : SUCCESS in 1h 45m 10s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e8f97611edba4e61bd58b7381bea292b : SUCCESS in 2h 11m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/80d2b3021fb647e8b4fa01f26cb3d990 : SUCCESS in 2h 04m 31s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/328b8c35a2f349be946dfdd674a0e285 : SUCCESS in 1h 57m 38s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e7c712097bcf48c2aa16cd31bff6f909 : SUCCESS in 1h 41m 06s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e5696f007025452ca1a1e218e8c865fd : TIMED_OUT in 3h 05m 18s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e1308a884a7b43b4b802ed65af84c280 : SUCCESS in 3h 03m 47s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7ae8f9d8384a4faba7a9b994ce9f00e9 : TIMED_OUT in 3h 07m 08s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/67cf913959f9485ebfe96907c5736586 : SUCCESS in 2h 21m 31s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2a198624f8434f4fa6242ff5457d7680 : SUCCESS in 59m 13s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/872a63a29b904167bb14f68c8cd01c79 : SUCCESS in 1h 09m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c301507e396d4847af5dfaca4abe8c91 : SUCCESS in 2m 52s","accounts_in_message":[],"_revision_number":3},{"id":"e613acb78b400a75e228642d597a9ac33e7e096b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-10-26 16:09:26.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c712d9c912d154da1e07644afd32452a4fbe3348","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 19:07:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af42d2bd0c6947d992ecc3666585dcef\n\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/41b88ae440a24d748559e5eaf9752d2e : SUCCESS in 1h 41m 39s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/20349ae674744fcab94577c9bcc69eac : SUCCESS in 1h 52m 24s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/0bf829c65d234efdb9685a046c453aae : SUCCESS in 1h 44m 10s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1c142fad812a45faa59983439d56fc8d : SUCCESS in 1h 36m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0b04c6c40acf47c08dff356ea65578d6 : SUCCESS in 1h 40m 55s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0663a234f2944ba1b1cb39c59f24471c : SUCCESS in 1h 49m 19s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d714ff7958a54e838671eca133cea322 : SUCCESS in 1h 39m 43s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8b69814f7c644449b87f9e118851e79d : SUCCESS in 2h 49m 25s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b93c051db7924616a0ccff8e5992fa31 : SUCCESS in 2h 32m 24s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/80d3d983d38f4bb9a894442a48e69f59 : SUCCESS in 2h 51m 23s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/77785e7d497846ab8c7545437cd7793b : SUCCESS in 2h 47m 06s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ad30f768ff884c5598b206d44249d1ee : SUCCESS in 1h 06m 05s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9c76a25a84a9495fb28f55ef93e9fa96 : SUCCESS in 1h 07m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b1bf89b7c5a41eb83a63e66fc24b715 : SUCCESS in 3m 16s","accounts_in_message":[],"_revision_number":3},{"id":"4b78c71f1b3c23548a14aae573112a7a8d6ec91d","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-10-26 20:38:57.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"548187fea88f5972fe46653cb924046121585bb1","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2023-10-27 10:56:56.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"2f4a2bd6ed85fc7a37610ede1cb8b21297431487","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-27 10:59:04.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"fca5923ed5a224fe25174fa9398197fb5d54a751","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-27 13:27:44.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9ca3ede1caba4909b823b28b01d914ee\n\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/88297e4a32a14276a3ff6fccb39c71e8 : SUCCESS in 1h 25m 00s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/991f49686d494f80af59a0e5558762ca : SUCCESS in 1h 33m 13s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bab9ef7a6940493a9fec6ebbc1d0d6f5 : FAILURE in 1h 07m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cd36e6df18d3477494aa0ac2c5fdd5de : SUCCESS in 1h 23m 57s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4fe800d525144f1ba8fb58a1ed855d25 : FAILURE in 1h 10m 13s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/245e269891a94e81962c3d07beb8740f : SUCCESS in 1h 26m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/766ea73f2cac4976a145525530b656d5 : SUCCESS in 2m 57s","accounts_in_message":[],"_revision_number":3},{"id":"fb82ad19f5392e5db08286c1a2ec3616ced3f766","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-10-30 13:11:56.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ea5f8a2fe18cab3589e83a17727fc5ea92880c19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-30 16:06:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57959688a628489396a99dc482c7a8a3\n\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/a8270fd8751c4d30bb0df7bdfc7634a9 : SUCCESS in 1h 25m 45s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/d789eae57572494a9b7b8c285b8de9f6 : SUCCESS in 1h 24m 55s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/df3cea0b09064ea18a5b5f6812ecbac8 : SUCCESS in 1h 45m 46s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b72638bb5d0f499194a39f28910ddb11 : SUCCESS in 1h 44m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a41d687bbcd74d1880a26be33a6040c4 : SUCCESS in 1h 37m 24s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5a09c4a350264189806c73a5b03ffe6e : SUCCESS in 1h 43m 32s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/482e3bded06c4b6f907bde5e20f521e7 : SUCCESS in 1h 37m 20s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7103e4ab26e047fe845988dedbc02d41 : SUCCESS in 2h 46m 49s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/92804e114ba64ad995eb660000e97abc : SUCCESS in 2h 19m 03s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ef170c9862284ae0904bbe4adc50673d : SUCCESS in 2h 35m 36s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ecb2efa5468a4d67b049f41773908b92 : SUCCESS in 2h 22m 16s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f6a784172ce949ca81cf91f0bfd36201 : SUCCESS in 57m 21s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/178a17c051ce49d4b57c5f8ff68b5bcf : SUCCESS in 56m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ff9e38ee5cb4ac09f2579887459a827 : SUCCESS in 2m 56s","accounts_in_message":[],"_revision_number":3},{"id":"8d80d8712f0fa46bebe194f0b367fa69242fab7c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-30 16:08:25.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8ce8a74d74c87e887bb9191a26025e2d7b0b5474","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-30 18:06:13.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26244efce9fc4a8db5efadab0ad809a0\n\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/cccbecb884404a59a08fe0214d37052d : SUCCESS in 1h 26m 32s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/2c95fadc0f6c48f2b566c67622ef7949 : SUCCESS in 1h 24m 57s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9b499d235bc949e5a434567a2815c696 : SUCCESS in 1h 47m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/79b19030a0d14d74bfe4cb40c290dbf7 : SUCCESS in 1h 31m 15s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c24a5a7a4f9b4039bb71568855c61145 : SUCCESS in 1h 46m 32s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c78ad93e63eb42018a3750f5cd34794e : SUCCESS in 1h 21m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31d2de63d85848cb944960d763a370c8 : SUCCESS in 3m 01s","accounts_in_message":[],"_revision_number":3},{"id":"b6dfd2ad2ec09a1d8b6f530c1dcac66e61b4f859","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-30 18:06:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"0bb31c3c82b65b5a1a09f4efc188c48bbf036b61","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-30 18:07:09.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41e702b1f7c34311995dd5203d796a44\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1146b7c77e7d42a9bb6e807a6c54ac9f : SUCCESS in 37s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3e64b8de73df02468dfd81a3dbae620d97226ff5","revisions":{"a16680a48eac8c9263e40b5645658d1e74e308f7":{"kind":"REWORK","_number":1,"created":"2023-10-20 10:42:50.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/22/898922/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_murano","ref":"refs/changes/22/898922/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/22/898922/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/22/898922/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/22/898922/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/22/898922/1"}}},"commit":{"parents":[{"commit":"27726e7643bb785da40279b60059364ec9f22935","subject":"Use proper galera port in configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/27726e7643bb785da40279b60059364ec9f22935"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-10-20 10:42:31.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-10-20 10:42:47.000000000","tz":120},"subject":"Add quorum queues support for service","message":"Add quorum queues support for service\n\nThis change implements and enables by default quorum support\nfor rabbitmq as well as providing default variables to globally tune\nit\u0027s behaviour.\n\nIn order to ensure upgrade path and ability to switch back to HA queues\nwe change vhost names with removing leading `/`, as enabling quorum\nrequires to remove exchange which is tricky thing to do with running\nservices.\n\nChange-Id: Ic9cf5ffae328b5dff3fb357a27628ffab7698b22\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/a16680a48eac8c9263e40b5645658d1e74e308f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/a16680a48eac8c9263e40b5645658d1e74e308f7"}]},"branch":"refs/heads/master"},"742774878c35fb17c38f196c849cc24659f21883":{"kind":"REWORK","_number":2,"created":"2023-10-20 12:35:22.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/22/898922/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_murano","ref":"refs/changes/22/898922/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/22/898922/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/22/898922/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/22/898922/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/22/898922/2"}}},"commit":{"parents":[{"commit":"27726e7643bb785da40279b60059364ec9f22935","subject":"Use proper galera port in configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/27726e7643bb785da40279b60059364ec9f22935"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-10-20 10:42:31.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-10-20 12:35:22.000000000","tz":0},"subject":"Add quorum queues support for service","message":"Add quorum queues support for service\n\nThis change implements and enables by default quorum support\nfor rabbitmq as well as providing default variables to globally tune\nit\u0027s behaviour.\n\nIn order to ensure upgrade path and ability to switch back to HA queues\nwe change vhost names with removing leading `/`, as enabling quorum\nrequires to remove exchange which is tricky thing to do with running\nservices.\n\nChange-Id: Ic9cf5ffae328b5dff3fb357a27628ffab7698b22\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/742774878c35fb17c38f196c849cc24659f21883"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/742774878c35fb17c38f196c849cc24659f21883"}]},"branch":"refs/heads/master"},"3e64b8de73df02468dfd81a3dbae620d97226ff5":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-10-25 10:26:00.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/22/898922/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_murano","ref":"refs/changes/22/898922/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/22/898922/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/22/898922/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/22/898922/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/22/898922/3"}}},"commit":{"parents":[{"commit":"27726e7643bb785da40279b60059364ec9f22935","subject":"Use proper galera port in configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/27726e7643bb785da40279b60059364ec9f22935"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-10-20 10:42:31.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-10-25 10:26:00.000000000","tz":0},"subject":"Add quorum queues support for service","message":"Add quorum queues support for service\n\nThis change implements and enables by default quorum support\nfor rabbitmq as well as providing default variables to globally tune\nit\u0027s behaviour.\n\nIn order to ensure upgrade path and ability to switch back to HA queues\nwe change vhost names with removing leading `/`, as enabling quorum\nrequires to remove exchange which is tricky thing to do with running\nservices.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-plugins/+/896017\nChange-Id: Ic9cf5ffae328b5dff3fb357a27628ffab7698b22\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/3e64b8de73df02468dfd81a3dbae620d97226ff5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/3e64b8de73df02468dfd81a3dbae620d97226ff5"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
