)]}'
{"id":"openstack%2Fopenstack-ansible-os_murano~916891","triplet_id":"openstack%2Fopenstack-ansible-os_murano~master~Iffcf7e865591ef76f62053f3ec3d022a90cfb331","project":"openstack/openstack-ansible-os_murano","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2024-05-13 14:21:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iffcf7e865591ef76f62053f3ec3d022a90cfb331","subject":"Preserve actual production playbook in examples","status":"MERGED","created":"2024-04-24 12:11:34.000000000","updated":"2024-05-13 14:23:02.000000000","submitted":"2024-05-13 14:21:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"916891","meta_rev_id":"7bdd3fd3a3d187a88db9aaa77aa562497c1cfcb1","_number":916891,"virtual_id_number":916891,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-13 14:21:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":2,"date":"2024-05-13 14:05:35.000000000","_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-05-13 13:34:46.000000000","_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":1,"date":"2024-05-13 14:05:35.000000000","_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-24 15:22:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-13 13:34:46.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2024-05-13 14:05:35.000000000","updated_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"}],"messages":[{"id":"a1d4e253d4d56a2a7eb0234242bfd571f77c2eae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-04-24 12:11:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"de047a99874d832bee529a5208969ee5989b55db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 15:22:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5e999e5135424f0793ec03d7eff7c385\n\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/fdc29bba332c4fc09ee0f7bcb1b47e8a : SUCCESS in 1h 25m 33s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/681aeef4cfc2402d97a736f6a87428b4 : SUCCESS in 1h 32m 12s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/98bf015735004493bcda38bca3ab4e69 : SUCCESS in 1h 31m 04s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9ec34978e9ad47b6a2f7e296dc3a3996 : SUCCESS in 1h 31m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7d946debff7646b7a0615536497d4dec : SUCCESS in 1h 23m 14s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f079bfdf9d5e42dd95d9f867700f8cff : SUCCESS in 1h 49m 40s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/51a25f5438aa41fa99859aea88f72b24 : SUCCESS in 1h 22m 58s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1008ac90194a441587b472f01da7a514 : SUCCESS in 1h 23m 44s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/91926b19e0bd4826a445f88fde4ac1ec : SUCCESS in 2h 34m 48s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b130f3e0d25846efaa6cfa979a13920a : TIMED_OUT in 3h 03m 42s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/63f110a759fe4ea39af6987f7d2e0c95 : SUCCESS in 2h 03m 00s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ceba0d83d4944b77a0c13e4e91fab747 : SUCCESS in 2h 50m 13s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8826dfce315b4fe5b3177fd53e0597c6 : SUCCESS in 2h 35m 21s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fc3aef461ef448eeae043ffd2894773f : SUCCESS in 2h 19m 14s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fd5845df86ca42fe8c3d9a393845ee6d : SUCCESS in 1h 16m 14s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6813e63d0c1445e58720f24a7ea99401 : SUCCESS in 1h 13m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2bdb74ce039243a5bc485fce32d568fc : SUCCESS in 2m 55s\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6b7d83f451c94971aec229bcf0000bc7 : SUCCESS in 13m 28s","accounts_in_message":[],"_revision_number":1},{"id":"5caae57c6ef2fdf16b9171728ef029e8b75100ab","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-04-25 10:04:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e995fda1baf65223ed6243c514b30617cd5d0941","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 13:10:03.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ae7aff9f699347d6925450b9a1388164\n\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/0c3525cf95214aa3a683a55d987d2112 : SUCCESS in 1h 25m 34s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/5de89cd9ceee4010a06fd3f49163631c : SUCCESS in 1h 21m 59s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/d1e58914edcd4f3098c27a1ea2b22ad3 : SUCCESS in 1h 37m 35s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fc7bddd65d434f14b4fda1d76172a72d : SUCCESS in 1h 25m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5a1abcf60e804072b2fe40efbadf82b9 : SUCCESS in 1h 30m 00s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bafe7edfe437475ebbee26102bb464ab : SUCCESS in 1h 28m 35s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f71c171d540043a3ba9b6798d37e631f : SUCCESS in 1h 33m 15s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2e442bde66004292b9bc9fc526e13bdb : FAILURE in 1h 35m 44s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fc47bdc6c35f4e59892930914738b50e : SUCCESS in 2h 16m 48s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/071652a8ccdb43368041d4faca715f10 : SUCCESS in 3h 00m 49s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/227ff8397da0484483066bed0db66f1d : SUCCESS in 2h 11m 49s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2fc096f0925b4e74b22c821cd721ad70 : SUCCESS in 2h 45m 03s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b6602eebb6b64173abc60da8b4d727b1 : SUCCESS in 2h 36m 37s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fe78bbfbaf4a4f78aa1bbc4aa610c3d4 : SUCCESS in 2h 12m 54s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/524e700fec1a406a94c2a6fac914f995 : SUCCESS in 1h 14m 29s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/00458a9df6eb4a688ab8faaf32849b24 : SUCCESS in 1h 03m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1ecb543e30d444c931db45ccbaee6f5 : SUCCESS in 4m 37s\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3a15370ac69d49d59035c853bc26a253 : SUCCESS in 14m 27s","accounts_in_message":[],"_revision_number":1},{"id":"73c0f52637296ccacce541b9980b83d8839ea855","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-05-04 12:15:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b50c6e9c8f57f317c3b1f1fba728fd4322249b7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-04 15:20:48.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e49fead072544114898478ff5e20692e\n\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/97921549e0664dde9a2d509088cf6a6f : FAILURE in 1h 19m 52s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/57eba175f78e4b898c90657c7facc595 : FAILURE in 1h 27m 19s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/1dffc1fd25e04f15b4bbd10069788205 : FAILURE in 1h 31m 05s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/04479a2a5beb4305b44dca5ca6142826 : FAILURE in 1h 29m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/807c71a95282474f8ab6c51d6933b80b : FAILURE in 1h 26m 34s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/87a83fe7bee8499b84fbbe965b90c6b6 : FAILURE in 1h 31m 24s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e56c8559e8444655bd4e81ac17cc4475 : FAILURE in 1h 11m 21s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4078d89df9794e2189c6af1321999fbc : FAILURE in 1h 26m 52s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9706486f1cb743e0bad5c74e6b4d6a73 : FAILURE in 2h 22m 31s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/562a1949bb9244a088247b7862b18c45 : FAILURE in 2h 57m 09s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c8694356a9354db6bc4dc3e3edfbe17e : FAILURE in 1h 31m 31s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b60c63a248084634aaaffdb461c328a2 : FAILURE in 1h 50m 28s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/90fe92bb6e8d45f0a42e868ecda783d6 : FAILURE in 2h 17m 17s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/23f4547a8b30478889fb995aa5736388 : FAILURE in 2h 05m 57s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b3ff270b85d748b0b0ec09b7c37174c5 : FAILURE in 24m 37s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ddd5b86bfb14490ea023114564b8f8a5 : FAILURE in 55m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75c011f61b8f48398ac8cc30bb915d47 : SUCCESS in 2m 41s\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/af34b8fd51d843e8a8f4a218d2311a32 : SUCCESS in 13m 42s","accounts_in_message":[],"_revision_number":1},{"id":"b0ebc63e6612e1fb6038809d0b82d7ea0de8d794","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-05-10 12:23:20.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"58bdc91a2eec8826998f97e045f7c2e8bf93a8e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-10 13:11:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03e89a6d6bd647408ab4f2e82eaf6153\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0581fb2e6c754c11aa38a04923fcae09 : SUCCESS in 4m 03s\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b7a1da66c2ae498d9a50f02ae01a3109 : SUCCESS in 12m 41s","accounts_in_message":[],"_revision_number":2},{"id":"d942650459ab78fc83c9bc3ea7d395a46704d0b2","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-05-13 13:34:46.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"39a7daff59bfc0fa0f29738cfe2b88f8f6acb7f7","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2024-05-13 14:05:35.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5f44786bb0b8c4473ba7b0bc61f80637b9476215","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 14:07:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5ed456e226d2884c9824609a3c6ac0f62d09b183","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 14:21:45.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6203baa42b654c10ac4cbd1539b2c026\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2956950479246d48ea833bd85b8b4d1 : SUCCESS in 2m 48s\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e29008d0e07b4ab0b6084e112afd0ef6 : SUCCESS in 13m 57s","accounts_in_message":[],"_revision_number":2},{"id":"abe40d9bc28394b9ca2b0134912b789e3d90bc1a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 14:21:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"7bdd3fd3a3d187a88db9aaa77aa562497c1cfcb1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 14:23:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5644476b796f4dcfb83dee506347dfc3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f0c042ec780468f9446e049cb27c366 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"73c63dac4e05c7d5fb697493ea39bd1792e0c0a6","revisions":{"86d24f9b90390723d8d24225f12389b2b3f542bc":{"kind":"REWORK","_number":1,"created":"2024-04-24 12:11:34.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/91/916891/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_murano","ref":"refs/changes/91/916891/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/91/916891/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/91/916891/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/91/916891/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/91/916891/1"}}},"commit":{"parents":[{"commit":"5722ebe72af71876cd3113c4d1335c64848a53b7","subject":"Add example playbook and linters job to the role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/5722ebe72af71876cd3113c4d1335c64848a53b7"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-04-24 12:11:27.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-04-24 12:11:27.000000000","tz":120},"subject":"Preserve actual production playbook in examples","message":"Preserve actual production playbook in examples\n\nSince Murano is moved to inactive state, we can not provide further\nreleases of the role either, which means that Sahara should be dropped\nfrom the integrated repo.\n\nWith that it\u0027s worth to preserve actuall playbook that was used for\ndeployment as example playbook, in case somebody needs to use old\nrole/service version in their deployments.\n\nChange-Id: Iffcf7e865591ef76f62053f3ec3d022a90cfb331\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/86d24f9b90390723d8d24225f12389b2b3f542bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/86d24f9b90390723d8d24225f12389b2b3f542bc"}]},"branch":"refs/heads/master"},"73c63dac4e05c7d5fb697493ea39bd1792e0c0a6":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-05-10 12:23:20.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/91/916891/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_murano","ref":"refs/changes/91/916891/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/91/916891/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/91/916891/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/91/916891/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/91/916891/2"}}},"commit":{"parents":[{"commit":"dfe1ea375a5261e2e1311d1ecd174abee15293ce","subject":"Disable integrated tests of Murano","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/dfe1ea375a5261e2e1311d1ecd174abee15293ce"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-04-24 12:11:27.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2024-05-10 12:23:20.000000000","tz":0},"subject":"Preserve actual production playbook in examples","message":"Preserve actual production playbook in examples\n\nSince Murano is moved to inactive state, we can not provide further\nreleases of the role either, which means that Sahara should be dropped\nfrom the integrated repo.\n\nWith that it\u0027s worth to preserve actuall playbook that was used for\ndeployment as example playbook, in case somebody needs to use old\nrole/service version in their deployments.\n\nChange-Id: Iffcf7e865591ef76f62053f3ec3d022a90cfb331\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/73c63dac4e05c7d5fb697493ea39bd1792e0c0a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/73c63dac4e05c7d5fb697493ea39bd1792e0c0a6"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
