)]}'
{"id":"openstack%2Fopenstack-ansible-os_murano~919178","triplet_id":"openstack%2Fopenstack-ansible-os_murano~master~I51173e3f0aac7894e903465e515a7e3aca6cf6b8","project":"openstack/openstack-ansible-os_murano","branch":"master","topic":"reno-eom-zed","hashtags":[],"change_id":"I51173e3f0aac7894e903465e515a7e3aca6cf6b8","subject":"reno: Update master for unmaintained/zed","status":"MERGED","created":"2024-05-10 18:56:27.000000000","updated":"2024-05-13 13:53:48.000000000","submitted":"2024-05-13 13:52:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"919178-reno-eom-zed","meta_rev_id":"27097e0e8095e0ea2ab4038be34b1b846a59c000","_number":919178,"virtual_id_number":919178,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-13 13:52:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2024-05-13 13:33:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2024-05-13 13:33:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-10 19:23:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-13 13:33:58.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"246b8cd5c0d685c5bd352e7690b13e9aedcdf86a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2024-05-10 18:56:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"867c48ebf816d3e87fa32afe268b260b61290a9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-10 19:23:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f5e222619d444d883607de8658d0b53\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3da5c12bd0874e91a6f13ed0dc82782f : SUCCESS in 2m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/abcac67401a64abba3c33c49b0544f9a : SUCCESS in 4m 19s\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/806b2ff16f104378885f28b994a4024e : SUCCESS in 17m 40s","accounts_in_message":[],"_revision_number":1},{"id":"661106faaeebbe9865166152a55f61307a7e6fde","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-05-13 13:33:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2d3004c0ae8240e3122cca52879e168eea0de759","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 13:36:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2e6436a232f36855efaee1ebe4f303c853a80ea3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 13:52:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0848d259331547709bd5939e74e4a13a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2279d2041ec94cb7b515b5bfb03bc6bf : SUCCESS in 2m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/83fce47e04ec4220b43b95e51b0ef74b : SUCCESS in 2m 22s\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bb1f579233a9495c85b1f5093ee87da3 : SUCCESS in 11m 32s","accounts_in_message":[],"_revision_number":1},{"id":"5975db1a89ef85f7982ec47a32f46ea9c146ad4a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 13:52:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"27097e0e8095e0ea2ab4038be34b1b846a59c000","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 13:53:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/762ae9ca5e5c4eb5a1654b727e19e065\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a87afe92039f4e4e99e21ee76834b3ab : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"39aeb27b7fe347ec64fcf5f5c0ec6130438dfd42","revisions":{"39aeb27b7fe347ec64fcf5f5c0ec6130438dfd42":{"kind":"REWORK","_number":1,"created":"2024-05-10 18:56:27.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/78/919178/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_murano","ref":"refs/changes/78/919178/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/78/919178/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/78/919178/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/78/919178/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_murano refs/changes/78/919178/1"}}},"commit":{"parents":[{"commit":"dfe1ea375a5261e2e1311d1ecd174abee15293ce","subject":"Disable integrated tests of Murano","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/dfe1ea375a5261e2e1311d1ecd174abee15293ce"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-05-10 18:56:26.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-05-10 18:56:26.000000000","tz":0},"subject":"reno: Update master for unmaintained/zed","message":"reno: Update master for unmaintained/zed\n\nUpdate the zed release notes configuration to build from\nunmaintained/zed.\n\nChange-Id: I51173e3f0aac7894e903465e515a7e3aca6cf6b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/39aeb27b7fe347ec64fcf5f5c0ec6130438dfd42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_murano/commit/39aeb27b7fe347ec64fcf5f5c0ec6130438dfd42"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
