)]}'
{"id":"openstack%2Fopenstack-ansible-os_neutron~486156","triplet_id":"openstack%2Fopenstack-ansible-os_neutron~master~Ia7ebd13be9ce7834679d439b7bda242805768ef8","project":"openstack/openstack-ansible-os_neutron","branch":"master","topic":"bp/goal-deploy-api-in-wsgi","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2022-06-08 11:19:01.000000000","reason":"Change was submitted"},"2799":{"account":{"_account_id":2799,"name":"Andy McCrae","email":"andy.mccrae@googlemail.com","username":"andrew-mccrae"},"last_update":"2022-06-08 11:19:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia7ebd13be9ce7834679d439b7bda242805768ef8","subject":"Implement uWSGI for neutron-api","status":"MERGED","created":"2017-07-21 14:12:08.000000000","updated":"2022-06-08 11:20:17.000000000","submitted":"2022-06-08 11:19:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":7,"has_review_started":true,"submission_id":"486156-bp/goal-deploy-api-in-wsgi","meta_rev_id":"3a26b2d860440f2fb06a3cc79498c1f774030c0d","_number":486156,"virtual_id_number":486156,"owner":{"_account_id":2799,"name":"Andy McCrae","email":"andy.mccrae@googlemail.com","username":"andrew-mccrae"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-08 11:19:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":2799,"name":"Andy McCrae","email":"andy.mccrae@googlemail.com","username":"andrew-mccrae"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},{"value":2,"date":"2022-06-08 08:13:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":2799,"name":"Andy McCrae","email":"andy.mccrae@googlemail.com","username":"andrew-mccrae"},{"value":2,"date":"2022-06-08 08:52:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":2799,"name":"Andy McCrae","email":"andy.mccrae@googlemail.com","username":"andrew-mccrae"},{"value":1,"date":"2022-06-08 08:52:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":2799,"name":"Andy McCrae","email":"andy.mccrae@googlemail.com","username":"andrew-mccrae"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":2799,"name":"Andy McCrae","email":"andy.mccrae@googlemail.com","username":"andrew-mccrae"},{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}],"CC":[{"_account_id":31792,"name":"Marcus Klein","email":"marcus.klein@open-xchange.com","username":"marcus.klein"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-07-24 14:54:23.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2017-12-06 17:29:56.000000000","updated_by":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"reviewer":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"state":"REVIEWER"},{"updated":"2017-12-06 17:34:06.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2020-03-30 11:04:33.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-11-07 16:30:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-09 07:41:02.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-11-09 23:32:41.000000000","updated_by":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"reviewer":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"state":"REVIEWER"},{"updated":"2021-02-03 20:14:50.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REMOVED"},{"updated":"2021-03-13 12:44:04.000000000","updated_by":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"reviewer":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"state":"REVIEWER"},{"updated":"2021-08-26 09:03:34.000000000","updated_by":{"_account_id":31792,"name":"Marcus Klein","email":"marcus.klein@open-xchange.com","username":"marcus.klein"},"reviewer":{"_account_id":31792,"name":"Marcus Klein","email":"marcus.klein@open-xchange.com","username":"marcus.klein"},"state":"CC"}],"messages":[{"id":"08269a1052e038657851a059db76debd58c54eea","author":{"_account_id":2799,"name":"Andy McCrae","email":"andy.mccrae@googlemail.com","username":"andrew-mccrae"},"date":"2017-07-21 14:12:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7689084414185ca705fea1cbe4fd6edce27946ea","author":{"_account_id":2799,"name":"Andy McCrae","email":"andy.mccrae@googlemail.com","username":"andrew-mccrae"},"date":"2017-07-21 14:21:16.000000000","message":"Patch Set 2: Published edit on patch set 1","accounts_in_message":[],"_revision_number":2},{"id":"9a84377f906ef76ad9a7a43e5f6eb7da50a02ead","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-21 15:12:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-openstack-ansible-os_neutron-docs-ubuntu-xenial http://docs-draft.openstack.org/56/486156/2/check/gate-openstack-ansible-os_neutron-docs-ubuntu-xenial/25b5c5e//doc/build/html/ : SUCCESS in 2m 07s\n- gate-openstack-ansible-os_neutron-linters-ubuntu-xenial http://logs.openstack.org/56/486156/2/check/gate-openstack-ansible-os_neutron-linters-ubuntu-xenial/b6dabba/ : SUCCESS in 2m 43s\n- gate-openstack-ansible-os_neutron-ansible-func-centos-7 http://logs.openstack.org/56/486156/2/check/gate-openstack-ansible-os_neutron-ansible-func-centos-7/82fc278/ : FAILURE in 47m 29s\n- gate-openstack-ansible-os_neutron-ansible-func-ubuntu-xenial http://logs.openstack.org/56/486156/2/check/gate-openstack-ansible-os_neutron-ansible-func-ubuntu-xenial/01116a4/ : FAILURE in 23m 35s\n- gate-openstack-ansible-os_neutron-ansible-func-opensuse-422-nv http://logs.openstack.org/56/486156/2/check/gate-openstack-ansible-os_neutron-ansible-func-opensuse-422-nv/a4cec18/ : FAILURE in 45m 21s (non-voting)\n- gate-openstack-ansible-os_neutron-ansible-calico-ubuntu-xenial http://logs.openstack.org/56/486156/2/check/gate-openstack-ansible-os_neutron-ansible-calico-ubuntu-xenial/493f993/ : FAILURE in 35m 30s\n- gate-openstack-ansible-os_neutron-ansible-func_ovs-ubuntu-xenial http://logs.openstack.org/56/486156/2/check/gate-openstack-ansible-os_neutron-ansible-func_ovs-ubuntu-xenial/c040587/ : FAILURE in 30m 50s\n- gate-openstack-ansible-os_neutron-ansible-upgrade-ubuntu-xenial http://logs.openstack.org/56/486156/2/check/gate-openstack-ansible-os_neutron-ansible-upgrade-ubuntu-xenial/a8145b0/ : FAILURE in 22m 44s","accounts_in_message":[],"_revision_number":2},{"id":"56b5a405851314237e21e1b377f7e475f38b8cef","author":{"_account_id":2799,"name":"Andy McCrae","email":"andy.mccrae@googlemail.com","username":"andrew-mccrae"},"date":"2017-07-22 11:08:48.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"640aa2cd37ee72efcda47caca318dc666117dd19","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-22 12:14:23.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-openstack-ansible-os_neutron-docs-ubuntu-xenial http://docs-draft.openstack.org/56/486156/3/check/gate-openstack-ansible-os_neutron-docs-ubuntu-xenial/b113c57//doc/build/html/ : SUCCESS in 37s\n- gate-openstack-ansible-os_neutron-linters-ubuntu-xenial http://logs.openstack.org/56/486156/3/check/gate-openstack-ansible-os_neutron-linters-ubuntu-xenial/3943511/ : SUCCESS in 1m 14s\n- gate-openstack-ansible-os_neutron-ansible-func-centos-7 http://logs.openstack.org/56/486156/3/check/gate-openstack-ansible-os_neutron-ansible-func-centos-7/266d373/ : FAILURE in 1h 05m 19s\n- gate-openstack-ansible-os_neutron-ansible-func-ubuntu-xenial http://logs.openstack.org/56/486156/3/check/gate-openstack-ansible-os_neutron-ansible-func-ubuntu-xenial/49a7bb7/ : FAILURE in 39m 23s\n- gate-openstack-ansible-os_neutron-ansible-func-opensuse-422-nv http://logs.openstack.org/56/486156/3/check/gate-openstack-ansible-os_neutron-ansible-func-opensuse-422-nv/6e587db/ : FAILURE in 50m 55s (non-voting)\n- gate-openstack-ansible-os_neutron-ansible-calico-ubuntu-xenial http://logs.openstack.org/56/486156/3/check/gate-openstack-ansible-os_neutron-ansible-calico-ubuntu-xenial/3cb8fd5/ : FAILURE in 42m 45s\n- gate-openstack-ansible-os_neutron-ansible-func_ovs-ubuntu-xenial http://logs.openstack.org/56/486156/3/check/gate-openstack-ansible-os_neutron-ansible-func_ovs-ubuntu-xenial/047a52b/ : FAILURE in 39m 55s\n- gate-openstack-ansible-os_neutron-ansible-upgrade-ubuntu-xenial http://logs.openstack.org/56/486156/3/check/gate-openstack-ansible-os_neutron-ansible-upgrade-ubuntu-xenial/832cc85/ : FAILURE in 27m 25s","accounts_in_message":[],"_revision_number":3},{"id":"27643d2114efb023bbdfed61f9e2f32ad9c65fa9","author":{"_account_id":2799,"name":"Andy McCrae","email":"andy.mccrae@googlemail.com","username":"andrew-mccrae"},"date":"2017-07-24 08:07:00.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"80dd1f6a8f89455eb93f805a22a4918e77ecd724","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-24 08:59:42.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-openstack-ansible-os_neutron-docs-ubuntu-xenial http://docs-draft.openstack.org/56/486156/4/check/gate-openstack-ansible-os_neutron-docs-ubuntu-xenial/c6836c3//doc/build/html/ : SUCCESS in 54s\n- gate-openstack-ansible-os_neutron-linters-ubuntu-xenial http://logs.openstack.org/56/486156/4/check/gate-openstack-ansible-os_neutron-linters-ubuntu-xenial/aa9cc04/ : SUCCESS in 1m 53s\n- gate-openstack-ansible-os_neutron-ansible-func-centos-7 http://logs.openstack.org/56/486156/4/check/gate-openstack-ansible-os_neutron-ansible-func-centos-7/5dfb2a0/ : FAILURE in 52m 28s\n- gate-openstack-ansible-os_neutron-ansible-func-ubuntu-xenial http://logs.openstack.org/56/486156/4/check/gate-openstack-ansible-os_neutron-ansible-func-ubuntu-xenial/dd4b479/ : FAILURE in 38m 16s\n- gate-openstack-ansible-os_neutron-ansible-func-opensuse-422-nv http://logs.openstack.org/56/486156/4/check/gate-openstack-ansible-os_neutron-ansible-func-opensuse-422-nv/dd6dafc/ : FAILURE in 45m 40s (non-voting)\n- gate-openstack-ansible-os_neutron-ansible-calico-ubuntu-xenial http://logs.openstack.org/56/486156/4/check/gate-openstack-ansible-os_neutron-ansible-calico-ubuntu-xenial/bc1469f/ : FAILURE in 47m 37s\n- gate-openstack-ansible-os_neutron-ansible-func_ovs-ubuntu-xenial http://logs.openstack.org/56/486156/4/check/gate-openstack-ansible-os_neutron-ansible-func_ovs-ubuntu-xenial/291fd19/ : FAILURE in 26m 08s\n- gate-openstack-ansible-os_neutron-ansible-upgrade-ubuntu-xenial http://logs.openstack.org/56/486156/4/check/gate-openstack-ansible-os_neutron-ansible-upgrade-ubuntu-xenial/b5b43ce/ : FAILURE in 42m 54s","accounts_in_message":[],"_revision_number":4},{"id":"0fa5e65d2cb61a630ed92352aabe933e67f1ceba","author":{"_account_id":2799,"name":"Andy McCrae","email":"andy.mccrae@googlemail.com","username":"andrew-mccrae"},"date":"2017-07-24 09:54:31.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"db6bc2d70a765f426296b32eb026aa8ec376425c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-24 10:51:01.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-openstack-ansible-os_neutron-docs-ubuntu-xenial http://docs-draft.openstack.org/56/486156/5/check/gate-openstack-ansible-os_neutron-docs-ubuntu-xenial/726fbe9//doc/build/html/ : SUCCESS in 1m 04s\n- gate-openstack-ansible-os_neutron-linters-ubuntu-xenial http://logs.openstack.org/56/486156/5/check/gate-openstack-ansible-os_neutron-linters-ubuntu-xenial/304a4a0/ : SUCCESS in 1m 58s\n- gate-openstack-ansible-os_neutron-ansible-func-centos-7 http://logs.openstack.org/56/486156/5/check/gate-openstack-ansible-os_neutron-ansible-func-centos-7/1b83830/ : FAILURE in 44m 31s\n- gate-openstack-ansible-os_neutron-ansible-func-ubuntu-xenial http://logs.openstack.org/56/486156/5/check/gate-openstack-ansible-os_neutron-ansible-func-ubuntu-xenial/3d773a0/ : FAILURE in 36m 55s\n- gate-openstack-ansible-os_neutron-ansible-func-opensuse-422-nv http://logs.openstack.org/56/486156/5/check/gate-openstack-ansible-os_neutron-ansible-func-opensuse-422-nv/bd132f9/ : FAILURE in 56m 17s (non-voting)\n- gate-openstack-ansible-os_neutron-ansible-calico-ubuntu-xenial http://logs.openstack.org/56/486156/5/check/gate-openstack-ansible-os_neutron-ansible-calico-ubuntu-xenial/07e1640/ : FAILURE in 44m 03s\n- gate-openstack-ansible-os_neutron-ansible-func_ovs-ubuntu-xenial http://logs.openstack.org/56/486156/5/check/gate-openstack-ansible-os_neutron-ansible-func_ovs-ubuntu-xenial/59053f3/ : FAILURE in 25m 16s\n- gate-openstack-ansible-os_neutron-ansible-upgrade-ubuntu-xenial http://logs.openstack.org/56/486156/5/check/gate-openstack-ansible-os_neutron-ansible-upgrade-ubuntu-xenial/7488772/ : FAILURE in 29m 29s","accounts_in_message":[],"_revision_number":5},{"id":"3e41c82a954dae723624ce6ccccc0fd9970551f9","author":{"_account_id":2799,"name":"Andy McCrae","email":"andy.mccrae@googlemail.com","username":"andrew-mccrae"},"date":"2017-07-24 12:19:34.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"1f3913857361ad5fd1268127d54b84be47106298","author":{"_account_id":2799,"name":"Andy McCrae","email":"andy.mccrae@googlemail.com","username":"andrew-mccrae"},"date":"2017-07-24 12:28:01.000000000","message":"Patch Set 6: Workflow-1","accounts_in_message":[],"_revision_number":6},{"id":"48d1fcc9988a41ec6993a73db67e1da8731fee87","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-24 13:25:32.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-openstack-ansible-os_neutron-docs-ubuntu-xenial http://docs-draft.openstack.org/56/486156/6/check/gate-openstack-ansible-os_neutron-docs-ubuntu-xenial/d935f7e//doc/build/html/ : SUCCESS in 1m 06s\n- gate-openstack-ansible-os_neutron-linters-ubuntu-xenial http://logs.openstack.org/56/486156/6/check/gate-openstack-ansible-os_neutron-linters-ubuntu-xenial/1a23e7f/ : SUCCESS in 1m 07s\n- gate-openstack-ansible-os_neutron-ansible-func-centos-7 http://logs.openstack.org/56/486156/6/check/gate-openstack-ansible-os_neutron-ansible-func-centos-7/82dab37/ : FAILURE in 1h 05m 33s\n- gate-openstack-ansible-os_neutron-ansible-func-ubuntu-xenial http://logs.openstack.org/56/486156/6/check/gate-openstack-ansible-os_neutron-ansible-func-ubuntu-xenial/2bc6b59/ : SUCCESS in 25m 23s\n- gate-openstack-ansible-os_neutron-ansible-func-opensuse-422-nv http://logs.openstack.org/56/486156/6/check/gate-openstack-ansible-os_neutron-ansible-func-opensuse-422-nv/57be813/ : FAILURE in 56m 27s (non-voting)\n- gate-openstack-ansible-os_neutron-ansible-calico-ubuntu-xenial http://logs.openstack.org/56/486156/6/check/gate-openstack-ansible-os_neutron-ansible-calico-ubuntu-xenial/c413e8f/ : FAILURE in 28m 52s\n- gate-openstack-ansible-os_neutron-ansible-func_ovs-ubuntu-xenial http://logs.openstack.org/56/486156/6/check/gate-openstack-ansible-os_neutron-ansible-func_ovs-ubuntu-xenial/f3ff6f6/ : SUCCESS in 39m 08s\n- gate-openstack-ansible-os_neutron-ansible-upgrade-ubuntu-xenial http://logs.openstack.org/56/486156/6/check/gate-openstack-ansible-os_neutron-ansible-upgrade-ubuntu-xenial/b8ba435/ : SUCCESS in 27m 06s","accounts_in_message":[],"_revision_number":6},{"id":"c8027d2a772bc6b5e1a1a277316de43a11714725","author":{"_account_id":2799,"name":"Andy McCrae","email":"andy.mccrae@googlemail.com","username":"andrew-mccrae"},"date":"2017-07-24 13:48:17.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"e4fa67c7884910e38c2afbca55fd3d5a1f9e272b","author":{"_account_id":2799,"name":"Andy McCrae","email":"andy.mccrae@googlemail.com","username":"andrew-mccrae"},"date":"2017-07-24 13:53:43.000000000","message":"Patch Set 7: Workflow-1\n\nthis requires a bit of new functionality.","accounts_in_message":[],"_revision_number":7},{"id":"7edfc39097e99e39ed1531ae392e4bc1b783e11a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-24 14:54:23.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-openstack-ansible-os_neutron-docs-ubuntu-xenial http://docs-draft.openstack.org/56/486156/7/check/gate-openstack-ansible-os_neutron-docs-ubuntu-xenial/0bd37df//doc/build/html/ : SUCCESS in 2m 08s\n- gate-openstack-ansible-os_neutron-linters-ubuntu-xenial http://logs.openstack.org/56/486156/7/check/gate-openstack-ansible-os_neutron-linters-ubuntu-xenial/8315f4f/ : SUCCESS in 2m 58s\n- gate-openstack-ansible-os_neutron-ansible-func-centos-7 http://logs.openstack.org/56/486156/7/check/gate-openstack-ansible-os_neutron-ansible-func-centos-7/d5e5f68/ : FAILURE in 1h 05m 54s\n- gate-openstack-ansible-os_neutron-ansible-func-ubuntu-xenial http://logs.openstack.org/56/486156/7/check/gate-openstack-ansible-os_neutron-ansible-func-ubuntu-xenial/fe98be4/ : SUCCESS in 43m 27s\n- gate-openstack-ansible-os_neutron-ansible-func-opensuse-422-nv http://logs.openstack.org/56/486156/7/check/gate-openstack-ansible-os_neutron-ansible-func-opensuse-422-nv/75b6b64/ : FAILURE in 57m 52s (non-voting)\n- gate-openstack-ansible-os_neutron-ansible-calico-ubuntu-xenial http://logs.openstack.org/56/486156/7/check/gate-openstack-ansible-os_neutron-ansible-calico-ubuntu-xenial/5f89a0a/ : FAILURE in 52m 07s\n- gate-openstack-ansible-os_neutron-ansible-func_ovs-ubuntu-xenial http://logs.openstack.org/56/486156/7/check/gate-openstack-ansible-os_neutron-ansible-func_ovs-ubuntu-xenial/bff4280/ : SUCCESS in 41m 58s\n- gate-openstack-ansible-os_neutron-ansible-upgrade-ubuntu-xenial http://logs.openstack.org/56/486156/7/check/gate-openstack-ansible-os_neutron-ansible-upgrade-ubuntu-xenial/6c57699/ : SUCCESS in 28m 29s","accounts_in_message":[],"_revision_number":7},{"id":"9411a5f7479bd5ba9ebd3032b05916b1c317bcda","author":{"_account_id":2799,"name":"Andy McCrae","email":"andy.mccrae@googlemail.com","username":"andrew-mccrae"},"date":"2017-08-11 13:45:30.000000000","message":"Patch Set 7:\n\nThis is blocked until https://review.openstack.org/#/c/439191/ can merge - since there is some conf file logic that doesn\u0027t work.","accounts_in_message":[],"_revision_number":7},{"id":"fb901902773ae7b76cd29b18200dd8df0a582416","author":{"_account_id":2799,"name":"Andy McCrae","email":"andy.mccrae@googlemail.com","username":"andrew-mccrae"},"date":"2017-08-11 13:45:35.000000000","message":"Patch Set 7: Code-Review-2","accounts_in_message":[],"_revision_number":7},{"id":"929ca74af4530a0d9417d59c4c59993778025ff3","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2017-12-01 03:26:36.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"161dc7ec462430890f062ae96e04034693bf4b01","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-01 04:45:18.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/56/486156/7/check/build-openstack-sphinx-docs/e1f71f6/html/ : SUCCESS in 1m 55s\n- openstack-ansible-linters http://logs.openstack.org/56/486156/7/check/openstack-ansible-linters/86a5f00/ : SUCCESS in 5m 33s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/56/486156/7/check/openstack-ansible-functional-centos-7/cf4d016/ : SUCCESS in 1h 15m 58s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/56/486156/7/check/openstack-ansible-functional-opensuse-423/86d5c36/ : SUCCESS in 37m 17s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/56/486156/7/check/openstack-ansible-functional-ubuntu-xenial/815333e/ : FAILURE in 40m 11s\n- openstack-ansible-upgrade-ubuntu-xenial http://logs.openstack.org/56/486156/7/check/openstack-ansible-upgrade-ubuntu-xenial/97b2e1d/ : FAILURE in 57m 41s\n- openstack-ansible-ovs-ubuntu-xenial-nv http://logs.openstack.org/56/486156/7/check/openstack-ansible-ovs-ubuntu-xenial-nv/6c2c3b2/ : FAILURE in 38m 42s (non-voting)\n- openstack-ansible-calico-ubuntu-xenial-nv http://logs.openstack.org/56/486156/7/check/openstack-ansible-calico-ubuntu-xenial-nv/f527a7d/ : FAILURE in 38m 42s (non-voting)\n- openstack-ansible-dragonflow-ubuntu-xenial-nv http://logs.openstack.org/56/486156/7/check/openstack-ansible-dragonflow-ubuntu-xenial-nv/7b215e3/ : FAILURE in 46m 19s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-xenial-nv http://logs.openstack.org/56/486156/7/check/openstack-ansible-opendaylight-ubuntu-xenial-nv/0d5d29a/ : FAILURE in 49m 19s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"e5a6d8bfcbb6a11d9cd03dc1ff9877c01b27dccf","author":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"date":"2017-12-06 16:01:51.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"f463f83acec788fb45561cffca544518a64b8bab","author":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"date":"2017-12-06 16:02:35.000000000","message":"Patch Set 8:\n\nRebased and added libxml2-dev to distro package list.","accounts_in_message":[],"_revision_number":8},{"id":"fd93ebe468520ffa7aa5ad9759084c44085644c7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-06 17:13:17.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/56/486156/8/check/build-openstack-sphinx-docs/fd987c2/html/ : SUCCESS in 2m 42s\n- openstack-ansible-linters http://logs.openstack.org/56/486156/8/check/openstack-ansible-linters/8a47090/ : SUCCESS in 5m 59s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/56/486156/8/check/openstack-ansible-functional-centos-7/340bc0e/ : SUCCESS in 58m 44s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/56/486156/8/check/openstack-ansible-functional-opensuse-423/7580442/ : SUCCESS in 37m 17s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/56/486156/8/check/openstack-ansible-functional-ubuntu-xenial/61c4d94/ : SUCCESS in 38m 20s\n- openstack-ansible-upgrade-ubuntu-xenial http://logs.openstack.org/56/486156/8/check/openstack-ansible-upgrade-ubuntu-xenial/e7eff3c/ : SUCCESS in 43m 51s\n- openstack-ansible-ovs-ubuntu-xenial-nv http://logs.openstack.org/56/486156/8/check/openstack-ansible-ovs-ubuntu-xenial-nv/b5654bc/ : SUCCESS in 37m 23s (non-voting)\n- openstack-ansible-calico-ubuntu-xenial-nv http://logs.openstack.org/56/486156/8/check/openstack-ansible-calico-ubuntu-xenial-nv/63eeda5/ : FAILURE in 39m 12s (non-voting)\n- openstack-ansible-dragonflow-ubuntu-xenial-nv http://logs.openstack.org/56/486156/8/check/openstack-ansible-dragonflow-ubuntu-xenial-nv/1b31c00/ : FAILURE in 39m 53s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-xenial-nv http://logs.openstack.org/56/486156/8/check/openstack-ansible-opendaylight-ubuntu-xenial-nv/714baf3/ : SUCCESS in 46m 04s (non-voting)\n- openstack-ansible-ovs-nsh-ubuntu-xenial http://logs.openstack.org/56/486156/8/check/openstack-ansible-ovs-nsh-ubuntu-xenial/cf7d922/ : SUCCESS in 48m 24s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"4ab4236682a46d7cc088662437e777844fdebc1b","author":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"date":"2017-12-06 17:29:56.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"eec71fb8e6888aa910173fed193bbcf8bf9f1e77","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2017-12-06 17:34:06.000000000","message":"Patch Set 8: Workflow+1 Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"99c152c2e567126c46937a5e6d7b2e587af72ddf","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-19 17:06:58.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"a5fdf9cb713702b81037e1e6fdec12702980aab6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 18:26:48.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/29f1502a9ecc40bebfa1a21734f2b464 : FAILURE in 4m 38s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/41f66bf6aabe4619b7cb7b0714ed3b8b : FAILURE in 1h 05m 31s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f65c4122050e42dfb028c3728a4d6fef : FAILURE in 1h 14m 00s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/aa4a04f987934c578fc286fec734e726 : FAILURE in 1h 01m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/048994f668cb4a7c9b72c4b261bceb06 : FAILURE in 51m 54s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/68d4185bf72c49548a5ef9e0e10e640d : FAILURE in 56m 58s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/7de97b628f744331be7d68375f12c2e6 : FAILURE in 28m 47s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b821e84f10654e57ba4738f118e32d2e : FAILURE in 43m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3db70f3b9f9f4767b675b01278bf82da : SUCCESS in 2m 55s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1ee3ad5f543e44d2a4601a53cc3027c9 : FAILURE in 6m 55s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a75a23c4ea914ce8877e4470288a6044 : FAILURE in 5m 34s\n- openstack-ansible-ovn-centos-7 https://zuul.opendev.org/t/openstack/build/4a98c4c7c6a741fca7a12e15c95e55e7 : FAILURE in 7m 50s (non-voting)\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d2e374cb51ec4af99d93dfa5c7c1d160 : FAILURE in 48m 43s\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f44901d2f2f244d3b0270c76e778bb43 : FAILURE in 6m 27s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a2e61d4e921e42d18a3224fa51da29f8 : FAILURE in 8m 12s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2ba8a092554e4b479ebcf306126ab50b : FAILURE in 6m 29s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e214a0e0ac8548b08f6dbcd627cc4059 : FAILURE in 6m 15s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4713e38bfcf045c2a7272f6e3c7519e8 : FAILURE in 5m 56s (non-voting)\n- openstack-ansible-ovs-fwaasv2-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/81b1bbc9449544caa169f73833ea5306 : FAILURE in 5m 42s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"4cab4be541c5abcd1bb7bf7477d69e5cc19e30f3","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-20 09:47:28.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"0cc552c85a6a9dde6e8b98664f238983167f9139","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 13:23:21.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/ac98e77fbb3d48de866082763fc90621 : SUCCESS in 8m 59s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/f3751ffe495e42c89575ef26cc821b79 : FAILURE in 1h 15m 45s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/1522c8a2c3c04ca4802f9f8bc69f25d1 : FAILURE in 1h 36m 15s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/ee821c5566f34a6c8b72483156fdc21c : FAILURE in 1h 14m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/607cfa9c8f7a44e79f4ae2aa95a27eea : FAILURE in 1h 19m 11s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ff6a5bbf8aa84d82b10ca3691a9fa19f : FAILURE in 1h 10m 59s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/f2b1f9037e4b4748b668f4b8624a9cbf : FAILURE in 30m 54s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ae9f052b444a4035b7cf41f127cc6e48 : FAILURE in 53m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cade23a157114fecaebf2dcc1e14112b : SUCCESS in 3m 30s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2713efe9f120408f857f83b01a1de55b : SUCCESS in 44m 39s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/980244a8c3f1489aad41f853363cc6f6 : TIMED_OUT in 1h 34m 51s\n- openstack-ansible-ovn-centos-7 https://zuul.opendev.org/t/openstack/build/553f9e85ef424c38a80833566ffee83b : FAILURE in 31m 38s (non-voting)\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5d7387d6ad80491aa9ede794661437b3 : TIMED_OUT in 3h 13m 02s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f685748d2e6b4dfe8baf8f2178955a77 : SUCCESS in 51m 44s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a27d49db377f4a6e8d1a031ba4c9fc86 : SUCCESS in 47m 09s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/02efbc1c76aa46a491b3e64a83997998 : SUCCESS in 47m 13s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/57ceaa6de5614a12850eba350641ec09 : SUCCESS in 41m 53s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4031316f22ba45a8942b640491fbe573 : SUCCESS in 51m 08s (non-voting)\n- openstack-ansible-ovs-fwaasv2-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eadd5744f35b4656b94b49719a689c36 : SUCCESS in 40m 51s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"e529037622c5581e5d3b11cf40913712bbc90222","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2019-12-09 02:32:50.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"a89892399b02d6e3bd73118e67a4910eceb89457","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 05:24:03.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/1d03564281a54e32abf73b9e05b84e0a : SUCCESS in 4m 30s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/f0ecce889acc4aa8a4bf5be2b6031831 : FAILURE in 53m 14s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/aae084aa54f344d794f0a1e0551d4877 : FAILURE in 1h 23m 41s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/bb4e34c9e4d043bb82cdfe14c3ef1b82 : FAILURE in 1h 16m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4386a57b46824568963e9d9485b5715f : FAILURE in 55m 56s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/781ef4655cf5444f8174c90afc721db2 : SUCCESS in 2h 46m 15s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c6e0da59025e41c297d957b979d95957 : SUCCESS in 1h 42m 44s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/cbae39d536d449a280f1cde02db71a48 : FAILURE in 56m 48s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/15fb350c9efb46f09a2f0afae02ed261 : FAILURE in 50m 10s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/49330996fb4b40aa8ddc6c435ea5342f : FAILURE in 56m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ece55453e744953b66da659dee22a5f : SUCCESS in 4m 40s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/96365ca0388f4c1cb04b2d0fb83a16d2 : SUCCESS in 41m 39s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3668c627212c443d826591ca73b81000 : TIMED_OUT in 1h 34m 07s\n- openstack-ansible-ovn-centos-7 https://zuul.opendev.org/t/openstack/build/08cd273dcc2f4c21bc10c769ab497b48 : FAILURE in 6m 09s (non-voting)\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b8a58b4648c2463790f1311c2b1a676e : FAILURE in 1h 13m 28s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ad7c3dadab304bc0b7f74d2227f8a435 : SUCCESS in 55m 50s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5fe2c861ce6a48bfaf571933b0a4325a : SUCCESS in 43m 29s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/95a1f04d270c481f9eb47e939ed8950e : SUCCESS in 57m 54s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7550f618ea4644719911af54e82666b3 : SUCCESS in 51m 21s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a6069a35584741b4b61f7624a760d378 : SUCCESS in 53m 50s (non-voting)\n- openstack-ansible-ovs-fwaasv2-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a15e19968f8e46108181f3e815056949 : SUCCESS in 47m 58s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"2381f2fd1125bbe4622f62c95b242899ee42f034","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-03-29 20:57:04.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"1541ce642b2f93a6a900cec29cca096ba9e62364","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-29 22:48:34.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/5500974f48f34fdc873e0d150fb51653 : SUCCESS in 7m 07s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/51fb00fa80094535bab40e173b2785da : FAILURE in 56m 40s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/15b1a22da05c482b8a91856b81f6542d : FAILURE in 1h 22m 57s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/7430dc545afe41f5b14c3957302f8d3a : RETRY_LIMIT in 5m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/afa42e4f7b254fe8927ef5f5b2366232 : FAILURE in 1h 12m 58s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/586c6827526c48ddbec1ac3649088876 : FAILURE in 16m 07s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8512224c465d4a42bd6c0dd9995c9b56 : SUCCESS in 1h 49m 52s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b0c857d570914f4aa42dcf3c173afb4d : FAILURE in 57m 01s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/a2087db23249488b837ff875b16b7c7d : RETRY_LIMIT in 7m 17s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/28b1ce9625b544b08b3ae625dfeff0a3 : FAILURE in 50m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f565afaf73b42d3857584e5ac7610c2 : SUCCESS in 5m 00s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5f0f42943304400da15d94d2301fa00b : SUCCESS in 45m 52s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/90603e6ad9a340a096a9369132473b2b : TIMED_OUT in 1h 36m 27s\n- openstack-ansible-ovn-centos-7 https://zuul.opendev.org/t/openstack/build/0de5454acb1d4a6b9c9c03b2c753f289 : FAILURE in 30m 23s (non-voting)\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9a480fa150f54052a97e01c7aa7e7107 : FAILURE in 1h 14m 11s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8f6bfba4c65b4869ac3dbf9a1c0bd2a6 : SUCCESS in 56m 26s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ddfa13167be948e58702c801fee3bc1f : SUCCESS in 55m 49s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/594c2b6d0e5d4b0dabfcf1416a8e159e : SUCCESS in 50m 32s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/775c8edccfad4018bc26047fe17e8b65 : SUCCESS in 45m 18s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2b42832a21ed4430861d7dfcd42feaf6 : SUCCESS in 49m 15s (non-voting)\n- openstack-ansible-ovs-fwaasv2-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8d39a7c1183c4781857b56b11adbcd80 : SUCCESS in 51m 20s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"ea5ddf5f9507253a5bf486043010e7c88d05b5ac","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-03-30 11:04:33.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.","accounts_in_message":[],"_revision_number":12},{"id":"b5a224ee525f9b7b851d9e96fa5498ab948da157","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 13:35:47.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/eac1befed5bd4855ae48b377e0955b1e : SUCCESS in 6m 21s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/523a54fcfda34222bc9720f22aecc31d : FAILURE in 1h 02m 21s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b9cd66e1649d4fb7a757116765a05b3e : FAILURE in 1h 29m 41s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/3200d7288f3c4a8ebac7408a74e708a4 : FAILURE in 1h 14m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5f8d61697f09475991b7044fa920605f : FAILURE in 1h 17m 48s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/151eb9e242b842f1a4e8c4d75051776a : FAILURE in 14m 10s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5ffe35a64ed74a46ab141a3be8d9321d : SUCCESS in 2h 27m 04s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/57668aec1af64729958a6344af1cc905 : FAILURE in 58m 55s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/2ba727c3feb248149e5c5ea82344cd5c : FAILURE in 35m 15s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0fdf728687b1483799fe94b135a1d578 : FAILURE in 54m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d309b422ed8f474ea3380500d80575b7 : SUCCESS in 5m 48s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1d149b22988246a58d96d70dff192f47 : SUCCESS in 44m 28s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/37759dd37d924f44bfca1efbff9214c1 : TIMED_OUT in 1h 36m 14s\n- openstack-ansible-ovn-centos-7 https://zuul.opendev.org/t/openstack/build/b874a56d37a64c25b8cb9cf5b8d56c4c : FAILURE in 29m 04s (non-voting)\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dc7d5293b6a542b0b15a9573d7b0ff0d : FAILURE in 1h 03m 23s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7ad0832147da4b26817e6e43a0817092 : SUCCESS in 1h 03m 32s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e20cfb37af5848f3b9b5063959039540 : SUCCESS in 56m 32s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d908b1121ed04a9b904226a9a0135152 : SUCCESS in 1h 03m 50s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e5375a31fa25466eaf6bb42d33e3abc9 : SUCCESS in 47m 51s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/25b3b0345c8f4d8381084bf8384603a3 : SUCCESS in 50m 33s (non-voting)\n- openstack-ansible-ovs-fwaasv2-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6e795aed1b504263bdafafbba1d7de6e : SUCCESS in 43m 33s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"8c521a31ac6878eb9b52253f141f05f7e44fc175","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2020-04-07 01:45:54.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"e15e3033c1398b1f26b155277a826e0284be2d5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 04:23:12.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/70bab9fbfc52421f9a296ba5e3d8b0f5 : SUCCESS in 7m 04s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/566f9621138e468e91beaeea3e9d95d4 : FAILURE in 1h 00m 25s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f272511c2cbe4c50839aaae2ecfeb767 : FAILURE in 1h 28m 21s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/9134670a522f4d99b6a20d38dcb3a0f0 : FAILURE in 1h 15m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8d858cffd7174046ae1f58e6bca8e816 : FAILURE in 1h 04m 16s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/316037641f7d443788122a889fa365d2 : SUCCESS in 2h 30m 09s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/397e5f17922247f393f3ddac528bbd78 : SUCCESS in 1h 51m 04s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/1d173d1232ef4608a70ac20f57cf9827 : FAILURE in 58m 04s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/1aa8023c54a84e918755fc83ece2d2b7 : FAILURE in 34m 26s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/03b0460c52214b84a7fad5e8a535191d : FAILURE in 51m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2799397005714b37a72c5dffc3d33913 : SUCCESS in 4m 48s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/565a8912115b4452a2bc50b60b66d619 : SUCCESS in 43m 44s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fa79c6a184fe44538046c177a5ea3914 : TIMED_OUT in 1h 36m 41s\n- openstack-ansible-ovn-centos-7 https://zuul.opendev.org/t/openstack/build/aad63c7fb9b24b8aad5e537abcaaf430 : FAILURE in 29m 34s (non-voting)\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/93cab3ed84184b468fb0ed361c3006e5 : FAILURE in 55m 14s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1e492adaa1ac43188a2c43dddb118dcf : SUCCESS in 48m 54s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/29365f39299e46d6a841e1cf8777bfe9 : SUCCESS in 48m 56s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a6a96ed682fd444fae13e9c7be2a23c8 : SUCCESS in 47m 14s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f2bd4419fe3e4c75ad164e3c5acab937 : SUCCESS in 41m 22s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/33ca1514a0674dbbb827f823a29ee38e : SUCCESS in 47m 23s (non-voting)\n- openstack-ansible-ovs-fwaasv2-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fd49ef5afbcb46c9b31a95f10df62909 : SUCCESS in 41m 00s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"9c38ebb33ee24507a98be6475b7c277d9db59298","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2020-10-26 14:23:55.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"3a961dd46e49add957110015ad95b3e7bee15eb7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-26 17:41:49.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/0f541457640b42aeb18a317378c0c05c : SUCCESS in 7m 54s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/f04fac2bc188470596e2e6552063854e : FAILURE in 1h 05m 47s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/c705855f893c4a75916bddcd111647c6 : FAILURE in 1h 30m 41s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4105bdfd6d864f1ab9705d4fa4b44e7f : FAILURE in 1h 50m 12s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8a47ed8738dc4b9785dd67bf8bc88082 : FAILURE in 1h 46m 17s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/1e57d72609ca464a9a7190d5e8ea345f : FAILURE in 2h 31m 32s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f9095b0c910a428ea47f3fbd7d5c280b : TIMED_OUT in 3h 01m 56s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/5c8dca7ef7254c0fa4b87c0d8214ea5d : FAILURE in 54m 03s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/79f74f1b1e2f4733a9a5699dd1ecf704 : FAILURE in 53m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6df380bcec074a98a24a308960a506d9 : SUCCESS in 5m 38s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6d8e2ba4d2b24b89a8c42a2e8db10b68 : FAILURE in 31m 25s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d20faef9489c4f1eaf05d597ec1f71fc : FAILURE in 30m 14s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/968c7f3718344ed78869266564814c08 : FAILURE in 1h 07m 27s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fcfda74e676a4717be1a3d1e3d85fa1f : FAILURE in 29m 17s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c0bde5a66cc74ebd82774b263fc7cd5a : FAILURE in 11m 21s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c62db6ebdb2643d4b2fe3265c40eb925 : FAILURE in 10m 08s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/54f7bc23153944e7b4d4520da5d4af84 : FAILURE in 10m 51s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2bc52dae6e3b440c881d9548ce732322 : FAILURE in 11m 24s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"5aa6ee26cd93b8e307a993aa6e12eed2b23ebd20","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2020-10-27 14:01:17.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"8cc98982d69fd268e86487406fab2a035272cfa5","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2020-10-27 16:10:47.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"082e7c0109bcac21a88f3171ebf8616a3326db12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-27 18:25:48.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/03718bd405f248dda7c67a82b1d7332c : SUCCESS in 12m 56s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/7c11d19bec9c4761b5fa61aa6519ca32 : SUCCESS in 1h 18m 39s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f517d4faa22548379c7b8aee1ea31f34 : SUCCESS in 1h 31m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d0b7d235faf7495b91dbcfb24a5d5703 : SUCCESS in 1h 25m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e30b6eb06b5e4cc2883493b934728912 : SUCCESS in 1h 17m 41s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/78bd9d53c4d64bf59d417755f0b3593d : SUCCESS in 1h 57m 31s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a1cf8ef11fdd4a5481ae2311b7eb8a97 : SUCCESS in 1h 10m 14s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/a3ee78a855a54d8da077df16e96c71cc : FAILURE in 1h 15m 40s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fddd7b9fa0df4719b408bdb1ee3eac7d : SUCCESS in 1h 44m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3520a32776504da6ada0553c47d19b1e : SUCCESS in 25m 57s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3cf5467c035e4243b1a9dcf0ce3aaa6e : SUCCESS in 45m 10s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f4f76a2789dd46a29ece1d98ace7167a : TIMED_OUT in 1h 36m 04s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/24dd4a7f72254f348f64c3921869ad58 : FAILURE in 1h 15m 03s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/15caa72bd890463ca5ec2aa62d8bc01d : SUCCESS in 48m 42s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/03ad44a660864d0d953e1ef8cd31fc92 : SUCCESS in 51m 14s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/85f08d8fe8d0404087c955bd671ccbcb : TIMED_OUT in 1h 38m 13s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/97fcf1b4b4ce4794b39d234b3828a0a1 : SUCCESS in 48m 14s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0a2763a51350418697280079f073a099 : SUCCESS in 48m 17s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"113940c7918132066476bee24136c2b9fa0774c0","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2020-10-27 20:09:11.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"56ff9e67ba2a0e08679ef2f9c5f3fb39138813fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-27 22:38:47.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/17e9adfbf4ee4c87a612ff7d67aefc2a : SUCCESS in 8m 42s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/638e08939bd74b8bb063735b01f2ab7e : FAILURE in 1h 12m 59s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/9690c66d566b4d8c8cda3b477eeb841d : SUCCESS in 1h 39m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b2478b76076c4c58bed6541a767f67d9 : SUCCESS in 1h 46m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/30a6134158a940f1ad9e3f15168d8d54 : SUCCESS in 1h 16m 07s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e56df070a70c4d3db51e5e84ea65bb39 : SUCCESS in 2h 16m 15s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/198e3ca94e274fe4b93bfc2821652b95 : SUCCESS in 1h 40m 17s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e77645907c8c426db42a4d4bb9522287 : FAILURE in 57m 38s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4d26ba86155c46348f70bb7051e25f9a : SUCCESS in 2h 02m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5689c2e1e8bc4bca8f4f8284bda9558a : SUCCESS in 5m 08s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e4c3d751b2874a74acacca539d747cb8 : SUCCESS in 47m 48s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8f9732e19ff24fa390c010de6562300d : TIMED_OUT in 1h 34m 52s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a80e42d570ab4bd1aafbd84d492f8510 : FAILURE in 1h 16m 46s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a0dd292bb41a4a7a81fa452a44d801b9 : SUCCESS in 50m 54s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b75f5a842a114aea9c9d549469775985 : SUCCESS in 52m 47s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9e4a06af9c194faca95ff40d10b494e9 : SUCCESS in 47m 35s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fa066b5f4c2549b0a461465e66ac0475 : SUCCESS in 49m 14s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/78fad2bc771649219e0e1d8893a8db38 : SUCCESS in 54m 50s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"b73ebce086999d78808015ebf146b22b783c7a06","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2020-10-28 15:07:20.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"d2e1ef0a06338b38515a622322b5110884ce58cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 17:44:40.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/903fe8dbf4a54f088e641428d944e67d : SUCCESS in 8m 13s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/5d16040f076d4ba48650298ada380ba2 : SUCCESS in 1h 14m 33s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/aeb527f7af504daea1f8a3c5672ff229 : SUCCESS in 1h 34m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6b8761186d0b44be9cb6beeca02f213e : SUCCESS in 1h 19m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5a462966d6204304a1e15e6daea167b9 : SUCCESS in 1h 17m 33s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/7e56af6f19d6415eab19e1987375ed33 : SUCCESS in 2h 19m 09s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c49f78b4014d41a1b5817774251cccf6 : SUCCESS in 1h 12m 21s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/eeedcbe092704c2290e1e96e93549f5c : FAILURE in 56m 22s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f3acedc9a2d94a9bbded0e195646b8ad : SUCCESS in 1h 01m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b485cf00905543188a773cf9b89522a4 : SUCCESS in 5m 59s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e6bbfdb171fa47ed9bef91080cb6caf9 : SUCCESS in 49m 37s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0d5353070c2c43e195ac729bd75830ae : TIMED_OUT in 1h 36m 02s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/96c337f6f6234481a9a76e6470117337 : FAILURE in 1h 51m 15s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9fd6b0c2c3e747d1922ee3471651ea3d : FAILURE in 4m 45s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/850b2c0948404ea5906488623cee57ec : SUCCESS in 55m 15s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/81a431b931a24cb0a3f9ca0d8a96e90d : SUCCESS in 52m 23s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c6f29060439e49d69ae3dddf3b6545d2 : FAILURE in 4m 53s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3385f48c23f64f72b7bf05715f25ba61 : FAILURE in 7m 56s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"944f6711ff23968a07bbcbc027f9ccf4d99760a4","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-11-06 12:21:25.000000000","message":"Patch Set 16: Patch Set 15 was rebased","accounts_in_message":[],"_revision_number":16},{"id":"6c3d50994b099edf51fadd079d17ae4e5d28033e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-06 14:31:40.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/b6c6c36b91bc4b21a939fe17d90b574f : SUCCESS in 9m 33s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/2e209da0855e458cbd73d25faf24ee6b : SUCCESS in 1h 10m 12s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/c4e19c22ae9a4034a9db4a5c10316080 : SUCCESS in 1h 26m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/10c449768cbc4d6ebd49adc84b4bdf58 : SUCCESS in 1h 23m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bd3cd0ce704d453ebd3181ecc0efe1aa : SUCCESS in 1h 14m 15s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/d96f496abbb446b3b672551022cd0e02 : SUCCESS in 2h 04m 03s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/57a9c313801e45f491a4764aa7eb087b : SUCCESS in 1h 31m 55s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/02279443c3c148c6a6abc31a645f8161 : FAILURE in 1h 01m 08s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/afd0d89defab451299b57af138390be0 : SUCCESS in 53m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef0303b763a24b71be4719a81cbf11f8 : SUCCESS in 6m 37s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b2ca5e29e4634b328831e2367c8a7b18 : SUCCESS in 51m 39s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/129d3094515a46f9930581fa0d700cb4 : TIMED_OUT in 1h 35m 57s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3ea791c93edd43588348165bcfa0e676 : FAILURE in 1h 15m 13s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/02dc759a5d9541f78d4d782b3a4c4bea : SUCCESS in 55m 28s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/21ee3c388e904a82a38b82386459688f : SUCCESS in 54m 06s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a920db5da4b94ce08584bf79c2cca59b : SUCCESS in 51m 42s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/430adc7efd5943afbe81fd70ccd23189 : SUCCESS in 51m 06s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/63c08a4841aa42bab99d8f5f18caba8f : SUCCESS in 58m 10s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"77971bede322282f7eb64c7cbd83989e0019a603","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-11-07 08:14:21.000000000","message":"Patch Set 16:\n\nrecheck","accounts_in_message":[],"_revision_number":16},{"id":"3ab952a52cab204893fba842a7025e1371f9520f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-11-07 08:17:57.000000000","message":"Patch Set 17: Published edit on patch set 16.","accounts_in_message":[],"_revision_number":17},{"id":"95e965d8978da6ee5d3fa7d9c493300173db1369","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-07 09:46:30.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/22a52b6d370d4ce295faa5e5a3d00785 : FAILURE in 8m 38s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/6575fcc0ed96401281cf5278ec6fd117 : FAILURE in 23m 34s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/085333cfb78844c393b9352b6f4ce434 : FAILURE in 29m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7b7d7576fa5647b98d810b2e728f0b29 : FAILURE in 23m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f4548d5f691142f0938303edd4b020fa : FAILURE in 24m 42s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/06e9c56dbcf14c3bb9db78b306358c7b : FAILURE in 1h 19m 50s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d99f1127b35e41998c41d02d0ec623b8 : FAILURE in 58m 53s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/dfdc457ce4d648fca155bffc1a9bc4d0 : FAILURE in 27m 21s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8a0b7d8b58eb4e6280a054baafdf7cf0 : FAILURE in 21m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ffdfdf261ec4cea90e14708e74ad87a : SUCCESS in 5m 36s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fa3948113426450bbffaad21d0ef3fb4 : FAILURE in 8m 10s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f49df5b87bcb44f7ae40498e6913ebff : FAILURE in 8m 31s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/054545aafdb34e098f69c787d53c5327 : FAILURE in 26m 24s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/48fce3f9f75740269b2c62edb74d2b01 : FAILURE in 8m 17s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1dd80bdd9e184f66a1df85170d383a27 : FAILURE in 8m 00s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/33cd9d0690964b39a9b3467c42d14304 : FAILURE in 7m 24s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/735772be7ca3496aa85daabbcb77f2f7 : FAILURE in 8m 30s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c18d0086814942f4a69f0b1f15058bf6 : FAILURE in 7m 59s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"c2df5f41c53d6774f36bb12600cd99cced53d998","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-11-07 14:31:55.000000000","message":"Patch Set 18: Published edit on patch set 17.","accounts_in_message":[],"_revision_number":18},{"id":"a6bb083f187771091aa9998f58ac14ce2614b39e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-11-07 14:32:00.000000000","message":"Patch Set 18: Code-Review+2","accounts_in_message":[],"_revision_number":18},{"id":"ad5d2ea7653070930681a17aa832013846fc96f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-07 16:30:29.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/a7d258212b6e4e9b897c787633fe23fe : SUCCESS in 7m 56s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/6ecd7570830b49b4815316595b8b91f8 : SUCCESS in 1h 12m 06s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/0b4d6b4c9748460eaa8b22f84f022812 : SUCCESS in 1h 20m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/615fc350c7e7497db7c3a75407a523df : SUCCESS in 1h 21m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/79220e0d5250479abc81fa3342b19c32 : SUCCESS in 1h 04m 20s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/d4fa9a2c8d5d4df5bd94f6686a5fa029 : SUCCESS in 1h 54m 28s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d83dd9581a8040eabed18f16580e347f : SUCCESS in 1h 24m 21s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/837772e8a60d44b88cc01d58cd6e0d30 : FAILURE in 58m 47s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/81f19f910f6941258a3025d82becabda : SUCCESS in 53m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d2dbd20a5c24c519d9bcc6410fa0275 : SUCCESS in 5m 41s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e3b49bfa3aac4c259edff084ffeafc5a : SUCCESS in 43m 09s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ba5898b50c9f4c468f3a275887c6a210 : TIMED_OUT in 1h 34m 50s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/71c3bf8d870b4d4e920cde33be095bd1 : FAILURE in 1h 05m 32s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d1b7cb10fb61408393b041663d9eee2e : SUCCESS in 51m 27s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/20a710553b014fa4b5c38301d7ff6a27 : SUCCESS in 45m 46s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9fc50051d44c45d79d41eedf3c8b8d38 : SUCCESS in 51m 23s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ead9b0425efd4707a7de4af852d8cfd5 : SUCCESS in 42m 12s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/acc9261696e34a65b5ad49cd6289d7bf : SUCCESS in 52m 34s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"eeb93a06426b09ba24a7aa65f24ca0ccf87a013a","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2020-11-08 21:23:18.000000000","message":"Patch Set 18:\n\nrecheck","accounts_in_message":[],"_revision_number":18},{"id":"8bd7e945ca3054c412c8249cc090e9cf53effc7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-08 23:40:25.000000000","message":"Patch Set 18:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/48f4260074f1483ebe21c64ea10b1a3e : SUCCESS in 7m 46s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/3a24dcf5181d4402985b6818f32ea1e3 : SUCCESS in 1h 03m 55s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/6b1d6bde0a674bd3b4dbf5fb8327a0fc : SUCCESS in 1h 22m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2f7aefc7b89e45229f94386f443fbfab : SUCCESS in 1h 10m 41s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1bd2d321c9e1466a8cdb9e0db182eb51 : SUCCESS in 1h 05m 45s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/043c08dc3e6d4bbb98362eae9508fb9e : SUCCESS in 1h 57m 23s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/845a4c2f89214084922573054874f5a4 : SUCCESS in 1h 29m 39s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/4798e9b061254c5a9b7c8974a7733664 : FAILURE in 55m 07s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/125b2593a3114ae4a7c53c33007f3fa6 : SUCCESS in 53m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ca7b7766ed84370b2cadd32f6316716 : SUCCESS in 4m 36s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d72f4111f8864fcaaea6e3c19d1184e2 : SUCCESS in 46m 20s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5815d932731d4564a5e61dff8addd953 : TIMED_OUT in 1h 35m 19s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e2a5b456a29446c191ec3b1c3cfbce44 : FAILURE in 1h 02m 16s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/114417df4f104f318ae96d09e738e2b6 : SUCCESS in 49m 02s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0664f05bc4f446818c2c657245b0c512 : SUCCESS in 41m 56s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8a27093c3f52416ab54aaeb3b36565b2 : SUCCESS in 47m 45s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/75921cbe32ca4488a07a6476beae5cc0 : SUCCESS in 46m 42s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c8023bf0ce714c508c22aabda95aaf12 : SUCCESS in 50m 16s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"fe72220c4516febb174d4ec87e00a55554aa3441","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-11-09 07:38:39.000000000","message":"Removed Code-Review+2 by Dmitriy Rabotyagov (noonedeadpunk) \u003cnoonedeadpunk@ya.ru\u003e\n","accounts_in_message":[],"_revision_number":18},{"id":"fe06a1d954f4539bbf95af3104b25f32db4cdd96","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-11-09 07:41:02.000000000","message":"Patch Set 18:\n\nseems these 2 jobs are just failing with uwsgi for some reason...\nAnd happens due to l3 agent registration failure https://zuul.opendev.org/t/openstack/build/4798e9b061254c5a9b7c8974a7733664/log/logs/host/neutron-l3-agent.service.journal-22-18-23.log.txt#63","accounts_in_message":[],"_revision_number":18},{"id":"3d9b5e8817270eba366e4cd0eb870eddc98eda14","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2020-11-09 17:44:07.000000000","message":"Patch Set 18:\n\n@noonedeadpunk: Odd, as I am unable to replicate the CentOS error with SCENARIO\u003d\"aio_distro_metal_neutron\" locally.","accounts_in_message":[],"_revision_number":18},{"id":"4210cc772716a528241d5631408bc7d354b448ab","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2020-11-09 23:32:41.000000000","message":"Patch Set 18:\n\nrecheck","accounts_in_message":[],"_revision_number":18},{"id":"0d574b12d4d33e940aeb92f2d571362eea49b743","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-10 01:49:45.000000000","message":"Patch Set 18:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/5b52b96e28514c9eaa63dffe6196911b : SUCCESS in 12m 25s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/ab1cbe6d3927496ea237f085c0e35297 : FAILURE in 1h 22m 23s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/cc779dcb1c87478e80670304b0a7fc69 : SUCCESS in 1h 24m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7159619e00774aeea83c2db5c4e2701e : SUCCESS in 1h 33m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ce13f2d9861647788b343bf3b55186f1 : SUCCESS in 1h 19m 45s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/4ff61fdf0fc344c19fe5b3d8060af1cb : SUCCESS in 2h 06m 32s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5e9507ebb7064dd99ba98a23bd9933d9 : SUCCESS in 1h 42m 36s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/5405e1d5c23a4666b8fff965e6c493de : FAILURE in 1h 01m 13s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/48685c2ece8541a2b656e2160cbe0434 : SUCCESS in 53m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa4d6488fb714560befa8ef76045c43d : SUCCESS in 6m 16s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cb49d46302744d6a9d802ebdd9acdeed : SUCCESS in 47m 28s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2cb0c666422344fc9da73b35c4a5e928 : TIMED_OUT in 1h 36m 29s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f355889d13d04191b072bdb0db3aeaf6 : FAILURE in 31m 19s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bd509deb6a3044e2aae591a0f74cec8f : SUCCESS in 1h 03m 25s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4818138f69c84d68a0cb981ce40bf8d4 : SUCCESS in 1h 00m 55s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/241f9f1ee6e3476abf7835597a599453 : SUCCESS in 1h 00m 08s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/203268df05ad4e25be82f92f3b8ec615 : SUCCESS in 58m 36s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/602df2a2a5f74543aa0de0dee940fe66 : SUCCESS in 1h 00m 46s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"11e3e4499caefae7b85c599c25bcb79e19eed734","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-12-02 12:55:12.000000000","message":"Patch Set 19: Patch Set 18 was rebased","accounts_in_message":[],"_revision_number":19},{"id":"ffeaf0250cd7ade7b573023b6a416eb6eb80ace6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 16:07:29.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/9b1ba7cde8c34bb48686311e44da1f8d : SUCCESS in 9m 46s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/053957bbadf948599a2d2f654b98333a : SUCCESS in 1h 10m 03s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e443570077794a5a9c2b50f1a0b9efbe : SUCCESS in 1h 31m 25s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/43c3736bf86a4e19b0973cb53d4985bc : SUCCESS in 1h 30m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa314611e3814f42b089eb562f7d3973 : SUCCESS in 1h 03m 36s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e4f89365f82a4203875324b0cd248b1a : SUCCESS in 2h 02m 42s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/432e72f0cb4c44c694374e273732a6b5 : SUCCESS in 2h 27m 52s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/10318bf2cce24442b109baf179db119f : FAILURE in 1h 05m 12s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2754f3f7245d4c16bac3be28eaa262fb : SUCCESS in 1h 03m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/972922a234114c10a38462beb028acc5 : SUCCESS in 5m 24s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/45306550764b41b287481994770e2e15 : SUCCESS in 50m 42s\n- openstack-ansible-ovs-centos-8 https://zuul.opendev.org/t/openstack/build/3a8814cd7d94419db0de7a45718be41d : SUCCESS in 54m 57s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/19e8f6e96608463b9f7808625a9948da : FAILURE in 48m 55s\n- openstack-ansible-ovn-centos-8 https://zuul.opendev.org/t/openstack/build/672e9a00602846bebdf79f06b54d306e : FAILURE in 48m 24s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e4c484c3566640de9869a5d60fa11285 : FAILURE in 1h 18m 58s\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fc2c51c94373400eb172573d4f7cdf9f : SUCCESS in 1h 01m 15s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b79347f3805f4aa8a150d100728f37a0 : TIMED_OUT in 1h 31m 37s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1d0b755fe28b40299a5144efad94ca20 : SUCCESS in 55m 14s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/33bfd32bd40e4bdf968ae3b2171da800 : SUCCESS in 47m 08s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/72284cfdcc4e49cd8635b9af8bdc73bb : SUCCESS in 1h 00m 01s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"bc5e8873ed503d8823226f0f23f52d8f6c26ac08","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-02-03 18:20:01.000000000","message":"Patch Set 20: Patch Set 19 was rebased","accounts_in_message":[],"_revision_number":20},{"id":"9701e90db71893e7fbcfc4a245b3b80fd1e6a75f","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2021-02-03 20:14:50.000000000","message":"Removed reviewer Kevin Carter.","accounts_in_message":[],"_revision_number":20},{"id":"766f5b0be7f95217f32a18bc7627d1dbdf256233","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 21:01:29.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/0231af6843b54a13ba4fba5746fad592 : SUCCESS in 10m 16s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/ab7d2752197540509c30b1500aabacd2 : FAILURE in 1h 44m 23s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/2a254cf68c49400091998c43b56a6341 : FAILURE in 1h 35m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9267f83bc6e14944a18701975f7071c2 : SUCCESS in 1h 26m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/185647ff630f4c5597ef6b3bbcd215ac : SUCCESS in 1h 23m 38s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f2488d906e8f40d9a0f22d351056ae73 : FAILURE in 2h 27m 15s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5e05e9cc44da4f6bb7d6c13e865b47c2 : SUCCESS in 2h 13m 56s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f5bdd2d10442452ba4ccd1a6f48c89af : FAILURE in 1h 02m 16s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4dc8276cd6e846d0b8b9d5f9084de97a : SUCCESS in 1h 17m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/934d1f7542404de28bd1cbba3ae3c06a : SUCCESS in 5m 39s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/924c3515d80b44c2adf54e48220a1e73 : SUCCESS in 53m 17s\n- openstack-ansible-ovs-centos-8 https://zuul.opendev.org/t/openstack/build/a160978babcc4571beea447e38454ea5 : FAILURE in 40m 29s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7bafe21e88f84d949a70f9bfd3831ebc : FAILURE in 35m 53s\n- openstack-ansible-ovn-centos-8 https://zuul.opendev.org/t/openstack/build/5c554dfb42b74b4f891e343a7ad00014 : FAILURE in 53m 14s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2def9adf59e84abba06257607738a703 : FAILURE in 1h 20m 01s\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e991ae353765430ba2e14d95367addff : FAILURE in 45m 47s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fff163c98ca84eaeaf38113b50b4cd12 : SUCCESS in 1h 00m 00s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/69151d8505ac45b2919811c13f71a5dc : SUCCESS in 52m 40s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/202c51752c79422a967ff3c51dc63bd7 : FAILURE in 38m 55s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/20be9a36a9fb4997b787a9a09990c8d2 : FAILURE in 36m 29s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"9642375b5d7bb4e53508fc43eaf482ca37cded5a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-02-07 19:33:40.000000000","message":"Patch Set 21: Patch Set 20 was rebased","accounts_in_message":[],"_revision_number":21},{"id":"623e3cfd8e817848f2809ba939b19663eb316f62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-07 21:23:31.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/ad2ec30fa5644351ae1239661db52e72 : SUCCESS in 7m 36s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/1c7bd973819e4aee917c98586a235729 : SUCCESS in 1h 03m 32s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e1eb36370ce14fbca610ad9163bab22e : SUCCESS in 1h 21m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d7578667b5cc4528ab4b21086c17c20c : SUCCESS in 1h 10m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0f69bdcf19ce4c72b832265a75361335 : SUCCESS in 1h 00m 15s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/bc0902f5a8f24d95bb574ee9cc0f76dd : SUCCESS in 1h 46m 48s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0c4d925567064041840d7d9a3c8c8956 : SUCCESS in 1h 30m 21s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/ab3a5ffdd59940beb1b400d29031f97c : FAILURE in 52m 34s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/710b4078bff94b63a4e99b45720dd31f : SUCCESS in 54m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aaba12d6bd68432c9ed5588943d410ed : SUCCESS in 4m 35s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b5b88fc47c9c4570b596cbe93c0bbf21 : SUCCESS in 50m 18s\n- openstack-ansible-ovs-centos-8 https://zuul.opendev.org/t/openstack/build/62b4a289a00f4cc296ec6df9bd1dc07c : SUCCESS in 52m 36s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/49cc86d3fff14e76be5996c28d664eb1 : FAILURE in 44m 39s\n- openstack-ansible-ovn-centos-8 https://zuul.opendev.org/t/openstack/build/83c1a72ff97d49d8aa40ab2b5234268a : FAILURE in 36m 27s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0d2dda66c56d4b2a8b5611958bf92fc6 : FAILURE in 1h 13m 03s\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4f71a31ea39e403db79ae78a970548ba : SUCCESS in 52m 42s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d464376da7e7448a9b8213a79974a2ad : POST_FAILURE in 21m 48s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c56c2dba6d8b4767af9cc6f307a42121 : SUCCESS in 53m 30s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dcacb2d19ad742359d61c9f976560850 : SUCCESS in 48m 53s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/51691f69ce7e4bbcb721d88f146ebe78 : SUCCESS in 51m 32s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"cfdd83150614e36ab8c5bb58701e2fd588901906","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-07 22:58:24.000000000","message":"Patch Set 21:\n\nIs this breaking the OVN job? https://zuul.opendev.org/t/openstack/build/49cc86d3fff14e76be5996c28d664eb1/log/logs/openstack/server1/neutron-server.service.journal.log.txt#2955","accounts_in_message":[],"_revision_number":21},{"id":"79f3b311f2733ef8f538190b461ebed4d1c18f81","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2021-03-11 14:35:47.000000000","message":"Patch Set 22: Patch Set 21 was rebased","accounts_in_message":[],"_revision_number":22},{"id":"b9e062b203dcb84c1949638ebbe440c2aac36bb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 17:53:09.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/55fbec7c93994ef5809b807c0d2f5a84 : SUCCESS in 9m 56s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/d25c552326c246aea29e6d90a1637761 : SUCCESS in 1h 08m 21s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/0749ce62434c4da6aec8f1728d9df77a : SUCCESS in 1h 37m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/211e1e29a7e64e5495e3ba60e736a859 : SUCCESS in 1h 18m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/00b3efb033084090bd296f1ff5a92613 : SUCCESS in 1h 05m 03s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/4a9900860ba548c3aea5b1d870f6b76c : SUCCESS in 1h 48m 36s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6fe0f21c668a4de0b2f29b4a94289125 : SUCCESS in 1h 41m 10s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/8df71e1adb0d4915ab0c2d6bf3e1a2b3 : FAILURE in 35m 40s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eb5d13c5dfbb4ef8a40fa6e99501ef90 : SUCCESS in 1h 02m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6f59a2ce6494194aacddadd62b64fc8 : SUCCESS in 6m 10s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2176d265883846269dd54ced25bafef5 : TIMED_OUT in 3h 03m 04s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6628e9c00e3648b4923ae45a602a0cff : FAILURE in 1h 29m 20s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5fda61b9f9e14f0a8060fded2cc72b1b : SUCCESS in 1h 01m 26s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/cfde326237694030b65a9c1acb30aec5 : SUCCESS in 1h 53m 32s\n- openstack-ansible-deploy-aio_metal_ovs-centos-8 https://zuul.opendev.org/t/openstack/build/c6ac9ecc77974b59a4ecf2338f30e4eb : SUCCESS in 1h 05m 19s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/c54d702dda2c49f3b997b4c3d332de97 : FAILURE in 1h 53m 04s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/72a6c0d9977140ffac0eb791bc8859a7 : FAILURE in 1h 00m 13s","accounts_in_message":[],"_revision_number":22},{"id":"bcacc549c569a22e2452f7ebbe37cf14962d133c","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2021-03-13 12:44:04.000000000","message":"Patch Set 22: Code-Review-1\n\n(4 comments)\n\nI am under the impression that the neutron-server consists of neutron-api and neutron-rpc-server, if that is the case, I think whatever\u0027s now called neutron-server be moved/renamed over to neutron-rpc-server and neutron-api be its own thing.","accounts_in_message":[],"_revision_number":22},{"id":"46015ad2517a9e1874b6c848eeb95cb5cd8e1434","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2021-03-13 12:48:15.000000000","message":"Patch Set 22:\n\nI\u0027m also excited to see if this works, I remember back in the days of pike it didn\u0027t, and we gave up on it. https://bugs.launchpad.net/neutron/+bug/1687896","accounts_in_message":[],"_revision_number":22},{"id":"4bbe3682937205979444e453621dfe63076a465e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2021-03-13 22:20:12.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"87602bd35b9b8721d989b249b4c57dc862f00ae1","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2021-03-13 22:41:07.000000000","message":"Patch Set 23:\n\nI backported this to ussuri and it seems to run fine. We could optionally do something like\n\nneutron_services:\n  \u0027neutron-server\u0027: {\u0027service_en\u0027: {{ not neutron_use_uwsgi }} }\n  \u0027neutron-api\u0027: {\u0027service_en\u0027: {{ neutron_use_uwsgi }} }\n  \u0027neutron-rpc-server\u0027: {\u0027service_en\u0027: {{ neutron_use_uwsgi }} }\n\nand conditionally split the service depending on neutron_use_uwsgi, in case there are still unknown issues with neutron-rpc-server.","accounts_in_message":[],"_revision_number":23},{"id":"afca93badfaeaed972e1c87ae2bde2ab8c1014c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2021-03-13 23:36:17.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"ea91fa8b27edb97edfae7a635dd0f78387c88f03","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2021-03-14 00:53:29.000000000","message":"Patch Set 24:\n\n\u003e Patch Set 23:\n\u003e \n\u003e I backported this to ussuri and it seems to run fine. We could optionally do something like\n\u003e \n\u003e neutron_services:\n\u003e   \u0027neutron-server\u0027: {\u0027service_en\u0027: {{ not neutron_use_uwsgi }} }\n\u003e   \u0027neutron-api\u0027: {\u0027service_en\u0027: {{ neutron_use_uwsgi }} }\n\u003e   \u0027neutron-rpc-server\u0027: {\u0027service_en\u0027: {{ neutron_use_uwsgi }} }\n\u003e \n\u003e and conditionally split the service depending on neutron_use_uwsgi, in case there are still unknown issues with neutron-rpc-server.\n\nright on. anxiously awaiting gating.","accounts_in_message":[],"_revision_number":24},{"id":"4a1acd3674ea10661f902cd312f39a13a04463c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-14 01:32:14.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/a572fb2afcb54917aab4552b7efbd4da : SUCCESS in 7m 20s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/8fc0b06ff59845c095846de8c5b840d8 : FAILURE in 1h 07m 55s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/9fa27b09c05e4d1893046d0347f3731b : FAILURE in 1h 23m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6afe97c898d54d4280448a0244b046d3 : FAILURE in 1h 13m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/896ecc0fa5554c63ba92c13297a0b6f3 : FAILURE in 1h 03m 27s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/2add5602ffff42edbfe2c790b31c9204 : SUCCESS in 1h 52m 00s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e66b3893633d4e7f9e23044582e7c104 : SUCCESS in 1h 33m 57s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/5c3e6b77ace8408ba21915d507ae9945 : FAILURE in 36m 49s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/38ab0832c5994d6a9511f7714f2192bb : SUCCESS in 57m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acbc84510d74447d92e1ca3a24b6bbaa : SUCCESS in 6m 18s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ca13b45feece48a28430ce5e94f80884 : FAILURE in 1h 14m 12s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc4cc68645c14a63ae10f734292f335a : FAILURE in 1h 48m 06s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a235d70d512d4a43861c96e5c7903497 : FAILURE in 1h 02m 24s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/f1a5d039bf7742ee94ea3749e9e87aef : FAILURE in 1h 48m 59s\n- openstack-ansible-deploy-aio_metal_ovs-centos-8 https://zuul.opendev.org/t/openstack/build/08e1038d14594e99bec9580204995ce0 : FAILURE in 1h 08m 49s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/f41c68855e5d4a19891e545d6d202d9f : FAILURE in 1h 43m 37s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/4dc18bd31f27443da13899a4cfc9d8b5 : FAILURE in 1h 03m 38s","accounts_in_message":[],"_revision_number":24},{"id":"9bec04b6a492e2b080b59631514389e7a410a91e","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2021-03-14 21:51:26.000000000","message":"Patch Set 24:\n\nHmm,\n\n  changed: [aio1] \u003d\u003e (item\u003d{\u0027service\u0027: {\u0027haproxy_service_name\u0027: \u0027neutron_server\u0027, \u0027haproxy_backend_nodes\u0027: [\u0027aio1_neutron_server_container-9ab61788\u0027], \u0027haproxy_port\u0027: 9696, \u0027haproxy_ssl\u0027: True, \u0027haproxy_ssl_all_vips\u0027: False, \u0027haproxy_balance_type\u0027: \u0027http\u0027, \u0027haproxy_backend_options\u0027: [\u0027httpchk GET / HTTP/1.0\\\\r\\\\nUser-agent:\\\\ osa-haproxy-healthcheck\u0027], \u0027haproxy_service_enabled\u0027: True}})\n\nEvery other service has\n\n  \u0027haproxy_bind\u0027: [\u0027172.29.236.101\u0027],\n\nbut here it\u0027s missing... Which is probably why it\u0027s failing on all.","accounts_in_message":[],"_revision_number":24},{"id":"aa393535ad3057e878edebda821bb6dff2fff9bd","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2021-03-15 10:40:19.000000000","message":"Patch Set 24:\n\nNope, I actually have no idea why neutron-api is failing on all tests. I\u0027m seeing no issues with neutron-api on ussuri with this patch. I do have issues with rpc-server not responding to dhcp-agent.","accounts_in_message":[],"_revision_number":24},{"id":"f3b6b346669a8539d16131478cf283c8c1de7c9a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-05-14 13:39:08.000000000","message":"Patch Set 25: Patch Set 24 was rebased","accounts_in_message":[],"_revision_number":25},{"id":"49485916854737d7c3508e8839efa29b5db13b05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 15:39:36.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/ddae2d64c8354b909ffa84dc05f4a616 : SUCCESS in 10m 39s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/b86a89f7bdee4b2dabe7aad97c1f57ed : SUCCESS in 57m 37s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/d3fd06e97ad64d6b8c8189ae9f02ba24 : SUCCESS in 1h 18m 08s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/f6c22a236415437dbdea2edcc78b69ba : SUCCESS in 1h 15m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1359ba7534d24d44bfbc288513129c13 : FAILURE in 46m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f2b4cba50a747b7a190ce93d4162f57 : SUCCESS in 1h 09m 53s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/ab24ca733bb846b1b72e4600a6d6f3a7 : SUCCESS in 1h 50m 52s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b0aa601332da4ddc90d8c2fc0081a8e5 : SUCCESS in 1h 37m 18s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/705008ee806e4ae5b5e17f6b6eb01a4b : FAILURE in 50m 03s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f18979ece0ce4e8d9c4412df8f6690cf : SUCCESS in 58m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7876c8ce4e2648bab17b63832e2bfecc : SUCCESS in 6m 23s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f2f3386de8c2443796ea0bca9f6ca9ca : FAILURE in 1h 21m 52s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8cadbbe3c8984b6da0166219d9f388ed : FAILURE in 1h 37m 29s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9ecfeab382544d4d8676f483ef00d616 : SUCCESS in 1h 08m 38s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/2056aa0411e046778fcf2818ce4d2b58 : FAILURE in 1h 54m 16s\n- openstack-ansible-deploy-aio_metal_ovs-centos-8 https://zuul.opendev.org/t/openstack/build/bd97e0633d5c406dba5bd2e0cb622009 : SUCCESS in 1h 17m 06s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/e4e4e9a83dbb4a30b49ac25c953e4925 : FAILURE in 1h 40m 01s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/60c1127c65b44e56a3bf6a3de9775691 : FAILURE in 1h 04m 34s","accounts_in_message":[],"_revision_number":25},{"id":"6fb7a6784aababa2a6fa5b7308db25d07565fe16","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-06-09 07:04:32.000000000","message":"Patch Set 26: Patch Set 25 was rebased","accounts_in_message":[],"_revision_number":26},{"id":"af931f2b751453aeba97694b505cd3e6b32abe5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 09:13:23.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/97cf9eff11e74457b608d039e9e9ed20 : SUCCESS in 16m 39s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/47a9361ff1c8423d9f5e07af15b52c09 : SUCCESS in 1h 15m 30s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/a9599892c34b414abd441b7c08fda790 : SUCCESS in 1h 33m 32s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/c55f2129875a4a31a6415a5107501333 : SUCCESS in 1h 22m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f21c3f3be5824ce789e2f5b8d8feade7 : SUCCESS in 1h 19m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1b524a9f2af6414a9a4e8d68d704feb0 : SUCCESS in 1h 11m 30s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/6b1aaf23f809466e9f9ec9ad1440294c : SUCCESS in 1h 53m 40s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c456dafb76a040eba53e1f902293dcc2 : SUCCESS in 1h 47m 11s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/c5ab6a835e58497c9afdeb47c90a7af4 : FAILURE in 57m 51s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/019b095c0d734201994cb8bf7dfa3f63 : SUCCESS in 1h 01m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/754532daf880477abf52605d5360e063 : SUCCESS in 7m 57s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5c1118d79bd24bda8701c30f8393ced9 : FAILURE in 1h 18m 10s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2f2f3d4f26974af89c61197d4c2d87c5 : FAILURE in 49m 46s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/199600ed8e42403ab66e0f667638b3e9 : SUCCESS in 1h 15m 04s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/10c1719ad1364378a41c6113d2667747 : SUCCESS in 2h 05m 54s\n- openstack-ansible-deploy-aio_metal_ovs-centos-8 https://zuul.opendev.org/t/openstack/build/e601447e57c34653beaf25d8e266c3c8 : SUCCESS in 1h 23m 45s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/113658b5c09341f69b0d66977362eade : FAILURE in 1h 52m 35s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/261193b15d0e417ea7788664e8426c20 : FAILURE in 1h 01m 53s","accounts_in_message":[],"_revision_number":26},{"id":"7b2e6d8aa33e2d2562b66abf557fb16c8f10695c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-06-23 10:01:50.000000000","message":"Uploaded patch set 27.","accounts_in_message":[],"_revision_number":27},{"id":"56acd90ca41a1c9b9117944d72230875c46b13df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 12:07:46.000000000","message":"Patch Set 27: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a2fde95edc6f4d91b00a4960271061db : SUCCESS in 19m 05s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/11d637ecb61445f78263ab68c7c97cbf : SUCCESS in 1h 15m 28s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/8d9d2d22bed94aeda104e752f4a64e8b : SUCCESS in 1h 20m 34s (non-voting)\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e845654eba9b4c419f30338b75ff4961 : SUCCESS in 1h 20m 49s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/1e09375935c74cdf9014db639f3b29c8 : SUCCESS in 1h 32m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2065ba565eff4c388881812646de0b26 : SUCCESS in 1h 24m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b8bfd9aea5e84fdc9f35d05142889793 : SUCCESS in 1h 13m 24s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/3fe64a13b0a74ee4b24f16c61240e133 : SUCCESS in 1h 54m 10s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a1c47757974a426388f5abbfd5d03694 : SUCCESS in 1h 43m 09s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/3d1a80ca2e144456953c6cd539472aec : FAILURE in 41m 04s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/509688de04994bb59622c6e8d7f5d377 : SUCCESS in 58m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/220fa5a1f85f4ba28a1e6aadd7dba035 : SUCCESS in 5m 59s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/844313bf914f47dabafd42236f5bd8fa : FAILURE in 1h 21m 49s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ccc6581a24a410ca768915d72ec3353 : FAILURE in 1h 45m 26s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f8d0f825ebea45a88a52727a9aeac329 : SUCCESS in 1h 10m 22s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/222fcc0e0110431191df3276e1801d05 : SUCCESS in 2h 01m 52s\n- openstack-ansible-deploy-aio_metal_ovs-centos-8 https://zuul.opendev.org/t/openstack/build/73ad87896f15450d9aa82d234f443608 : SUCCESS in 1h 21m 24s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/5e4f63ef546d44a7bc2840b4d63e682b : FAILURE in 1h 46m 00s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/e2639557c6e44f9da62c3497773980cd : FAILURE in 1h 04m 01s","accounts_in_message":[],"_revision_number":27},{"id":"fa7b98783aeefd63e6521cd09f25a36ee6af5572","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-08-25 10:58:10.000000000","message":"Patch Set 28: Patch Set 27 was rebased","accounts_in_message":[],"_revision_number":28},{"id":"e26d120756bdd7653a59b3781523e339b96e648f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 13:19:28.000000000","message":"Patch Set 28: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6db1e9f5a17d4a8aa1c7cdd32b13035f : SUCCESS in 15m 19s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/c0df91fa49c24db987a8c7f800f00eda : SUCCESS in 1h 18m 05s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/20d8524862c94b04aea93715e9e7f478 : SUCCESS in 1h 20m 28s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/ec2740b2d1c64e059391ce13a659d013 : SUCCESS in 1h 30m 20s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/907c8182be7f4ed0867d83d85d4aecfc : FAILURE in 1h 30m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f703128f634d4a5eb2cc558bf121814e : SUCCESS in 1h 28m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c45a1f886d5245cbbc108206efe6f352 : SUCCESS in 1h 19m 49s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/eff60741e22443acb39110b6640471aa : SUCCESS in 2h 00m 23s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6bcb33e95efd411eacae522feaeb2983 : SUCCESS in 1h 52m 23s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/416c29f9d0bd4bbd90023ee776c5fc73 : FAILURE in 40m 48s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6529a0bd858f4467b770e263a2b4d172 : SUCCESS in 1h 00m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5bdd663a2ca45adabd0cf6512315eee : SUCCESS in 6m 15s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-focal https://zuul.opendev.org/t/openstack/build/858b8cca244d40d6a0fdd26e35b1033a : FAILURE in 1h 18m 15s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c9e94b4fd33e4200bc37e634e148e1ed : FAILURE in 1h 39m 20s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/071e495b0fbe4a57b75fa0fe4ab4ff5d : SUCCESS in 1h 03m 34s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/ac26cc6518f64261846974e2e104372b : FAILURE in 2h 08m 03s\n- openstack-ansible-deploy-aio_metal_ovs-centos-8-stream https://zuul.opendev.org/t/openstack/build/88aa0dbd3b0848ec92a43e207f84c353 : FAILURE in 1h 24m 47s\n- openstack-ansible-deploy-aio_ovn_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/2f2df70c07c647e49bf463be69bda961 : FAILURE in 2h 17m 30s (non-voting)\n- openstack-ansible-deploy-aio_ovn_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/9fe3eda648f24462b450af09e48c20c5 : FAILURE in 1h 28m 23s (non-voting)\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/2d116b6d249d43fe8ac243c21aa36bb2 : FAILURE in 1h 49m 14s (non-voting)\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/fef85063fb084f3c997faf54191cb59e : FAILURE in 1h 17m 57s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"b663fff0f9b34e9e982cfa234252050f53925bd4","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2021-08-25 14:21:48.000000000","message":"Patch Set 28:\n\n(3 comments)\n\nJust voicing my opinion, we\u0027re moving the api portion out of server. So I think that part deserves its own section. The neutron-rpc-server is still the neutron-server.\n\nDoes that make more sense?","accounts_in_message":[],"_revision_number":28},{"id":"3cdf8d58d9312302c970ec0a1cce3076eefe0fbe","author":{"_account_id":31792,"name":"Marcus Klein","email":"marcus.klein@open-xchange.com","username":"marcus.klein"},"date":"2021-08-26 09:03:34.000000000","message":"Patch Set 28:\n\nI tested this with U in my staging environment and it solves my problem, that response to / contains wrong href scheme of http, although haproxy in front is only listening to https.","accounts_in_message":[],"_revision_number":28},{"id":"7a6c51a495f143770a607c53b1d14274f895e7c4","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-08-26 11:07:20.000000000","message":"Patch Set 28:\n\n(1 comment)\n\n\u003e Patch Set 28:\n\u003e \n\u003e (3 comments)\n\u003e \n\u003e Just voicing my opinion, we\u0027re moving the api portion out of server. So I think that part deserves its own section. The neutron-rpc-server is still the neutron-server.\n\u003e \n\u003e Does that make more sense?\n\nThe problem of that approach is deployers that are upgrading. As then there can be a confusion about what neutron-server is. As well as it would be harder to switch back and forth or at least this path would be not crystal clear.\n\nAlso current way we preserve neutron-server service to be API - either with uwsgi or with native way. And neutron-rpc service will be enabled only when we do uwsgi, otherwise it will be stopped and disabled.","accounts_in_message":[],"_revision_number":28},{"id":"5af1e5ad70194b307e0e04ddb85797e713dd8789","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-08-26 14:12:42.000000000","message":"Uploaded patch set 29.","accounts_in_message":[],"_revision_number":29},{"id":"f3f1e337881fdf0dfae28bfb15d2927ea074b881","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2021-08-26 16:18:12.000000000","message":"Patch Set 29:\n\n\"name service differently with same exec\" would be my point also. I similarly find it weird having neutron-api start in under the neutron-server service. Would there be a third option?\n\nneutron_services:\n  neutron-api:\n    uwsgi_app: neutron_use_uwsgi\n  neutron-server:\n    service_en: !neutron_use_uwsgi\n  neutron-rpc-server:\n    service_en: neutron_use_uwsgi\n\nthe neutron-api gets filtered out by `filtered_neutron_services`,\nneutron-rpc-server starts on use_uwsgi\nand\nneutron-server starts when use_uwsgi is false\n\nI admit it gets a litte complicated, but the list in `neutron_services:` would atleast make more sense. They all belong to the neutron_server group, and will all start in the same neutron container anyways.","accounts_in_message":[],"_revision_number":29},{"id":"aca4953aaf83dee1ee7f1e72281c2a2bb81c1665","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-08-26 16:56:06.000000000","message":"Patch Set 29:\n\n\u003e Patch Set 29:\n\u003e \n\u003e \"name service differently with same exec\" would be my point also. I similarly find it weird having neutron-api start in under the neutron-server service. Would there be a third option?\n\u003e \n\u003e neutron_services:\n\u003e   neutron-api:\n\u003e     uwsgi_app: neutron_use_uwsgi\n\u003e   neutron-server:\n\u003e     service_en: !neutron_use_uwsgi\n\u003e   neutron-rpc-server:\n\u003e     service_en: neutron_use_uwsgi\n\u003e \n\u003e the neutron-api gets filtered out by `filtered_neutron_services`,\n\u003e neutron-rpc-server starts on use_uwsgi\n\u003e and\n\u003e neutron-server starts when use_uwsgi is false\n\u003e \n\u003e I admit it gets a litte complicated, but the list in `neutron_services:` would atleast make more sense. They all belong to the neutron_server group, and will all start in the same neutron container anyways.\n\nThen we would need to do the same with all other services that are done like that. Ie https://opendev.org/openstack/openstack-ansible-os_glance/src/branch/master/defaults/main.yml#L275\n\nIf apply this logic, we would need to name all services that run with uwsgi by name of the wsgi scripts, which doesn\u0027t really make sense to me. And then same API service would have different name depedning if it\u0027s run with uwsgi or not.\n\nSo, I\u0027d preffer to name services like the official documantation reference them, which states only neutron-server as for service that serves API [1][2]. So replacing neutron-server with neutron-api service would be at least pretty confusing.\n\n\n[1] https://docs.openstack.org/neutron/latest/install/controller-install-ubuntu.html#finalize-installation\n[2] https://docs.openstack.org/neutron/latest/install/common/get-started-networking.html","accounts_in_message":[],"_revision_number":29},{"id":"8a873ad0eb8127e0c2d987eb454bdb6029400c1a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-08-26 16:58:39.000000000","message":"Patch Set 30: Patch Set 29 was rebased","accounts_in_message":[],"_revision_number":30},{"id":"aabb03820a04d1dddb22fc295c7ff79aa549bd4e","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2021-08-26 18:07:00.000000000","message":"Patch Set 30:\n\nThanks for taking the time to discuss this\n\n\u003e Then we would need to do the same with all other services that are done like that. Ie https://opendev.org/openstack/openstack-ansible-os_glance/src/branch/master/defaults/main.yml#L275\n\nGlance is only one service (uwsgi), and it\u0027s called glance-api\n\nHave a look at cinder;\nhttps://opendev.org/openstack/openstack-ansible-os_cinder/src/branch/master/defaults/main.yml#L346\nOr nova;\nhttps://opendev.org/openstack/openstack-ansible-os_nova/src/branch/master/defaults/main.yml#L426\nOr keystone;\nhttps://opendev.org/openstack/openstack-ansible-os_keystone/src/branch/master/defaults/main.yml#L543\n\nYou see they ALL got their own -api service.\n\n\u003e If apply this logic, we would need to name all services that run with uwsgi by name of the wsgi scripts, which doesn\u0027t really make sense to me.\n\nSeems to me only keystone has the weird keystone-wsgi-public name, the others are plainly called -api. But yes that happens to be the case for keystone, glance, cinder and nova. cinder and nova have multiple services and the api gets to be its own service.\n\n\u003e And then same API service would have different name depedning if it\u0027s run with uwsgi or not.\n\nNeutron is the last of the \"big-five\" to actually split off it\u0027s api portion. How is the rpc-server more worthy of it\u0027s own service name?\n\n\u003e So, I\u0027d preffer to name services like the official documantation reference them, which states only neutron-server as for service that serves API [1][2]. So replacing neutron-server with neutron-api service would be at least pretty confusing.\n\nIn the case we\u0027re deploying them as separate services it would be just as, if not more important that the rpc-server gets a restart, which is equally undocumented and confusing.\n\n\nI think we both have made good arguments and points on this, would be good if we got a third core\u0027s opinion on the color of this bikeshed ;)","accounts_in_message":[],"_revision_number":30},{"id":"48ef68f950d542aceb42ee3676cab67d0fac1ec1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 19:16:45.000000000","message":"Patch Set 30: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/197919780b61481e9bb23ffc1bc3975b : SUCCESS in 18m 04s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/05d850ae1ca3428c86ae93d1e0d6434f : SUCCESS in 1h 21m 45s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/4081feaeec344675b94f354750716d59 : SUCCESS in 1h 19m 19s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f57db11bf419426d90d1c04427cc1e2d : SUCCESS in 1h 32m 55s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/3376fe548b8a4ada93987c94330219b2 : FAILURE in 1h 34m 24s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2cd4542966434c71badddd9e7e7df79a : SUCCESS in 1h 35m 16s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/edacf0f0c57440dfa58e732290f68a1b : SUCCESS in 1h 15m 42s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/109f3ee64932459b9d2ed15e18e3d713 : FAILURE in 21m 12s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/90ec90aa04514a96b17fb2c6f0cb0b0e : FAILURE in 22m 13s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/c6d553c7911044b294bbbea3f12ade61 : FAILURE in 40m 57s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2d15538304d8448b83de99b24b13224b : SUCCESS in 1h 15m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6b2000eed23469cbb247b8c53f73836 : SUCCESS in 8m 12s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d335177c598c4b0482cd56f3b6131009 : FAILURE in 1h 22m 26s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ebd7bcf85d4246dfbb2917cb9951b25b : FAILURE in 1h 49m 37s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4b3d2ee07d5e4770b1eab8d108df1121 : SUCCESS in 1h 12m 50s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/519196a0572d463bb78ee3fc5632b7f5 : FAILURE in 2h 03m 24s\n- openstack-ansible-deploy-aio_metal_ovs-centos-8-stream https://zuul.opendev.org/t/openstack/build/8f31943305de4462ba9ab397f0c39278 : FAILURE in 1h 25m 21s\n- openstack-ansible-deploy-aio_ovn_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/9dc22a6fe3ea47558fcac3ae9bdd8d48 : FAILURE in 2h 09m 58s (non-voting)\n- openstack-ansible-deploy-aio_ovn_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/7c74076a805a4cc99d2e90cdebf8e9cc : FAILURE in 1h 22m 58s (non-voting)\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/93e6769fd03347a0a65bb0526a696e53 : FAILURE in 1h 50m 00s (non-voting)\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/ab5e9c4f1cd14974a17b797a4630ecdd : FAILURE in 1h 17m 50s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"f2967fa423e8422be0ed47207bd5d7426dfe327e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-08-27 08:24:32.000000000","message":"Patch Set 30:\n\nrecheck","accounts_in_message":[],"_revision_number":30},{"id":"4273ac9b7c9dba3af15be2418ab737bca5f9c5d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-08-27 08:38:55.000000000","message":"Patch Set 31: Patch Set 30 was rebased","accounts_in_message":[],"_revision_number":31},{"id":"8e72761fc893ee95bff676e5de6461f4b29d5553","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-27 11:50:53.000000000","message":"Patch Set 31: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e031a15e95ac4182a43228449ff7c2d1 : SUCCESS in 17m 44s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/bd282c2cec0e47bdb10fbda70344b9f4 : SUCCESS in 1h 16m 38s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/ecee747ad17345bf937d9c3d36f78827 : SUCCESS in 1h 14m 59s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f66c326b0a0d4148a7a0d153cdee3111 : SUCCESS in 1h 44m 04s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/dbb933e8513d4bdca32d6c6a8f70e373 : FAILURE in 1h 44m 43s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9a2777779d8b4e38afcf8700e846bf4e : SUCCESS in 1h 20m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/230c2133c26a4e0abd653c659df57007 : SUCCESS in 1h 10m 40s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/8ac18a6678b34516a1362a1ddc3e7a00 : SUCCESS in 2h 02m 57s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/70ff034d26d04479a8fb6e8b1e55d000 : SUCCESS in 1h 42m 53s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/3d8a63eebb4141aa82096e513863244f : FAILURE in 41m 03s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8ba49eabe0c2477285207c1c184d51b2 : SUCCESS in 55m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aaea9429dcb6451ea71191c46ad33a64 : SUCCESS in 5m 56s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cb23c4282daa4ecdbaa4ad85747534a8 : FAILURE in 1h 13m 41s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/426593fb268748e08b285e8a8af785e5 : SUCCESS in 1h 44m 20s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f566265c7b6b40938be16fae719d1672 : SUCCESS in 1h 07m 00s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/fbbe5bb542ab469d9c08ccaa1e317210 : FAILURE in 2h 15m 07s\n- openstack-ansible-deploy-aio_metal_ovs-centos-8-stream https://zuul.opendev.org/t/openstack/build/d0cf687081be432b83c087caf549bf5c : FAILURE in 1h 26m 52s\n- openstack-ansible-deploy-aio_ovn_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/8067cb4905364b59b4f15b3848ee3fc7 : FAILURE in 2h 09m 56s\n- openstack-ansible-deploy-aio_ovn_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/ef6f2030a73d4654afe20572a361ea61 : FAILURE in 1h 25m 39s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/fd51b5049e61424bb19f5597002bb731 : FAILURE in 1h 51m 58s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/7dc8fa5b1de34410a02f95e4d6824c83 : FAILURE in 1h 11m 55s","accounts_in_message":[],"_revision_number":31},{"id":"626f613966002322c88260a121c0d010912f3119","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-06 17:29:41.000000000","message":"Patch Set 32: Patch Set 31 was rebased","accounts_in_message":[],"_revision_number":32},{"id":"bd9e463301ee43e31eb1ad1c3b77331fe9363a20","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-06 19:12:03.000000000","message":"Patch Set 33: Commit message was updated.","accounts_in_message":[],"_revision_number":33},{"id":"61ac7b7b16754b5c3025fbc25a57187643aa2b4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-06 21:42:10.000000000","message":"Patch Set 33: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2c13a0be9c104a07b396b5eab0c8849e : SUCCESS in 16m 31s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/9bf8d7cd9f3d4e0783849d12e5133312 : SUCCESS in 1h 25m 15s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/03293d9dec4947d8a52dab10e9814e0d : SUCCESS in 1h 08m 37s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/46d482c09bc049e38792f3f21786560e : SUCCESS in 1h 19m 16s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/b668b6125a3547edb50e87f06cd93443 : SUCCESS in 1h 36m 16s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/42cccd184f244536b41871c48b31bb41 : SUCCESS in 1h 49m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7e25da93a58f44d8915b33b52d995ecc : SUCCESS in 1h 18m 10s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/5f60a3cccee44aa39d4efa13c26d0f35 : FAILURE in 2h 15m 47s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/433cccf941c54fbbb22dc0a06bfb6cd4 : SUCCESS in 1h 41m 20s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/f42588dd47fb4cc4b3ee9055ab3ff4ed : FAILURE in 50m 07s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f06d8907a1724b7d8285a836d82ca4aa : SUCCESS in 54m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a12a16bc6884ee7861dc2609c4394b4 : SUCCESS in 5m 59s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-focal https://zuul.opendev.org/t/openstack/build/961a46aed35c40339980c4d9f75705a4 : FAILURE in 1h 19m 36s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/30e64ae3e5ce45f8b8be562fb00949e5 : SUCCESS in 2h 04m 41s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/004000409cfc456799528cbb4ce85857 : SUCCESS in 1h 16m 00s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/846f00b4a51540898709732248d9eeb5 : SUCCESS in 2h 11m 26s\n- openstack-ansible-deploy-aio_metal_ovs-centos-8-stream https://zuul.opendev.org/t/openstack/build/8f4dc8b62df94ee8a5e32edc75492f77 : SUCCESS in 1h 16m 43s\n- openstack-ansible-deploy-aio_ovn_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/438ee9da18fc48f593bcf2339dd83058 : FAILURE in 2h 22m 43s\n- openstack-ansible-deploy-aio_ovn_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/252360f8a71844d2ac1fa6caa7a2294e : FAILURE in 1h 16m 37s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/7989cb6094fc4415b321603b46b2fcfa : FAILURE in 2h 05m 13s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/9ac242757c00471aafa97ffcfd43968a : FAILURE in 1h 27m 53s","accounts_in_message":[],"_revision_number":33},{"id":"58227c17259c65ddf89fc9c725986725ed50bd27","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-06 21:46:27.000000000","message":"Patch Set 34: Published edit on patch set 33.","accounts_in_message":[],"_revision_number":34},{"id":"99ef66742e6bad437a37798c0661796a38045fbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-07 00:29:22.000000000","message":"Patch Set 34: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/663b16caf95e422e869c7a210402069f : SUCCESS in 14m 44s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/5e87d7af592c4c19ae0333c1607d3dc6 : SUCCESS in 1h 40m 41s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/d96f73621b5541c7be971164948be72e : SUCCESS in 1h 23m 18s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/bdbd121fc56d48cba9fc8848148d479f : SUCCESS in 1h 18m 10s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/9a5f8e8d7f864454a3b90fd929ff9f3d : SUCCESS in 1h 28m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/102507afaa3841d3acaf7e3e05011a78 : SUCCESS in 1h 31m 44s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ad922877d69048f08e4fe9510546148f : SUCCESS in 1h 35m 23s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/80f124df21be46e58374481689c84a59 : FAILURE in 1h 53m 08s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d5b20cd7a21f4af6a63914160f6bfc84 : SUCCESS in 1h 53m 18s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/8f0d565028c54c088c9d0439d8e27faa : FAILURE in 36m 24s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f2850399051c47ca8c72753eab81fdcf : SUCCESS in 1h 03m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f2691b0862e44d8827f2581660542a2 : SUCCESS in 6m 04s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b05bef69d05b42fa9275be98d10b63fc : FAILURE in 1h 25m 47s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2ce87695524f4c78b2c839522d291e62 : SUCCESS in 2h 05m 20s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fea8d1b50e974bfebed52e9089eb9671 : SUCCESS in 1h 16m 55s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/b44481c6d3624f5daa908653397ea1d7 : SUCCESS in 2h 36m 56s\n- openstack-ansible-deploy-aio_metal_ovs-centos-8-stream https://zuul.opendev.org/t/openstack/build/9922035db2cd42fca8735860b191a687 : SUCCESS in 1h 31m 45s\n- openstack-ansible-deploy-aio_ovn_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/0cedfe1a848a4f35899e4b092f88fa09 : SUCCESS in 2h 10m 03s\n- openstack-ansible-deploy-aio_ovn_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/08430defb38d4001bc2cd1115ce3ba00 : SUCCESS in 1h 27m 03s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/f816ab177e914aa0bcadab4155ee8d88 : SUCCESS in 1h 44m 34s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/89227fe61b36400881df9d4fd959c740 : SUCCESS in 1h 22m 41s","accounts_in_message":[],"_revision_number":34},{"id":"c19d37589d17b036c6dedfeb41cd08a40934eae4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-07 06:02:58.000000000","message":"Patch Set 35: Published edit on patch set 34.","accounts_in_message":[],"_revision_number":35},{"id":"60aa728f5bc279cc3ac5303dd62bf12fb989672e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-07 09:16:58.000000000","message":"Patch Set 35: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6791c3fbdb9c436c92db42787f8e31cb : SUCCESS in 15m 48s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/2d50d05dbb1146db97b2c34615f80f61 : SUCCESS in 1h 34m 28s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/c1b6b358a8ab41938908e0ed1ce73a33 : SUCCESS in 1h 10m 26s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/9d03d8cfbb66464fa91986ea61fb3a99 : SUCCESS in 1h 39m 36s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/734ab0a93c134ca8b5385a1fc1717ef1 : FAILURE in 43m 08s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9f350d74e14f4932885f410ab4265b3c : SUCCESS in 1h 23m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/de296a7e8bf94028897a187d1639f6cc : SUCCESS in 1h 27m 31s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/66a421b474da4c68b644cd3061dc8611 : FAILURE in 42m 48s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/36615d22a0fb4288a75beb07b911c897 : TIMED_OUT in 3h 07m 58s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/6c928f98feec4cc39707b5cef47870bc : FAILURE in 46m 05s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7bf5b35b027b4817a3f6ab932fc9d427 : SUCCESS in 1h 00m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00e12aa81df64fc5aa92ccb15efbc501 : SUCCESS in 6m 40s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-focal https://zuul.opendev.org/t/openstack/build/70902eeccb7944d5a2a2870597e2b59e : SUCCESS in 1h 30m 24s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6e65bfc4696744d3b7abb4364289b063 : SUCCESS in 2h 05m 06s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7c8bdd3d7f1d45779892267bb1213014 : SUCCESS in 1h 22m 13s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/d26ff21e3df847218afef9998f42b669 : SUCCESS in 1h 55m 11s\n- openstack-ansible-deploy-aio_metal_ovs-centos-8-stream https://zuul.opendev.org/t/openstack/build/7bf7bcc300774013b3ebc64fdbedab3d : SUCCESS in 1h 23m 02s\n- openstack-ansible-deploy-aio_ovn_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/52f443dfea7d4b2588ccc4084bb15b04 : SUCCESS in 2h 06m 30s\n- openstack-ansible-deploy-aio_ovn_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/99eacc0ca4de4c82a972d48e1a9fddcc : SUCCESS in 1h 20m 33s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/180b0daa14b3465985d672f6895df1d3 : SUCCESS in 2h 09m 46s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/a281179ffc164285a6975e1f6c547937 : SUCCESS in 1h 21m 57s","accounts_in_message":[],"_revision_number":35},{"id":"76ec61d655e57f2eda653ecf0cc844fc890542d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-10 12:11:29.000000000","message":"Patch Set 36: Published edit on patch set 35.","accounts_in_message":[],"_revision_number":36},{"id":"b97de2048d6d8631e54b4d6c19c15bca386048b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-10 12:11:49.000000000","message":"Patch Set 37: Published edit on patch set 36.","accounts_in_message":[],"_revision_number":37},{"id":"2c84c495117d6ea41f9b9f799911849bafa98dab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-10 14:25:01.000000000","message":"Patch Set 37: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c54f842de4c948dd95d419c559867fb6 : SUCCESS in 14m 59s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/4b09a39987364c50910047b4a4f8729f : SUCCESS in 1h 40m 44s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/7c06fdf6913d4d999901da6cae7ca5b3 : SUCCESS in 1h 18m 11s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/9a266437c251495abf607ed67be9f4e8 : SUCCESS in 1h 23m 31s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/67c13fbf1a544a69a5e41e6517994d3e : SUCCESS in 1h 29m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eaf3f0c71aa74020b8b155d7b8eac7b1 : SUCCESS in 1h 31m 34s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1b98c7d7b1734c58973c70a1458089f2 : SUCCESS in 1h 32m 58s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/ac1916be8b2f4e99a7660cd5ab086e57 : FAILURE in 2h 03m 54s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/906ff5872ffa4804b12bf36936f30b2b : SUCCESS in 1h 58m 04s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/28b14f87fbf94b309bb2c5ab43fa71df : FAILURE in 38m 18s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a18d3590f5084de7bbf112305f6831b4 : SUCCESS in 1h 03m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd6e18be871e4d30b2c96399f4d08c34 : SUCCESS in 7m 03s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c73eb58e91c34d2d98c0790b902734a0 : SUCCESS in 1h 27m 27s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5583b144b4604139a62fba8a911321ad : SUCCESS in 1h 58m 04s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/44adea94f7aa42e3b4004154ba19e58d : SUCCESS in 1h 29m 58s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/a0129fb5d9484bd6bba5aa740e1a6cd0 : SUCCESS in 2h 00m 32s\n- openstack-ansible-deploy-aio_metal_ovs-centos-8-stream https://zuul.opendev.org/t/openstack/build/9550f7afc211459486e3ac32b8ee3f35 : SUCCESS in 1h 18m 01s\n- openstack-ansible-deploy-aio_ovn_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/b080412482d94be496a223e81800fda3 : SUCCESS in 2h 06m 46s\n- openstack-ansible-deploy-aio_ovn_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/47ff0b53a0974343a0176a22b315a59a : SUCCESS in 1h 19m 16s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/a72047c55fa841eaaf1282c037edf59b : SUCCESS in 1h 57m 35s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/579cc39485f3485797d4b22fa7bd8585 : SUCCESS in 1h 20m 58s","accounts_in_message":[],"_revision_number":37},{"id":"a0b42aff7c31047940a3a7a5f424b85e3f33e842","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-03-21 11:59:01.000000000","message":"Uploaded patch set 38.","accounts_in_message":[],"_revision_number":38},{"id":"3a94c6c37f96becd14db886ef63774448418a547","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-21 14:29:36.000000000","message":"Patch Set 38: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dd9be7b4e9874642a5aa84d6192f5c8a : SUCCESS in 13m 22s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/f57d9e9c0b994dfd830791cee6985048 : SUCCESS in 1h 10m 47s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/49b1f029dc254b83b1b4dc05ab9a652b : SUCCESS in 1h 20m 25s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/be03654caf8346e3b297be162c4a49d8 : SUCCESS in 1h 25m 50s\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/05a10e1fc59b49c0b2e450967c4184a2 : SUCCESS in 1h 19m 14s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/280b3f0a76374c93a9cfdea148894d7a : FAILURE in 1h 22m 56s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b267602cf904ea9874a932857342ed7 : SUCCESS in 2h 23m 27s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/665ab218ecfa42e29dae22132b0d5588 : FAILURE in 32m 27s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bf3451f358ce4b3c8440c1d90ea994e1 : SUCCESS in 57m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5ee9d3cb03c4a118ab7a11a271c06ea : SUCCESS in 6m 41s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c120a59ca861487999ec186d05ab8f52 : FAILURE in 1h 21m 37s (non-voting)\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b676c3972f7b442b96585719ef9ee742 : SUCCESS in 2h 21m 42s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/efad0979595e429ca781390c8cc866f4 : SUCCESS in 1h 21m 44s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/1cfd4e04ca084656b85d5813ce05664a : SUCCESS in 1h 59m 12s\n- openstack-ansible-deploy-aio_metal_ovs-centos-8-stream https://zuul.opendev.org/t/openstack/build/3d58a8d87ba142329fdf2f16a9c9db1e : SUCCESS in 1h 20m 12s\n- openstack-ansible-deploy-aio_ovn_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/0ac3be7bd7ed4c14a1c621ec8d6fed2e : SUCCESS in 2h 03m 15s\n- openstack-ansible-deploy-aio_ovn_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/6155dc4fba1d49a197cc6378247102a6 : SUCCESS in 1h 18m 40s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/b722f2ea478c4f00a30c71bc4a5653c6 : SUCCESS in 2h 05m 36s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/88dfcec15167441ca28668b7255a55ae : SUCCESS in 1h 22m 55s","accounts_in_message":[],"_revision_number":38},{"id":"2a4aa1dbb46f0869005880baf4e797bf94063d40","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-05-27 18:34:38.000000000","message":"Patch Set 39: Patch Set 38 was rebased","accounts_in_message":[],"_revision_number":39},{"id":"d6ad14c8c8c62dce052965b6344d9c604d794ac5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-27 21:04:16.000000000","message":"Patch Set 39: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/78c3ec5e2b4b47bb9f3665c9b651e112 : SUCCESS in 12m 57s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/a0a9c9e1600c440faaf19a356b70cdde : SUCCESS in 1h 19m 02s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/5c39b3ec0acb482a8d6ab8baf2baa7cb : SUCCESS in 1h 35m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a4163a53594c40718ed18aa244d91aea : SUCCESS in 1h 31m 34s\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/674360ec0e1c431eac4af59c9f11862a : SUCCESS in 1h 29m 18s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/896a1ef5ed8e4336a620937c17512bd7 : FAILURE in 1h 39m 55s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c42fd595273243289bc0f328fb232f3b : SUCCESS in 1h 59m 29s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/c67d2e3f75e345199175b88ba7bbeb90 : FAILURE in 1h 04m 19s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/41ce936760da445f95e4c88b5eee343c : SUCCESS in 1h 06m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04d158c98b9344f3aeb01d526b301dc1 : SUCCESS in 5m 16s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b8616f8254034bbcb850b79f9af1470a : FAILURE in 1h 21m 28s (non-voting)\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0d65905890de41b49330efa26e37cad4 : SUCCESS in 2h 12m 35s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1edffd352af74aebbaf456591fd75c11 : SUCCESS in 1h 28m 32s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/8f19cb8f42ca4f24b25f71799589f553 : SUCCESS in 2h 00m 58s\n- openstack-ansible-deploy-aio_metal_ovs-centos-8-stream https://zuul.opendev.org/t/openstack/build/f5f02e50e1754fee922d4a965ae20723 : SUCCESS in 1h 25m 57s\n- openstack-ansible-deploy-aio_ovn_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/79966217b4704ffdbc8719f733fe42ca : SUCCESS in 2h 22m 17s\n- openstack-ansible-deploy-aio_ovn_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/a74f0775df2e4e4da1837a2f55cf41ab : SUCCESS in 1h 24m 03s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/56f85ebf14624a6d84260b9f5bfd6839 : SUCCESS in 2h 19m 22s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/b1dca6e08e514b72bf0a2b3dd364a28e : SUCCESS in 1h 21m 51s","accounts_in_message":[],"_revision_number":39},{"id":"286f2bce32808d269728a581867f13c05d5039f0","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-06-08 08:13:46.000000000","message":"Patch Set 39: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":39},{"id":"a2771621a92a5d9c7070ef774785ee00a3764d75","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-06-08 08:52:06.000000000","message":"Patch Set 39: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":39},{"id":"b2da716d557ebbc45ee23f4819521ecaea96419d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-08 08:54:24.000000000","message":"Patch Set 39: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":39},{"id":"7df34c544c04dba79c4f85a300968e7724927e54","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-08 11:19:01.000000000","message":"Patch Set 39: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ebbbc335074f497ba86833877c1014c3 : SUCCESS in 18m 37s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/fd44990dff0b4201891077b402fe285f : SUCCESS in 1h 19m 14s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/52b86490c50049ecab03bb7e55fbb71a : SUCCESS in 1h 28m 14s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/1bc036e3c6244cefad72d05b277b2800 : SUCCESS in 1h 31m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd5392e2ce144c7c9fffd44074fe3637 : SUCCESS in 1h 23m 47s\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/f12258dcaabe43d7ab62932460fcc134 : SUCCESS in 1h 24m 19s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f2fead21c5b243feaa4d04e2132ac087 : SUCCESS in 1h 03m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b34b221e1c34ae5af14445f8c215934 : SUCCESS in 5m 53s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/076fc4ccb2534ef887f50e2000d8a8de : SUCCESS in 2h 05m 03s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c286c73d35f94d21bc0ddfaa0dfdd167 : SUCCESS in 1h 21m 30s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/1a9d352d9b4345b289f11ed7f7351cb0 : SUCCESS in 2h 06m 08s\n- openstack-ansible-deploy-aio_metal_ovs-centos-8-stream https://zuul.opendev.org/t/openstack/build/86895b28935f4ad49f7e94c6933d79b7 : SUCCESS in 1h 24m 48s\n- openstack-ansible-deploy-aio_ovn_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/e34dddb449384ddea5f1ac40ad7d989e : SUCCESS in 2h 16m 04s\n- openstack-ansible-deploy-aio_ovn_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/f0207237fa95445fa33effd5295139bb : SUCCESS in 1h 24m 08s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/a22cd56646224f4b86ad136078afa287 : SUCCESS in 2h 13m 27s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/908b59d0d87c41dfa7180ee0999c27cd : SUCCESS in 1h 33m 49s","accounts_in_message":[],"_revision_number":39},{"id":"1020713c6b3216aadc00e5bc98547710c2845515","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-08 11:19:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":39},{"id":"3a26b2d860440f2fb06a3cc79498c1f774030c0d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-08 11:20:17.000000000","message":"Patch Set 39:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a64faae78f74603bc1d67b2718cc00a : SUCCESS in 51s","accounts_in_message":[],"_revision_number":39}],"current_revision_number":39,"current_revision":"bdd06c84488c7008c0dcb3eaca4effc8d38fcaeb","revisions":{"256c220eac6b25015af6fe03d4b1d9642e17d711":{"kind":"REWORK","_number":1,"created":"2017-07-21 14:12:08.000000000","uploader":{"_account_id":2799,"name":"Andy McCrae","email":"andy.mccrae@googlemail.com","username":"andrew-mccrae"},"ref":"refs/changes/56/486156/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/56/486156/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/1"}}},"commit":{"parents":[{"commit":"b25a6fb29ebcbf17b5c2ad8593f5f546e11e57a2","subject":"Update URLs in documents according to document migration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/b25a6fb29ebcbf17b5c2ad8593f5f546e11e57a2"}]}],"author":{"name":"Andy McCrae","email":"andy.mccrae@gmail.com","date":"2017-07-21 14:10:18.000000000","tz":60},"committer":{"name":"Andy McCrae","email":"andy.mccrae@gmail.com","date":"2017-07-21 14:11:32.000000000","tz":60},"subject":"Implement uWSGI for neutron-api","message":"Implement uWSGI for neutron-api\n\nAs part of the Pike goals we are moving api services to run as WSGI\napps. neutron-server service is set up as a wsgi app, and this patch\nmoves it over.\n\nSince this is just a drop in replacement for the existing eventlet\nservice, operators an deployers should notice no difference.\n\nAdditionally, for Neutron, the logging method has been reworked for the\ninit scripts to match up with other roles.\n\nChange-Id: Ia7ebd13be9ce7834679d439b7bda242805768ef8\nImplements: blueprint goal-deploy-api-in-wsgi\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/256c220eac6b25015af6fe03d4b1d9642e17d711"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/256c220eac6b25015af6fe03d4b1d9642e17d711"}]},"branch":"refs/heads/master"},"432ba474cdc952edad7814adb0ea3b17e9f91cce":{"kind":"REWORK","_number":2,"created":"2017-07-21 14:21:16.000000000","uploader":{"_account_id":2799,"name":"Andy McCrae","email":"andy.mccrae@googlemail.com","username":"andrew-mccrae"},"ref":"refs/changes/56/486156/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/56/486156/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/2"}}},"commit":{"parents":[{"commit":"b25a6fb29ebcbf17b5c2ad8593f5f546e11e57a2","subject":"Update URLs in documents according to document migration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/b25a6fb29ebcbf17b5c2ad8593f5f546e11e57a2"}]}],"author":{"name":"Andy McCrae","email":"andy.mccrae@gmail.com","date":"2017-07-21 14:10:18.000000000","tz":60},"committer":{"name":"Andy McCrae","email":"andy.mccrae@googlemail.com","date":"2017-07-21 14:21:10.000000000","tz":0},"subject":"Implement uWSGI for neutron-api","message":"Implement uWSGI for neutron-api\n\nAs part of the Pike goals we are moving api services to run as WSGI\napps. neutron-server service is set up as a wsgi app, and this patch\nmoves it over.\n\nSince this is just a drop in replacement for the existing eventlet\nservice, operators an deployers should notice no difference.\n\nAdditionally, for Neutron, the logging method has been reworked for the\ninit scripts to match up with other roles.\n\nChange-Id: Ia7ebd13be9ce7834679d439b7bda242805768ef8\nImplements: blueprint goal-deploy-api-in-wsgi\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/432ba474cdc952edad7814adb0ea3b17e9f91cce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/432ba474cdc952edad7814adb0ea3b17e9f91cce"}]},"branch":"refs/heads/master"},"afb509eb0e0d47768f29a4eb94f8326f07b5c81c":{"kind":"REWORK","_number":3,"created":"2017-07-22 11:08:48.000000000","uploader":{"_account_id":2799,"name":"Andy McCrae","email":"andy.mccrae@googlemail.com","username":"andrew-mccrae"},"ref":"refs/changes/56/486156/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/56/486156/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/3"}}},"commit":{"parents":[{"commit":"b25a6fb29ebcbf17b5c2ad8593f5f546e11e57a2","subject":"Update URLs in documents according to document migration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/b25a6fb29ebcbf17b5c2ad8593f5f546e11e57a2"}]}],"author":{"name":"Andy McCrae","email":"andy.mccrae@gmail.com","date":"2017-07-21 14:10:18.000000000","tz":60},"committer":{"name":"Andy McCrae","email":"andy.mccrae@gmail.com","date":"2017-07-22 11:08:42.000000000","tz":60},"subject":"Implement uWSGI for neutron-api","message":"Implement uWSGI for neutron-api\n\nAs part of the Pike goals we are moving api services to run as WSGI\napps. neutron-server service is set up as a wsgi app, and this patch\nmoves it over.\n\nSince this is just a drop in replacement for the existing eventlet\nservice, operators an deployers should notice no difference.\n\nAdditionally, for Neutron, the logging method has been reworked for the\ninit scripts to match up with other roles.\n\nChange-Id: Ia7ebd13be9ce7834679d439b7bda242805768ef8\nImplements: blueprint goal-deploy-api-in-wsgi\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/afb509eb0e0d47768f29a4eb94f8326f07b5c81c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/afb509eb0e0d47768f29a4eb94f8326f07b5c81c"}]},"branch":"refs/heads/master"},"230e5c62248c1c1219f39b23b5021b6b1cd60d3b":{"kind":"REWORK","_number":4,"created":"2017-07-24 08:07:00.000000000","uploader":{"_account_id":2799,"name":"Andy McCrae","email":"andy.mccrae@googlemail.com","username":"andrew-mccrae"},"ref":"refs/changes/56/486156/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/56/486156/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/4"}}},"commit":{"parents":[{"commit":"b25a6fb29ebcbf17b5c2ad8593f5f546e11e57a2","subject":"Update URLs in documents according to document migration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/b25a6fb29ebcbf17b5c2ad8593f5f546e11e57a2"}]}],"author":{"name":"Andy McCrae","email":"andy.mccrae@gmail.com","date":"2017-07-21 14:10:18.000000000","tz":60},"committer":{"name":"Andy McCrae","email":"andy.mccrae@gmail.com","date":"2017-07-24 08:06:56.000000000","tz":60},"subject":"Implement uWSGI for neutron-api","message":"Implement uWSGI for neutron-api\n\nAs part of the Pike goals we are moving api services to run as WSGI\napps. neutron-server service is set up as a wsgi app, and this patch\nmoves it over.\n\nSince this is just a drop in replacement for the existing eventlet\nservice, operators an deployers should notice no difference.\n\nAdditionally, for Neutron, the logging method has been reworked for the\ninit scripts to match up with other roles.\n\nChange-Id: Ia7ebd13be9ce7834679d439b7bda242805768ef8\nImplements: blueprint goal-deploy-api-in-wsgi\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/230e5c62248c1c1219f39b23b5021b6b1cd60d3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/230e5c62248c1c1219f39b23b5021b6b1cd60d3b"}]},"branch":"refs/heads/master"},"f5a08f58d90332cc96fdae0ed8fd608e0c2b874b":{"kind":"REWORK","_number":5,"created":"2017-07-24 09:54:31.000000000","uploader":{"_account_id":2799,"name":"Andy McCrae","email":"andy.mccrae@googlemail.com","username":"andrew-mccrae"},"ref":"refs/changes/56/486156/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/56/486156/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/5"}}},"commit":{"parents":[{"commit":"b25a6fb29ebcbf17b5c2ad8593f5f546e11e57a2","subject":"Update URLs in documents according to document migration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/b25a6fb29ebcbf17b5c2ad8593f5f546e11e57a2"}]}],"author":{"name":"Andy McCrae","email":"andy.mccrae@gmail.com","date":"2017-07-21 14:10:18.000000000","tz":60},"committer":{"name":"Andy McCrae","email":"andy.mccrae@gmail.com","date":"2017-07-24 09:54:26.000000000","tz":60},"subject":"Implement uWSGI for neutron-api","message":"Implement uWSGI for neutron-api\n\nAs part of the Pike goals we are moving api services to run as WSGI\napps. neutron-server service is set up as a wsgi app, and this patch\nmoves it over.\n\nSince this is just a drop in replacement for the existing eventlet\nservice, operators an deployers should notice no difference.\n\nAdditionally, for Neutron, the logging method has been reworked for the\ninit scripts to match up with other roles.\n\nChange-Id: Ia7ebd13be9ce7834679d439b7bda242805768ef8\nImplements: blueprint goal-deploy-api-in-wsgi\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/f5a08f58d90332cc96fdae0ed8fd608e0c2b874b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/f5a08f58d90332cc96fdae0ed8fd608e0c2b874b"}]},"branch":"refs/heads/master"},"49c82dd44c3837b366df880bd15aa568d180be4b":{"kind":"REWORK","_number":6,"created":"2017-07-24 12:19:34.000000000","uploader":{"_account_id":2799,"name":"Andy McCrae","email":"andy.mccrae@googlemail.com","username":"andrew-mccrae"},"ref":"refs/changes/56/486156/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/56/486156/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/6"}}},"commit":{"parents":[{"commit":"b25a6fb29ebcbf17b5c2ad8593f5f546e11e57a2","subject":"Update URLs in documents according to document migration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/b25a6fb29ebcbf17b5c2ad8593f5f546e11e57a2"}]}],"author":{"name":"Andy McCrae","email":"andy.mccrae@gmail.com","date":"2017-07-21 14:10:18.000000000","tz":60},"committer":{"name":"Andy McCrae","email":"andy.mccrae@gmail.com","date":"2017-07-24 12:19:28.000000000","tz":60},"subject":"Implement uWSGI for neutron-api","message":"Implement uWSGI for neutron-api\n\nAs part of the Pike goals we are moving api services to run as WSGI\napps. neutron-server service is set up as a wsgi app, and this patch\nmoves it over.\n\nSince this is just a drop in replacement for the existing eventlet\nservice, operators an deployers should notice no difference.\n\nAdditionally, for Neutron, the logging method has been reworked for the\ninit scripts to match up with other roles.\n\nChange-Id: Ia7ebd13be9ce7834679d439b7bda242805768ef8\nImplements: blueprint goal-deploy-api-in-wsgi\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/49c82dd44c3837b366df880bd15aa568d180be4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/49c82dd44c3837b366df880bd15aa568d180be4b"}]},"branch":"refs/heads/master"},"66d4b20035a4e6be21d0e9929db70c26f301ec0c":{"kind":"REWORK","_number":7,"created":"2017-07-24 13:48:17.000000000","uploader":{"_account_id":2799,"name":"Andy McCrae","email":"andy.mccrae@googlemail.com","username":"andrew-mccrae"},"ref":"refs/changes/56/486156/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/56/486156/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/7"}}},"commit":{"parents":[{"commit":"b25a6fb29ebcbf17b5c2ad8593f5f546e11e57a2","subject":"Update URLs in documents according to document migration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/b25a6fb29ebcbf17b5c2ad8593f5f546e11e57a2"}]}],"author":{"name":"Andy McCrae","email":"andy.mccrae@gmail.com","date":"2017-07-21 14:10:18.000000000","tz":60},"committer":{"name":"Andy McCrae","email":"andy.mccrae@gmail.com","date":"2017-07-24 13:48:05.000000000","tz":60},"subject":"Implement uWSGI for neutron-api","message":"Implement uWSGI for neutron-api\n\nAs part of the Pike goals we are moving api services to run as WSGI\napps. neutron-server service is set up as a wsgi app, and this patch\nmoves it over.\n\nSince this is just a drop in replacement for the existing eventlet\nservice, operators an deployers should notice no difference.\n\nAdditionally, for Neutron, the logging method has been reworked for the\ninit scripts to match up with other roles.\n\nChange-Id: Ia7ebd13be9ce7834679d439b7bda242805768ef8\nImplements: blueprint goal-deploy-api-in-wsgi\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/66d4b20035a4e6be21d0e9929db70c26f301ec0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/66d4b20035a4e6be21d0e9929db70c26f301ec0c"}]},"branch":"refs/heads/master"},"13a82118bf936af9c63cd6856b9dbbad11a5c8b7":{"kind":"REWORK","_number":8,"created":"2017-12-06 16:01:51.000000000","uploader":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"ref":"refs/changes/56/486156/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/56/486156/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/8"}}},"commit":{"parents":[{"commit":"962cd92243641092412b6ef09a41bbf5e698c4a1","subject":"Merge \"Allow the experimental trigger of the integrated repo\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/962cd92243641092412b6ef09a41bbf5e698c4a1"}]}],"author":{"name":"Andy McCrae","email":"andy.mccrae@gmail.com","date":"2017-07-21 14:10:18.000000000","tz":60},"committer":{"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","date":"2017-12-06 16:00:38.000000000","tz":-480},"subject":"Implement uWSGI for neutron-api","message":"Implement uWSGI for neutron-api\n\nAs part of the Pike goals we are moving api services to run as WSGI\napps. neutron-server service is set up as a wsgi app, and this patch\nmoves it over.\n\nSince this is just a drop in replacement for the existing eventlet\nservice, operators an deployers should notice no difference.\n\nAdditionally, for Neutron, the logging method has been reworked for the\ninit scripts to match up with other roles.\n\nChange-Id: Ia7ebd13be9ce7834679d439b7bda242805768ef8\nImplements: blueprint goal-deploy-api-in-wsgi\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/13a82118bf936af9c63cd6856b9dbbad11a5c8b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/13a82118bf936af9c63cd6856b9dbbad11a5c8b7"}]},"branch":"refs/heads/master"},"5e81ec0561ac1c8f1e49001981c8956b0d46b441":{"kind":"REWORK","_number":9,"created":"2019-09-19 17:06:58.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/56/486156/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/56/486156/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/9"}}},"commit":{"parents":[{"commit":"9f06d548abf350c869193033afda6bdf366aa9ab","subject":"Merge \"Allow venv python interpreter to be overridden\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/9f06d548abf350c869193033afda6bdf366aa9ab"}]}],"author":{"name":"Andy McCrae","email":"andy.mccrae@gmail.com","date":"2017-07-21 14:10:18.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-09-19 17:06:30.000000000","tz":180},"subject":"Implement uWSGI for neutron-api","message":"Implement uWSGI for neutron-api\n\nAs part of the Pike goals we are moving api services to run as WSGI\napps. neutron-server service is set up as a wsgi app, and this patch\nmoves it over.\n\nSince this is just a drop in replacement for the existing eventlet\nservice, operators an deployers should notice no difference.\n\nChange-Id: Ia7ebd13be9ce7834679d439b7bda242805768ef8\nImplements: blueprint goal-deploy-api-in-wsgi\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/5e81ec0561ac1c8f1e49001981c8956b0d46b441"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/5e81ec0561ac1c8f1e49001981c8956b0d46b441"}]},"branch":"refs/heads/master"},"b36fd0e0a3784f002bc47dc9947aabaeb589f20f":{"kind":"REWORK","_number":10,"created":"2019-09-20 09:47:28.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/56/486156/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/56/486156/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/10"}}},"commit":{"parents":[{"commit":"9f06d548abf350c869193033afda6bdf366aa9ab","subject":"Merge \"Allow venv python interpreter to be overridden\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/9f06d548abf350c869193033afda6bdf366aa9ab"}]}],"author":{"name":"Andy McCrae","email":"andy.mccrae@gmail.com","date":"2017-07-21 14:10:18.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-09-20 09:47:13.000000000","tz":180},"subject":"Implement uWSGI for neutron-api","message":"Implement uWSGI for neutron-api\n\nAs part of the Pike goals we are moving api services to run as WSGI\napps. neutron-server service is set up as a wsgi app, and this patch\nmoves it over.\n\nSince this is just a drop in replacement for the existing eventlet\nservice, operators an deployers should notice no difference.\n\nChange-Id: Ia7ebd13be9ce7834679d439b7bda242805768ef8\nImplements: blueprint goal-deploy-api-in-wsgi\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/b36fd0e0a3784f002bc47dc9947aabaeb589f20f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/b36fd0e0a3784f002bc47dc9947aabaeb589f20f"}]},"branch":"refs/heads/master"},"58ad9417bd9f12f843da8be2a45b62a95219b26a":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2019-12-09 02:32:50.000000000","uploader":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"ref":"refs/changes/56/486156/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/56/486156/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/11"}}},"commit":{"parents":[{"commit":"05c45ad94f32c3b861d5163b4cec7d648f2d47fd","subject":"Merge \"Adds DPDK support for Open vSwitch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/05c45ad94f32c3b861d5163b4cec7d648f2d47fd"}]}],"author":{"name":"Andy McCrae","email":"andy.mccrae@gmail.com","date":"2017-07-21 14:10:18.000000000","tz":60},"committer":{"name":"James Denton","email":"james.denton@outlook.com","date":"2019-12-09 02:32:50.000000000","tz":0},"subject":"Implement uWSGI for neutron-api","message":"Implement uWSGI for neutron-api\n\nAs part of the Pike goals we are moving api services to run as WSGI\napps. neutron-server service is set up as a wsgi app, and this patch\nmoves it over.\n\nSince this is just a drop in replacement for the existing eventlet\nservice, operators an deployers should notice no difference.\n\nChange-Id: Ia7ebd13be9ce7834679d439b7bda242805768ef8\nImplements: blueprint goal-deploy-api-in-wsgi\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/58ad9417bd9f12f843da8be2a45b62a95219b26a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/58ad9417bd9f12f843da8be2a45b62a95219b26a"}]},"branch":"refs/heads/master"},"a8fcc2e1d12d64d1b7c1fbc4fa410c4d2b4b7d7b":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2020-03-30 11:04:33.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/56/486156/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/56/486156/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/12"}}},"commit":{"parents":[{"commit":"bcc7c496be7cbe4aca15450c90a1312f96c3b731","subject":"Add systemd-devel to neutron_devel_distro_packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/bcc7c496be7cbe4aca15450c90a1312f96c3b731"}]}],"author":{"name":"Andy McCrae","email":"andy.mccrae@gmail.com","date":"2017-07-21 14:10:18.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-03-30 11:04:31.000000000","tz":60},"subject":"Implement uWSGI for neutron-api","message":"Implement uWSGI for neutron-api\n\nAs part of the Pike goals we are moving api services to run as WSGI\napps. neutron-server service is set up as a wsgi app, and this patch\nmoves it over.\n\nSince this is just a drop in replacement for the existing eventlet\nservice, operators an deployers should notice no difference.\n\nChange-Id: Ia7ebd13be9ce7834679d439b7bda242805768ef8\nImplements: blueprint goal-deploy-api-in-wsgi\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/a8fcc2e1d12d64d1b7c1fbc4fa410c4d2b4b7d7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/a8fcc2e1d12d64d1b7c1fbc4fa410c4d2b4b7d7b"}]},"branch":"refs/heads/master"},"0551bcfbc062b0c3180af44371b27b03692165b5":{"kind":"REWORK","_number":13,"created":"2020-10-26 14:23:55.000000000","uploader":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"ref":"refs/changes/56/486156/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/56/486156/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/13"}}},"commit":{"parents":[{"commit":"2b655a857ca0d7666feb35a987bb666faa754968","subject":"Merge \"Updated from OpenStack Ansible Tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/2b655a857ca0d7666feb35a987bb666faa754968"}]}],"author":{"name":"Andy McCrae","email":"andy.mccrae@gmail.com","date":"2017-07-21 14:10:18.000000000","tz":60},"committer":{"name":"James Denton","email":"james.denton@rackspace.com","date":"2020-10-26 14:23:52.000000000","tz":-300},"subject":"Implement uWSGI for neutron-api","message":"Implement uWSGI for neutron-api\n\nAs part of the Pike goals we are moving api services to run as WSGI\napps. neutron-server service is set up as a wsgi app, and this patch\nmoves it over.\n\nSince this is just a drop in replacement for the existing eventlet\nservice, operators an deployers should notice no difference.\n\nChange-Id: Ia7ebd13be9ce7834679d439b7bda242805768ef8\nImplements: blueprint goal-deploy-api-in-wsgi\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/0551bcfbc062b0c3180af44371b27b03692165b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/0551bcfbc062b0c3180af44371b27b03692165b5"}]},"branch":"refs/heads/master"},"3275b5b7f1c9af369bf3580039bb6f6b8d964b5a":{"kind":"REWORK","_number":14,"created":"2020-10-27 14:01:17.000000000","uploader":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"ref":"refs/changes/56/486156/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/56/486156/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/14"}}},"commit":{"parents":[{"commit":"2b655a857ca0d7666feb35a987bb666faa754968","subject":"Merge \"Updated from OpenStack Ansible Tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/2b655a857ca0d7666feb35a987bb666faa754968"}]}],"author":{"name":"Andy McCrae","email":"andy.mccrae@gmail.com","date":"2017-07-21 14:10:18.000000000","tz":60},"committer":{"name":"James Denton","email":"james.denton@rackspace.com","date":"2020-10-27 14:01:17.000000000","tz":0},"subject":"Implement uWSGI for neutron-api","message":"Implement uWSGI for neutron-api\n\nAs part of the Pike goals we are moving api services to run as WSGI\napps. neutron-server service is set up as a wsgi app, and this patch\nmoves it over.\n\nSince this is just a drop in replacement for the existing eventlet\nservice, operators an deployers should notice no difference.\n\nChange-Id: Ia7ebd13be9ce7834679d439b7bda242805768ef8\nImplements: blueprint goal-deploy-api-in-wsgi\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/3275b5b7f1c9af369bf3580039bb6f6b8d964b5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/3275b5b7f1c9af369bf3580039bb6f6b8d964b5a"}]},"branch":"refs/heads/master"},"c755e4403a872faccb03ebbad89e95234599707a":{"kind":"REWORK","_number":15,"created":"2020-10-27 16:10:47.000000000","uploader":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"ref":"refs/changes/56/486156/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/56/486156/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/15"}}},"commit":{"parents":[{"commit":"2b655a857ca0d7666feb35a987bb666faa754968","subject":"Merge \"Updated from OpenStack Ansible Tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/2b655a857ca0d7666feb35a987bb666faa754968"}]}],"author":{"name":"Andy McCrae","email":"andy.mccrae@gmail.com","date":"2017-07-21 14:10:18.000000000","tz":60},"committer":{"name":"James Denton","email":"james.denton@rackspace.com","date":"2020-10-27 16:10:40.000000000","tz":0},"subject":"Implement uWSGI for neutron-api","message":"Implement uWSGI for neutron-api\n\nAs part of the Pike goals we are moving api services to run as WSGI\napps. neutron-server service is set up as a wsgi app, and this patch\nmoves it over.\n\nSince this is just a drop in replacement for the existing eventlet\nservice, operators an deployers should notice no difference.\n\nChange-Id: Ia7ebd13be9ce7834679d439b7bda242805768ef8\nImplements: blueprint goal-deploy-api-in-wsgi\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/c755e4403a872faccb03ebbad89e95234599707a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/c755e4403a872faccb03ebbad89e95234599707a"}]},"branch":"refs/heads/master"},"ce4e6df5a1badf5cb326ba34c71d3f6dbf466007":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2020-11-06 12:21:25.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/56/486156/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/56/486156/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/16"}}},"commit":{"parents":[{"commit":"e3f04ef863c1af90143655e7640df8b943ff12ff","subject":"Enable notifications when Designate is enabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/e3f04ef863c1af90143655e7640df8b943ff12ff"}]}],"author":{"name":"Andy McCrae","email":"andy.mccrae@gmail.com","date":"2017-07-21 14:10:18.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-11-06 12:21:25.000000000","tz":0},"subject":"Implement uWSGI for neutron-api","message":"Implement uWSGI for neutron-api\n\nAs part of the Pike goals we are moving api services to run as WSGI\napps. neutron-server service is set up as a wsgi app, and this patch\nmoves it over.\n\nSince this is just a drop in replacement for the existing eventlet\nservice, operators an deployers should notice no difference.\n\nChange-Id: Ia7ebd13be9ce7834679d439b7bda242805768ef8\nImplements: blueprint goal-deploy-api-in-wsgi\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/ce4e6df5a1badf5cb326ba34c71d3f6dbf466007"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/ce4e6df5a1badf5cb326ba34c71d3f6dbf466007"}]},"branch":"refs/heads/master"},"0c48af09b1c4ba1f2a0c8f6a4cef3f26b63a5f07":{"kind":"REWORK","_number":17,"created":"2020-11-07 08:17:57.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/56/486156/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/56/486156/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/17"}}},"commit":{"parents":[{"commit":"e3f04ef863c1af90143655e7640df8b943ff12ff","subject":"Enable notifications when Designate is enabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/e3f04ef863c1af90143655e7640df8b943ff12ff"}]}],"author":{"name":"Andy McCrae","email":"andy.mccrae@gmail.com","date":"2017-07-21 14:10:18.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-11-07 08:17:53.000000000","tz":0},"subject":"Implement uWSGI for neutron-api","message":"Implement uWSGI for neutron-api\n\nAs part of the Pike goals we are moving api services to run as WSGI\napps. neutron-server service is set up as a wsgi app, and this patch\nmoves it over.\n\nSince this is just a drop in replacement for the existing eventlet\nservice, operators an deployers should notice no difference.\n\nChange-Id: Ia7ebd13be9ce7834679d439b7bda242805768ef8\nImplements: blueprint goal-deploy-api-in-wsgi\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/0c48af09b1c4ba1f2a0c8f6a4cef3f26b63a5f07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/0c48af09b1c4ba1f2a0c8f6a4cef3f26b63a5f07"}]},"branch":"refs/heads/master"},"0a3b566edb88dd48064703bb938c68ae769bf723":{"kind":"REWORK","_number":18,"created":"2020-11-07 14:31:55.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/56/486156/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/56/486156/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/18"}}},"commit":{"parents":[{"commit":"e3f04ef863c1af90143655e7640df8b943ff12ff","subject":"Enable notifications when Designate is enabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/e3f04ef863c1af90143655e7640df8b943ff12ff"}]}],"author":{"name":"Andy McCrae","email":"andy.mccrae@gmail.com","date":"2017-07-21 14:10:18.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-11-07 14:31:52.000000000","tz":0},"subject":"Implement uWSGI for neutron-api","message":"Implement uWSGI for neutron-api\n\nAs part of the Pike goals we are moving api services to run as WSGI\napps. neutron-server service is set up as a wsgi app, and this patch\nmoves it over.\n\nSince this is just a drop in replacement for the existing eventlet\nservice, operators an deployers should notice no difference.\n\nChange-Id: Ia7ebd13be9ce7834679d439b7bda242805768ef8\nImplements: blueprint goal-deploy-api-in-wsgi\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/0a3b566edb88dd48064703bb938c68ae769bf723"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/0a3b566edb88dd48064703bb938c68ae769bf723"}]},"branch":"refs/heads/master"},"9fd0deb8832131054e768a638f318a826754094e":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2020-12-02 12:55:12.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/56/486156/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/56/486156/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/19"}}},"commit":{"parents":[{"commit":"b5e4535dbe86e6affbe61ca3371a9093b44c0672","subject":"Merge \"Reduce number of processes on small systems\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/b5e4535dbe86e6affbe61ca3371a9093b44c0672"}]}],"author":{"name":"Andy McCrae","email":"andy.mccrae@gmail.com","date":"2017-07-21 14:10:18.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-12-02 12:55:12.000000000","tz":0},"subject":"Implement uWSGI for neutron-api","message":"Implement uWSGI for neutron-api\n\nAs part of the Pike goals we are moving api services to run as WSGI\napps. neutron-server service is set up as a wsgi app, and this patch\nmoves it over.\n\nSince this is just a drop in replacement for the existing eventlet\nservice, operators an deployers should notice no difference.\n\nChange-Id: Ia7ebd13be9ce7834679d439b7bda242805768ef8\nImplements: blueprint goal-deploy-api-in-wsgi\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/9fd0deb8832131054e768a638f318a826754094e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/9fd0deb8832131054e768a638f318a826754094e"}]},"branch":"refs/heads/master","description":"Rebase"},"8c34340c4b6f6334e0815ef3cc04980fb37b5238":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2021-02-03 18:20:01.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/56/486156/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/56/486156/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/20"}}},"commit":{"parents":[{"commit":"4b080bb3f940128c6b6572e634ed4070e88d81e0","subject":"L3 agent cleanup_on_shutdown","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/4b080bb3f940128c6b6572e634ed4070e88d81e0"}]}],"author":{"name":"Andy McCrae","email":"andy.mccrae@gmail.com","date":"2017-07-21 14:10:18.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-02-03 18:20:01.000000000","tz":0},"subject":"Implement uWSGI for neutron-api","message":"Implement uWSGI for neutron-api\n\nAs part of the Pike goals we are moving api services to run as WSGI\napps. neutron-server service is set up as a wsgi app, and this patch\nmoves it over.\n\nSince this is just a drop in replacement for the existing eventlet\nservice, operators an deployers should notice no difference.\n\nChange-Id: Ia7ebd13be9ce7834679d439b7bda242805768ef8\nImplements: blueprint goal-deploy-api-in-wsgi\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/8c34340c4b6f6334e0815ef3cc04980fb37b5238"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/8c34340c4b6f6334e0815ef3cc04980fb37b5238"}]},"branch":"refs/heads/master","description":"Rebase"},"49c456ef407880376f2399259c6fb4c59a44c6e0":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2021-02-07 19:33:40.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/56/486156/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/56/486156/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/21"}}},"commit":{"parents":[{"commit":"f441b212f5786f07472c103b36aa6224b19bbf25","subject":"Fix neutron_keepalived_no_track default logic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/f441b212f5786f07472c103b36aa6224b19bbf25"}]}],"author":{"name":"Andy McCrae","email":"andy.mccrae@gmail.com","date":"2017-07-21 14:10:18.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-02-07 19:33:40.000000000","tz":0},"subject":"Implement uWSGI for neutron-api","message":"Implement uWSGI for neutron-api\n\nAs part of the Pike goals we are moving api services to run as WSGI\napps. neutron-server service is set up as a wsgi app, and this patch\nmoves it over.\n\nSince this is just a drop in replacement for the existing eventlet\nservice, operators an deployers should notice no difference.\n\nChange-Id: Ia7ebd13be9ce7834679d439b7bda242805768ef8\nImplements: blueprint goal-deploy-api-in-wsgi\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/49c456ef407880376f2399259c6fb4c59a44c6e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/49c456ef407880376f2399259c6fb4c59a44c6e0"}]},"branch":"refs/heads/master","description":"Rebase"},"f509594a0df66e57aee12fd614e6ada8262eee69":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2021-03-11 14:35:47.000000000","uploader":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"ref":"refs/changes/56/486156/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/56/486156/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/22"}}},"commit":{"parents":[{"commit":"fc34321dc63c1514c0b59393031b9df63e8455c7","subject":"Merge \"Fix conditional for ovs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/fc34321dc63c1514c0b59393031b9df63e8455c7"}]}],"author":{"name":"Andy McCrae","email":"andy.mccrae@gmail.com","date":"2017-07-21 14:10:18.000000000","tz":60},"committer":{"name":"James Denton","email":"james.denton@outlook.com","date":"2021-03-11 14:35:47.000000000","tz":0},"subject":"Implement uWSGI for neutron-api","message":"Implement uWSGI for neutron-api\n\nAs part of the Pike goals we are moving api services to run as WSGI\napps. neutron-server service is set up as a wsgi app, and this patch\nmoves it over.\n\nSince this is just a drop in replacement for the existing eventlet\nservice, operators an deployers should notice no difference.\n\nChange-Id: Ia7ebd13be9ce7834679d439b7bda242805768ef8\nImplements: blueprint goal-deploy-api-in-wsgi\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/f509594a0df66e57aee12fd614e6ada8262eee69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/f509594a0df66e57aee12fd614e6ada8262eee69"}]},"branch":"refs/heads/master","description":"Rebase"},"a47f72db0ad83ebf1586c0c5fe7688eedef0a099":{"kind":"REWORK","_number":23,"created":"2021-03-13 22:20:12.000000000","uploader":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"ref":"refs/changes/56/486156/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/56/486156/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/23"}}},"commit":{"parents":[{"commit":"cd9d64d434ed7faa517f18198c447f71c9a727e3","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/cd9d64d434ed7faa517f18198c447f71c9a727e3"}]}],"author":{"name":"Andy McCrae","email":"andy.mccrae@gmail.com","date":"2017-07-21 14:10:18.000000000","tz":60},"committer":{"name":"Erik Berg","email":"openstack@slipsprogrammor.no","date":"2021-03-13 22:20:09.000000000","tz":60},"subject":"Implement uWSGI for neutron-api","message":"Implement uWSGI for neutron-api\n\nAs part of the Pike goals we are moving api services to run as WSGI\napps. neutron-server service is set up as a wsgi app, and this patch\nmoves it over.\n\nSince this is just a drop in replacement for the existing eventlet\nservice, operators an deployers should notice no difference.\n\nChange-Id: Ia7ebd13be9ce7834679d439b7bda242805768ef8\nImplements: blueprint goal-deploy-api-in-wsgi\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/a47f72db0ad83ebf1586c0c5fe7688eedef0a099"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/a47f72db0ad83ebf1586c0c5fe7688eedef0a099"}]},"branch":"refs/heads/master"},"29b1fecbaf6a751b6b4b77d990d29c303655ddb7":{"kind":"REWORK","_number":24,"created":"2021-03-13 23:36:17.000000000","uploader":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"ref":"refs/changes/56/486156/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/56/486156/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/24"}}},"commit":{"parents":[{"commit":"5ce6fa9daf14a108e7406ac385e123c3285e9cdf","subject":"Merge \"[reno] Stop publishing release notes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/5ce6fa9daf14a108e7406ac385e123c3285e9cdf"}]}],"author":{"name":"Andy McCrae","email":"andy.mccrae@gmail.com","date":"2017-07-21 14:10:18.000000000","tz":60},"committer":{"name":"Erik Berg","email":"openstack@slipsprogrammor.no","date":"2021-03-13 23:36:15.000000000","tz":60},"subject":"Implement uWSGI for neutron-api","message":"Implement uWSGI for neutron-api\n\nAs part of the Pike goals we are moving api services to run as WSGI\napps. neutron-server service is set up as a wsgi app, and this patch\nmoves it over.\n\nSince this is just a drop in replacement for the existing eventlet\nservice, operators an deployers should notice no difference.\n\nChange-Id: Ia7ebd13be9ce7834679d439b7bda242805768ef8\nImplements: blueprint goal-deploy-api-in-wsgi\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/29b1fecbaf6a751b6b4b77d990d29c303655ddb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/29b1fecbaf6a751b6b4b77d990d29c303655ddb7"}]},"branch":"refs/heads/master"},"760e9b0e52302e2d7c9bf2eb538b7599cf09f9e9":{"kind":"TRIVIAL_REBASE","_number":25,"created":"2021-05-14 13:39:08.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/56/486156/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/56/486156/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/25"}}},"commit":{"parents":[{"commit":"4e123a134db6f5675d25abcfcb732263270c9f68","subject":"Merge \"Updated from OpenStack Ansible Tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/4e123a134db6f5675d25abcfcb732263270c9f68"}]}],"author":{"name":"Andy McCrae","email":"andy.mccrae@gmail.com","date":"2017-07-21 14:10:18.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-05-14 13:39:08.000000000","tz":0},"subject":"Implement uWSGI for neutron-api","message":"Implement uWSGI for neutron-api\n\nAs part of the Pike goals we are moving api services to run as WSGI\napps. neutron-server service is set up as a wsgi app, and this patch\nmoves it over.\n\nSince this is just a drop in replacement for the existing eventlet\nservice, operators an deployers should notice no difference.\n\nChange-Id: Ia7ebd13be9ce7834679d439b7bda242805768ef8\nImplements: blueprint goal-deploy-api-in-wsgi\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/760e9b0e52302e2d7c9bf2eb538b7599cf09f9e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/760e9b0e52302e2d7c9bf2eb538b7599cf09f9e9"}]},"branch":"refs/heads/master","description":"Rebase"},"bde0a16a434b5b7ec0ec0b81735e52728f00831d":{"kind":"TRIVIAL_REBASE","_number":26,"created":"2021-06-09 07:04:32.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/56/486156/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/56/486156/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/26"}}},"commit":{"parents":[{"commit":"c42711e1cf407e7830f6c4263482dc06b432683a","subject":"Merge \"Use neutron_conf_dir for absent policy removal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/c42711e1cf407e7830f6c4263482dc06b432683a"}]}],"author":{"name":"Andy McCrae","email":"andy.mccrae@gmail.com","date":"2017-07-21 14:10:18.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-06-09 07:04:32.000000000","tz":0},"subject":"Implement uWSGI for neutron-api","message":"Implement uWSGI for neutron-api\n\nAs part of the Pike goals we are moving api services to run as WSGI\napps. neutron-server service is set up as a wsgi app, and this patch\nmoves it over.\n\nSince this is just a drop in replacement for the existing eventlet\nservice, operators an deployers should notice no difference.\n\nChange-Id: Ia7ebd13be9ce7834679d439b7bda242805768ef8\nImplements: blueprint goal-deploy-api-in-wsgi\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/bde0a16a434b5b7ec0ec0b81735e52728f00831d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/bde0a16a434b5b7ec0ec0b81735e52728f00831d"}]},"branch":"refs/heads/master","description":"Rebase"},"04bc729f8fc01b2f5361f4af4d16b3ebe9565460":{"kind":"REWORK","_number":27,"created":"2021-06-23 10:01:50.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/56/486156/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/56/486156/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/27"}}},"commit":{"parents":[{"commit":"8f7a948dcba2194b6dded2c70b7f98ac78426208","subject":"Merge \"Add ovn clustering support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/8f7a948dcba2194b6dded2c70b7f98ac78426208"}]}],"author":{"name":"Andy McCrae","email":"andy.mccrae@gmail.com","date":"2017-07-21 14:10:18.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-06-23 10:01:35.000000000","tz":180},"subject":"Implement uWSGI for neutron-api","message":"Implement uWSGI for neutron-api\n\nAs part of the Pike goals we are moving api services to run as WSGI\napps. neutron-server service is set up as a wsgi app, and this patch\nmoves it over.\n\nSince this is just a drop in replacement for the existing eventlet\nservice, operators an deployers should notice no difference.\n\nChange-Id: Ia7ebd13be9ce7834679d439b7bda242805768ef8\nImplements: blueprint goal-deploy-api-in-wsgi\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/04bc729f8fc01b2f5361f4af4d16b3ebe9565460"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/04bc729f8fc01b2f5361f4af4d16b3ebe9565460"}]},"branch":"refs/heads/master"},"5464c6b6057f2377d5b43c75875f084451216cec":{"kind":"TRIVIAL_REBASE","_number":28,"created":"2021-08-25 10:58:10.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/56/486156/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/56/486156/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/28"}}},"commit":{"parents":[{"commit":"454866176f6444c60471ab2c34dc53203a7dbb95","subject":"Use list of cluster member for ovn ml2 agent to directly connect","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/454866176f6444c60471ab2c34dc53203a7dbb95"}]}],"author":{"name":"Andy McCrae","email":"andy.mccrae@gmail.com","date":"2017-07-21 14:10:18.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-08-25 10:58:10.000000000","tz":0},"subject":"Implement uWSGI for neutron-api","message":"Implement uWSGI for neutron-api\n\nAs part of the Pike goals we are moving api services to run as WSGI\napps. neutron-server service is set up as a wsgi app, and this patch\nmoves it over.\n\nSince this is just a drop in replacement for the existing eventlet\nservice, operators an deployers should notice no difference.\n\nChange-Id: Ia7ebd13be9ce7834679d439b7bda242805768ef8\nImplements: blueprint goal-deploy-api-in-wsgi\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/5464c6b6057f2377d5b43c75875f084451216cec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/5464c6b6057f2377d5b43c75875f084451216cec"}]},"branch":"refs/heads/master","description":"Rebase"},"2eb8808f91f5ed0f0880e6668931580b36c5887d":{"kind":"REWORK","_number":29,"created":"2021-08-26 14:12:42.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/56/486156/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/56/486156/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/29"}}},"commit":{"parents":[{"commit":"454866176f6444c60471ab2c34dc53203a7dbb95","subject":"Use list of cluster member for ovn ml2 agent to directly connect","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/454866176f6444c60471ab2c34dc53203a7dbb95"}]}],"author":{"name":"Andy McCrae","email":"andy.mccrae@gmail.com","date":"2017-07-21 14:10:18.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-08-26 14:12:31.000000000","tz":180},"subject":"Implement uWSGI for neutron-api","message":"Implement uWSGI for neutron-api\n\nAs part of the Pike goals we are moving api services to run as WSGI\napps. neutron-server service is set up as a wsgi app, and this patch\nmoves it over.\n\nSince this is just a drop in replacement for the existing eventlet\nservice, operators an deployers should notice no difference.\n\nChange-Id: Ia7ebd13be9ce7834679d439b7bda242805768ef8\nImplements: blueprint goal-deploy-api-in-wsgi\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/2eb8808f91f5ed0f0880e6668931580b36c5887d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/2eb8808f91f5ed0f0880e6668931580b36c5887d"}]},"branch":"refs/heads/master"},"b1d3d75763043bd8fce3345e212ad9665f66a298":{"kind":"TRIVIAL_REBASE","_number":30,"created":"2021-08-26 16:58:39.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/56/486156/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/56/486156/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/30"}}},"commit":{"parents":[{"commit":"016b2ab2aa858253e8a62157655ea0e7b1d5adc7","subject":"Merge \"Exclude neutron from venv constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/016b2ab2aa858253e8a62157655ea0e7b1d5adc7"}]}],"author":{"name":"Andy McCrae","email":"andy.mccrae@gmail.com","date":"2017-07-21 14:10:18.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-08-26 16:58:39.000000000","tz":0},"subject":"Implement uWSGI for neutron-api","message":"Implement uWSGI for neutron-api\n\nAs part of the Pike goals we are moving api services to run as WSGI\napps. neutron-server service is set up as a wsgi app, and this patch\nmoves it over.\n\nSince this is just a drop in replacement for the existing eventlet\nservice, operators an deployers should notice no difference.\n\nChange-Id: Ia7ebd13be9ce7834679d439b7bda242805768ef8\nImplements: blueprint goal-deploy-api-in-wsgi\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/b1d3d75763043bd8fce3345e212ad9665f66a298"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/b1d3d75763043bd8fce3345e212ad9665f66a298"}]},"branch":"refs/heads/master","description":"Rebase"},"3408f31706271ba4b14aefc0e2c177cd24083f65":{"kind":"TRIVIAL_REBASE","_number":31,"created":"2021-08-27 08:38:55.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/56/486156/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/56/486156/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/31"}}},"commit":{"parents":[{"commit":"af7e91c20b5ba36d06a1103c7f86dca0b38d81f9","subject":"Merge \"Set OVN jobs to voting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/af7e91c20b5ba36d06a1103c7f86dca0b38d81f9"}]}],"author":{"name":"Andy McCrae","email":"andy.mccrae@gmail.com","date":"2017-07-21 14:10:18.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-08-27 08:38:55.000000000","tz":0},"subject":"Implement uWSGI for neutron-api","message":"Implement uWSGI for neutron-api\n\nAs part of the Pike goals we are moving api services to run as WSGI\napps. neutron-server service is set up as a wsgi app, and this patch\nmoves it over.\n\nSince this is just a drop in replacement for the existing eventlet\nservice, operators an deployers should notice no difference.\n\nChange-Id: Ia7ebd13be9ce7834679d439b7bda242805768ef8\nImplements: blueprint goal-deploy-api-in-wsgi\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/3408f31706271ba4b14aefc0e2c177cd24083f65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/3408f31706271ba4b14aefc0e2c177cd24083f65"}]},"branch":"refs/heads/master","description":"Rebase"},"d61fca32c567cf3ed6e85a4805e1dbaae36409e9":{"kind":"TRIVIAL_REBASE","_number":32,"created":"2021-12-06 17:29:41.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/56/486156/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/56/486156/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/32"}}},"commit":{"parents":[{"commit":"8752d02a9f7cc1230d6b4d578bb6c36fb7ca6ae5","subject":"Merge \"Drop designate notifications topic\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/8752d02a9f7cc1230d6b4d578bb6c36fb7ca6ae5"}]}],"author":{"name":"Andy McCrae","email":"andy.mccrae@gmail.com","date":"2017-07-21 14:10:18.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-12-06 17:29:41.000000000","tz":0},"subject":"Implement uWSGI for neutron-api","message":"Implement uWSGI for neutron-api\n\nAs part of the Pike goals we are moving api services to run as WSGI\napps. neutron-server service is set up as a wsgi app, and this patch\nmoves it over.\n\nSince this is just a drop in replacement for the existing eventlet\nservice, operators an deployers should notice no difference.\n\nChange-Id: Ia7ebd13be9ce7834679d439b7bda242805768ef8\nImplements: blueprint goal-deploy-api-in-wsgi\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/d61fca32c567cf3ed6e85a4805e1dbaae36409e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/d61fca32c567cf3ed6e85a4805e1dbaae36409e9"}]},"branch":"refs/heads/master","description":"Rebase"},"a41f039b8a89f05da97fd97bd259fbc3771062cb":{"kind":"NO_CODE_CHANGE","_number":33,"created":"2021-12-06 19:12:03.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/56/486156/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/56/486156/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/33"}}},"commit":{"parents":[{"commit":"8752d02a9f7cc1230d6b4d578bb6c36fb7ca6ae5","subject":"Merge \"Drop designate notifications topic\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/8752d02a9f7cc1230d6b4d578bb6c36fb7ca6ae5"}]}],"author":{"name":"Andy McCrae","email":"andy.mccrae@gmail.com","date":"2017-07-21 14:10:18.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-12-06 19:12:03.000000000","tz":0},"subject":"Implement uWSGI for neutron-api","message":"Implement uWSGI for neutron-api\n\nAs part of the Pike goals we are moving api services to run as WSGI\napps. neutron-server service is set up as a wsgi app, and this patch\nmoves it over.\n\nSince this is just a drop in replacement for the existing eventlet\nservice, operators an deployers should notice no difference.\n\nChange-Id: Ia7ebd13be9ce7834679d439b7bda242805768ef8\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/820586\nImplements: blueprint goal-deploy-api-in-wsgi\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/a41f039b8a89f05da97fd97bd259fbc3771062cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/a41f039b8a89f05da97fd97bd259fbc3771062cb"}]},"branch":"refs/heads/master","description":"Edit commit message"},"1dd42814c0d35206d3af4b868eb35a2ff4fea265":{"kind":"REWORK","_number":34,"created":"2021-12-06 21:46:27.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/56/486156/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/56/486156/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/34"}}},"commit":{"parents":[{"commit":"8752d02a9f7cc1230d6b4d578bb6c36fb7ca6ae5","subject":"Merge \"Drop designate notifications topic\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/8752d02a9f7cc1230d6b4d578bb6c36fb7ca6ae5"}]}],"author":{"name":"Andy McCrae","email":"andy.mccrae@gmail.com","date":"2017-07-21 14:10:18.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-12-06 21:46:27.000000000","tz":0},"subject":"Implement uWSGI for neutron-api","message":"Implement uWSGI for neutron-api\n\nAs part of the Pike goals we are moving api services to run as WSGI\napps. neutron-server service is set up as a wsgi app, and this patch\nmoves it over.\n\nSince this is just a drop in replacement for the existing eventlet\nservice, operators an deployers should notice no difference.\n\nChange-Id: Ia7ebd13be9ce7834679d439b7bda242805768ef8\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/820586\nImplements: blueprint goal-deploy-api-in-wsgi\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/1dd42814c0d35206d3af4b868eb35a2ff4fea265"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/1dd42814c0d35206d3af4b868eb35a2ff4fea265"}]},"branch":"refs/heads/master"},"8548fcda8c1fc197df53334c8e11130fb0052bbd":{"kind":"REWORK","_number":35,"created":"2021-12-07 06:02:58.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/56/486156/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/56/486156/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/35"}}},"commit":{"parents":[{"commit":"8752d02a9f7cc1230d6b4d578bb6c36fb7ca6ae5","subject":"Merge \"Drop designate notifications topic\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/8752d02a9f7cc1230d6b4d578bb6c36fb7ca6ae5"}]}],"author":{"name":"Andy McCrae","email":"andy.mccrae@gmail.com","date":"2017-07-21 14:10:18.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-12-07 06:02:58.000000000","tz":0},"subject":"Implement uWSGI for neutron-api","message":"Implement uWSGI for neutron-api\n\nAs part of the Pike goals we are moving api services to run as WSGI\napps. neutron-server service is set up as a wsgi app, and this patch\nmoves it over.\n\nSince this is just a drop in replacement for the existing eventlet\nservice, operators an deployers should notice no difference.\n\nChange-Id: Ia7ebd13be9ce7834679d439b7bda242805768ef8\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/820586\nImplements: blueprint goal-deploy-api-in-wsgi\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/8548fcda8c1fc197df53334c8e11130fb0052bbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/8548fcda8c1fc197df53334c8e11130fb0052bbd"}]},"branch":"refs/heads/master"},"7b0178670c366c58ca726548bee3395cb662eb30":{"kind":"REWORK","_number":36,"created":"2021-12-10 12:11:29.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/56/486156/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/56/486156/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/36"}}},"commit":{"parents":[{"commit":"8752d02a9f7cc1230d6b4d578bb6c36fb7ca6ae5","subject":"Merge \"Drop designate notifications topic\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/8752d02a9f7cc1230d6b4d578bb6c36fb7ca6ae5"}]}],"author":{"name":"Andy McCrae","email":"andy.mccrae@gmail.com","date":"2017-07-21 14:10:18.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-12-10 12:11:28.000000000","tz":0},"subject":"Implement uWSGI for neutron-api","message":"Implement uWSGI for neutron-api\n\nAs part of the Pike goals we are moving api services to run as WSGI\napps. neutron-server service is set up as a wsgi app, and this patch\nmoves it over.\n\nSince this is just a drop in replacement for the existing eventlet\nservice, operators an deployers should notice no difference.\n\nChange-Id: Ia7ebd13be9ce7834679d439b7bda242805768ef8\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/820586\nImplements: blueprint goal-deploy-api-in-wsgi\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/7b0178670c366c58ca726548bee3395cb662eb30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/7b0178670c366c58ca726548bee3395cb662eb30"}]},"branch":"refs/heads/master"},"16de36da7463953cf843a3c334923227c4392aa5":{"kind":"REWORK","_number":37,"created":"2021-12-10 12:11:49.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/56/486156/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/56/486156/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/37"}}},"commit":{"parents":[{"commit":"8752d02a9f7cc1230d6b4d578bb6c36fb7ca6ae5","subject":"Merge \"Drop designate notifications topic\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/8752d02a9f7cc1230d6b4d578bb6c36fb7ca6ae5"}]}],"author":{"name":"Andy McCrae","email":"andy.mccrae@gmail.com","date":"2017-07-21 14:10:18.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-12-10 12:11:49.000000000","tz":0},"subject":"Implement uWSGI for neutron-api","message":"Implement uWSGI for neutron-api\n\nAs part of the Pike goals we are moving api services to run as WSGI\napps. neutron-server service is set up as a wsgi app, and this patch\nmoves it over.\n\nSince this is just a drop in replacement for the existing eventlet\nservice, operators an deployers should notice no difference.\n\nChange-Id: Ia7ebd13be9ce7834679d439b7bda242805768ef8\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/820586\nImplements: blueprint goal-deploy-api-in-wsgi\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/16de36da7463953cf843a3c334923227c4392aa5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/16de36da7463953cf843a3c334923227c4392aa5"}]},"branch":"refs/heads/master"},"ec468b089f2ca04b7b0564d0a3ea47d3458c47a4":{"kind":"REWORK","_number":38,"created":"2022-03-21 11:59:01.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/56/486156/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/56/486156/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/38"}}},"commit":{"parents":[{"commit":"01951cd77b7edfc1008da14c1afd4e260fcf432f","subject":"Add configuration option for heartbeat_in_pthread","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/01951cd77b7edfc1008da14c1afd4e260fcf432f"}]}],"author":{"name":"Andy McCrae","email":"andy.mccrae@gmail.com","date":"2017-07-21 14:10:18.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-03-21 11:57:51.000000000","tz":60},"subject":"Implement uWSGI for neutron-api","message":"Implement uWSGI for neutron-api\n\nAs part of the Pike goals we are moving api services to run as WSGI\napps. neutron-server service is set up as a wsgi app, and this patch\nmoves it over.\n\nSince this is just a drop in replacement for the existing eventlet\nservice, operators an deployers should notice no difference.\n\nChange-Id: Ia7ebd13be9ce7834679d439b7bda242805768ef8\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/820586\nImplements: blueprint goal-deploy-api-in-wsgi\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/ec468b089f2ca04b7b0564d0a3ea47d3458c47a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/ec468b089f2ca04b7b0564d0a3ea47d3458c47a4"}]},"branch":"refs/heads/master"},"bdd06c84488c7008c0dcb3eaca4effc8d38fcaeb":{"kind":"TRIVIAL_REBASE","_number":39,"created":"2022-05-27 18:34:38.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/56/486156/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/56/486156/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/486156/39"}}},"commit":{"parents":[{"commit":"c5ee3450546d8577110f9bb2ae13721bdbacfdb4","subject":"Merge \"Support dns_domain_keywords extension\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/c5ee3450546d8577110f9bb2ae13721bdbacfdb4"}]}],"author":{"name":"Andy McCrae","email":"andy.mccrae@gmail.com","date":"2017-07-21 14:10:18.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-05-27 18:34:38.000000000","tz":0},"subject":"Implement uWSGI for neutron-api","message":"Implement uWSGI for neutron-api\n\nAs part of the Pike goals we are moving api services to run as WSGI\napps. neutron-server service is set up as a wsgi app, and this patch\nmoves it over.\n\nSince this is just a drop in replacement for the existing eventlet\nservice, operators an deployers should notice no difference.\n\nChange-Id: Ia7ebd13be9ce7834679d439b7bda242805768ef8\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/820586\nImplements: blueprint goal-deploy-api-in-wsgi\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/bdd06c84488c7008c0dcb3eaca4effc8d38fcaeb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/bdd06c84488c7008c0dcb3eaca4effc8d38fcaeb"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
