)]}'
{"id":"openstack%2Fopenstack-ansible-os_neutron~528692","triplet_id":"openstack%2Fopenstack-ansible-os_neutron~master~Ic3e8eee3c97a5df3e988c302a8111952385859bf","project":"openstack/openstack-ansible-os_neutron","branch":"master","topic":"dnm-testing-opensuse-job","hashtags":[],"change_id":"Ic3e8eee3c97a5df3e988c302a8111952385859bf","subject":"Testing neutron jobs","status":"ABANDONED","created":"2017-12-18 11:13:12.000000000","updated":"2020-10-13 17:45:18.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"459d3daaf7fa2e12ed127c8304223b7439ef9e8b","_number":528692,"virtual_id_number":528692,"owner":{"_account_id":23163,"name":"Markos Chandras (hwoarang)","email":"markos@chandras.me","username":"mchandras"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":23163,"name":"Markos Chandras (hwoarang)","email":"markos@chandras.me","username":"mchandras"},{"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},{"value":1,"date":"2019-05-20 22:20:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23163,"name":"Markos Chandras (hwoarang)","email":"markos@chandras.me","username":"mchandras"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":23163,"name":"Markos Chandras (hwoarang)","email":"markos@chandras.me","username":"mchandras"},{"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":23163,"name":"Markos Chandras (hwoarang)","email":"markos@chandras.me","username":"mchandras"},{"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23163,"name":"Markos Chandras (hwoarang)","email":"markos@chandras.me","username":"mchandras"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-12-19 08:13:21.000000000","updated_by":{"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},"reviewer":{"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},"state":"REVIEWER"},{"updated":"2019-05-20 22:20:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"93146ddfa0be7de6115f7885ce284147229a34b8","author":{"_account_id":23163,"name":"Markos Chandras (hwoarang)","email":"markos@chandras.me","username":"mchandras"},"date":"2017-12-18 11:13:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2a9846cbe1053f3669663557e2083812c276b01f","author":{"_account_id":23163,"name":"Markos Chandras (hwoarang)","email":"markos@chandras.me","username":"mchandras"},"date":"2017-12-18 11:29:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"db6c36a85f37667f4d40335cb40eea4442ff95ee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-18 11:34:41.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"dceddd0291bdddb25b18c1f81523c60b35b42d88","author":{"_account_id":23163,"name":"Markos Chandras (hwoarang)","email":"markos@chandras.me","username":"mchandras"},"date":"2017-12-18 11:44:32.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c8a458e888fbf44d93a927c19bdf3ad2a6eab9fc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-18 12:00:30.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"46703533aec3e75141b67b9d537812e2e84be150","author":{"_account_id":23163,"name":"Markos Chandras (hwoarang)","email":"markos@chandras.me","username":"mchandras"},"date":"2017-12-18 12:00:40.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"be3857c5d022389309527b490f5779c4fe29a952","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-18 12:13:00.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"edff8e1da86869d561b57c6d42df3b40aadb0e63","author":{"_account_id":23163,"name":"Markos Chandras (hwoarang)","email":"markos@chandras.me","username":"mchandras"},"date":"2017-12-18 12:13:01.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"4be2b866e0bb7a1d549c224813e91054ef4f56dd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-18 12:43:25.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/92/528692/2/check/build-openstack-sphinx-docs/709f216/html/ : SUCCESS in 2m 18s\n- openstack-ansible-linters http://logs.openstack.org/92/528692/2/check/openstack-ansible-linters/76284c3/ : FAILURE in 4m 51s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/92/528692/2/check/openstack-ansible-functional-centos-7/14e5e50/ : FAILURE in 5m 42s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/92/528692/2/check/openstack-ansible-functional-opensuse-423/9f037b7/ : FAILURE in 7m 43s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/92/528692/2/check/openstack-ansible-functional-ubuntu-xenial/9cbf68e/ : FAILURE in 5m 29s\n- openstack-ansible-upgrade-ubuntu-xenial http://logs.openstack.org/92/528692/2/check/openstack-ansible-upgrade-ubuntu-xenial/2572efa/ : FAILURE in 25m 05s\n- openstack-ansible-ovs-ubuntu-xenial-nv http://logs.openstack.org/92/528692/2/check/openstack-ansible-ovs-ubuntu-xenial-nv/1a683a5/ : FAILURE in 4m 38s (non-voting)\n- openstack-ansible-calico-ubuntu-xenial-nv http://logs.openstack.org/92/528692/2/check/openstack-ansible-calico-ubuntu-xenial-nv/d78c6a7/ : FAILURE in 7m 11s (non-voting)\n- openstack-ansible-dragonflow-ubuntu-xenial-nv http://logs.openstack.org/92/528692/2/check/openstack-ansible-dragonflow-ubuntu-xenial-nv/bb8f79b/ : FAILURE in 3m 53s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-xenial-nv http://logs.openstack.org/92/528692/2/check/openstack-ansible-opendaylight-ubuntu-xenial-nv/2b583b1/ : FAILURE in 7m 16s (non-voting)\n- openstack-ansible-ovs-nsh-ubuntu-xenial http://logs.openstack.org/92/528692/2/check/openstack-ansible-ovs-nsh-ubuntu-xenial/2e51d4d/ : FAILURE in 4m 46s (non-voting)\n- openstack-ansible-neutron-ssl-nv http://logs.openstack.org/92/528692/2/check/openstack-ansible-neutron-ssl-nv/482be23/ : FAILURE in 5m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7d48614011172f4d2bd51edeea60796d063222bd","author":{"_account_id":23163,"name":"Markos Chandras (hwoarang)","email":"markos@chandras.me","username":"mchandras"},"date":"2017-12-18 14:17:23.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"12981cda584dae470c90256e2f69aa838b22f238","author":{"_account_id":23163,"name":"Markos Chandras (hwoarang)","email":"markos@chandras.me","username":"mchandras"},"date":"2017-12-18 14:28:20.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"be98ab53f485c765d07ec404dd48c36d51b42a2c","author":{"_account_id":23163,"name":"Markos Chandras (hwoarang)","email":"markos@chandras.me","username":"mchandras"},"date":"2017-12-18 14:30:08.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"dccb5a9d2b6d59f8942570949b64c7fce6eb68f9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-18 14:30:12.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"5ec7c5115ea680a2d96f4104c5ef2d3d58d2b32b","author":{"_account_id":23163,"name":"Markos Chandras (hwoarang)","email":"markos@chandras.me","username":"mchandras"},"date":"2017-12-18 14:35:19.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"22a67f21d7a2da989f334fa790b738458aae74be","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-18 14:46:37.000000000","message":"Patch Set 4: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":4},{"id":"5b62d77de1fc8669de3e20124f223bccc18d06ef","author":{"_account_id":23163,"name":"Markos Chandras (hwoarang)","email":"markos@chandras.me","username":"mchandras"},"date":"2017-12-18 14:48:09.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"c397a9bef9c12b2bfb4b1d51ab30267ad80a4fcb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-18 15:01:28.000000000","message":"Patch Set 4:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":4},{"id":"e53972e37b16f28ef4ac3fefb8302d1eaee36301","author":{"_account_id":23163,"name":"Markos Chandras (hwoarang)","email":"markos@chandras.me","username":"mchandras"},"date":"2017-12-18 15:01:37.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"1c5fbcbfc670f0af6a8c4bf3f1cc37a26962eede","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2017-12-18 15:23:24.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"0c64b3ade91658c7a5023268b85caccf7ab1745a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-18 15:39:48.000000000","message":"Patch Set 5: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":5},{"id":"6ae3c257cf8f740621e875be5ede8c085d17c561","author":{"_account_id":23163,"name":"Markos Chandras (hwoarang)","email":"markos@chandras.me","username":"mchandras"},"date":"2017-12-18 16:49:53.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"4119dba33688f1df1ff0db8cfc4320cfbfd002bf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-18 17:55:30.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/92/528692/5/check/build-openstack-sphinx-docs/2a20593/html/ : SUCCESS in 2m 06s\n- openstack-ansible-linters http://logs.openstack.org/92/528692/5/check/openstack-ansible-linters/faa5c7d/ : SUCCESS in 7m 00s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/92/528692/5/check/openstack-ansible-functional-centos-7/590d64d/ : SUCCESS in 1h 03m 36s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/92/528692/5/check/openstack-ansible-functional-opensuse-423/0cf95e0/ : SUCCESS in 30m 44s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/92/528692/5/check/openstack-ansible-functional-ubuntu-xenial/c5512e3/ : SUCCESS in 43m 25s\n- openstack-ansible-upgrade-ubuntu-xenial http://logs.openstack.org/92/528692/5/check/openstack-ansible-upgrade-ubuntu-xenial/12ed730/ : SUCCESS in 43m 14s\n- openstack-ansible-ovs-ubuntu-xenial-nv http://logs.openstack.org/92/528692/5/check/openstack-ansible-ovs-ubuntu-xenial-nv/0158f37/ : SUCCESS in 33m 34s (non-voting)\n- openstack-ansible-calico-ubuntu-xenial-nv http://logs.openstack.org/92/528692/5/check/openstack-ansible-calico-ubuntu-xenial-nv/87270a6/ : FAILURE in 37m 30s (non-voting)\n- openstack-ansible-dragonflow-ubuntu-xenial-nv http://logs.openstack.org/92/528692/5/check/openstack-ansible-dragonflow-ubuntu-xenial-nv/0281a0e/ : FAILURE in 35m 58s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-xenial-nv http://logs.openstack.org/92/528692/5/check/openstack-ansible-opendaylight-ubuntu-xenial-nv/511cd90/ : SUCCESS in 41m 33s (non-voting)\n- openstack-ansible-ovs-nsh-ubuntu-xenial http://logs.openstack.org/92/528692/5/check/openstack-ansible-ovs-nsh-ubuntu-xenial/c4a7eca/ : SUCCESS in 49m 54s (non-voting)\n- openstack-ansible-neutron-ssl-nv http://logs.openstack.org/92/528692/5/check/openstack-ansible-neutron-ssl-nv/51d34e5/ : SUCCESS in 38m 14s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"986b71b68c1025613628f1231734ed5e13478569","author":{"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},"date":"2017-12-19 08:10:30.000000000","message":"Patch Set 5:\n\nany idea why linters works here but does not work in my patch? I still get the Error when trying to get requirement for VCS system Command \"git config --get-regexp remote\\..*\\.url\" failed with error code 1","accounts_in_message":[],"_revision_number":5},{"id":"26fb5d91287154ae65600b42e082c9e497ebbc7e","author":{"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},"date":"2017-12-19 08:13:21.000000000","message":"Patch Set 5:\n\n\u003e any idea why linters works here but does not work in my patch? I\n \u003e still get the Error when trying to get requirement for VCS system\n \u003e Command \"git config --get-regexp remote\\..*\\.url\" failed with error\n \u003e code 1\n\nSorry, that is not the problem. The problem is that in my patch it does \"ANSIBLE0006 systemctl used in place of systemd module\" and then \"ERROR: InvocationError: \u0027/bin/bash -c /home/zuul/src/git.openstack.org/openstack/openstack-ansible-os_neutron/tests/common/test-ansible-lint.sh\u0027\" :(","accounts_in_message":[],"_revision_number":5},{"id":"a454e8c90f6da88d91de5654f537041c9facc25c","author":{"_account_id":23163,"name":"Markos Chandras (hwoarang)","email":"markos@chandras.me","username":"mchandras"},"date":"2017-12-19 09:30:37.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"8566cf5f16a8b4d706b06ecfa3befd3dad8cb30e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-19 09:42:25.000000000","message":"Patch Set 5: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":5},{"id":"1c9ea178fc1bb17e203a797e9fa88a9cdfbac455","author":{"_account_id":23163,"name":"Markos Chandras (hwoarang)","email":"markos@chandras.me","username":"mchandras"},"date":"2017-12-19 09:45:13.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"bb1372e2dd5c495b70bac7909810eaf91b2c8919","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-19 09:53:25.000000000","message":"Patch Set 5:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":5},{"id":"539d005a027acf8757c4db3ae76d9af56d821c30","author":{"_account_id":23163,"name":"Markos Chandras (hwoarang)","email":"markos@chandras.me","username":"mchandras"},"date":"2017-12-19 09:55:10.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"7e215fc8b0cb4dcbab04f8d5da3c600f6a946211","author":{"_account_id":23163,"name":"Markos Chandras (hwoarang)","email":"markos@chandras.me","username":"mchandras"},"date":"2017-12-19 10:11:33.000000000","message":"Patch Set 5:\n\n\u003e \u003e any idea why linters works here but does not work in my patch? I\n \u003e \u003e still get the Error when trying to get requirement for VCS system\n \u003e \u003e Command \"git config --get-regexp remote\\..*\\.url\" failed with\n \u003e error\n \u003e \u003e code 1\n \u003e \n \u003e Sorry, that is not the problem. The problem is that in my patch it\n \u003e does \"ANSIBLE0006 systemctl used in place of systemd module\" and\n \u003e then \"ERROR: InvocationError: \u0027/bin/bash -c /home/zuul/src/git.openstack.org/openstack/openstack-ansible-os_neutron/tests/common/test-ansible-lint.sh\u0027\"\n \u003e :(\n\nPlease add a \u0027Depends-On: I49c01fb63054e45bae5ae45a89cce986579959de\u0027 on your patch if you want to test it against the linter fixes.","accounts_in_message":[],"_revision_number":5},{"id":"5c359b773651fd248257e00555332b6c6ef08f65","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-19 11:09:03.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/92/528692/5/check/build-openstack-sphinx-docs/ec35fbe/html/ : SUCCESS in 2m 55s\n- openstack-ansible-linters http://logs.openstack.org/92/528692/5/check/openstack-ansible-linters/9e5d5c3/ : FAILURE in 4m 31s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/92/528692/5/check/openstack-ansible-functional-centos-7/bbf8467/ : SUCCESS in 48m 43s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/92/528692/5/check/openstack-ansible-functional-opensuse-423/bb9053e/ : SUCCESS in 36m 28s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/92/528692/5/check/openstack-ansible-functional-ubuntu-xenial/a4f034d/ : FAILURE in 7m 07s\n- openstack-ansible-upgrade-ubuntu-xenial http://logs.openstack.org/92/528692/5/check/openstack-ansible-upgrade-ubuntu-xenial/649abc1/ : SUCCESS in 37m 22s\n- openstack-ansible-ovs-ubuntu-xenial-nv http://logs.openstack.org/92/528692/5/check/openstack-ansible-ovs-ubuntu-xenial-nv/89a6aec/ : SUCCESS in 40m 07s (non-voting)\n- openstack-ansible-calico-ubuntu-xenial-nv http://logs.openstack.org/92/528692/5/check/openstack-ansible-calico-ubuntu-xenial-nv/c980168/ : FAILURE in 50m 26s (non-voting)\n- openstack-ansible-dragonflow-ubuntu-xenial-nv http://logs.openstack.org/92/528692/5/check/openstack-ansible-dragonflow-ubuntu-xenial-nv/e5b57d9/ : FAILURE in 44m 04s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-xenial-nv http://logs.openstack.org/92/528692/5/check/openstack-ansible-opendaylight-ubuntu-xenial-nv/85ebd66/ : SUCCESS in 46m 32s (non-voting)\n- openstack-ansible-ovs-nsh-ubuntu-xenial http://logs.openstack.org/92/528692/5/check/openstack-ansible-ovs-nsh-ubuntu-xenial/d18299e/ : SUCCESS in 1h 09m 50s (non-voting)\n- openstack-ansible-neutron-ssl-nv http://logs.openstack.org/92/528692/5/check/openstack-ansible-neutron-ssl-nv/50155ce/ : SUCCESS in 38m 22s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2c22efa297810df7faac6b2f56a0ddf803b75c8e","author":{"_account_id":23163,"name":"Markos Chandras (hwoarang)","email":"markos@chandras.me","username":"mchandras"},"date":"2017-12-19 11:14:40.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"9dafaf5fc7be2b33bf44599cdd91ebfb59ba44b5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-19 12:07:27.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/92/528692/5/check/build-openstack-sphinx-docs/528987f/html/ : SUCCESS in 2m 13s\n- openstack-ansible-linters http://logs.openstack.org/92/528692/5/check/openstack-ansible-linters/da78f2d/ : FAILURE in 5m 02s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/92/528692/5/check/openstack-ansible-functional-centos-7/52d30fc/ : SUCCESS in 46m 42s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/92/528692/5/check/openstack-ansible-functional-opensuse-423/df8085f/ : SUCCESS in 48m 00s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/92/528692/5/check/openstack-ansible-functional-ubuntu-xenial/2f1fb05/ : SUCCESS in 37m 21s\n- openstack-ansible-upgrade-ubuntu-xenial http://logs.openstack.org/92/528692/5/check/openstack-ansible-upgrade-ubuntu-xenial/9d6f8d7/ : SUCCESS in 35m 42s\n- openstack-ansible-ovs-ubuntu-xenial-nv http://logs.openstack.org/92/528692/5/check/openstack-ansible-ovs-ubuntu-xenial-nv/48b222c/ : SUCCESS in 35m 09s (non-voting)\n- openstack-ansible-calico-ubuntu-xenial-nv http://logs.openstack.org/92/528692/5/check/openstack-ansible-calico-ubuntu-xenial-nv/f43efc7/ : FAILURE in 36m 50s (non-voting)\n- openstack-ansible-dragonflow-ubuntu-xenial-nv http://logs.openstack.org/92/528692/5/check/openstack-ansible-dragonflow-ubuntu-xenial-nv/81aa655/ : FAILURE in 50m 00s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-xenial-nv http://logs.openstack.org/92/528692/5/check/openstack-ansible-opendaylight-ubuntu-xenial-nv/f50345c/ : SUCCESS in 33m 59s (non-voting)\n- openstack-ansible-ovs-nsh-ubuntu-xenial http://logs.openstack.org/92/528692/5/check/openstack-ansible-ovs-nsh-ubuntu-xenial/b5ac670/ : SUCCESS in 40m 29s (non-voting)\n- openstack-ansible-neutron-ssl-nv http://logs.openstack.org/92/528692/5/check/openstack-ansible-neutron-ssl-nv/9952cf6/ : SUCCESS in 28m 54s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"adac26724fc8dd189df72d6f75117dcabb6d7cc6","author":{"_account_id":23163,"name":"Markos Chandras (hwoarang)","email":"markos@chandras.me","username":"mchandras"},"date":"2018-02-13 08:50:59.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"ee266e8fb95251cf8c5d8decab5720d926a46f84","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-13 09:40:28.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/92/528692/6/check/build-openstack-sphinx-docs/0ee6a87/html/ : SUCCESS in 2m 16s\n- openstack-ansible-linters http://logs.openstack.org/92/528692/6/check/openstack-ansible-linters/05fae8a/ : SUCCESS in 5m 08s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/92/528692/6/check/openstack-ansible-functional-centos-7/3b12d40/ : FAILURE in 12m 22s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/92/528692/6/check/openstack-ansible-functional-opensuse-423/fe41221/ : SUCCESS in 34m 34s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/92/528692/6/check/openstack-ansible-functional-ubuntu-xenial/ea5ef22/ : SUCCESS in 32m 53s\n- openstack-ansible-upgrade-ubuntu-xenial http://logs.openstack.org/92/528692/6/check/openstack-ansible-upgrade-ubuntu-xenial/3669099/ : FAILURE in 36m 33s\n- openstack-ansible-ovs-ubuntu-xenial-nv http://logs.openstack.org/92/528692/6/check/openstack-ansible-ovs-ubuntu-xenial-nv/8e65889/ : SUCCESS in 35m 03s (non-voting)\n- openstack-ansible-calico-ubuntu-xenial-nv http://logs.openstack.org/92/528692/6/check/openstack-ansible-calico-ubuntu-xenial-nv/063364d/ : FAILURE in 36m 03s (non-voting)\n- openstack-ansible-dragonflow-ubuntu-xenial-nv http://logs.openstack.org/92/528692/6/check/openstack-ansible-dragonflow-ubuntu-xenial-nv/14d9fb5/ : FAILURE in 41m 21s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-xenial-nv http://logs.openstack.org/92/528692/6/check/openstack-ansible-opendaylight-ubuntu-xenial-nv/ab3e962/ : SUCCESS in 46m 50s (non-voting)\n- openstack-ansible-ovs-nsh-ubuntu-xenial http://logs.openstack.org/92/528692/6/check/openstack-ansible-ovs-nsh-ubuntu-xenial/38c4fff/ : SUCCESS in 39m 04s (non-voting)\n- openstack-ansible-neutron-ssl-nv http://logs.openstack.org/92/528692/6/check/openstack-ansible-neutron-ssl-nv/f78a31f/ : SUCCESS in 34m 58s (non-voting)\n- openstack-ansible-odl-sfc-ubuntu-xenial http://logs.openstack.org/92/528692/6/check/openstack-ansible-odl-sfc-ubuntu-xenial/15b0615/ : SUCCESS in 44m 16s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"68428fe8fbeba4f4c34801748863c376cb3956f1","author":{"_account_id":23163,"name":"Markos Chandras (hwoarang)","email":"markos@chandras.me","username":"mchandras"},"date":"2018-03-27 15:29:30.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"ddafa600ecdb2cf8f6161dd7eb5d2816c8b2517f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-27 16:17:37.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/92/528692/7/check/build-openstack-sphinx-docs/83aa9c4/html/ : SUCCESS in 2m 16s\n- openstack-ansible-linters http://logs.openstack.org/92/528692/7/check/openstack-ansible-linters/1b4fb71/ : SUCCESS in 5m 21s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/92/528692/7/check/openstack-ansible-functional-centos-7/607969b/ : FAILURE in 45m 32s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/92/528692/7/check/openstack-ansible-functional-opensuse-423/14bde66/ : FAILURE in 35m 29s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/92/528692/7/check/openstack-ansible-functional-ubuntu-xenial/a00d2c8/ : SUCCESS in 38m 56s\n- openstack-ansible-upgrade-ubuntu-xenial http://logs.openstack.org/92/528692/7/check/openstack-ansible-upgrade-ubuntu-xenial/fb3a0c8/ : SUCCESS in 43m 48s\n- openstack-ansible-ovs-ubuntu-xenial-nv http://logs.openstack.org/92/528692/7/check/openstack-ansible-ovs-ubuntu-xenial-nv/24e3296/ : SUCCESS in 37m 22s (non-voting)\n- openstack-ansible-calico-ubuntu-xenial-nv http://logs.openstack.org/92/528692/7/check/openstack-ansible-calico-ubuntu-xenial-nv/c10a3d9/ : FAILURE in 35m 45s (non-voting)\n- openstack-ansible-dragonflow-ubuntu-xenial-nv http://logs.openstack.org/92/528692/7/check/openstack-ansible-dragonflow-ubuntu-xenial-nv/0778233/ : FAILURE in 26m 25s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-xenial-nv http://logs.openstack.org/92/528692/7/check/openstack-ansible-opendaylight-ubuntu-xenial-nv/e0f37cf/ : FAILURE in 25m 36s (non-voting)\n- openstack-ansible-ovs-nsh-ubuntu-xenial http://logs.openstack.org/92/528692/7/check/openstack-ansible-ovs-nsh-ubuntu-xenial/09592a1/ : FAILURE in 26m 55s (non-voting)\n- openstack-ansible-neutron-ssl-nv http://logs.openstack.org/92/528692/7/check/openstack-ansible-neutron-ssl-nv/329bc9f/ : SUCCESS in 40m 01s (non-voting)\n- openstack-ansible-odl-sfc-ubuntu-xenial http://logs.openstack.org/92/528692/7/check/openstack-ansible-odl-sfc-ubuntu-xenial/e692c64/ : FAILURE in 25m 17s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-xenial http://logs.openstack.org/92/528692/7/check/openstack-ansible-opendaylight-bgpvpn-ubuntu-xenial/317ae48/ : FAILURE in 25m 23s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"99441bcde486f907db86f59fa3318c53be70f8ee","author":{"_account_id":23163,"name":"Markos Chandras (hwoarang)","email":"markos@chandras.me","username":"mchandras"},"date":"2018-09-20 07:39:53.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"ec2ea1c5eefcb92d974eab08d3d7646be595ed6c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-20 09:59:41.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/92/528692/8/check/build-openstack-sphinx-docs/b5a90e2/html/ : SUCCESS in 2m 50s\n- openstack-ansible-linters http://logs.openstack.org/92/528692/8/check/openstack-ansible-linters/1ecd9b5/ : SUCCESS in 6m 36s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/92/528692/8/check/openstack-ansible-functional-centos-7/902b86c/ : SUCCESS in 45m 24s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/92/528692/8/check/openstack-ansible-functional-opensuse-150/8352d0f/ : SUCCESS in 1h 20m 55s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/92/528692/8/check/openstack-ansible-functional-opensuse-423/b00ec6b/ : SUCCESS in 38m 56s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/92/528692/8/check/openstack-ansible-functional-ubuntu-bionic/3ff66b7/ : SUCCESS in 42m 53s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/92/528692/8/check/openstack-ansible-functional-ubuntu-xenial/d878f03/ : SUCCESS in 37m 53s\n- openstack-ansible-ovs-ubuntu-xenial http://logs.openstack.org/92/528692/8/check/openstack-ansible-ovs-ubuntu-xenial/c855b14/ : SUCCESS in 37m 32s\n- openstack-ansible-ovn-ubuntu-xenial http://logs.openstack.org/92/528692/8/check/openstack-ansible-ovn-ubuntu-xenial/385b0c3/ : SUCCESS in 37m 47s\n- openstack-ansible-ovn-centos-7 http://logs.openstack.org/92/528692/8/check/openstack-ansible-ovn-centos-7/18c9add/ : SUCCESS in 45m 07s\n- openstack-ansible-calico-ubuntu-xenial http://logs.openstack.org/92/528692/8/check/openstack-ansible-calico-ubuntu-xenial/6f5ee74/ : FAILURE in 39m 46s (non-voting)\n- openstack-ansible-dragonflow-ubuntu-xenial http://logs.openstack.org/92/528692/8/check/openstack-ansible-dragonflow-ubuntu-xenial/86d8916/ : FAILURE in 48m 40s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-xenial http://logs.openstack.org/92/528692/8/check/openstack-ansible-opendaylight-ubuntu-xenial/326a569/ : SUCCESS in 1h 00m 28s\n- openstack-ansible-ovs-nsh-ubuntu-xenial http://logs.openstack.org/92/528692/8/check/openstack-ansible-ovs-nsh-ubuntu-xenial/50c3b62/ : SUCCESS in 46m 41s\n- openstack-ansible-odl-sfc-ubuntu-xenial http://logs.openstack.org/92/528692/8/check/openstack-ansible-odl-sfc-ubuntu-xenial/c22a50a/ : SUCCESS in 55m 02s\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-xenial http://logs.openstack.org/92/528692/8/check/openstack-ansible-opendaylight-bgpvpn-ubuntu-xenial/ce18823/ : SUCCESS in 54m 23s (non-voting)\n- openstack-ansible-functional-distro_install-ubuntu-xenial http://logs.openstack.org/92/528692/8/check/openstack-ansible-functional-distro_install-ubuntu-xenial/9bb02f5/ : SUCCESS in 39m 15s\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/92/528692/8/check/openstack-ansible-functional-distro_install-centos-7/6a474f5/ : SUCCESS in 48m 06s (non-voting)\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/92/528692/8/check/openstack-ansible-functional-distro_install-opensuse-423/aac9cc5/ : SUCCESS in 36m 32s","accounts_in_message":[],"_revision_number":8},{"id":"89bfdef415853d43071b2cd9ab3a3ab1eced46ec","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-20 20:47:18.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"fc94fa6b732b989ebca09d50c96d711ca8fa6521","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 22:20:54.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/92/528692/9/check/openstack-ansible-linters/835d7df/ : SUCCESS in 4m 29s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/92/528692/9/check/openstack-ansible-deploy-aio_metal-debian-stable/981d2fa/ : SUCCESS in 58m 34s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/92/528692/9/check/openstack-ansible-deploy-aio_metal-centos-7/4b023a4/ : SUCCESS in 1h 17m 59s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/92/528692/9/check/openstack-ansible-deploy-aio_metal-opensuse-150/9d10428/ : SUCCESS in 1h 04m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/92/528692/9/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/64bc236/ : SUCCESS in 53m 53s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/92/528692/9/check/openstack-ansible-deploy-aio_distro_metal-centos-7/05b8e5a/ : SUCCESS in 59m 15s\n- openstack-tox-docs http://logs.openstack.org/92/528692/9/check/openstack-tox-docs/d1ccaec/html/ : SUCCESS in 2m 26s\n- openstack-ansible-ovs-ubuntu-bionic http://logs.openstack.org/92/528692/9/check/openstack-ansible-ovs-ubuntu-bionic/f76135a/ : SUCCESS in 36m 20s\n- openstack-ansible-ovn-ubuntu-bionic http://logs.openstack.org/92/528692/9/check/openstack-ansible-ovn-ubuntu-bionic/c6b877f/ : SUCCESS in 49m 05s\n- openstack-ansible-ovn-centos-7 http://logs.openstack.org/92/528692/9/check/openstack-ansible-ovn-centos-7/3ebf8b4/ : FAILURE in 37m 26s (non-voting)\n- openstack-ansible-calico-ubuntu-bionic http://logs.openstack.org/92/528692/9/check/openstack-ansible-calico-ubuntu-bionic/615da84/ : FAILURE in 36m 39s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic http://logs.openstack.org/92/528692/9/check/openstack-ansible-opendaylight-ubuntu-bionic/8d6e1a8/ : SUCCESS in 59m 05s\n- openstack-ansible-ovs-nsh-ubuntu-bionic http://logs.openstack.org/92/528692/9/check/openstack-ansible-ovs-nsh-ubuntu-bionic/ec9c3e6/ : SUCCESS in 56m 41s\n- openstack-ansible-odl-sfc-ubuntu-bionic http://logs.openstack.org/92/528692/9/check/openstack-ansible-odl-sfc-ubuntu-bionic/b2ad3a7/ : SUCCESS in 54m 32s\n- openstack-ansible-ovs-sfc-ubuntu-bionic http://logs.openstack.org/92/528692/9/check/openstack-ansible-ovs-sfc-ubuntu-bionic/88ad76e/ : SUCCESS in 38m 47s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic http://logs.openstack.org/92/528692/9/check/openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic/4ede4e1/ : SUCCESS in 1h 05m 43s (non-voting)\n- openstack-ansible-ovs-fwaasv2-ubuntu-bionic http://logs.openstack.org/92/528692/9/check/openstack-ansible-ovs-fwaasv2-ubuntu-bionic/c8e335d/ : SUCCESS in 36m 23s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"9f59f5e09b26c6ba2bd2ab47d51ba4473210f823","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-13 17:45:18.000000000","message":"Abandoned\n\nDNM patch from long ago, lets abandon","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"02ad9e2b83422be5d95f2e32afe7aa4441d4c609","revisions":{"314e63c170c1455cb266f0a07d01a86b23a9ebea":{"kind":"REWORK","_number":1,"created":"2017-12-18 11:13:12.000000000","uploader":{"_account_id":23163,"name":"Markos Chandras (hwoarang)","email":"markos@chandras.me","username":"mchandras"},"ref":"refs/changes/92/528692/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/92/528692/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/92/528692/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/92/528692/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/92/528692/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/92/528692/1"}}},"commit":{"parents":[{"commit":"7dce3277b04a1bacb015c7885bdb8db24f4d766f","subject":"Add MySQL connection SSL support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/7dce3277b04a1bacb015c7885bdb8db24f4d766f"}]}],"author":{"name":"Markos Chandras","email":"mchandras@suse.de","date":"2017-12-18 11:12:38.000000000","tz":0},"committer":{"name":"Markos Chandras","email":"mchandras@suse.de","date":"2017-12-18 11:12:38.000000000","tz":0},"subject":"DNM - Testing lint fixes","message":"DNM - Testing lint fixes\n\nChange-Id: Ic3e8eee3c97a5df3e988c302a8111952385859bf\nDepends-On: I49c01fb63054e45bae5ae45a89cce986579959de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/314e63c170c1455cb266f0a07d01a86b23a9ebea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/314e63c170c1455cb266f0a07d01a86b23a9ebea"}]},"branch":"refs/heads/master"},"55cd05a61cf54e5806f20581bce96bd81e11e199":{"kind":"REWORK","_number":2,"created":"2017-12-18 11:29:28.000000000","uploader":{"_account_id":23163,"name":"Markos Chandras (hwoarang)","email":"markos@chandras.me","username":"mchandras"},"ref":"refs/changes/92/528692/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/92/528692/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/92/528692/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/92/528692/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/92/528692/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/92/528692/2"}}},"commit":{"parents":[{"commit":"7dce3277b04a1bacb015c7885bdb8db24f4d766f","subject":"Add MySQL connection SSL support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/7dce3277b04a1bacb015c7885bdb8db24f4d766f"}]}],"author":{"name":"Markos Chandras","email":"mchandras@suse.de","date":"2017-12-18 11:12:38.000000000","tz":0},"committer":{"name":"Markos Chandras","email":"mchandras@suse.de","date":"2017-12-18 11:29:18.000000000","tz":0},"subject":"DNM - Testing lint fixes","message":"DNM - Testing lint fixes\n\nChange-Id: Ic3e8eee3c97a5df3e988c302a8111952385859bf\nDepends-On: I49c01fb63054e45bae5ae45a89cce986579959de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/55cd05a61cf54e5806f20581bce96bd81e11e199"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/55cd05a61cf54e5806f20581bce96bd81e11e199"}]},"branch":"refs/heads/master"},"ccbc431a5f4138add3305fac452ed2a276b5042a":{"kind":"REWORK","_number":3,"created":"2017-12-18 14:28:20.000000000","uploader":{"_account_id":23163,"name":"Markos Chandras (hwoarang)","email":"markos@chandras.me","username":"mchandras"},"ref":"refs/changes/92/528692/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/92/528692/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/92/528692/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/92/528692/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/92/528692/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/92/528692/3"}}},"commit":{"parents":[{"commit":"7dce3277b04a1bacb015c7885bdb8db24f4d766f","subject":"Add MySQL connection SSL support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/7dce3277b04a1bacb015c7885bdb8db24f4d766f"}]}],"author":{"name":"Markos Chandras","email":"mchandras@suse.de","date":"2017-12-18 11:12:38.000000000","tz":0},"committer":{"name":"Markos Chandras","email":"mchandras@suse.de","date":"2017-12-18 14:27:10.000000000","tz":0},"subject":"DNM - Testing lint fixes","message":"DNM - Testing lint fixes\n\nChange-Id: Ic3e8eee3c97a5df3e988c302a8111952385859bf\nDepends-On: I49c01fb63054e45bae5ae45a89cce986579959de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/ccbc431a5f4138add3305fac452ed2a276b5042a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/ccbc431a5f4138add3305fac452ed2a276b5042a"}]},"branch":"refs/heads/master"},"0518a70759f9e1f522978cdf4beee45f9e22f2c8":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2017-12-18 14:35:19.000000000","uploader":{"_account_id":23163,"name":"Markos Chandras (hwoarang)","email":"markos@chandras.me","username":"mchandras"},"ref":"refs/changes/92/528692/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/92/528692/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/92/528692/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/92/528692/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/92/528692/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/92/528692/4"}}},"commit":{"parents":[{"commit":"bdc0b3172da84c0f9e6361a0b90016aeab365324","subject":"Merge \"Add and document the Octavia proxy plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/bdc0b3172da84c0f9e6361a0b90016aeab365324"}]}],"author":{"name":"Markos Chandras","email":"mchandras@suse.de","date":"2017-12-18 11:12:38.000000000","tz":0},"committer":{"name":"Markos Chandras (hwoarang)","email":"mchandras@suse.de","date":"2017-12-18 14:35:19.000000000","tz":0},"subject":"DNM - Testing lint fixes","message":"DNM - Testing lint fixes\n\nChange-Id: Ic3e8eee3c97a5df3e988c302a8111952385859bf\nDepends-On: I49c01fb63054e45bae5ae45a89cce986579959de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/0518a70759f9e1f522978cdf4beee45f9e22f2c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/0518a70759f9e1f522978cdf4beee45f9e22f2c8"}]},"branch":"refs/heads/master"},"62b3433d897b689d85b562b0c0d8097a5dc41fbf":{"kind":"REWORK","_number":5,"created":"2017-12-18 15:23:24.000000000","uploader":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"ref":"refs/changes/92/528692/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/92/528692/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/92/528692/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/92/528692/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/92/528692/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/92/528692/5"}}},"commit":{"parents":[{"commit":"bdc0b3172da84c0f9e6361a0b90016aeab365324","subject":"Merge \"Add and document the Octavia proxy plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/bdc0b3172da84c0f9e6361a0b90016aeab365324"}]}],"author":{"name":"Markos Chandras","email":"mchandras@suse.de","date":"2017-12-18 11:12:38.000000000","tz":0},"committer":{"name":"Jesse Pretorius","email":"jesse.pretorius@rackspace.co.uk","date":"2017-12-18 15:23:00.000000000","tz":0},"subject":"DNM - Testing lint fixes","message":"DNM - Testing lint fixes\n\nThis should fail the lint test.\n\nChange-Id: Ic3e8eee3c97a5df3e988c302a8111952385859bf\nDepends-On: I49c01fb63054e45bae5ae45a89cce986579959de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/62b3433d897b689d85b562b0c0d8097a5dc41fbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/62b3433d897b689d85b562b0c0d8097a5dc41fbf"}]},"branch":"refs/heads/master"},"8d4ca26767702232c754e249c58ea5cf5213ea8c":{"kind":"REWORK","_number":6,"created":"2018-02-13 08:50:59.000000000","uploader":{"_account_id":23163,"name":"Markos Chandras (hwoarang)","email":"markos@chandras.me","username":"mchandras"},"ref":"refs/changes/92/528692/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/92/528692/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/92/528692/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/92/528692/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/92/528692/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/92/528692/6"}}},"commit":{"parents":[{"commit":"293bec299811ff32ce5a62fdd18429ab8bb88734","subject":"Merge \"Update the url links in doc file of openstack-ansible-os_neutron\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/293bec299811ff32ce5a62fdd18429ab8bb88734"}]}],"author":{"name":"Markos Chandras","email":"mchandras@suse.de","date":"2018-02-13 08:50:27.000000000","tz":0},"committer":{"name":"Markos Chandras","email":"mchandras@suse.de","date":"2018-02-13 08:50:27.000000000","tz":0},"subject":"Testing neutron jobs","message":"Testing neutron jobs\n\nChange-Id: Ic3e8eee3c97a5df3e988c302a8111952385859bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/8d4ca26767702232c754e249c58ea5cf5213ea8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/8d4ca26767702232c754e249c58ea5cf5213ea8c"}]},"branch":"refs/heads/master"},"63d32eadfb6d30ba85ff10da1f79be75522c8283":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2018-03-27 15:29:30.000000000","uploader":{"_account_id":23163,"name":"Markos Chandras (hwoarang)","email":"markos@chandras.me","username":"mchandras"},"ref":"refs/changes/92/528692/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/92/528692/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/92/528692/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/92/528692/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/92/528692/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/92/528692/7"}}},"commit":{"parents":[{"commit":"2e86cbd4d4d7c3f06b0955272845d7075d1e3427","subject":"add networking-bgpvpn repo correctly for the tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/2e86cbd4d4d7c3f06b0955272845d7075d1e3427"}]}],"author":{"name":"Markos Chandras","email":"mchandras@suse.de","date":"2018-02-13 08:50:27.000000000","tz":0},"committer":{"name":"Markos Chandras (hwoarang)","email":"mchandras@suse.de","date":"2018-03-27 15:29:30.000000000","tz":0},"subject":"Testing neutron jobs","message":"Testing neutron jobs\n\nChange-Id: Ic3e8eee3c97a5df3e988c302a8111952385859bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/63d32eadfb6d30ba85ff10da1f79be75522c8283"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/63d32eadfb6d30ba85ff10da1f79be75522c8283"}]},"branch":"refs/heads/master"},"4cdf9052bce65d695a0074d0150664aaf3fe0896":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2018-09-20 07:39:53.000000000","uploader":{"_account_id":23163,"name":"Markos Chandras (hwoarang)","email":"markos@chandras.me","username":"mchandras"},"ref":"refs/changes/92/528692/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/92/528692/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/92/528692/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/92/528692/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/92/528692/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/92/528692/8"}}},"commit":{"parents":[{"commit":"6b4c0199182150df5ba0d850510efbfb512bf7ff","subject":"Remove galera_client meta-dependency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/6b4c0199182150df5ba0d850510efbfb512bf7ff"}]}],"author":{"name":"Markos Chandras","email":"mchandras@suse.de","date":"2018-02-13 08:50:27.000000000","tz":0},"committer":{"name":"Markos Chandras (hwoarang)","email":"mchandras@suse.de","date":"2018-09-20 07:39:53.000000000","tz":0},"subject":"Testing neutron jobs","message":"Testing neutron jobs\n\nChange-Id: Ic3e8eee3c97a5df3e988c302a8111952385859bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/4cdf9052bce65d695a0074d0150664aaf3fe0896"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/4cdf9052bce65d695a0074d0150664aaf3fe0896"}]},"branch":"refs/heads/master"},"02ad9e2b83422be5d95f2e32afe7aa4441d4c609":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2019-05-20 20:47:18.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/92/528692/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/92/528692/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/92/528692/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/92/528692/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/92/528692/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/92/528692/9"}}},"commit":{"parents":[{"commit":"9f619e3f059d2bf8db930ee23b8847456775708a","subject":"Remove LBaaS v2 plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/9f619e3f059d2bf8db930ee23b8847456775708a"}]}],"author":{"name":"Markos Chandras","email":"mchandras@suse.de","date":"2018-02-13 08:50:27.000000000","tz":0},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-05-20 20:47:18.000000000","tz":0},"subject":"Testing neutron jobs","message":"Testing neutron jobs\n\nChange-Id: Ic3e8eee3c97a5df3e988c302a8111952385859bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/02ad9e2b83422be5d95f2e32afe7aa4441d4c609"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/02ad9e2b83422be5d95f2e32afe7aa4441d4c609"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
