)]}'
{"id":"openstack%2Fopenstack-ansible-os_neutron~631644","triplet_id":"openstack%2Fopenstack-ansible-os_neutron~master~Ie0a4a8c9001eb1c90f06b2f16a6be078cb9176e6","project":"openstack/openstack-ansible-os_neutron","branch":"master","topic":"vpp-deploy","hashtags":[],"change_id":"Ie0a4a8c9001eb1c90f06b2f16a6be078cb9176e6","subject":"[WIP] Deploy Vector Packet Processing (VPP) Platform for Neutron","status":"ABANDONED","created":"2019-01-17 21:15:57.000000000","updated":"2022-04-27 14:26:10.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1ae13483cce7e082c7b0567ed71e82d37a919ee6","_number":631644,"virtual_id_number":631644,"owner":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":-1,"date":"2019-05-04 16:56:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-04 16:56:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8bc83424cbebb4c0f1fd43610d111fa9e295d848","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2019-01-17 21:15:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"68b21c113c1962e6d0e4d3a7ccf2425224bc93c4","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2019-01-17 21:16:08.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"b045f309fb5caa259c2db8b418d5b0f75ab0ee53","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-17 22:36:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/44/631644/1/check/openstack-ansible-linters/acfb747/ : SUCCESS in 7m 29s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/44/631644/1/check/openstack-ansible-functional-centos-7/6ae8291/ : FAILURE in 34m 50s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/44/631644/1/check/openstack-ansible-functional-opensuse-150/8eeaa41/ : FAILURE in 39m 28s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/44/631644/1/check/openstack-ansible-functional-opensuse-423/c31f637/ : FAILURE in 26m 43s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/44/631644/1/check/openstack-ansible-functional-ubuntu-bionic/2095b2f/ : SUCCESS in 38m 36s\n- openstack-tox-docs http://logs.openstack.org/44/631644/1/check/openstack-tox-docs/7374440/html/ : SUCCESS in 3m 46s\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/44/631644/1/check/openstack-ansible-functional-distro_install-centos-7/90ad4d4/ : FAILURE in 26m 44s\n- openstack-ansible-functional-distro_install-opensuse-150 http://logs.openstack.org/44/631644/1/check/openstack-ansible-functional-distro_install-opensuse-150/c2cda06/ : FAILURE in 35m 46s (non-voting)\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/44/631644/1/check/openstack-ansible-functional-distro_install-opensuse-423/f89868b/ : FAILURE in 25m 48s (non-voting)\n- openstack-ansible-functional-distro_install-ubuntu-bionic http://logs.openstack.org/44/631644/1/check/openstack-ansible-functional-distro_install-ubuntu-bionic/2d418a0/ : SUCCESS in 44m 05s (non-voting)\n- openstack-ansible-ovs-ubuntu-xenial http://logs.openstack.org/44/631644/1/check/openstack-ansible-ovs-ubuntu-xenial/42aa6e1/ : SUCCESS in 36m 53s\n- openstack-ansible-ovn-ubuntu-xenial http://logs.openstack.org/44/631644/1/check/openstack-ansible-ovn-ubuntu-xenial/ae6cdc0/ : SUCCESS in 37m 19s\n- openstack-ansible-ovn-centos-7 http://logs.openstack.org/44/631644/1/check/openstack-ansible-ovn-centos-7/410cce7/ : FAILURE in 28m 16s\n- openstack-ansible-calico-ubuntu-xenial http://logs.openstack.org/44/631644/1/check/openstack-ansible-calico-ubuntu-xenial/c90478b/ : FAILURE in 17m 08s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-xenial http://logs.openstack.org/44/631644/1/check/openstack-ansible-opendaylight-ubuntu-xenial/fab46de/ : SUCCESS in 50m 17s\n- openstack-ansible-ovs-nsh-ubuntu-xenial http://logs.openstack.org/44/631644/1/check/openstack-ansible-ovs-nsh-ubuntu-xenial/e43e813/ : SUCCESS in 1h 00m 43s\n- openstack-ansible-odl-sfc-ubuntu-xenial http://logs.openstack.org/44/631644/1/check/openstack-ansible-odl-sfc-ubuntu-xenial/a05428d/ : SUCCESS in 1h 15m 33s\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-xenial http://logs.openstack.org/44/631644/1/check/openstack-ansible-opendaylight-bgpvpn-ubuntu-xenial/1c98a13/ : SUCCESS in 52m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"01a439d75dcf012398b06015120665748ae55058","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2019-01-21 19:30:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7d2513398d7fc5c80d3ed30f12ce55b4b4094307","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2019-01-21 19:30:53.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"82863171eb1298d4e8f50e2bb374055d6c4e2192","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2019-01-23 04:17:15.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0f6a0d9ab7450b85c918c40f57b41eac4e56de5d","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2019-01-23 04:17:30.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"ac1c2e7a5d04a2c0ac8868da3caa54508ea3d05b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-23 05:24:41.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/44/631644/3/check/openstack-ansible-linters/3038be8/ : SUCCESS in 4m 36s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/44/631644/3/check/openstack-ansible-functional-centos-7/339785b/ : FAILURE in 28m 43s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/44/631644/3/check/openstack-ansible-functional-opensuse-150/fe4b2bb/ : FAILURE in 43m 33s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/44/631644/3/check/openstack-ansible-functional-opensuse-423/053ba17/ : FAILURE in 23m 53s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/44/631644/3/check/openstack-ansible-functional-ubuntu-bionic/799d8f9/ : SUCCESS in 36m 53s\n- openstack-tox-docs http://logs.openstack.org/44/631644/3/check/openstack-tox-docs/0105a78/html/ : SUCCESS in 3m 22s\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/44/631644/3/check/openstack-ansible-functional-distro_install-centos-7/2188864/ : FAILURE in 33m 14s\n- openstack-ansible-functional-distro_install-opensuse-150 http://logs.openstack.org/44/631644/3/check/openstack-ansible-functional-distro_install-opensuse-150/3519717/ : FAILURE in 40m 08s (non-voting)\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/44/631644/3/check/openstack-ansible-functional-distro_install-opensuse-423/2017b05/ : FAILURE in 22m 06s (non-voting)\n- openstack-ansible-functional-distro_install-ubuntu-bionic http://logs.openstack.org/44/631644/3/check/openstack-ansible-functional-distro_install-ubuntu-bionic/1519c8f/ : SUCCESS in 57m 35s (non-voting)\n- openstack-ansible-ovs-ubuntu-xenial http://logs.openstack.org/44/631644/3/check/openstack-ansible-ovs-ubuntu-xenial/1624bc9/ : SUCCESS in 46m 10s\n- openstack-ansible-ovn-ubuntu-xenial http://logs.openstack.org/44/631644/3/check/openstack-ansible-ovn-ubuntu-xenial/34a2ae9/ : SUCCESS in 39m 27s\n- openstack-ansible-ovn-centos-7 http://logs.openstack.org/44/631644/3/check/openstack-ansible-ovn-centos-7/e0da7d7/ : FAILURE in 27m 43s\n- openstack-ansible-calico-ubuntu-xenial http://logs.openstack.org/44/631644/3/check/openstack-ansible-calico-ubuntu-xenial/e30777b/ : FAILURE in 16m 07s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-xenial http://logs.openstack.org/44/631644/3/check/openstack-ansible-opendaylight-ubuntu-xenial/a993eba/ : SUCCESS in 1h 03m 34s\n- openstack-ansible-ovs-nsh-ubuntu-xenial http://logs.openstack.org/44/631644/3/check/openstack-ansible-ovs-nsh-ubuntu-xenial/48ed8f0/ : SUCCESS in 49m 36s\n- openstack-ansible-odl-sfc-ubuntu-xenial http://logs.openstack.org/44/631644/3/check/openstack-ansible-odl-sfc-ubuntu-xenial/f631a56/ : SUCCESS in 52m 18s\n- openstack-ansible-ovs-sfc-ubuntu-xenial http://logs.openstack.org/44/631644/3/check/openstack-ansible-ovs-sfc-ubuntu-xenial/a4285a6/ : SUCCESS in 43m 59s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-xenial http://logs.openstack.org/44/631644/3/check/openstack-ansible-opendaylight-bgpvpn-ubuntu-xenial/46b9d21/ : SUCCESS in 1h 04m 42s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"91751e8627fae4cb187b8c85b339442de88e6e94","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2019-01-23 16:54:23.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"85d727bd09bca5f9d8b4a9792038be54e5b9508c","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2019-01-23 16:56:28.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"9d54dcf94d65686558ccd2e75f82001c7ef7c634","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2019-01-23 16:56:49.000000000","message":"Patch Set 5: Workflow-1","accounts_in_message":[],"_revision_number":5},{"id":"d91c1f5c99d22977d249e8ee945610115575e0bb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-23 18:11:08.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/44/631644/5/check/openstack-ansible-linters/4c3e4c0/ : FAILURE in 11m 01s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/44/631644/5/check/openstack-ansible-functional-centos-7/e146bc1/ : FAILURE in 33m 28s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/44/631644/5/check/openstack-ansible-functional-opensuse-150/d898132/ : FAILURE in 46m 03s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/44/631644/5/check/openstack-ansible-functional-opensuse-423/88383e1/ : FAILURE in 35m 29s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/44/631644/5/check/openstack-ansible-functional-ubuntu-bionic/c521977/ : SUCCESS in 41m 29s\n- openstack-tox-docs http://logs.openstack.org/44/631644/5/check/openstack-tox-docs/f6c7a2a/html/ : SUCCESS in 5m 33s\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/44/631644/5/check/openstack-ansible-functional-distro_install-centos-7/93e66c5/ : FAILURE in 26m 42s\n- openstack-ansible-functional-distro_install-opensuse-150 http://logs.openstack.org/44/631644/5/check/openstack-ansible-functional-distro_install-opensuse-150/6fb6a95/ : FAILURE in 45m 01s (non-voting)\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/44/631644/5/check/openstack-ansible-functional-distro_install-opensuse-423/9d905d8/ : FAILURE in 27m 12s (non-voting)\n- openstack-ansible-functional-distro_install-ubuntu-bionic http://logs.openstack.org/44/631644/5/check/openstack-ansible-functional-distro_install-ubuntu-bionic/05c468a/ : SUCCESS in 48m 00s (non-voting)\n- openstack-ansible-ovs-ubuntu-xenial http://logs.openstack.org/44/631644/5/check/openstack-ansible-ovs-ubuntu-xenial/3c1e9fa/ : SUCCESS in 54m 12s\n- openstack-ansible-ovn-ubuntu-xenial http://logs.openstack.org/44/631644/5/check/openstack-ansible-ovn-ubuntu-xenial/b0efa65/ : SUCCESS in 39m 49s\n- openstack-ansible-ovn-centos-7 http://logs.openstack.org/44/631644/5/check/openstack-ansible-ovn-centos-7/b3c63df/ : FAILURE in 28m 14s\n- openstack-ansible-calico-ubuntu-xenial http://logs.openstack.org/44/631644/5/check/openstack-ansible-calico-ubuntu-xenial/83834a5/ : FAILURE in 20m 08s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-xenial http://logs.openstack.org/44/631644/5/check/openstack-ansible-opendaylight-ubuntu-xenial/1f656f1/ : SUCCESS in 58m 54s\n- openstack-ansible-ovs-nsh-ubuntu-xenial http://logs.openstack.org/44/631644/5/check/openstack-ansible-ovs-nsh-ubuntu-xenial/5acc466/ : SUCCESS in 52m 56s\n- openstack-ansible-odl-sfc-ubuntu-xenial http://logs.openstack.org/44/631644/5/check/openstack-ansible-odl-sfc-ubuntu-xenial/7b8993a/ : SUCCESS in 58m 16s\n- openstack-ansible-ovs-sfc-ubuntu-xenial http://logs.openstack.org/44/631644/5/check/openstack-ansible-ovs-sfc-ubuntu-xenial/7213496/ : SUCCESS in 38m 50s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-xenial http://logs.openstack.org/44/631644/5/check/openstack-ansible-opendaylight-bgpvpn-ubuntu-xenial/2e55286/ : SUCCESS in 1h 00m 53s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ed47f718ee0b5a560ee477a2ae8b8cb4e2f3de9a","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2019-01-24 15:22:11.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"126f89ad5c01a50c10de5829c35c2f42850402c0","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2019-01-24 15:54:51.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"98cb5c4681b6ec354f4cee37781b8990d77586e3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-24 17:11:41.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/44/631644/7/check/openstack-ansible-linters/4d2fd2e/ : FAILURE in 7m 32s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/44/631644/7/check/openstack-ansible-functional-centos-7/a097190/ : FAILURE in 30m 51s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/44/631644/7/check/openstack-ansible-functional-opensuse-150/b3deffd/ : FAILURE in 43m 09s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/44/631644/7/check/openstack-ansible-functional-opensuse-423/f6ad687/ : FAILURE in 26m 47s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/44/631644/7/check/openstack-ansible-functional-ubuntu-bionic/8447286/ : SUCCESS in 40m 14s\n- openstack-tox-docs http://logs.openstack.org/44/631644/7/check/openstack-tox-docs/108eda4/html/ : SUCCESS in 3m 51s\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/44/631644/7/check/openstack-ansible-functional-distro_install-centos-7/d3e333b/ : FAILURE in 30m 58s\n- openstack-ansible-functional-distro_install-opensuse-150 http://logs.openstack.org/44/631644/7/check/openstack-ansible-functional-distro_install-opensuse-150/3ed002d/ : FAILURE in 43m 52s (non-voting)\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/44/631644/7/check/openstack-ansible-functional-distro_install-opensuse-423/eb5e54e/ : FAILURE in 25m 48s (non-voting)\n- openstack-ansible-functional-distro_install-ubuntu-bionic http://logs.openstack.org/44/631644/7/check/openstack-ansible-functional-distro_install-ubuntu-bionic/8d5921b/ : SUCCESS in 50m 10s (non-voting)\n- openstack-ansible-ovs-ubuntu-xenial http://logs.openstack.org/44/631644/7/check/openstack-ansible-ovs-ubuntu-xenial/331cc8b/ : SUCCESS in 40m 48s\n- openstack-ansible-ovn-ubuntu-xenial http://logs.openstack.org/44/631644/7/check/openstack-ansible-ovn-ubuntu-xenial/5b86472/ : SUCCESS in 41m 55s\n- openstack-ansible-ovn-centos-7 http://logs.openstack.org/44/631644/7/check/openstack-ansible-ovn-centos-7/5fc703b/ : FAILURE in 28m 55s\n- openstack-ansible-calico-ubuntu-xenial http://logs.openstack.org/44/631644/7/check/openstack-ansible-calico-ubuntu-xenial/a20ce0e/ : FAILURE in 19m 30s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-xenial http://logs.openstack.org/44/631644/7/check/openstack-ansible-opendaylight-ubuntu-xenial/73d7852/ : SUCCESS in 55m 59s\n- openstack-ansible-ovs-nsh-ubuntu-xenial http://logs.openstack.org/44/631644/7/check/openstack-ansible-ovs-nsh-ubuntu-xenial/6a313d4/ : SUCCESS in 59m 13s\n- openstack-ansible-odl-sfc-ubuntu-xenial http://logs.openstack.org/44/631644/7/check/openstack-ansible-odl-sfc-ubuntu-xenial/21f1a1b/ : SUCCESS in 54m 08s\n- openstack-ansible-ovs-sfc-ubuntu-xenial http://logs.openstack.org/44/631644/7/check/openstack-ansible-ovs-sfc-ubuntu-xenial/fc3eca1/ : SUCCESS in 37m 26s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-xenial http://logs.openstack.org/44/631644/7/check/openstack-ansible-opendaylight-bgpvpn-ubuntu-xenial/d48436a/ : SUCCESS in 1h 12m 17s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"7a48d1385979b415bc4b13a29461bf2525f6f7a4","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2019-01-25 16:25:55.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"22f816a702f16e9c6a9ed98603575d46f5c1c3d2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-25 17:33:06.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/44/631644/8/check/openstack-ansible-linters/c73cea3/ : FAILURE in 6m 20s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/44/631644/8/check/openstack-ansible-functional-centos-7/660d8a6/ : FAILURE in 32m 06s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/44/631644/8/check/openstack-ansible-functional-opensuse-150/24352b7/ : FAILURE in 43m 52s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/44/631644/8/check/openstack-ansible-functional-opensuse-423/e13dff3/ : FAILURE in 31m 05s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/44/631644/8/check/openstack-ansible-functional-ubuntu-bionic/81f142b/ : SUCCESS in 42m 05s\n- openstack-tox-docs http://logs.openstack.org/44/631644/8/check/openstack-tox-docs/15e1884/html/ : SUCCESS in 3m 18s\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/44/631644/8/check/openstack-ansible-functional-distro_install-centos-7/6ee08c6/ : FAILURE in 27m 19s\n- openstack-ansible-functional-distro_install-opensuse-150 http://logs.openstack.org/44/631644/8/check/openstack-ansible-functional-distro_install-opensuse-150/ae9d882/ : FAILURE in 39m 09s (non-voting)\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/44/631644/8/check/openstack-ansible-functional-distro_install-opensuse-423/a546b44/ : FAILURE in 26m 22s (non-voting)\n- openstack-ansible-functional-distro_install-ubuntu-bionic http://logs.openstack.org/44/631644/8/check/openstack-ansible-functional-distro_install-ubuntu-bionic/21abc04/ : SUCCESS in 45m 58s (non-voting)\n- openstack-ansible-ovs-ubuntu-xenial http://logs.openstack.org/44/631644/8/check/openstack-ansible-ovs-ubuntu-xenial/096b613/ : SUCCESS in 39m 34s\n- openstack-ansible-ovn-ubuntu-xenial http://logs.openstack.org/44/631644/8/check/openstack-ansible-ovn-ubuntu-xenial/b0193c8/ : SUCCESS in 37m 18s\n- openstack-ansible-ovn-centos-7 http://logs.openstack.org/44/631644/8/check/openstack-ansible-ovn-centos-7/219117e/ : FAILURE in 33m 01s\n- openstack-ansible-calico-ubuntu-xenial http://logs.openstack.org/44/631644/8/check/openstack-ansible-calico-ubuntu-xenial/8e41602/ : FAILURE in 44m 34s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-xenial http://logs.openstack.org/44/631644/8/check/openstack-ansible-opendaylight-ubuntu-xenial/3987e75/ : SUCCESS in 56m 12s\n- openstack-ansible-ovs-nsh-ubuntu-xenial http://logs.openstack.org/44/631644/8/check/openstack-ansible-ovs-nsh-ubuntu-xenial/799a65d/ : SUCCESS in 57m 18s\n- openstack-ansible-odl-sfc-ubuntu-xenial http://logs.openstack.org/44/631644/8/check/openstack-ansible-odl-sfc-ubuntu-xenial/0805af7/ : SUCCESS in 56m 04s\n- openstack-ansible-ovs-sfc-ubuntu-xenial http://logs.openstack.org/44/631644/8/check/openstack-ansible-ovs-sfc-ubuntu-xenial/16995c6/ : SUCCESS in 46m 26s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-xenial http://logs.openstack.org/44/631644/8/check/openstack-ansible-opendaylight-bgpvpn-ubuntu-xenial/e246cf0/ : SUCCESS in 52m 44s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"744b854f4b48e938db9fde12935bf779eeeac9cb","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2019-01-28 18:37:45.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"2aaafeb3cf19bb3af867ad95f9818a2ac04443ec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-28 20:20:47.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/44/631644/9/check/openstack-ansible-linters/501b9ad/ : FAILURE in 8m 52s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/44/631644/9/check/openstack-ansible-functional-centos-7/bdb8659/ : TIMED_OUT in 1h 34m 09s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/44/631644/9/check/openstack-ansible-functional-opensuse-150/794f4c7/ : FAILURE in 38m 24s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/44/631644/9/check/openstack-ansible-functional-opensuse-423/57b0f41/ : FAILURE in 26m 42s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/44/631644/9/check/openstack-ansible-functional-ubuntu-bionic/03e58e6/ : SUCCESS in 37m 49s\n- openstack-tox-docs http://logs.openstack.org/44/631644/9/check/openstack-tox-docs/a03d49e/html/ : SUCCESS in 4m 19s\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/44/631644/9/check/openstack-ansible-functional-distro_install-centos-7/9ea29d8/ : FAILURE in 32m 25s\n- openstack-ansible-functional-distro_install-opensuse-150 http://logs.openstack.org/44/631644/9/check/openstack-ansible-functional-distro_install-opensuse-150/d8f5ba8/ : FAILURE in 49m 38s (non-voting)\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/44/631644/9/check/openstack-ansible-functional-distro_install-opensuse-423/389c4c5/ : FAILURE in 28m 23s (non-voting)\n- openstack-ansible-functional-distro_install-ubuntu-bionic http://logs.openstack.org/44/631644/9/check/openstack-ansible-functional-distro_install-ubuntu-bionic/6c117be/ : SUCCESS in 44m 51s (non-voting)\n- openstack-ansible-ovs-ubuntu-xenial http://logs.openstack.org/44/631644/9/check/openstack-ansible-ovs-ubuntu-xenial/7c04282/ : SUCCESS in 37m 16s\n- openstack-ansible-ovn-ubuntu-xenial http://logs.openstack.org/44/631644/9/check/openstack-ansible-ovn-ubuntu-xenial/431bc9d/ : SUCCESS in 40m 19s\n- openstack-ansible-ovn-centos-7 http://logs.openstack.org/44/631644/9/check/openstack-ansible-ovn-centos-7/cb38c87/ : FAILURE in 28m 04s\n- openstack-ansible-calico-ubuntu-xenial http://logs.openstack.org/44/631644/9/check/openstack-ansible-calico-ubuntu-xenial/8d6c911/ : FAILURE in 17m 57s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-xenial http://logs.openstack.org/44/631644/9/check/openstack-ansible-opendaylight-ubuntu-xenial/4c40944/ : SUCCESS in 1h 10m 02s\n- openstack-ansible-ovs-nsh-ubuntu-xenial http://logs.openstack.org/44/631644/9/check/openstack-ansible-ovs-nsh-ubuntu-xenial/60c79db/ : FAILURE in 1h 26m 14s\n- openstack-ansible-odl-sfc-ubuntu-xenial http://logs.openstack.org/44/631644/9/check/openstack-ansible-odl-sfc-ubuntu-xenial/acfa59d/ : SUCCESS in 54m 33s\n- openstack-ansible-ovs-sfc-ubuntu-xenial http://logs.openstack.org/44/631644/9/check/openstack-ansible-ovs-sfc-ubuntu-xenial/c671832/ : SUCCESS in 41m 21s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-xenial http://logs.openstack.org/44/631644/9/check/openstack-ansible-opendaylight-bgpvpn-ubuntu-xenial/af480ff/ : SUCCESS in 1h 06m 34s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"7032d305c647d573a9b8f6f8e658bf8ff0b946f2","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2019-02-26 03:15:14.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"20fe8b7b02745622af6fc1ae2f74bc14bd2b30a2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-26 04:05:47.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/44/631644/10/check/openstack-ansible-linters/96b55dd/ : FAILURE in 6m 12s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/44/631644/10/check/openstack-ansible-functional-centos-7/d106fb1/ : FAILURE in 27m 32s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/44/631644/10/check/openstack-ansible-functional-opensuse-150/89fbfdd/ : FAILURE in 39m 50s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/44/631644/10/check/openstack-ansible-functional-opensuse-423/440b935/ : FAILURE in 24m 23s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/44/631644/10/check/openstack-ansible-functional-ubuntu-bionic/85d0965/ : FAILURE in 34m 18s\n- openstack-tox-docs http://logs.openstack.org/44/631644/10/check/openstack-tox-docs/804cc14/html/ : SUCCESS in 2m 50s\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/44/631644/10/check/openstack-ansible-functional-distro_install-centos-7/6a4cd1d/ : FAILURE in 27m 34s\n- openstack-ansible-functional-distro_install-opensuse-150 http://logs.openstack.org/44/631644/10/check/openstack-ansible-functional-distro_install-opensuse-150/fdeb65d/ : FAILURE in 37m 07s (non-voting)\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/44/631644/10/check/openstack-ansible-functional-distro_install-opensuse-423/09a38b2/ : FAILURE in 23m 43s (non-voting)\n- openstack-ansible-functional-distro_install-ubuntu-bionic http://logs.openstack.org/44/631644/10/check/openstack-ansible-functional-distro_install-ubuntu-bionic/aad5b3a/ : FAILURE in 37m 43s (non-voting)\n- openstack-ansible-ovs-ubuntu-xenial http://logs.openstack.org/44/631644/10/check/openstack-ansible-ovs-ubuntu-xenial/46614dc/ : FAILURE in 32m 06s\n- openstack-ansible-ovn-ubuntu-xenial http://logs.openstack.org/44/631644/10/check/openstack-ansible-ovn-ubuntu-xenial/0d17483/ : FAILURE in 36m 23s\n- openstack-ansible-ovn-centos-7 http://logs.openstack.org/44/631644/10/check/openstack-ansible-ovn-centos-7/f83878d/ : FAILURE in 26m 17s\n- openstack-ansible-calico-ubuntu-xenial http://logs.openstack.org/44/631644/10/check/openstack-ansible-calico-ubuntu-xenial/188bdea/ : FAILURE in 17m 49s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-xenial http://logs.openstack.org/44/631644/10/check/openstack-ansible-opendaylight-ubuntu-xenial/7ebf53c/ : FAILURE in 45m 43s\n- openstack-ansible-ovs-nsh-ubuntu-xenial http://logs.openstack.org/44/631644/10/check/openstack-ansible-ovs-nsh-ubuntu-xenial/199b41d/ : FAILURE in 15m 40s\n- openstack-ansible-odl-sfc-ubuntu-xenial http://logs.openstack.org/44/631644/10/check/openstack-ansible-odl-sfc-ubuntu-xenial/9e60800/ : FAILURE in 47m 40s\n- openstack-ansible-ovs-sfc-ubuntu-xenial http://logs.openstack.org/44/631644/10/check/openstack-ansible-ovs-sfc-ubuntu-xenial/2fadb17/ : FAILURE in 36m 18s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-xenial http://logs.openstack.org/44/631644/10/check/openstack-ansible-opendaylight-bgpvpn-ubuntu-xenial/079d075/ : FAILURE in 45m 48s (non-voting)\n- openstack-ansible-ovs-fwaasv2-ubuntu-bionic http://logs.openstack.org/44/631644/10/check/openstack-ansible-ovs-fwaasv2-ubuntu-bionic/ab9af72/ : FAILURE in 37m 47s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"b648f45152537c89d2b52171a6cdf72d719df01a","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2019-02-26 12:07:24.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"417d34abe0ec6c3c983918e0953cf7986faf4ac0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-26 13:34:29.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/44/631644/10/check/openstack-ansible-linters/68f2136/ : FAILURE in 6m 47s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/44/631644/10/check/openstack-ansible-functional-centos-7/0ac2d4f/ : FAILURE in 29m 14s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/44/631644/10/check/openstack-ansible-functional-opensuse-150/43ad8bf/ : FAILURE in 46m 54s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/44/631644/10/check/openstack-ansible-functional-opensuse-423/569e2ae/ : FAILURE in 28m 39s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/44/631644/10/check/openstack-ansible-functional-ubuntu-bionic/2b8d61a/ : FAILURE in 40m 25s\n- openstack-tox-docs http://logs.openstack.org/44/631644/10/check/openstack-tox-docs/ee05969/html/ : SUCCESS in 3m 27s\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/44/631644/10/check/openstack-ansible-functional-distro_install-centos-7/b110680/ : FAILURE in 26m 15s\n- openstack-ansible-functional-distro_install-opensuse-150 http://logs.openstack.org/44/631644/10/check/openstack-ansible-functional-distro_install-opensuse-150/61a643b/ : FAILURE in 43m 28s (non-voting)\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/44/631644/10/check/openstack-ansible-functional-distro_install-opensuse-423/50ea78c/ : FAILURE in 24m 22s (non-voting)\n- openstack-ansible-functional-distro_install-ubuntu-bionic http://logs.openstack.org/44/631644/10/check/openstack-ansible-functional-distro_install-ubuntu-bionic/26e6a14/ : FAILURE in 41m 57s (non-voting)\n- openstack-ansible-ovs-ubuntu-xenial http://logs.openstack.org/44/631644/10/check/openstack-ansible-ovs-ubuntu-xenial/b194d86/ : FAILURE in 48m 31s\n- openstack-ansible-ovn-ubuntu-xenial http://logs.openstack.org/44/631644/10/check/openstack-ansible-ovn-ubuntu-xenial/9e5034c/ : FAILURE in 44m 35s\n- openstack-ansible-ovn-centos-7 http://logs.openstack.org/44/631644/10/check/openstack-ansible-ovn-centos-7/f4f7d98/ : FAILURE in 28m 01s\n- openstack-ansible-calico-ubuntu-xenial http://logs.openstack.org/44/631644/10/check/openstack-ansible-calico-ubuntu-xenial/a86067c/ : FAILURE in 25m 28s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-xenial http://logs.openstack.org/44/631644/10/check/openstack-ansible-opendaylight-ubuntu-xenial/dcba51d/ : FAILURE in 48m 27s\n- openstack-ansible-ovs-nsh-ubuntu-xenial http://logs.openstack.org/44/631644/10/check/openstack-ansible-ovs-nsh-ubuntu-xenial/4272592/ : FAILURE in 52m 40s\n- openstack-ansible-odl-sfc-ubuntu-xenial http://logs.openstack.org/44/631644/10/check/openstack-ansible-odl-sfc-ubuntu-xenial/fe37c93/ : FAILURE in 49m 41s\n- openstack-ansible-ovs-sfc-ubuntu-xenial http://logs.openstack.org/44/631644/10/check/openstack-ansible-ovs-sfc-ubuntu-xenial/2c0ef42/ : FAILURE in 33m 13s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-xenial http://logs.openstack.org/44/631644/10/check/openstack-ansible-opendaylight-bgpvpn-ubuntu-xenial/7d0d9e4/ : FAILURE in 51m 44s (non-voting)\n- openstack-ansible-ovs-fwaasv2-ubuntu-bionic http://logs.openstack.org/44/631644/10/check/openstack-ansible-ovs-fwaasv2-ubuntu-bionic/d11ed15/ : FAILURE in 35m 08s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"6cc3a61bb36f659a5df120fee2b23a799271aa28","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2019-05-03 15:11:38.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"eefdae6ba312d24bc3b3ec7c7acd370334f14366","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2019-05-03 15:12:16.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"8dff1c6fb90d98ac17b9efe5eec325c22805a5ba","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2019-05-03 16:04:04.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"2ce2e81bdd56a87403f5312a0c6a2dcec8ecf949","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2019-05-03 16:06:00.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"887d779677ac5fc529f3b3cf4b24c51bd8f2f26d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 17:19:44.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/44/631644/14/check/openstack-ansible-linters/1e3c122/ : FAILURE in 6m 21s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/44/631644/14/check/openstack-ansible-deploy-aio_metal-debian-stable/37d95d2/ : SUCCESS in 57m 35s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/44/631644/14/check/openstack-ansible-deploy-aio_metal-centos-7/948210c/ : SUCCESS in 1h 11m 20s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/44/631644/14/check/openstack-ansible-deploy-aio_metal-opensuse-150/30ad3d0/ : SUCCESS in 57m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/44/631644/14/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/edef7a8/ : SUCCESS in 51m 48s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/44/631644/14/check/openstack-ansible-deploy-aio_distro_metal-centos-7/bf3a54f/ : FAILURE in 43m 56s\n- openstack-tox-docs http://logs.openstack.org/44/631644/14/check/openstack-tox-docs/8d2326a/html/ : SUCCESS in 2m 45s\n- openstack-ansible-ovs-ubuntu-bionic http://logs.openstack.org/44/631644/14/check/openstack-ansible-ovs-ubuntu-bionic/7c26272/ : SUCCESS in 39m 10s\n- openstack-ansible-ovn-ubuntu-bionic http://logs.openstack.org/44/631644/14/check/openstack-ansible-ovn-ubuntu-bionic/c891a2e/ : SUCCESS in 38m 07s\n- openstack-ansible-ovn-centos-7 http://logs.openstack.org/44/631644/14/check/openstack-ansible-ovn-centos-7/a37127a/ : FAILURE in 8m 36s\n- openstack-ansible-calico-ubuntu-bionic http://logs.openstack.org/44/631644/14/check/openstack-ansible-calico-ubuntu-bionic/fae89b7/ : FAILURE in 39m 20s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic http://logs.openstack.org/44/631644/14/check/openstack-ansible-opendaylight-ubuntu-bionic/db90fbf/ : SUCCESS in 56m 22s\n- openstack-ansible-ovs-nsh-ubuntu-bionic http://logs.openstack.org/44/631644/14/check/openstack-ansible-ovs-nsh-ubuntu-bionic/c8416ed/ : SUCCESS in 56m 52s\n- openstack-ansible-odl-sfc-ubuntu-bionic http://logs.openstack.org/44/631644/14/check/openstack-ansible-odl-sfc-ubuntu-bionic/c54ff4c/ : SUCCESS in 54m 08s\n- openstack-ansible-ovs-sfc-ubuntu-bionic http://logs.openstack.org/44/631644/14/check/openstack-ansible-ovs-sfc-ubuntu-bionic/e2ab061/ : SUCCESS in 38m 17s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic http://logs.openstack.org/44/631644/14/check/openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic/9756961/ : FAILURE in 56m 32s (non-voting)\n- openstack-ansible-ovs-fwaasv2-ubuntu-bionic http://logs.openstack.org/44/631644/14/check/openstack-ansible-ovs-fwaasv2-ubuntu-bionic/db1c140/ : FAILURE in 33m 45s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"6008ec6e52670e2adc1c2711195f558f1a16559e","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2019-05-04 14:48:10.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"ba36aeb49ae7ccdb5559c9ab6895109e679dc980","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2019-05-04 15:41:05.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"25cbb26124e0e5ca5ab8453c45036f707dd183b4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 16:56:41.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/44/631644/15/check/openstack-ansible-linters/8777c7b/ : FAILURE in 8m 38s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/44/631644/15/check/openstack-ansible-deploy-aio_metal-debian-stable/00b1807/ : SUCCESS in 58m 51s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/44/631644/15/check/openstack-ansible-deploy-aio_metal-centos-7/502448b/ : SUCCESS in 1h 13m 09s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/44/631644/15/check/openstack-ansible-deploy-aio_metal-opensuse-150/9897481/ : SUCCESS in 1h 03m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/44/631644/15/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/28a7768/ : SUCCESS in 56m 15s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/44/631644/15/check/openstack-ansible-deploy-aio_distro_metal-centos-7/6f7d605/ : FAILURE in 39m 55s\n- openstack-tox-docs http://logs.openstack.org/44/631644/15/check/openstack-tox-docs/9453838/html/ : SUCCESS in 2m 22s\n- openstack-ansible-ovs-ubuntu-bionic http://logs.openstack.org/44/631644/15/check/openstack-ansible-ovs-ubuntu-bionic/a83e53d/ : SUCCESS in 35m 54s\n- openstack-ansible-ovn-ubuntu-bionic http://logs.openstack.org/44/631644/15/check/openstack-ansible-ovn-ubuntu-bionic/9d25607/ : SUCCESS in 38m 48s\n- openstack-ansible-ovn-centos-7 http://logs.openstack.org/44/631644/15/check/openstack-ansible-ovn-centos-7/5aaf0b1/ : FAILURE in 30m 46s\n- openstack-ansible-calico-ubuntu-bionic http://logs.openstack.org/44/631644/15/check/openstack-ansible-calico-ubuntu-bionic/df151b3/ : FAILURE in 36m 17s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic http://logs.openstack.org/44/631644/15/check/openstack-ansible-opendaylight-ubuntu-bionic/358599a/ : SUCCESS in 51m 57s\n- openstack-ansible-ovs-nsh-ubuntu-bionic http://logs.openstack.org/44/631644/15/check/openstack-ansible-ovs-nsh-ubuntu-bionic/7aeeb9f/ : SUCCESS in 53m 07s\n- openstack-ansible-odl-sfc-ubuntu-bionic http://logs.openstack.org/44/631644/15/check/openstack-ansible-odl-sfc-ubuntu-bionic/c37309a/ : FAILURE in 9m 27s\n- openstack-ansible-ovs-sfc-ubuntu-bionic http://logs.openstack.org/44/631644/15/check/openstack-ansible-ovs-sfc-ubuntu-bionic/460739b/ : SUCCESS in 37m 55s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic http://logs.openstack.org/44/631644/15/check/openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic/5e716e3/ : FAILURE in 44m 55s (non-voting)\n- openstack-ansible-ovs-fwaasv2-ubuntu-bionic http://logs.openstack.org/44/631644/15/check/openstack-ansible-ovs-fwaasv2-ubuntu-bionic/7718c80/ : FAILURE in 36m 36s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"1ae13483cce7e082c7b0567ed71e82d37a919ee6","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-04-27 14:26:10.000000000","message":"Abandoned\n\n\u003cjamesdenton\u003e so, feel free to close","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"a817800adf9ba221094b4cf74aec3fc0a13313f4","revisions":{"5d0091e07bafa3d374f277b3331722a32c627482":{"kind":"REWORK","_number":1,"created":"2019-01-17 21:15:57.000000000","uploader":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"ref":"refs/changes/44/631644/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/44/631644/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/44/631644/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/44/631644/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/44/631644/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/44/631644/1"}}},"commit":{"parents":[{"commit":"9b6a7bf9febea3d64695a879c0b057d16d957932","subject":"Merge \"Fix whitespace in neutron.conf template\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/9b6a7bf9febea3d64695a879c0b057d16d957932"}]}],"author":{"name":"James Denton","email":"james.denton@rackspace.com","date":"2019-01-17 21:12:21.000000000","tz":0},"committer":{"name":"James Denton","email":"james.denton@rackspace.com","date":"2019-01-17 21:15:56.000000000","tz":0},"subject":"[WIP] Deploy Vector Packet Processing (VPP) Platform for Neutron","message":"[WIP] Deploy Vector Packet Processing (VPP) Platform for Neutron\n\nThis patch set aims to deploy and configure VPP and the corresponding\nML2 driver for Neutron, as well as etcd and other dependencies.\nExperimental - Do Not Merge!\n\nChange-Id: Ie0a4a8c9001eb1c90f06b2f16a6be078cb9176e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/5d0091e07bafa3d374f277b3331722a32c627482"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/5d0091e07bafa3d374f277b3331722a32c627482"}]},"branch":"refs/heads/master"},"b32def7d3d2119f10e55bdf58599dc5853e09049":{"kind":"REWORK","_number":2,"created":"2019-01-21 19:30:45.000000000","uploader":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"ref":"refs/changes/44/631644/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/44/631644/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/44/631644/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/44/631644/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/44/631644/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/44/631644/2"}}},"commit":{"parents":[{"commit":"9b6a7bf9febea3d64695a879c0b057d16d957932","subject":"Merge \"Fix whitespace in neutron.conf template\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/9b6a7bf9febea3d64695a879c0b057d16d957932"}]}],"author":{"name":"James Denton","email":"james.denton@rackspace.com","date":"2019-01-17 21:12:21.000000000","tz":0},"committer":{"name":"James Denton","email":"james.denton@rackspace.com","date":"2019-01-21 19:30:30.000000000","tz":0},"subject":"[WIP] Deploy Vector Packet Processing (VPP) Platform for Neutron","message":"[WIP] Deploy Vector Packet Processing (VPP) Platform for Neutron\n\nThis patch set aims to deploy and configure VPP and the corresponding\nML2 driver for Neutron, as well as etcd and other dependencies.\nExperimental - Do Not Merge!\n\nChange-Id: Ie0a4a8c9001eb1c90f06b2f16a6be078cb9176e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/b32def7d3d2119f10e55bdf58599dc5853e09049"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/b32def7d3d2119f10e55bdf58599dc5853e09049"}]},"branch":"refs/heads/master"},"ceb530d2dd0751485aef64cbdbc8d65e96816872":{"kind":"REWORK","_number":3,"created":"2019-01-23 04:17:15.000000000","uploader":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"ref":"refs/changes/44/631644/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/44/631644/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/44/631644/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/44/631644/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/44/631644/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/44/631644/3"}}},"commit":{"parents":[{"commit":"9b6a7bf9febea3d64695a879c0b057d16d957932","subject":"Merge \"Fix whitespace in neutron.conf template\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/9b6a7bf9febea3d64695a879c0b057d16d957932"}]}],"author":{"name":"James Denton","email":"james.denton@rackspace.com","date":"2019-01-17 21:12:21.000000000","tz":0},"committer":{"name":"James Denton","email":"james.denton@rackspace.com","date":"2019-01-23 04:17:14.000000000","tz":0},"subject":"[WIP] Deploy Vector Packet Processing (VPP) Platform for Neutron","message":"[WIP] Deploy Vector Packet Processing (VPP) Platform for Neutron\n\nThis patch set aims to deploy and configure VPP and the corresponding\nML2 driver for Neutron, as well as etcd and other dependencies.\nExperimental - Do Not Merge!\n\nChange-Id: Ie0a4a8c9001eb1c90f06b2f16a6be078cb9176e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/ceb530d2dd0751485aef64cbdbc8d65e96816872"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/ceb530d2dd0751485aef64cbdbc8d65e96816872"}]},"branch":"refs/heads/master"},"d1e671fc60d4f1c1dfcee6d5471777eb9d9f99a8":{"kind":"REWORK","_number":4,"created":"2019-01-23 16:54:23.000000000","uploader":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"ref":"refs/changes/44/631644/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/44/631644/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/44/631644/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/44/631644/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/44/631644/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/44/631644/4"}}},"commit":{"parents":[{"commit":"9b6a7bf9febea3d64695a879c0b057d16d957932","subject":"Merge \"Fix whitespace in neutron.conf template\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/9b6a7bf9febea3d64695a879c0b057d16d957932"}]}],"author":{"name":"James Denton","email":"james.denton@rackspace.com","date":"2019-01-17 21:12:21.000000000","tz":0},"committer":{"name":"James Denton","email":"james.denton@rackspace.com","date":"2019-01-23 16:54:07.000000000","tz":0},"subject":"[WIP] Deploy Vector Packet Processing (VPP) Platform for Neutron","message":"[WIP] Deploy Vector Packet Processing (VPP) Platform for Neutron\n\nThis patch set aims to deploy and configure VPP and the corresponding\nML2 driver for Neutron, as well as etcd and other dependencies.\nExperimental - Do Not Merge!\n\nChange-Id: Ie0a4a8c9001eb1c90f06b2f16a6be078cb9176e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/d1e671fc60d4f1c1dfcee6d5471777eb9d9f99a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/d1e671fc60d4f1c1dfcee6d5471777eb9d9f99a8"}]},"branch":"refs/heads/master"},"7a00b5809e5a1485307d8d66caa1b12863c76d1b":{"kind":"REWORK","_number":5,"created":"2019-01-23 16:56:28.000000000","uploader":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"ref":"refs/changes/44/631644/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/44/631644/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/44/631644/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/44/631644/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/44/631644/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/44/631644/5"}}},"commit":{"parents":[{"commit":"9b6a7bf9febea3d64695a879c0b057d16d957932","subject":"Merge \"Fix whitespace in neutron.conf template\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/9b6a7bf9febea3d64695a879c0b057d16d957932"}]}],"author":{"name":"James Denton","email":"james.denton@rackspace.com","date":"2019-01-17 21:12:21.000000000","tz":0},"committer":{"name":"James Denton","email":"james.denton@rackspace.com","date":"2019-01-23 16:56:20.000000000","tz":0},"subject":"[WIP] Deploy Vector Packet Processing (VPP) Platform for Neutron","message":"[WIP] Deploy Vector Packet Processing (VPP) Platform for Neutron\n\nThis patch set aims to deploy and configure VPP and the corresponding\nML2 driver for Neutron, as well as etcd and other dependencies.\nExperimental - Do Not Merge!\n\nChange-Id: Ie0a4a8c9001eb1c90f06b2f16a6be078cb9176e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/7a00b5809e5a1485307d8d66caa1b12863c76d1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/7a00b5809e5a1485307d8d66caa1b12863c76d1b"}]},"branch":"refs/heads/master"},"b6257cfb92a9da967ff5c1e2f82cc006331c96ae":{"kind":"REWORK","_number":6,"created":"2019-01-24 15:22:11.000000000","uploader":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"ref":"refs/changes/44/631644/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/44/631644/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/44/631644/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/44/631644/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/44/631644/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/44/631644/6"}}},"commit":{"parents":[{"commit":"9b6a7bf9febea3d64695a879c0b057d16d957932","subject":"Merge \"Fix whitespace in neutron.conf template\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/9b6a7bf9febea3d64695a879c0b057d16d957932"}]}],"author":{"name":"James Denton","email":"james.denton@rackspace.com","date":"2019-01-17 21:12:21.000000000","tz":0},"committer":{"name":"James Denton","email":"james.denton@rackspace.com","date":"2019-01-24 15:21:59.000000000","tz":0},"subject":"[WIP] Deploy Vector Packet Processing (VPP) Platform for Neutron","message":"[WIP] Deploy Vector Packet Processing (VPP) Platform for Neutron\n\nThis patch set aims to deploy and configure VPP and the corresponding\nML2 driver for Neutron, as well as etcd and other dependencies.\nExperimental - Do Not Merge!\n\nChange-Id: Ie0a4a8c9001eb1c90f06b2f16a6be078cb9176e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/b6257cfb92a9da967ff5c1e2f82cc006331c96ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/b6257cfb92a9da967ff5c1e2f82cc006331c96ae"}]},"branch":"refs/heads/master"},"14e9a02950eba7c6f8d999d659a885a3080a7953":{"kind":"REWORK","_number":7,"created":"2019-01-24 15:54:51.000000000","uploader":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"ref":"refs/changes/44/631644/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/44/631644/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/44/631644/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/44/631644/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/44/631644/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/44/631644/7"}}},"commit":{"parents":[{"commit":"9b6a7bf9febea3d64695a879c0b057d16d957932","subject":"Merge \"Fix whitespace in neutron.conf template\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/9b6a7bf9febea3d64695a879c0b057d16d957932"}]}],"author":{"name":"James Denton","email":"james.denton@rackspace.com","date":"2019-01-17 21:12:21.000000000","tz":0},"committer":{"name":"James Denton","email":"james.denton@rackspace.com","date":"2019-01-24 15:54:41.000000000","tz":0},"subject":"[WIP] Deploy Vector Packet Processing (VPP) Platform for Neutron","message":"[WIP] Deploy Vector Packet Processing (VPP) Platform for Neutron\n\nThis patch set aims to deploy and configure VPP and the corresponding\nML2 driver for Neutron, as well as etcd and other dependencies.\nExperimental - Do Not Merge!\n\nChange-Id: Ie0a4a8c9001eb1c90f06b2f16a6be078cb9176e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/14e9a02950eba7c6f8d999d659a885a3080a7953"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/14e9a02950eba7c6f8d999d659a885a3080a7953"}]},"branch":"refs/heads/master"},"c565525efadc45fbaf8ed5b746f73797de1fd7b8":{"kind":"REWORK","_number":8,"created":"2019-01-25 16:25:55.000000000","uploader":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"ref":"refs/changes/44/631644/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/44/631644/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/44/631644/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/44/631644/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/44/631644/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/44/631644/8"}}},"commit":{"parents":[{"commit":"e96c4448eac75690a5f93bf49336bef684b7a387","subject":"Merge \"Provide support for ovs-sfc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/e96c4448eac75690a5f93bf49336bef684b7a387"}]}],"author":{"name":"James Denton","email":"james.denton@rackspace.com","date":"2019-01-17 21:12:21.000000000","tz":0},"committer":{"name":"James Denton","email":"james.denton@rackspace.com","date":"2019-01-25 16:24:59.000000000","tz":0},"subject":"[WIP] Deploy Vector Packet Processing (VPP) Platform for Neutron","message":"[WIP] Deploy Vector Packet Processing (VPP) Platform for Neutron\n\nThis patch set aims to deploy and configure VPP and the corresponding\nML2 driver for Neutron as well other dependencies.\n\nExperimental - Do Not Merge!\n\nDepends-On: https://review.openstack.org/#/c/620386/\nChange-Id: Ie0a4a8c9001eb1c90f06b2f16a6be078cb9176e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/c565525efadc45fbaf8ed5b746f73797de1fd7b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/c565525efadc45fbaf8ed5b746f73797de1fd7b8"}]},"branch":"refs/heads/master"},"e2f9dfb133e447e0aff0159faf4ff13e31df6e9a":{"kind":"REWORK","_number":9,"created":"2019-01-28 18:37:45.000000000","uploader":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"ref":"refs/changes/44/631644/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/44/631644/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/44/631644/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/44/631644/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/44/631644/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/44/631644/9"}}},"commit":{"parents":[{"commit":"e96c4448eac75690a5f93bf49336bef684b7a387","subject":"Merge \"Provide support for ovs-sfc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/e96c4448eac75690a5f93bf49336bef684b7a387"}]}],"author":{"name":"James Denton","email":"james.denton@rackspace.com","date":"2019-01-17 21:12:21.000000000","tz":0},"committer":{"name":"James Denton","email":"james.denton@rackspace.com","date":"2019-01-28 18:37:44.000000000","tz":0},"subject":"[WIP] Deploy Vector Packet Processing (VPP) Platform for Neutron","message":"[WIP] Deploy Vector Packet Processing (VPP) Platform for Neutron\n\nThis patch set aims to deploy and configure VPP and the corresponding\nML2 driver for Neutron as well other dependencies.\n\nExperimental - Do Not Merge!\n\nChange-Id: Ie0a4a8c9001eb1c90f06b2f16a6be078cb9176e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/e2f9dfb133e447e0aff0159faf4ff13e31df6e9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/e2f9dfb133e447e0aff0159faf4ff13e31df6e9a"}]},"branch":"refs/heads/master"},"21763b335dadd4efce06130729506a08fa1923e6":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2019-02-26 03:15:14.000000000","uploader":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"ref":"refs/changes/44/631644/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/44/631644/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/44/631644/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/44/631644/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/44/631644/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/44/631644/10"}}},"commit":{"parents":[{"commit":"5e5a206c01598d4f14b719a4b521ff0de65c1eba","subject":"Remove the private option from include_role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/5e5a206c01598d4f14b719a4b521ff0de65c1eba"}]}],"author":{"name":"James Denton","email":"james.denton@rackspace.com","date":"2019-01-17 21:12:21.000000000","tz":0},"committer":{"name":"James Denton","email":"james.denton@rackspace.com","date":"2019-02-26 03:15:14.000000000","tz":0},"subject":"[WIP] Deploy Vector Packet Processing (VPP) Platform for Neutron","message":"[WIP] Deploy Vector Packet Processing (VPP) Platform for Neutron\n\nThis patch set aims to deploy and configure VPP and the corresponding\nML2 driver for Neutron as well other dependencies.\n\nExperimental - Do Not Merge!\n\nChange-Id: Ie0a4a8c9001eb1c90f06b2f16a6be078cb9176e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/21763b335dadd4efce06130729506a08fa1923e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/21763b335dadd4efce06130729506a08fa1923e6"}]},"branch":"refs/heads/master"},"87d20ffe19bb277e6087bab234b64f5508af653e":{"kind":"REWORK","_number":11,"created":"2019-05-03 15:11:38.000000000","uploader":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"ref":"refs/changes/44/631644/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/44/631644/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/44/631644/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/44/631644/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/44/631644/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/44/631644/11"}}},"commit":{"parents":[{"commit":"d2076235d6b026e2d5dbb0967baef25f33b5a32c","subject":"Merge \"Swap special CI jobs from xenial to bionic\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/d2076235d6b026e2d5dbb0967baef25f33b5a32c"}]}],"author":{"name":"James Denton","email":"james.denton@outlook.com","date":"2019-01-17 21:12:21.000000000","tz":0},"committer":{"name":"James Denton","email":"james.denton@outlook.com","date":"2019-05-03 15:11:23.000000000","tz":0},"subject":"[WIP] Deploy Vector Packet Processing (VPP) Platform for Neutron","message":"[WIP] Deploy Vector Packet Processing (VPP) Platform for Neutron\n\nThis patch set aims to deploy and configure VPP and the corresponding\nML2 driver for Neutron as well other dependencies.\n\nExperimental - Do Not Merge!\n\nChange-Id: Ie0a4a8c9001eb1c90f06b2f16a6be078cb9176e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/87d20ffe19bb277e6087bab234b64f5508af653e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/87d20ffe19bb277e6087bab234b64f5508af653e"}]},"branch":"refs/heads/master"},"102748437b585f44a1cd230717eb05540aa2bdac":{"kind":"REWORK","_number":12,"created":"2019-05-03 15:12:16.000000000","uploader":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"ref":"refs/changes/44/631644/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/44/631644/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/44/631644/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/44/631644/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/44/631644/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/44/631644/12"}}},"commit":{"parents":[{"commit":"d2076235d6b026e2d5dbb0967baef25f33b5a32c","subject":"Merge \"Swap special CI jobs from xenial to bionic\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/d2076235d6b026e2d5dbb0967baef25f33b5a32c"}]}],"author":{"name":"James Denton","email":"james.denton@outlook.com","date":"2019-01-17 21:12:21.000000000","tz":0},"committer":{"name":"James Denton","email":"james.denton@outlook.com","date":"2019-05-03 15:12:02.000000000","tz":0},"subject":"[WIP] Deploy Vector Packet Processing (VPP) Platform for Neutron","message":"[WIP] Deploy Vector Packet Processing (VPP) Platform for Neutron\n\nThis patch set aims to deploy and configure VPP and the corresponding\nML2 driver for Neutron as well other dependencies.\n\nExperimental - Do Not Merge!\n\nChange-Id: Ie0a4a8c9001eb1c90f06b2f16a6be078cb9176e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/102748437b585f44a1cd230717eb05540aa2bdac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/102748437b585f44a1cd230717eb05540aa2bdac"}]},"branch":"refs/heads/master"},"e2bf1a10c2e37a08914b8782c00d32cf329b404b":{"kind":"REWORK","_number":13,"created":"2019-05-03 16:04:04.000000000","uploader":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"ref":"refs/changes/44/631644/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/44/631644/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/44/631644/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/44/631644/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/44/631644/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/44/631644/13"}}},"commit":{"parents":[{"commit":"d2076235d6b026e2d5dbb0967baef25f33b5a32c","subject":"Merge \"Swap special CI jobs from xenial to bionic\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/d2076235d6b026e2d5dbb0967baef25f33b5a32c"}]}],"author":{"name":"James Denton","email":"james.denton@outlook.com","date":"2019-01-17 21:12:21.000000000","tz":0},"committer":{"name":"James Denton","email":"james.denton@outlook.com","date":"2019-05-03 16:03:52.000000000","tz":0},"subject":"[WIP] Deploy Vector Packet Processing (VPP) Platform for Neutron","message":"[WIP] Deploy Vector Packet Processing (VPP) Platform for Neutron\n\nThis patch set aims to deploy and configure VPP and the corresponding\nML2 driver for Neutron as well other dependencies.\n\nExperimental - Do Not Merge!\n\nChange-Id: Ie0a4a8c9001eb1c90f06b2f16a6be078cb9176e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/e2bf1a10c2e37a08914b8782c00d32cf329b404b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/e2bf1a10c2e37a08914b8782c00d32cf329b404b"}]},"branch":"refs/heads/master"},"fd989aa7f6578c914b03c715602a72a30ecf388f":{"kind":"REWORK","_number":14,"created":"2019-05-03 16:06:00.000000000","uploader":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"ref":"refs/changes/44/631644/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/44/631644/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/44/631644/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/44/631644/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/44/631644/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/44/631644/14"}}},"commit":{"parents":[{"commit":"d2076235d6b026e2d5dbb0967baef25f33b5a32c","subject":"Merge \"Swap special CI jobs from xenial to bionic\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/d2076235d6b026e2d5dbb0967baef25f33b5a32c"}]}],"author":{"name":"James Denton","email":"james.denton@outlook.com","date":"2019-01-17 21:12:21.000000000","tz":0},"committer":{"name":"James Denton","email":"james.denton@outlook.com","date":"2019-05-03 16:05:52.000000000","tz":0},"subject":"[WIP] Deploy Vector Packet Processing (VPP) Platform for Neutron","message":"[WIP] Deploy Vector Packet Processing (VPP) Platform for Neutron\n\nThis patch set aims to deploy and configure VPP and the corresponding\nML2 driver for Neutron as well other dependencies.\n\nExperimental - Do Not Merge!\n\nChange-Id: Ie0a4a8c9001eb1c90f06b2f16a6be078cb9176e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/fd989aa7f6578c914b03c715602a72a30ecf388f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/fd989aa7f6578c914b03c715602a72a30ecf388f"}]},"branch":"refs/heads/master"},"a817800adf9ba221094b4cf74aec3fc0a13313f4":{"kind":"REWORK","_number":15,"created":"2019-05-04 15:41:05.000000000","uploader":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"ref":"refs/changes/44/631644/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/44/631644/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/44/631644/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/44/631644/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/44/631644/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/44/631644/15"}}},"commit":{"parents":[{"commit":"d2076235d6b026e2d5dbb0967baef25f33b5a32c","subject":"Merge \"Swap special CI jobs from xenial to bionic\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/d2076235d6b026e2d5dbb0967baef25f33b5a32c"}]}],"author":{"name":"James Denton","email":"james.denton@outlook.com","date":"2019-01-17 21:12:21.000000000","tz":0},"committer":{"name":"James Denton","email":"james.denton@outlook.com","date":"2019-05-04 15:40:57.000000000","tz":0},"subject":"[WIP] Deploy Vector Packet Processing (VPP) Platform for Neutron","message":"[WIP] Deploy Vector Packet Processing (VPP) Platform for Neutron\n\nThis patch set aims to deploy and configure VPP and the corresponding\nML2 driver for Neutron as well other dependencies.\n\nExperimental - Do Not Merge!\n\nChange-Id: Ie0a4a8c9001eb1c90f06b2f16a6be078cb9176e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/a817800adf9ba221094b4cf74aec3fc0a13313f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/a817800adf9ba221094b4cf74aec3fc0a13313f4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
