)]}'
{"id":"openstack%2Fopenstack-ansible-os_neutron~652641","triplet_id":"openstack%2Fopenstack-ansible-os_neutron~master~I37d4bb79f4441932008a0f43c4ab3dead5fc96d6","project":"openstack/openstack-ansible-os_neutron","branch":"master","topic":"652641","hashtags":[],"change_id":"I37d4bb79f4441932008a0f43c4ab3dead5fc96d6","subject":"Swap special CI jobs from xenial to bionic","status":"MERGED","created":"2019-04-15 12:47:21.000000000","updated":"2019-04-24 09:11:23.000000000","submitted":"2019-04-24 09:11:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"652641-1556097083958-11ac1cc1","meta_rev_id":"14b0aac2faaa6acc57a4bfe02c94bba75892159e","_number":652641,"virtual_id_number":652641,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},{"value":2,"date":"2019-04-24 09:11:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"all":[{"value":2,"date":"2019-04-17 13:02:50.000000000","_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":2,"date":"2019-04-24 07:27:59.000000000","_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":1,"date":"2019-04-24 07:28:06.000000000","_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-17 13:02:50.000000000","updated_by":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"reviewer":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"state":"REVIEWER"},{"updated":"2019-04-24 07:28:06.000000000","updated_by":{"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},"reviewer":{"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},"state":"REVIEWER"},{"updated":"2019-04-24 09:11:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"87b4d5b83f255bc75e8ff67217f7738a3cd70a01","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-04-15 12:47:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"55579e2d5fa4ef989961a5ccb3c26dd8a4d5402d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-15 14:20:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/41/652641/1/check/openstack-ansible-linters/98f871a/ : SUCCESS in 5m 20s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/41/652641/1/check/openstack-ansible-deploy-aio_metal-debian-stable/b0eb90e/ : SUCCESS in 54m 01s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/41/652641/1/check/openstack-ansible-deploy-aio_metal-centos-7/7ed2ae0/ : SUCCESS in 1h 16m 48s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/41/652641/1/check/openstack-ansible-deploy-aio_metal-opensuse-150/2f9e5e3/ : SUCCESS in 1h 13m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/41/652641/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/e157a72/ : SUCCESS in 1h 00m 36s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/41/652641/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/5771fa5/ : SUCCESS in 56m 54s\n- openstack-tox-docs http://logs.openstack.org/41/652641/1/check/openstack-tox-docs/a871fa2/html/ : SUCCESS in 3m 07s\n- openstack-ansible-ovs-ubuntu-bionic http://logs.openstack.org/41/652641/1/check/openstack-ansible-ovs-ubuntu-bionic/883632f/ : SUCCESS in 38m 18s\n- openstack-ansible-ovn-ubuntu-bionic http://logs.openstack.org/41/652641/1/check/openstack-ansible-ovn-ubuntu-bionic/37242fe/ : SUCCESS in 44m 48s\n- openstack-ansible-ovn-centos-7 http://logs.openstack.org/41/652641/1/check/openstack-ansible-ovn-centos-7/a8d01bd/ : SUCCESS in 40m 35s\n- openstack-ansible-calico-ubuntu-bionic http://logs.openstack.org/41/652641/1/check/openstack-ansible-calico-ubuntu-bionic/518036a/ : FAILURE in 46m 05s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic http://logs.openstack.org/41/652641/1/check/openstack-ansible-opendaylight-ubuntu-bionic/c7f2f7b/ : FAILURE in 28m 38s\n- openstack-ansible-ovs-nsh-ubuntu-bionic http://logs.openstack.org/41/652641/1/check/openstack-ansible-ovs-nsh-ubuntu-bionic/6a07a47/ : FAILURE in 31m 13s\n- openstack-ansible-odl-sfc-ubuntu-bionic http://logs.openstack.org/41/652641/1/check/openstack-ansible-odl-sfc-ubuntu-bionic/c9ea98b/ : FAILURE in 30m 58s\n- openstack-ansible-ovs-sfc-ubuntu-bionic http://logs.openstack.org/41/652641/1/check/openstack-ansible-ovs-sfc-ubuntu-bionic/642ab5c/ : SUCCESS in 43m 51s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic http://logs.openstack.org/41/652641/1/check/openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic/809017b/ : FAILURE in 28m 48s (non-voting)\n- openstack-ansible-ovs-fwaasv2-ubuntu-bionic http://logs.openstack.org/41/652641/1/check/openstack-ansible-ovs-fwaasv2-ubuntu-bionic/e8bb91c/ : FAILURE in 38m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"548d0b36f1f0343f97d072bbe76edc85eecf6292","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-04-15 16:44:14.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"4b40a34ab99c3ccb1a2adc95e8313c4effd0f9ce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-15 18:22:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/41/652641/2/check/openstack-ansible-linters/b6612c2/ : SUCCESS in 6m 27s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/41/652641/2/check/openstack-ansible-deploy-aio_metal-debian-stable/ef86c23/ : SUCCESS in 1h 03m 27s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/41/652641/2/check/openstack-ansible-deploy-aio_metal-centos-7/71a5d10/ : SUCCESS in 1h 21m 05s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/41/652641/2/check/openstack-ansible-deploy-aio_metal-opensuse-150/475b662/ : SUCCESS in 1h 05m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/41/652641/2/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/d1c2c67/ : SUCCESS in 58m 13s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/41/652641/2/check/openstack-ansible-deploy-aio_distro_metal-centos-7/65335cb/ : SUCCESS in 55m 04s\n- openstack-tox-docs http://logs.openstack.org/41/652641/2/check/openstack-tox-docs/e4240af/html/ : SUCCESS in 2m 42s\n- openstack-ansible-ovs-ubuntu-bionic http://logs.openstack.org/41/652641/2/check/openstack-ansible-ovs-ubuntu-bionic/672dbd7/ : SUCCESS in 38m 40s\n- openstack-ansible-ovn-ubuntu-bionic http://logs.openstack.org/41/652641/2/check/openstack-ansible-ovn-ubuntu-bionic/86a3036/ : SUCCESS in 38m 21s\n- openstack-ansible-ovn-centos-7 http://logs.openstack.org/41/652641/2/check/openstack-ansible-ovn-centos-7/57bfbe1/ : SUCCESS in 41m 00s\n- openstack-ansible-calico-ubuntu-bionic http://logs.openstack.org/41/652641/2/check/openstack-ansible-calico-ubuntu-bionic/4028c5c/ : FAILURE in 40m 18s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic http://logs.openstack.org/41/652641/2/check/openstack-ansible-opendaylight-ubuntu-bionic/12ee052/ : FAILURE in 25m 56s\n- openstack-ansible-ovs-nsh-ubuntu-bionic http://logs.openstack.org/41/652641/2/check/openstack-ansible-ovs-nsh-ubuntu-bionic/dd8af75/ : FAILURE in 24m 26s\n- openstack-ansible-odl-sfc-ubuntu-bionic http://logs.openstack.org/41/652641/2/check/openstack-ansible-odl-sfc-ubuntu-bionic/3bb53a4/ : FAILURE in 25m 09s\n- openstack-ansible-ovs-sfc-ubuntu-bionic http://logs.openstack.org/41/652641/2/check/openstack-ansible-ovs-sfc-ubuntu-bionic/615dd41/ : SUCCESS in 40m 27s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic finger://ze01.openstack.org/b50dcc1e714945c691bfca450d8503c3 : RETRY_LIMIT in 15m 20s (non-voting)\n- openstack-ansible-ovs-fwaasv2-ubuntu-bionic http://logs.openstack.org/41/652641/2/check/openstack-ansible-ovs-fwaasv2-ubuntu-bionic/5966dab/ : FAILURE in 35m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bfcea8448d988f1f411619b06a4570f2b4cec245","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-04-16 13:01:12.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"65feefbc5ee2de886847b384dfcc8bc506305bba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-16 14:26:17.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/41/652641/3/check/openstack-ansible-linters/edbfa24/ : SUCCESS in 6m 08s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/41/652641/3/check/openstack-ansible-deploy-aio_metal-debian-stable/4587fe0/ : SUCCESS in 56m 35s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/41/652641/3/check/openstack-ansible-deploy-aio_metal-centos-7/e58cc22/ : SUCCESS in 1h 22m 49s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/41/652641/3/check/openstack-ansible-deploy-aio_metal-opensuse-150/8e3138e/ : SUCCESS in 56m 01s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/41/652641/3/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/f0bec4e/ : SUCCESS in 52m 07s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/41/652641/3/check/openstack-ansible-deploy-aio_distro_metal-centos-7/73ccd07/ : SUCCESS in 51m 18s\n- openstack-tox-docs http://logs.openstack.org/41/652641/3/check/openstack-tox-docs/021ffce/html/ : SUCCESS in 3m 30s\n- openstack-ansible-ovs-ubuntu-bionic http://logs.openstack.org/41/652641/3/check/openstack-ansible-ovs-ubuntu-bionic/556e79a/ : SUCCESS in 36m 13s\n- openstack-ansible-ovn-ubuntu-bionic http://logs.openstack.org/41/652641/3/check/openstack-ansible-ovn-ubuntu-bionic/d5150d1/ : SUCCESS in 37m 54s\n- openstack-ansible-ovn-centos-7 http://logs.openstack.org/41/652641/3/check/openstack-ansible-ovn-centos-7/bf3b671/ : SUCCESS in 40m 40s\n- openstack-ansible-calico-ubuntu-bionic http://logs.openstack.org/41/652641/3/check/openstack-ansible-calico-ubuntu-bionic/ac74b05/ : FAILURE in 38m 55s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic http://logs.openstack.org/41/652641/3/check/openstack-ansible-opendaylight-ubuntu-bionic/1655889/ : SUCCESS in 1h 03m 10s\n- openstack-ansible-ovs-nsh-ubuntu-bionic http://logs.openstack.org/41/652641/3/check/openstack-ansible-ovs-nsh-ubuntu-bionic/5f45d1d/ : SUCCESS in 1h 08m 57s\n- openstack-ansible-odl-sfc-ubuntu-bionic http://logs.openstack.org/41/652641/3/check/openstack-ansible-odl-sfc-ubuntu-bionic/bb5a80e/ : SUCCESS in 1h 05m 15s\n- openstack-ansible-ovs-sfc-ubuntu-bionic http://logs.openstack.org/41/652641/3/check/openstack-ansible-ovs-sfc-ubuntu-bionic/7dfb9ce/ : SUCCESS in 39m 05s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic http://logs.openstack.org/41/652641/3/check/openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic/25ed063/ : FAILURE in 57m 13s (non-voting)\n- openstack-ansible-ovs-fwaasv2-ubuntu-bionic http://logs.openstack.org/41/652641/3/check/openstack-ansible-ovs-fwaasv2-ubuntu-bionic/1f71849/ : FAILURE in 42m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"43465d091f88f71f3ee9e39d78e82036e0f4ab39","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-04-17 13:02:50.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"7940df6b6230f62fd9a63874b9f14c2a46cbe255","author":{"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},"date":"2019-04-24 07:27:59.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"68f14df6b9614587e1d859fc5ca74294488faa01","author":{"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},"date":"2019-04-24 07:28:06.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"d2d211fc49a2dc04d99786401872e903ae9915a2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 07:29:20.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2f1285f8c4c864051c7e9c79e66d65bd82ac21e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 09:11:23.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/41/652641/3/gate/openstack-ansible-linters/fada355/ : SUCCESS in 6m 40s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/41/652641/3/gate/openstack-ansible-deploy-aio_metal-debian-stable/85a70e4/ : SUCCESS in 54m 20s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/41/652641/3/gate/openstack-ansible-deploy-aio_metal-centos-7/cda538c/ : SUCCESS in 1h 13m 27s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/41/652641/3/gate/openstack-ansible-deploy-aio_metal-opensuse-150/7b27ab4/ : SUCCESS in 57m 44s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/41/652641/3/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/42d4f2b/ : SUCCESS in 52m 08s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/41/652641/3/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/6587bc1/ : SUCCESS in 53m 12s\n- openstack-tox-docs http://logs.openstack.org/41/652641/3/gate/openstack-tox-docs/39f5755/html/ : SUCCESS in 3m 08s\n- openstack-ansible-ovs-ubuntu-bionic http://logs.openstack.org/41/652641/3/gate/openstack-ansible-ovs-ubuntu-bionic/715fe7c/ : SUCCESS in 37m 24s\n- openstack-ansible-ovn-ubuntu-bionic http://logs.openstack.org/41/652641/3/gate/openstack-ansible-ovn-ubuntu-bionic/13ca94a/ : SUCCESS in 38m 40s\n- openstack-ansible-ovn-centos-7 http://logs.openstack.org/41/652641/3/gate/openstack-ansible-ovn-centos-7/1ee33ee/ : SUCCESS in 41m 41s\n- openstack-ansible-opendaylight-ubuntu-bionic http://logs.openstack.org/41/652641/3/gate/openstack-ansible-opendaylight-ubuntu-bionic/dbe2a54/ : SUCCESS in 52m 49s\n- openstack-ansible-ovs-nsh-ubuntu-bionic http://logs.openstack.org/41/652641/3/gate/openstack-ansible-ovs-nsh-ubuntu-bionic/0bb24a3/ : SUCCESS in 55m 06s\n- openstack-ansible-odl-sfc-ubuntu-bionic http://logs.openstack.org/41/652641/3/gate/openstack-ansible-odl-sfc-ubuntu-bionic/0f3df1c/ : SUCCESS in 53m 35s","accounts_in_message":[],"_revision_number":3},{"id":"9d3beafffa233a7cc30e3a92562c4b7ae06ac8d8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 09:11:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d6ba36d85ba6f50bc5609639b8247b56e41d009a","revisions":{"f5ad75c788943bf61ccb6d6a7ffe73e224452c3a":{"kind":"REWORK","_number":1,"created":"2019-04-15 12:47:21.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/41/652641/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/41/652641/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/41/652641/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/41/652641/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/41/652641/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/41/652641/1"}}},"commit":{"parents":[{"commit":"7cfc21c1f475568c8f4d316c8ca8c1e5c1614779","subject":"Merge \"Avoid distro installing unused services\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/7cfc21c1f475568c8f4d316c8ca8c1e5c1614779"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-04-15 12:43:23.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-04-15 12:44:33.000000000","tz":60},"subject":"Swap special CI jobs from xenial to bionic","message":"Swap special CI jobs from xenial to bionic\n\nThe neutron role currently tests a whole bunch of network setups\non xenial, which is no longer supported for OSA since Rocky. Switch\nthese jobs over to bionic\n\nChange-Id: I37d4bb79f4441932008a0f43c4ab3dead5fc96d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/f5ad75c788943bf61ccb6d6a7ffe73e224452c3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/f5ad75c788943bf61ccb6d6a7ffe73e224452c3a"}]},"branch":"refs/heads/master"},"48b12613faee74a8470cab14f83bfd58c47f610e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-04-15 16:44:14.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/41/652641/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/41/652641/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/41/652641/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/41/652641/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/41/652641/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/41/652641/2"}}},"commit":{"parents":[{"commit":"110c3b371a2b4d3abd8c0bc2e80eef7eb1b7d202","subject":"Force the use of odl xenial repo on bionic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/110c3b371a2b4d3abd8c0bc2e80eef7eb1b7d202"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-04-15 12:43:23.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-04-15 16:43:45.000000000","tz":60},"subject":"Swap special CI jobs from xenial to bionic","message":"Swap special CI jobs from xenial to bionic\n\nThe neutron role currently tests a whole bunch of network setups\non xenial, which is no longer supported for OSA since Rocky. Switch\nthese jobs over to bionic\n\nChange-Id: I37d4bb79f4441932008a0f43c4ab3dead5fc96d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/48b12613faee74a8470cab14f83bfd58c47f610e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/48b12613faee74a8470cab14f83bfd58c47f610e"}]},"branch":"refs/heads/master"},"d6ba36d85ba6f50bc5609639b8247b56e41d009a":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-04-16 13:01:12.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/41/652641/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/41/652641/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/41/652641/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/41/652641/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/41/652641/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/41/652641/3"}}},"commit":{"parents":[{"commit":"97ed8a2ab61dd8eb20113f72a2c7e4e88f1144c2","subject":"Merge \"Fix calico service unit user/group\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/97ed8a2ab61dd8eb20113f72a2c7e4e88f1144c2"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-04-15 12:43:23.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-04-16 13:01:12.000000000","tz":0},"subject":"Swap special CI jobs from xenial to bionic","message":"Swap special CI jobs from xenial to bionic\n\nThe neutron role currently tests a whole bunch of network setups\non xenial, which is no longer supported for OSA since Rocky. Switch\nthese jobs over to bionic\n\nChange-Id: I37d4bb79f4441932008a0f43c4ab3dead5fc96d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/d6ba36d85ba6f50bc5609639b8247b56e41d009a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/d6ba36d85ba6f50bc5609639b8247b56e41d009a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
