)]}'
{"id":"openstack%2Fopenstack-ansible-os_neutron~657262","triplet_id":"openstack%2Fopenstack-ansible-os_neutron~master~I94f1a1d0d87d6d23e33ea82bbab7f3ec2a0668d3","project":"openstack/openstack-ansible-os_neutron","branch":"master","hashtags":[],"change_id":"I94f1a1d0d87d6d23e33ea82bbab7f3ec2a0668d3","subject":"Replace git.openstack.org URLs with opendev.org URLs","status":"MERGED","created":"2019-05-05 13:04:18.000000000","updated":"2019-09-20 12:05:16.000000000","submitted":"2019-09-20 12:03:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657262-1568981020839-805a2d6b","meta_rev_id":"53d541f9c0bbca47a9f445cc2504b04896434767","_number":657262,"virtual_id_number":657262,"owner":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-09-03 08:13:03.000000000","_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"date":"2019-09-14 16:44:28.000000000","_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-09-20 12:03:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":2,"date":"2019-09-13 15:48:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-09-10 12:38:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":1,"date":"2019-09-13 15:48:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-15 00:35:37.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-09-03 08:13:03.000000000","updated_by":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"reviewer":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"state":"REVIEWER"},{"updated":"2019-09-10 12:38:46.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-09-13 15:48:51.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-09-20 12:03:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6b38f4a96bff745d48e23804062e26eb0647ce36","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-05-05 13:04:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"75a24aac027c541933f2f191fdf9b90a4a897b5a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-05 14:36:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/62/657262/1/check/openstack-ansible-linters/11cd01f/ : SUCCESS in 9m 20s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/62/657262/1/check/openstack-ansible-deploy-aio_metal-debian-stable/2a2e444/ : SUCCESS in 1h 00m 21s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/62/657262/1/check/openstack-ansible-deploy-aio_metal-centos-7/3aa51aa/ : SUCCESS in 1h 21m 08s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/62/657262/1/check/openstack-ansible-deploy-aio_metal-opensuse-150/f53b35d/ : SUCCESS in 1h 12m 22s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/62/657262/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/95da589/ : SUCCESS in 49m 51s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/62/657262/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/f0f4169/ : FAILURE in 39m 46s\n- openstack-tox-docs http://logs.openstack.org/62/657262/1/check/openstack-tox-docs/f264079/html/ : SUCCESS in 2m 12s\n- openstack-ansible-ovs-ubuntu-bionic http://logs.openstack.org/62/657262/1/check/openstack-ansible-ovs-ubuntu-bionic/4ad3c42/ : SUCCESS in 45m 34s\n- openstack-ansible-ovn-ubuntu-bionic http://logs.openstack.org/62/657262/1/check/openstack-ansible-ovn-ubuntu-bionic/600ff92/ : SUCCESS in 37m 13s\n- openstack-ansible-ovn-centos-7 http://logs.openstack.org/62/657262/1/check/openstack-ansible-ovn-centos-7/bb72660/ : SUCCESS in 42m 21s\n- openstack-ansible-calico-ubuntu-bionic http://logs.openstack.org/62/657262/1/check/openstack-ansible-calico-ubuntu-bionic/8e1ab6f/ : FAILURE in 38m 38s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic http://logs.openstack.org/62/657262/1/check/openstack-ansible-opendaylight-ubuntu-bionic/fe9d70c/ : SUCCESS in 50m 38s\n- openstack-ansible-ovs-nsh-ubuntu-bionic http://logs.openstack.org/62/657262/1/check/openstack-ansible-ovs-nsh-ubuntu-bionic/dd34c16/ : SUCCESS in 57m 16s\n- openstack-ansible-odl-sfc-ubuntu-bionic finger://ze01.openstack.org/8a05311fe46c4e6f9f3de87539eaecc3 : RETRY_LIMIT in 12m 48s\n- openstack-ansible-ovs-sfc-ubuntu-bionic http://logs.openstack.org/62/657262/1/check/openstack-ansible-ovs-sfc-ubuntu-bionic/eb2a563/ : SUCCESS in 40m 16s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic http://logs.openstack.org/62/657262/1/check/openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic/81987c2/ : FAILURE in 50m 47s (non-voting)\n- openstack-ansible-ovs-fwaasv2-ubuntu-bionic http://logs.openstack.org/62/657262/1/check/openstack-ansible-ovs-fwaasv2-ubuntu-bionic/a408892/ : FAILURE in 34m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e7b6d8394e39f8d2233e1cab7b0e49e8b7c37664","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-05-06 11:24:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d130c4204320657d49275fc716fa948378cd5b08","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-06 12:36:10.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/62/657262/2/check/openstack-ansible-linters/3aa5bd6/ : SUCCESS in 4m 34s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/62/657262/2/check/openstack-ansible-deploy-aio_metal-debian-stable/cfa702a/ : SUCCESS in 56m 08s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/62/657262/2/check/openstack-ansible-deploy-aio_metal-centos-7/e791b3d/ : SUCCESS in 1h 06m 19s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/62/657262/2/check/openstack-ansible-deploy-aio_metal-opensuse-150/e9b426d/ : SUCCESS in 1h 04m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/62/657262/2/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/3912546/ : SUCCESS in 56m 28s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/62/657262/2/check/openstack-ansible-deploy-aio_distro_metal-centos-7/8e4cb4e/ : FAILURE in 38m 58s\n- openstack-tox-docs http://logs.openstack.org/62/657262/2/check/openstack-tox-docs/c4dc852/html/ : SUCCESS in 2m 48s\n- build-openstack-releasenotes http://logs.openstack.org/62/657262/2/check/build-openstack-releasenotes/5c2bce3/html/ : SUCCESS in 2m 38s\n- openstack-ansible-ovs-ubuntu-bionic http://logs.openstack.org/62/657262/2/check/openstack-ansible-ovs-ubuntu-bionic/c163fcd/ : SUCCESS in 36m 44s\n- openstack-ansible-ovn-ubuntu-bionic http://logs.openstack.org/62/657262/2/check/openstack-ansible-ovn-ubuntu-bionic/552dc1f/ : SUCCESS in 35m 52s\n- openstack-ansible-ovn-centos-7 http://logs.openstack.org/62/657262/2/check/openstack-ansible-ovn-centos-7/a3c7dc4/ : SUCCESS in 41m 47s\n- openstack-ansible-calico-ubuntu-bionic http://logs.openstack.org/62/657262/2/check/openstack-ansible-calico-ubuntu-bionic/45f415d/ : FAILURE in 37m 57s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic http://logs.openstack.org/62/657262/2/check/openstack-ansible-opendaylight-ubuntu-bionic/dc04280/ : SUCCESS in 51m 20s\n- openstack-ansible-ovs-nsh-ubuntu-bionic http://logs.openstack.org/62/657262/2/check/openstack-ansible-ovs-nsh-ubuntu-bionic/25da028/ : SUCCESS in 54m 40s\n- openstack-ansible-odl-sfc-ubuntu-bionic http://logs.openstack.org/62/657262/2/check/openstack-ansible-odl-sfc-ubuntu-bionic/b364f2f/ : SUCCESS in 54m 27s\n- openstack-ansible-ovs-sfc-ubuntu-bionic http://logs.openstack.org/62/657262/2/check/openstack-ansible-ovs-sfc-ubuntu-bionic/84862df/ : SUCCESS in 38m 44s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic http://logs.openstack.org/62/657262/2/check/openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic/81182e7/ : FAILURE in 47m 12s (non-voting)\n- openstack-ansible-ovs-fwaasv2-ubuntu-bionic http://logs.openstack.org/62/657262/2/check/openstack-ansible-ovs-fwaasv2-ubuntu-bionic/6f50449/ : FAILURE in 32m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9806712f84721c7bbaf5118878912ec545d52510","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-08-15 00:35:37.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"5524f866844a190835a35058d4a0ae80a947a58f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 02:47:28.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/c8c45380b06c4f6184ea44ab7a9f40b3 : SUCCESS in 5m 15s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/b4cba47d91f24c4abd477e7e18f6570c : SUCCESS in 1h 08m 03s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/463f2fee0155421ab1d5dfbbdea9e0dc : SUCCESS in 1h 32m 48s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/0276bdffb997406abdb272af95b17f78 : FAILURE in 1h 05m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f710feab745d4b1f8b6323d8dc200c6e : SUCCESS in 59m 14s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/2015812fcd9740c09fab30f090450188 : SUCCESS in 1h 01m 50s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/0917d30cd39b4239893db841c54f397f : SUCCESS in 53m 00s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5512f8fa15934e09a2dcfe56d5b154fe : SUCCESS in 42m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa3763401ddb43a88eada4fd57577c27 : SUCCESS in 4m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d4ae0ea8e304ccd86fbccd73ae2e739 : SUCCESS in 3m 12s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0b231e7c93994e72a792f060cfa7c97e : FAILURE in 36m 02s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/402da8182ac74d2080d7ec9414dd6687 : FAILURE in 35m 26s\n- openstack-ansible-ovn-centos-7 https://zuul.opendev.org/t/openstack/build/129f9cc220f242e495fffab16b2243f8 : FAILURE in 30m 31s (non-voting)\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/88a36b4253194afe9d09fc48002fb700 : FAILURE in 1h 15m 09s\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ada72dc017034a44aeee016c09cf10de : FAILURE in 47m 38s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/753d313e43b344a9a1a51a63f76e9c35 : FAILURE in 40m 28s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/85b73a87ef1c472d924a3dac56843d45 : FAILURE in 39m 48s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8fba80ee4a0e4a87a8e2dc794e0fa547 : FAILURE in 37m 57s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e347fab842424b9592203b45c0a16965 : FAILURE in 40m 49s (non-voting)\n- openstack-ansible-ovs-fwaasv2-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f19e0988bfee46f8910803e1bdcf6409 : FAILURE in 37m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b8ce1c66ee750b6e0e50a5345e269e58b3abcce8","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-08-16 21:05:55.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"2f937ecc3fa19763d01ec6347dd1a63810feb81f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 22:44:39.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/bdada6b9ecf74b6cbd85b3f4d4c0fb2f : SUCCESS in 7m 50s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/b8f0b0af24a64ef7843d3eb334da5d3c : SUCCESS in 1h 21m 17s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/337fa809125c4797b09068c572f638aa : SUCCESS in 1h 22m 34s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/fa3ea79b5a0b4cb0bb9f9850bdea9eaa : FAILURE in 1h 04m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dd85a32a911143da930d82af8d3c9566 : SUCCESS in 1h 05m 52s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a29e3ec52ad9409593014b6b3519239a : SUCCESS in 1h 07m 29s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/60518b905b7b4d83be7fc51b317e6067 : SUCCESS in 59m 23s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7c81b9aea31042d1b886bd0a0eefc0de : SUCCESS in 1h 08m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee1770bc798b47adb1fd9b8b1632f284 : SUCCESS in 3m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b6983984874f462ca013ca855b5aa756 : SUCCESS in 3m 49s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0f433284f68c4fbaaa5021357c9a0798 : SUCCESS in 44m 52s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dac9b22b30304eda85d878628226dc6e : SUCCESS in 39m 00s\n- openstack-ansible-ovn-centos-7 https://zuul.opendev.org/t/openstack/build/f9a8e2f2398d44e5b2296d3f44f6e79f : FAILURE in 37m 43s (non-voting)\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6416909f8b814cab99fec133da992b7d : FAILURE in 1h 10m 05s\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bf546edba69a4d8d85296ce4bc57f5cb : SUCCESS in 45m 14s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ecf9f8151db34586bc3ca20bc0375b6f : SUCCESS in 43m 54s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/60f4a48ba74c45ed8be5701936a1cb36 : SUCCESS in 49m 27s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eb100a7cea2b4f84853de5b4791ea5e2 : SUCCESS in 42m 17s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ead8b437c81441a4b6093f673e22586f : SUCCESS in 48m 05s (non-voting)\n- openstack-ansible-ovs-fwaasv2-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/53f7573426a54e56be22484c3590b4c4 : SUCCESS in 50m 16s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0dcbb9f27c021b22519760efd041ecd1e40b6cad","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2019-09-03 08:13:03.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"084b8dbedaf0215a0ba27048b32de518af660606","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 10:54:02.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d2bdc9600ea24b34acb70f2d7f8c3026 : SUCCESS in 6m 42s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/30d5c50440a54ca48d9b52b2b4de441d : SUCCESS in 1h 07m 16s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/bbf62eb5f2ff478898147422c2a9e0fb : SUCCESS in 1h 56m 57s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/efec9227cc574293b3eee8cebe2a3d4c : FAILURE in 36m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b9f02746ef5d492cb9bedec079e58b65 : SUCCESS in 1h 04m 57s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/9bd2550e9a0144759271d10a4ae37e62 : SUCCESS in 1h 04m 17s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/b2020414e7144913b03d5e15d596b3f0 : FAILURE in 31m 01s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dca5c961390347e0965d71318714ec40 : SUCCESS in 49m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28dd888e72824b8bbaf10632f84ea83e : SUCCESS in 3m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7500308fbfd94666847d4a25be89dc66 : SUCCESS in 3m 36s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8e128ff4988742e0927c9314f697df8f : SUCCESS in 44m 49s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7526285d147b426e919b5e063137d3ac : SUCCESS in 37m 33s\n- openstack-ansible-ovn-centos-7 https://zuul.opendev.org/t/openstack/build/5d1f847998c04363b235b14ee539d471 : FAILURE in 40m 55s (non-voting)\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c3caddb7edcf40c9a199ccfe8e6f000d : SUCCESS in 1h 09m 45s\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/188e8775fb40412dabee2b7b95f0863b : SUCCESS in 1h 23m 29s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b0166ecc69f04023ba16da36e48046b5 : SUCCESS in 1h 13m 12s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d067e09b6820472ba97ed9c9ef9512dd : SUCCESS in 1h 24m 16s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/80ac434726a844d189827213d1104666 : SUCCESS in 45m 09s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/21da532e07a14c38b2073ae6e2165d30 : FAILURE in 46m 53s (non-voting)\n- openstack-ansible-ovs-fwaasv2-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/19db7a7f78724d8b9d3a24f9235ad4db : SUCCESS in 44m 56s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"493d771261d2746912c00191dd2a5f668f2435c3","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-09-10 06:12:35.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"d9d9e14eb234f520a60f73c75fed48a125fb3ffc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 08:54:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/929763ef7feb4a348657fc4477a11e8b : SUCCESS in 5m 32s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/6ae5e9644cdd40af8a6f1dff1b6a2959 : SUCCESS in 1h 14m 45s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b7de7abb32674695adee91a6b618b23f : SUCCESS in 1h 37m 44s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/0e20f49eafed493a90a174af000802ff : SUCCESS in 1h 26m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a0b6434299f2433791c05e22824d4583 : SUCCESS in 1h 08m 35s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/08e20afeaf1744d49544825b226e119f : SUCCESS in 1h 14m 17s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/2c8733b58b954bf19ee8785d800b176e : FAILURE in 39m 29s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e981023ea5734d5c8cac6fd694fbf09b : SUCCESS in 1h 06m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52f9566d6bec4efcaa93021f1f0c4f9f : SUCCESS in 4m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2a6627ad027543cd849aa58f80d69db1 : SUCCESS in 3m 36s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ccdd56a37f9349379818f5faae7ed30d : SUCCESS in 41m 20s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8eda9a0ee6034ffb930978bde5147abf : SUCCESS in 38m 34s\n- openstack-ansible-ovn-centos-7 https://zuul.opendev.org/t/openstack/build/2e9a416fc46743c08c9eb57d2fc97d9f : FAILURE in 31m 54s (non-voting)\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a60590bae5b94c5499fc4c582c9a28aa : SUCCESS in 1h 10m 39s\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f5c19221a62447e690119edef509e6a0 : SUCCESS in 43m 44s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c32372e442be4d00a61374197cdec1f1 : SUCCESS in 46m 58s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f0eaae1b893c421e84c8c3799f0475f3 : SUCCESS in 42m 58s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/287940ce8abe448994c513edd8d3a189 : SUCCESS in 34m 30s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e3ae0bce128d4ee18dab45773ca33b81 : SUCCESS in 43m 43s (non-voting)\n- openstack-ansible-ovs-fwaasv2-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d9ebe3d1d6bc4783b97c868452bfaa81 : SUCCESS in 39m 03s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1d2ff52837c6b73d4e4b0d92a347707655d7e548","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-09-10 12:38:46.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"7e70f5e5cacb5e3224d8837d7d3b3fce82bfeaa2","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-13 15:48:51.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"8025ce96ff7d3e7d2a1dceca9ffacd95b9d0e4c0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 15:49:04.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0e217e761606ad7c780f8a4494d9415fc36f3e80","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 18:32:08.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/ffdae7bc627b40f3bd2f680a1b77dab5 : SUCCESS in 7m 13s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/ce43703dd27d4e6fabe0022aa3c9f84f : SUCCESS in 1h 19m 45s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/eacd777f99ec4a1a89b3365241f267f5 : SUCCESS in 1h 37m 22s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/ba6f7bfb851b45559d38cbe4d42c376b : SUCCESS in 1h 27m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7807bb300a20409db1b591254b21e33f : SUCCESS in 1h 09m 04s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f82894f4524a4395b9bd6596d1a8cb38 : SUCCESS in 1h 15m 27s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/43a53ba31f094c61bd68b8da3d97af19 : SUCCESS in 54m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa727569c5df4f488225d9b8892e6117 : SUCCESS in 3m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b19b8069e8994fabad927c925cb1f4b2 : SUCCESS in 3m 21s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d58587eb02a44bafb3ecdfb0c3c0bea8 : FAILURE in 25m 02s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f28bddd2ba3544bfb69aa24520cb0260 : RETRY_LIMIT in 2m 27s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/84ee09b96d5640708c4d6194a93f39ad : FAILURE in 1h 11m 00s\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fed79170725c49d68579cd50da83430c : RETRY_LIMIT in 3m 30s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6ade4b0c8cbb41768edefac6cc599c4e : FAILURE in 25m 50s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0d3dfc4cbdd346078e7d414042df348d : RETRY_LIMIT in 17m 33s","accounts_in_message":[],"_revision_number":3},{"id":"21e0dfba4be9f1d2304ebda2a9d1c22b7c5b3517","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-09-14 16:44:28.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"3a5a089bdc190c54a4935ef204ac2dbeb99dc6d3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 18:31:58.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/e5ebabaab2b74d499610bffc0bd50d3f : SUCCESS in 4m 49s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/7c38ca36d9fd49efb86f9b754f2aed25 : SUCCESS in 1h 22m 31s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7e7b6a6ef9d845e7ad6ac5bee478fd13 : SUCCESS in 1h 41m 34s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/0a080a56eb3c40f284653cf72357d99c : SUCCESS in 1h 29m 25s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/58b73f9877774a12b86a790482b6a7f1 : SUCCESS in 1h 08m 55s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b81af5e31542458c92fa08c8ed01cf6d : FAILURE in 32m 34s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/738a9d78bbcc4602ac9b47813eb9b293 : FAILURE in 27m 28s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1ad9635972214c24939b02b1092f09a3 : SUCCESS in 52m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bc730dc13704a199b55418b90e355b7 : SUCCESS in 4m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/be5499cc9a80413ea111c9ac1eea86e1 : SUCCESS in 3m 55s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c17777c0117c4c3c9d234f9052507e4e : FAILURE in 25m 41s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c0a9cd80495f404697aa7f8709f66820 : FAILURE in 25m 07s\n- openstack-ansible-ovn-centos-7 https://zuul.opendev.org/t/openstack/build/3f1ac17d052748e6b74ace2b94769595 : FAILURE in 31m 00s (non-voting)\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/530febf9e3054f8d89340ab38dff36c6 : FAILURE in 59m 52s\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b9ad7af7b40a42d3978c9ffd0affb187 : FAILURE in 25m 53s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b74cc93130614d2799a8418e3bc42aed : FAILURE in 25m 25s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/48cb8c011ab44ccfb517a5090fea3b53 : FAILURE in 24m 06s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/01663ea5c946491b937a5c7b3967d887 : FAILURE in 24m 35s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/46bfdc04d44549bebdfbad253cd4b6cf : FAILURE in 24m 40s (non-voting)\n- openstack-ansible-ovs-fwaasv2-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6c1216d5b54c4a2f94242f53b921fdff : FAILURE in 24m 02s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8244e343b5a7c78fbd92361ad00458a059346b0d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-19 10:05:24.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"1c5b2c97717e9f013e2412b90004df1513373db3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 11:58:19.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f1ac7ba196e54530941e6b7e953c7386 : SUCCESS in 6m 40s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/6c9f71d223a942a0afe7bb0ce5fed5c1 : SUCCESS in 1h 12m 35s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e5f0b064745a44ab8dec1c956092c35d : SUCCESS in 1h 44m 56s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/69952c71b6254190b420067608da6bd6 : SUCCESS in 1h 28m 25s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/77b5454d2028498a9beac91382c18cef : SUCCESS in 1h 01m 45s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/fa3f7e142bc14052a074b4be6b215fec : SUCCESS in 1h 13m 41s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/9f4732dd54544f8c82dfb83b03ae182e : FAILURE in 28m 26s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a1752bb97a574747b16e3521d6a501b8 : SUCCESS in 49m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef26341c438e4f709fad926dcf4e1e22 : SUCCESS in 3m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e93a4a34231e4cf1ab17328441f5d569 : SUCCESS in 3m 31s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d9dfc18408824f6b94a01a4aa050ac35 : SUCCESS in 41m 52s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fff296797b704bd8bbdc985d5fba2389 : SUCCESS in 38m 17s\n- openstack-ansible-ovn-centos-7 https://zuul.opendev.org/t/openstack/build/b61148a099d34ab281b58e668ad96de2 : FAILURE in 31m 07s (non-voting)\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/126584053ee84b29b94142d9ccc6faba : FAILURE in 1h 11m 16s\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2b9f40da85904ab7853f440af658a159 : SUCCESS in 42m 22s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ca5ebf2f25624758848456fd68c40399 : SUCCESS in 47m 23s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ba56a7cd31a540f8a881e508627f0c49 : SUCCESS in 44m 44s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/546a0530dcd34b0b97d374a3b52e2983 : SUCCESS in 40m 35s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ba56c2bf39994ab0af48f777acf16633 : SUCCESS in 42m 14s (non-voting)\n- openstack-ansible-ovs-fwaasv2-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d80b532278594052bb627768309085dd : SUCCESS in 38m 44s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"14d87b66d9100fcc67f3f3c14d1fad3b064deb86","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-09-20 07:04:44.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"74971bd16c6813af7467c50cc1e55cfac58f8607","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 10:20:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/1feb1f6314534eb9b5417ee38e78f235 : SUCCESS in 6m 12s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/59729a7798b948c6a6c3290d3b6cd6b5 : SUCCESS in 1h 29m 56s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0f1e48d7e45b4a35b50ff22bf72c5125 : SUCCESS in 1h 57m 23s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/10dd4c4016274852a8942d2ab1c1ab20 : SUCCESS in 1h 28m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/92320b774dc349268434fcf787160c4c : SUCCESS in 1h 21m 53s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/37f0e80f9cb14f99b313995d763ce5a4 : SUCCESS in 1h 08m 59s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/0b350f16a2ab49c4bfb9f934c03735d0 : FAILURE in 30m 11s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fd3d9fd4e1654e7bb4bc0ff5665cb029 : SUCCESS in 1h 12m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7898862e920548d79f3a6cdcd312dd69 : SUCCESS in 3m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d87c61c8761d42f896f672fb6bbb6826 : SUCCESS in 3m 28s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d6861d17a54e4979997b9c80a78a4a4d : SUCCESS in 42m 33s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0c773c1a40c24469bf0538d2f2765c2e : SUCCESS in 40m 36s\n- openstack-ansible-ovn-centos-7 https://zuul.opendev.org/t/openstack/build/c6e194559c5f4c9680a403d4d24fb128 : FAILURE in 32m 36s (non-voting)\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9480e058384d4a5fb588b7e6526426f4 : FAILURE in 1h 29m 17s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7df7b2cb0ad04655957173de892a28de : SUCCESS in 49m 16s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/180fc179748e490787a8c841a6e192eb : SUCCESS in 46m 29s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/784cde96882b4a66806d378ac10c82c5 : SUCCESS in 51m 50s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9c718203c1b24fb2a6826e25dd724216 : SUCCESS in 42m 17s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b60ae7a554e84e72ab7aa778de0115e9 : SUCCESS in 47m 33s (non-voting)\n- openstack-ansible-ovs-fwaasv2-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fc3aab7cbc084e5cba2ad244e538a2c9 : SUCCESS in 45m 08s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2d8beda16400e1f45d0498c38353f0ce20ff93ec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 10:20:17.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"fd286658057224c2dd44de403071b001c30f335b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 12:03:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"ef8d83ec026ef294465e7116454724137e2c7afb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 12:03:40.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/3e4d8a9067ba403d873b6dc88594b3d8 : SUCCESS in 7m 08s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/cfeafda18d534c1d926526ccec348f7e : SUCCESS in 1h 36m 29s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d4ee20fcf59a4cfdb44ec5e9ca26d4b9 : SUCCESS in 1h 34m 34s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/af830b5c2cb94ff986195196057f6882 : SUCCESS in 1h 31m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6a1321a5e2704f899ab94ee2f115d038 : SUCCESS in 1h 08m 51s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/211e8d42b8e542e68cb32bd55e26e52b : SUCCESS in 1h 13m 36s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8c47d5f58aa64e908dc8d1aede83c5ad : SUCCESS in 1h 07m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e02e113bfd424e6ea8b0923cddc746bf : SUCCESS in 3m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/858357e31dcf4b41b1079029192e06e4 : SUCCESS in 5m 12s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7550213da04445ac8ead31334dda6238 : SUCCESS in 40m 47s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/83599a0d69bc4e2aa5f5ad14549aa967 : SUCCESS in 42m 28s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/342389a5f5064040ac530d746eae7234 : FAILURE in 1h 04m 57s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/019e4c3e777e4f23b37f7702bc3f1adf : SUCCESS in 46m 09s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fa603312d24f4add95eea92e308dcbd7 : SUCCESS in 43m 05s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2ee9c3f2fdc94e2ba3761ddd4696d4dd : SUCCESS in 50m 18s","accounts_in_message":[],"_revision_number":3},{"id":"53d541f9c0bbca47a9f445cc2504b04896434767","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 12:05:16.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6602bbc1fb5f41fc9bc68297e4321746 : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/50a9fb97aa3a4250a1093a2cad5dda49 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f7d5690c2b28a1a98ff81798e62ac836d8be5f93","revisions":{"3ea206233a60b93413df69c263b1686e2cd2cc78":{"kind":"REWORK","_number":1,"created":"2019-05-05 13:04:18.000000000","uploader":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"ref":"refs/changes/62/657262/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/62/657262/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/62/657262/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/62/657262/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/62/657262/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/62/657262/1"}}},"commit":{"parents":[{"commit":"d2076235d6b026e2d5dbb0967baef25f33b5a32c","subject":"Merge \"Swap special CI jobs from xenial to bionic\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/d2076235d6b026e2d5dbb0967baef25f33b5a32c"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2019-05-05 13:03:59.000000000","tz":480},"committer":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2019-05-05 13:04:15.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I94f1a1d0d87d6d23e33ea82bbab7f3ec2a0668d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/3ea206233a60b93413df69c263b1686e2cd2cc78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/3ea206233a60b93413df69c263b1686e2cd2cc78"}]},"branch":"refs/heads/master"},"50891be5191bf0e2433ee559dd24341e88faa322":{"kind":"REWORK","_number":2,"created":"2019-05-06 11:24:48.000000000","uploader":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"ref":"refs/changes/62/657262/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/62/657262/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/62/657262/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/62/657262/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/62/657262/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/62/657262/2"}}},"commit":{"parents":[{"commit":"d2076235d6b026e2d5dbb0967baef25f33b5a32c","subject":"Merge \"Swap special CI jobs from xenial to bionic\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/d2076235d6b026e2d5dbb0967baef25f33b5a32c"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2019-05-05 13:03:59.000000000","tz":480},"committer":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2019-05-06 11:24:27.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I94f1a1d0d87d6d23e33ea82bbab7f3ec2a0668d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/50891be5191bf0e2433ee559dd24341e88faa322"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/50891be5191bf0e2433ee559dd24341e88faa322"}]},"branch":"refs/heads/master"},"f7d5690c2b28a1a98ff81798e62ac836d8be5f93":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-08-16 21:05:55.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/62/657262/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/62/657262/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/62/657262/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/62/657262/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/62/657262/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/62/657262/3"}}},"commit":{"parents":[{"commit":"d2076235d6b026e2d5dbb0967baef25f33b5a32c","subject":"Merge \"Swap special CI jobs from xenial to bionic\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/d2076235d6b026e2d5dbb0967baef25f33b5a32c"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2019-05-05 13:03:59.000000000","tz":480},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-08-16 21:05:52.000000000","tz":0},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nDepends-On: https://review.opendev.org/676996\nChange-Id: I94f1a1d0d87d6d23e33ea82bbab7f3ec2a0668d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/f7d5690c2b28a1a98ff81798e62ac836d8be5f93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/f7d5690c2b28a1a98ff81798e62ac836d8be5f93"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
