)]}'
{"id":"openstack%2Fopenstack-ansible-os_neutron~658032","triplet_id":"openstack%2Fopenstack-ansible-os_neutron~master~I2f4686b929542973fedf4d3b1ba60efa693eb835","project":"openstack/openstack-ansible-os_neutron","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I2f4686b929542973fedf4d3b1ba60efa693eb835","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-05-09 11:35:10.000000000","updated":"2019-05-09 14:34:11.000000000","submitted":"2019-05-09 14:34:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658032-1557412451877-cb52d0ca","meta_rev_id":"37e7a73fbf66d8d54a82747e6bdad4f42438c6fc","_number":658032,"virtual_id_number":658032,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":2,"date":"2019-05-09 14:34:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":2,"date":"2019-05-09 12:57:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":1,"date":"2019-05-09 12:57:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-09 12:57:25.000000000","updated_by":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"reviewer":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"state":"REVIEWER"},{"updated":"2019-05-09 14:34:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7d41454cbeaec0c3638cd07fc8906e329472d735","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-05-09 11:35:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1f18d5efaee98ce4e80136239af0f6635423f32e","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-09 12:57:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ff9ecda3bfc8327e0c4df1746ee2cbed95ee5588","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 13:10:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/32/658032/1/check/openstack-ansible-linters/0a4ce7f/ : SUCCESS in 8m 07s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/32/658032/1/check/openstack-ansible-deploy-aio_metal-debian-stable/ce3a6dc/ : SUCCESS in 1h 01m 48s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/32/658032/1/check/openstack-ansible-deploy-aio_metal-centos-7/5619ad6/ : SUCCESS in 1h 16m 53s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/32/658032/1/check/openstack-ansible-deploy-aio_metal-opensuse-150/4a4a12d/ : SUCCESS in 1h 04m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/32/658032/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/c6c101b/ : SUCCESS in 55m 01s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/32/658032/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/25b79ac/ : FAILURE in 41m 31s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/32/658032/1/check/openstack-tox-docs/7434e40/html/ : SUCCESS in 2m 19s\n- build-openstack-releasenotes http://logs.openstack.org/32/658032/1/check/build-openstack-releasenotes/93a17c6/html/ : SUCCESS in 2m 14s\n- openstack-ansible-ovs-ubuntu-bionic http://logs.openstack.org/32/658032/1/check/openstack-ansible-ovs-ubuntu-bionic/9abb49d/ : SUCCESS in 48m 56s\n- openstack-ansible-ovn-ubuntu-bionic http://logs.openstack.org/32/658032/1/check/openstack-ansible-ovn-ubuntu-bionic/30866e1/ : SUCCESS in 48m 32s\n- openstack-ansible-ovn-centos-7 http://logs.openstack.org/32/658032/1/check/openstack-ansible-ovn-centos-7/5087c88/ : SUCCESS in 38m 56s\n- openstack-ansible-calico-ubuntu-bionic finger://ze05.openstack.org/4fede31b4004435ca153381266fcba4b : RETRY_LIMIT in 13m 56s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic http://logs.openstack.org/32/658032/1/check/openstack-ansible-opendaylight-ubuntu-bionic/1c63c3d/ : SUCCESS in 51m 45s\n- openstack-ansible-ovs-nsh-ubuntu-bionic http://logs.openstack.org/32/658032/1/check/openstack-ansible-ovs-nsh-ubuntu-bionic/83d8e29/ : SUCCESS in 48m 24s\n- openstack-ansible-odl-sfc-ubuntu-bionic http://logs.openstack.org/32/658032/1/check/openstack-ansible-odl-sfc-ubuntu-bionic/2f6cfa3/ : SUCCESS in 51m 02s\n- openstack-ansible-ovs-sfc-ubuntu-bionic http://logs.openstack.org/32/658032/1/check/openstack-ansible-ovs-sfc-ubuntu-bionic/246908e/ : SUCCESS in 35m 57s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic http://logs.openstack.org/32/658032/1/check/openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic/7623f60/ : SUCCESS in 55m 57s (non-voting)\n- openstack-ansible-ovs-fwaasv2-ubuntu-bionic http://logs.openstack.org/32/658032/1/check/openstack-ansible-ovs-fwaasv2-ubuntu-bionic/c18b108/ : SUCCESS in 37m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"892385794087e6fa8369cacb167d409f840dd9ad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 13:11:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2b2454b6ef2eac3d4169cc0af4c24a39b1b3a6cf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 14:34:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e23e658b5a1481124ebd0df030fbbb4a7c1717ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 14:34:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/32/658032/1/gate/openstack-ansible-linters/c81b084/ : SUCCESS in 5m 25s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/32/658032/1/gate/openstack-ansible-deploy-aio_metal-debian-stable/fed1beb/ : SUCCESS in 56m 53s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/32/658032/1/gate/openstack-ansible-deploy-aio_metal-centos-7/fd6e9ba/ : SUCCESS in 1h 19m 18s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/32/658032/1/gate/openstack-ansible-deploy-aio_metal-opensuse-150/9fbc4b6/ : SUCCESS in 1h 08m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/32/658032/1/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/a9aff3f/ : SUCCESS in 58m 34s\n- openstack-tox-docs http://logs.openstack.org/32/658032/1/gate/openstack-tox-docs/2eeb7d8/html/ : SUCCESS in 2m 59s\n- build-openstack-releasenotes http://logs.openstack.org/32/658032/1/gate/build-openstack-releasenotes/d0f52e6/html/ : SUCCESS in 2m 54s\n- openstack-ansible-ovs-ubuntu-bionic http://logs.openstack.org/32/658032/1/gate/openstack-ansible-ovs-ubuntu-bionic/cc5c0f5/ : SUCCESS in 36m 09s\n- openstack-ansible-ovn-ubuntu-bionic http://logs.openstack.org/32/658032/1/gate/openstack-ansible-ovn-ubuntu-bionic/73d301d/ : SUCCESS in 37m 44s\n- openstack-ansible-ovn-centos-7 http://logs.openstack.org/32/658032/1/gate/openstack-ansible-ovn-centos-7/07c592f/ : SUCCESS in 41m 26s\n- openstack-ansible-opendaylight-ubuntu-bionic http://logs.openstack.org/32/658032/1/gate/openstack-ansible-opendaylight-ubuntu-bionic/74bc896/ : SUCCESS in 53m 02s\n- openstack-ansible-ovs-nsh-ubuntu-bionic http://logs.openstack.org/32/658032/1/gate/openstack-ansible-ovs-nsh-ubuntu-bionic/c828c55/ : SUCCESS in 57m 14s\n- openstack-ansible-odl-sfc-ubuntu-bionic http://logs.openstack.org/32/658032/1/gate/openstack-ansible-odl-sfc-ubuntu-bionic/3e8fdf6/ : SUCCESS in 1h 08m 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"423a63c53c1671028c9dac7ad38ba3ac534272d6","revisions":{"423a63c53c1671028c9dac7ad38ba3ac534272d6":{"kind":"REWORK","_number":1,"created":"2019-05-09 11:35:10.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/32/658032/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/32/658032/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/32/658032/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/32/658032/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/32/658032/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/32/658032/1"}}},"commit":{"parents":[{"commit":"d2076235d6b026e2d5dbb0967baef25f33b5a32c","subject":"Merge \"Swap special CI jobs from xenial to bionic\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/d2076235d6b026e2d5dbb0967baef25f33b5a32c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-09 11:35:09.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-09 11:35:09.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I2f4686b929542973fedf4d3b1ba60efa693eb835\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/423a63c53c1671028c9dac7ad38ba3ac534272d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/423a63c53c1671028c9dac7ad38ba3ac534272d6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
