)]}'
{"id":"openstack%2Fopenstack-ansible-os_neutron~658261","triplet_id":"openstack%2Fopenstack-ansible-os_neutron~master~I3a71da978f74e02a1adc022f73a4eac1285be950","project":"openstack/openstack-ansible-os_neutron","branch":"master","topic":"osa/el8-support","hashtags":[],"change_id":"I3a71da978f74e02a1adc022f73a4eac1285be950","subject":"[redhat] Update variables to be version agnostic","status":"MERGED","created":"2019-05-10 04:46:39.000000000","updated":"2019-09-20 15:25:12.000000000","submitted":"2019-09-20 15:23:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658261-1568993011810-f233d7c7","meta_rev_id":"2e5ada61463752d719e6e1285fb146e5e7db49a2","_number":658261,"virtual_id_number":658261,"owner":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-09-20 15:23:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-20 07:04:07.000000000","_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":2,"date":"2019-09-20 09:48:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-09-20 13:26:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":1,"date":"2019-09-20 13:26:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-10 18:33:22.000000000","updated_by":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"reviewer":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"state":"REVIEWER"},{"updated":"2019-09-20 07:04:07.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-09-20 09:48:54.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-09-20 15:23:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dca7859205500d2984875f431a2d9472db6044ef","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-05-10 04:46:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cf91871ae06add942014a0d34994fec1399f697e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 06:20:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/61/658261/1/check/openstack-ansible-linters/4a786b4/ : SUCCESS in 5m 36s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/61/658261/1/check/openstack-ansible-deploy-aio_metal-debian-stable/217403c/ : SUCCESS in 1h 03m 33s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/61/658261/1/check/openstack-ansible-deploy-aio_metal-centos-7/9ded213/ : SUCCESS in 1h 17m 51s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/61/658261/1/check/openstack-ansible-deploy-aio_metal-opensuse-150/a54d6d9/ : SUCCESS in 1h 01m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/61/658261/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/2ad6441/ : SUCCESS in 1h 12m 07s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/61/658261/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/6c5761f/ : FAILURE in 39m 11s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/61/658261/1/check/openstack-tox-docs/0315fcb/html/ : SUCCESS in 2m 05s\n- openstack-ansible-ovs-ubuntu-bionic http://logs.openstack.org/61/658261/1/check/openstack-ansible-ovs-ubuntu-bionic/e20f29c/ : SUCCESS in 41m 04s\n- openstack-ansible-ovn-ubuntu-bionic http://logs.openstack.org/61/658261/1/check/openstack-ansible-ovn-ubuntu-bionic/db60069/ : SUCCESS in 35m 53s\n- openstack-ansible-ovn-centos-7 http://logs.openstack.org/61/658261/1/check/openstack-ansible-ovn-centos-7/034d6d4/ : SUCCESS in 42m 06s\n- openstack-ansible-calico-ubuntu-bionic http://logs.openstack.org/61/658261/1/check/openstack-ansible-calico-ubuntu-bionic/de12836/ : FAILURE in 37m 20s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic http://logs.openstack.org/61/658261/1/check/openstack-ansible-opendaylight-ubuntu-bionic/ebf5f33/ : SUCCESS in 52m 45s\n- openstack-ansible-ovs-nsh-ubuntu-bionic finger://ze02.openstack.org/22b93ba027ec4675bc17cb99dab3b3ee : RETRY_LIMIT in 11m 57s\n- openstack-ansible-odl-sfc-ubuntu-bionic http://logs.openstack.org/61/658261/1/check/openstack-ansible-odl-sfc-ubuntu-bionic/d416008/ : SUCCESS in 58m 58s\n- openstack-ansible-ovs-sfc-ubuntu-bionic http://logs.openstack.org/61/658261/1/check/openstack-ansible-ovs-sfc-ubuntu-bionic/9489e68/ : SUCCESS in 36m 27s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic http://logs.openstack.org/61/658261/1/check/openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic/23329ac/ : SUCCESS in 51m 15s (non-voting)\n- openstack-ansible-ovs-fwaasv2-ubuntu-bionic http://logs.openstack.org/61/658261/1/check/openstack-ansible-ovs-fwaasv2-ubuntu-bionic/3f193bc/ : SUCCESS in 40m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b6fae1eda65ec69ce1613335cf987330610c9730","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-05-10 16:22:06.000000000","message":"Patch Set 1:\n\nrecheck\n\nrandom retry_timeout failure","accounts_in_message":[],"_revision_number":1},{"id":"bbc214082f209114b70d99a71a6c3fa178098086","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 17:48:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/61/658261/1/check/openstack-ansible-linters/426579c/ : SUCCESS in 5m 57s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/61/658261/1/check/openstack-ansible-deploy-aio_metal-debian-stable/dd5ac1d/ : SUCCESS in 57m 57s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/61/658261/1/check/openstack-ansible-deploy-aio_metal-centos-7/9dd0f55/ : SUCCESS in 1h 19m 14s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/61/658261/1/check/openstack-ansible-deploy-aio_metal-opensuse-150/6340cca/ : SUCCESS in 1h 07m 01s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/61/658261/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/5982875/ : SUCCESS in 1h 07m 02s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/61/658261/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/72e116b/ : FAILURE in 43m 00s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/61/658261/1/check/openstack-tox-docs/9817b41/html/ : SUCCESS in 2m 10s\n- openstack-ansible-ovs-ubuntu-bionic http://logs.openstack.org/61/658261/1/check/openstack-ansible-ovs-ubuntu-bionic/56779dd/ : SUCCESS in 36m 20s\n- openstack-ansible-ovn-ubuntu-bionic http://logs.openstack.org/61/658261/1/check/openstack-ansible-ovn-ubuntu-bionic/d8ac488/ : SUCCESS in 34m 59s\n- openstack-ansible-ovn-centos-7 http://logs.openstack.org/61/658261/1/check/openstack-ansible-ovn-centos-7/891b8c3/ : SUCCESS in 40m 01s\n- openstack-ansible-calico-ubuntu-bionic http://logs.openstack.org/61/658261/1/check/openstack-ansible-calico-ubuntu-bionic/e8c6d96/ : FAILURE in 45m 26s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic http://logs.openstack.org/61/658261/1/check/openstack-ansible-opendaylight-ubuntu-bionic/8d126b4/ : SUCCESS in 52m 57s\n- openstack-ansible-ovs-nsh-ubuntu-bionic http://logs.openstack.org/61/658261/1/check/openstack-ansible-ovs-nsh-ubuntu-bionic/ba210f6/ : SUCCESS in 53m 24s\n- openstack-ansible-odl-sfc-ubuntu-bionic http://logs.openstack.org/61/658261/1/check/openstack-ansible-odl-sfc-ubuntu-bionic/ec65c28/ : SUCCESS in 51m 00s\n- openstack-ansible-ovs-sfc-ubuntu-bionic http://logs.openstack.org/61/658261/1/check/openstack-ansible-ovs-sfc-ubuntu-bionic/7f801d6/ : SUCCESS in 41m 58s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic http://logs.openstack.org/61/658261/1/check/openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic/d4c4ed6/ : SUCCESS in 51m 37s (non-voting)\n- openstack-ansible-ovs-fwaasv2-ubuntu-bionic http://logs.openstack.org/61/658261/1/check/openstack-ansible-ovs-fwaasv2-ubuntu-bionic/5c0bdd7/ : SUCCESS in 38m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"986d45e41681129340caf4c9b6f9b4fc8e6860a9","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-05-10 18:33:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"027e12553445f6fa60aa46fcc72ba906a7ea69a8","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-05-15 21:27:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5a07d777d859b140a6aa7e2d36de256f2d2a4f25","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 21:27:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6932f22ed05b8c54c0da5aec179a30401c1ab781","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 23:00:02.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/61/658261/1/gate/openstack-ansible-linters/6ce7e39/ : SUCCESS in 5m 50s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/61/658261/1/gate/openstack-ansible-deploy-aio_metal-debian-stable/f66b16c/ : SUCCESS in 55m 26s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/61/658261/1/gate/openstack-ansible-deploy-aio_metal-centos-7/11d816b/ : SUCCESS in 1h 25m 40s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/61/658261/1/gate/openstack-ansible-deploy-aio_metal-opensuse-150/0f6f7b2/ : SUCCESS in 1h 04m 34s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/61/658261/1/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/f81c73c/ : SUCCESS in 58m 49s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/61/658261/1/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/dc8c597/ : SUCCESS in 52m 04s\n- openstack-tox-docs http://logs.openstack.org/61/658261/1/gate/openstack-tox-docs/e0a1656/html/ : SUCCESS in 2m 22s\n- openstack-ansible-ovs-ubuntu-bionic http://logs.openstack.org/61/658261/1/gate/openstack-ansible-ovs-ubuntu-bionic/924e000/ : SUCCESS in 32m 46s\n- openstack-ansible-ovn-ubuntu-bionic http://logs.openstack.org/61/658261/1/gate/openstack-ansible-ovn-ubuntu-bionic/ca42dba/ : SUCCESS in 33m 10s\n- openstack-ansible-ovn-centos-7 http://logs.openstack.org/61/658261/1/gate/openstack-ansible-ovn-centos-7/ce153c9/ : FAILURE in 10m 05s\n- openstack-ansible-opendaylight-ubuntu-bionic http://logs.openstack.org/61/658261/1/gate/openstack-ansible-opendaylight-ubuntu-bionic/bcf5fd2/ : FAILURE in 53m 25s\n- openstack-ansible-ovs-nsh-ubuntu-bionic http://logs.openstack.org/61/658261/1/gate/openstack-ansible-ovs-nsh-ubuntu-bionic/ebc5356/ : FAILURE in 48m 38s\n- openstack-ansible-odl-sfc-ubuntu-bionic http://logs.openstack.org/61/658261/1/gate/openstack-ansible-odl-sfc-ubuntu-bionic/cc183a5/ : FAILURE in 49m 53s","accounts_in_message":[],"_revision_number":1},{"id":"97ebff66322a550537a0cb84dac2a278a992ba6d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-19 16:40:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8948cb2e723d5728857c57e65594a19cc727a69b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 18:26:47.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/4d74a01a272b4e488cea93225ef9ccf8 : SUCCESS in 7m 30s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/a15e00036a704ea6a8df7cd1a94a2a89 : SUCCESS in 1h 15m 20s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/4cfb26d23ce84cf282cf388fe04083a6 : SUCCESS in 1h 38m 06s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/26d27f8802054c6db503b65ff222b088 : SUCCESS in 1h 20m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e9d3a343992c4a658cf34524442707c7 : SUCCESS in 1h 07m 24s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3d21b3d42f1545fc904f9811a46e4393 : SUCCESS in 1h 20m 51s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/b321a64873f34da38b2cbf9b6cf0f5ed : FAILURE in 29m 50s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9c9b53b3cfa64340b593b5c80b98a1d4 : SUCCESS in 50m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c94e3e5aebf4401adc770693d4ea2d0 : SUCCESS in 4m 06s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a88b71f5dd9b4302b8b77619825ca350 : SUCCESS in 45m 51s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5a3ecebe6bc24a4da24e693b0dbbcab3 : SUCCESS in 44m 50s\n- openstack-ansible-ovn-centos-7 https://zuul.opendev.org/t/openstack/build/315039e0cf9247789f44045767ae4323 : FAILURE in 32m 41s (non-voting)\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1a76be8fd8d8462c87dc1957b427bff6 : FAILURE in 1h 04m 07s\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0d1122755732400083472799a21620b1 : SUCCESS in 1h 01m 14s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c2e3fb37ee9147fd888953798151022f : SUCCESS in 40m 04s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/75e94b36b2a648549259d4bf607a9a13 : SUCCESS in 47m 52s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ff7e16c4039a46d3a556f1e53355e112 : SUCCESS in 51m 40s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1a7e992478fc48318e8722a251d96eb4 : SUCCESS in 50m 25s (non-voting)\n- openstack-ansible-ovs-fwaasv2-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f07a9b3520ad48c59161da091d350322 : SUCCESS in 53m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"880256c29ae25441b0706451a51073f2fe0c603f","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-09-20 07:04:07.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"b25aaee013dc18946a02d7f28758c39baad5931f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 09:17:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/b4f521c8e265461a8a72da7ed617badd : SUCCESS in 5m 00s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/04bf26ea92a44602b34fad559eeb54dc : SUCCESS in 1h 07m 26s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/47d43aeaa53f4c7999d8e8a0dfe7e19b : SUCCESS in 1h 44m 17s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/4bd2448e1a5a4b5c97910db9d57394a1 : SUCCESS in 1h 29m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eaf3ba6a04294c4d82e323cbe266dd62 : SUCCESS in 1h 30m 53s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7f3f5a79382147bd893de1385e18c244 : SUCCESS in 1h 11m 48s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/3aab0e483c0f46519e6c65cb7315090b : FAILURE in 28m 55s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7dd7e07d11be4e33ab74c284bae30a1b : SUCCESS in 58m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e291ee77a684912a24ce5e7753092f5 : SUCCESS in 4m 16s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e59c25bb76d34c9c91c0fda5d3dd6fee : SUCCESS in 54m 57s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1441fcadd2104cc5ae3c7e7aeb33bd45 : SUCCESS in 39m 39s\n- openstack-ansible-ovn-centos-7 https://zuul.opendev.org/t/openstack/build/dd988f5f0d30473ea898f4a6a46279a8 : FAILURE in 32m 17s (non-voting)\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/da748937f49a40a28d19ed27b8afeb4c : FAILURE in 1h 08m 31s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3ac08482b94844d986df944ab726c89d : SUCCESS in 45m 39s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ebf4e0c202fb40de93108c8d315c9063 : SUCCESS in 50m 47s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/daf6b430b4e74173bac6980b6302e013 : SUCCESS in 51m 01s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f1791623dad444138fc791eb40a4ec5d : SUCCESS in 47m 33s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7828336adb1b491594b8363d116f53a2 : SUCCESS in 52m 27s (non-voting)\n- openstack-ansible-ovs-fwaasv2-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/61805aafb14e4b3181d2c80716460a47 : SUCCESS in 43m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0ba591c776cd0f89346190eab087bfeceff2a54b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-20 09:48:54.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2f1ddcb06e1008c895125eb4120c5f7b3051e373","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-09-20 13:26:23.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3e7bd91faa073c8ed9de0607e23ac5308f156c9e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 13:26:37.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"debbc9d3538cd3e93d49a9397a88721176b3c6a7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 15:23:31.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/3a608fa9230b441f8d5f6782e5a65063 : SUCCESS in 5m 57s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/e72657eb0403400fa141939ce5bfc03c : SUCCESS in 1h 38m 11s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c5cdf6f2ea0e4715b99d30937dad1b64 : SUCCESS in 1h 47m 32s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/6979a195b47249db8e81e0dcf8b1fe2a : SUCCESS in 1h 33m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f8142cb3633a4867a448f672268b9adf : SUCCESS in 1h 23m 19s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/bfbbe3de5c78492c8002491c934b09d3 : SUCCESS in 1h 16m 03s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9a8db12caf95471c9a26d7e635e023b2 : SUCCESS in 52m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca2b6731113044848086c251d1b1d4fa : SUCCESS in 3m 31s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/128fe8cea6864bd5ba09d760b0c72b94 : SUCCESS in 41m 48s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/681532e536024846bbf458b21c3c254b : SUCCESS in 41m 37s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5038ee7978ac44c8bd301107b073fbd8 : FAILURE in 1h 12m 14s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ae3ca18a83094848bf0e71dd3ba89ede : SUCCESS in 49m 42s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e971a75eb8234d7eb13c24b9b0e54fc1 : SUCCESS in 51m 13s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f3da94bd59674cf9b2adffc3de0f4f02 : SUCCESS in 48m 47s","accounts_in_message":[],"_revision_number":2},{"id":"1fcd105220e9bcd3c1677f654109d6024d28ac8b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 15:23:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"dfc0cb5a1fcf146701d16c7abc383547c727255e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 15:25:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cf5b583996a42ad8a6682b01a6d118f : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ce3cfff4d6dab76f99d386e3158f836617d39606","revisions":{"9cee04691fc46de5957ce12c12fc0be881f2f599":{"kind":"REWORK","_number":1,"created":"2019-05-10 04:46:39.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/61/658261/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/61/658261/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/61/658261/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/61/658261/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/61/658261/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/61/658261/1"}}},"commit":{"parents":[{"commit":"423a63c53c1671028c9dac7ad38ba3ac534272d6","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/423a63c53c1671028c9dac7ad38ba3ac534272d6"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2019-05-10 04:45:22.000000000","tz":-240},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2019-05-10 04:45:22.000000000","tz":-240},"subject":"[redhat] Update variables to be version agnostic","message":"[redhat] Update variables to be version agnostic\n\nThe RedHat-based operating systems such as CentOS have a pretty\nstable list of packages, therefore, we don\u0027t need to pin it by\nversion and we can instead move to a much more generic redhat.yml\nwhich will support a bigger range of systems.\n\nChange-Id: I3a71da978f74e02a1adc022f73a4eac1285be950\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/9cee04691fc46de5957ce12c12fc0be881f2f599"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/9cee04691fc46de5957ce12c12fc0be881f2f599"}]},"branch":"refs/heads/master"},"ce3cfff4d6dab76f99d386e3158f836617d39606":{"kind":"REWORK","_number":2,"created":"2019-09-19 16:40:57.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/61/658261/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/61/658261/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/61/658261/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/61/658261/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/61/658261/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/61/658261/2"}}},"commit":{"parents":[{"commit":"9f06d548abf350c869193033afda6bdf366aa9ab","subject":"Merge \"Allow venv python interpreter to be overridden\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/9f06d548abf350c869193033afda6bdf366aa9ab"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2019-05-10 04:45:22.000000000","tz":-240},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-09-19 16:39:49.000000000","tz":180},"subject":"[redhat] Update variables to be version agnostic","message":"[redhat] Update variables to be version agnostic\n\nThe RedHat-based operating systems such as CentOS have a pretty\nstable list of packages, therefore, we don\u0027t need to pin it by\nversion and we can instead move to a much more generic redhat.yml\nwhich will support a bigger range of systems.\n\nChange-Id: I3a71da978f74e02a1adc022f73a4eac1285be950\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/ce3cfff4d6dab76f99d386e3158f836617d39606"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/ce3cfff4d6dab76f99d386e3158f836617d39606"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
