)]}'
{"id":"openstack%2Fopenstack-ansible-os_neutron~659428","triplet_id":"openstack%2Fopenstack-ansible-os_neutron~master~I024777fa1d4d9880ea904a0f1d8a786b6adb2987","project":"openstack/openstack-ansible-os_neutron","branch":"master","hashtags":[],"change_id":"I024777fa1d4d9880ea904a0f1d8a786b6adb2987","subject":"Update neutron_bgp_speaker_driver","status":"MERGED","created":"2019-05-16 03:39:12.000000000","updated":"2019-05-24 20:33:22.000000000","submitted":"2019-05-24 20:33:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659428-1558730002931-50d017d9","meta_rev_id":"98f1d7cf3879c4ad729148e4a94958790f4fafe3","_number":659428,"virtual_id_number":659428,"owner":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-05-21 13:46:22.000000000","_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-05-24 20:33:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},{"value":2,"date":"2019-05-24 16:39:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-24 19:00:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-24 19:00:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-24 16:39:39.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-05-24 19:00:50.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-05-24 20:33:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"af9bb56a70370abe933fc8c849cbd599ff69a099","author":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"date":"2019-05-16 03:39:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3bef5fc919a65c05c3702dae40305493fc680438","author":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"date":"2019-05-16 03:41:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4b0d8fc1088d8f169bf0e279d09518d834bed2b0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 05:19:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/28/659428/2/check/openstack-ansible-linters/83b8eba/ : SUCCESS in 4m 27s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/28/659428/2/check/openstack-ansible-deploy-aio_metal-debian-stable/b8d36d0/ : SUCCESS in 1h 05m 05s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/28/659428/2/check/openstack-ansible-deploy-aio_metal-centos-7/9aec325/ : SUCCESS in 1h 36m 41s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/28/659428/2/check/openstack-ansible-deploy-aio_metal-opensuse-150/bc9cfaa/ : SUCCESS in 1h 15m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/28/659428/2/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/1800b74/ : SUCCESS in 1h 00m 20s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/28/659428/2/check/openstack-ansible-deploy-aio_distro_metal-centos-7/6c1307e/ : SUCCESS in 1h 02m 49s\n- openstack-tox-docs http://logs.openstack.org/28/659428/2/check/openstack-tox-docs/b344f63/html/ : SUCCESS in 2m 35s\n- build-openstack-releasenotes http://logs.openstack.org/28/659428/2/check/build-openstack-releasenotes/0f5b430/html/ : SUCCESS in 2m 59s\n- openstack-ansible-ovs-ubuntu-bionic http://logs.openstack.org/28/659428/2/check/openstack-ansible-ovs-ubuntu-bionic/01c1d1b/ : SUCCESS in 36m 23s\n- openstack-ansible-ovn-ubuntu-bionic http://logs.openstack.org/28/659428/2/check/openstack-ansible-ovn-ubuntu-bionic/66f0727/ : SUCCESS in 39m 06s\n- openstack-ansible-ovn-centos-7 http://logs.openstack.org/28/659428/2/check/openstack-ansible-ovn-centos-7/83aef1b/ : FAILURE in 40m 32s\n- openstack-ansible-calico-ubuntu-bionic http://logs.openstack.org/28/659428/2/check/openstack-ansible-calico-ubuntu-bionic/7d9da05/ : FAILURE in 39m 57s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic http://logs.openstack.org/28/659428/2/check/openstack-ansible-opendaylight-ubuntu-bionic/51dbd29/ : FAILURE in 50m 07s\n- openstack-ansible-ovs-nsh-ubuntu-bionic http://logs.openstack.org/28/659428/2/check/openstack-ansible-ovs-nsh-ubuntu-bionic/e5665af/ : FAILURE in 48m 05s\n- openstack-ansible-odl-sfc-ubuntu-bionic http://logs.openstack.org/28/659428/2/check/openstack-ansible-odl-sfc-ubuntu-bionic/9fb8314/ : FAILURE in 47m 23s\n- openstack-ansible-ovs-sfc-ubuntu-bionic http://logs.openstack.org/28/659428/2/check/openstack-ansible-ovs-sfc-ubuntu-bionic/988bc0b/ : SUCCESS in 35m 15s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic http://logs.openstack.org/28/659428/2/check/openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic/81d3a17/ : FAILURE in 49m 31s (non-voting)\n- openstack-ansible-ovs-fwaasv2-ubuntu-bionic http://logs.openstack.org/28/659428/2/check/openstack-ansible-ovs-fwaasv2-ubuntu-bionic/bf86ee0/ : SUCCESS in 35m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"94249e8bbeb9fe4fd9d449da86c3f3a5b4e72c4f","author":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"date":"2019-05-21 13:46:22.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"720f00a7f6b69844f8f7f70f2208295ebcd2f81f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 15:12:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/28/659428/2/check/openstack-ansible-linters/29968c1/ : SUCCESS in 5m 40s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/28/659428/2/check/openstack-ansible-deploy-aio_metal-debian-stable/bf59da5/ : SUCCESS in 1h 04m 35s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/28/659428/2/check/openstack-ansible-deploy-aio_metal-centos-7/74deeb4/ : SUCCESS in 1h 19m 59s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/28/659428/2/check/openstack-ansible-deploy-aio_metal-opensuse-150/a48eb7e/ : SUCCESS in 1h 06m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/28/659428/2/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/0a266b6/ : SUCCESS in 54m 31s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/28/659428/2/check/openstack-ansible-deploy-aio_distro_metal-centos-7/cfcd4d0/ : SUCCESS in 59m 15s\n- openstack-tox-docs http://logs.openstack.org/28/659428/2/check/openstack-tox-docs/0c6d3e2/html/ : SUCCESS in 3m 01s\n- build-openstack-releasenotes http://logs.openstack.org/28/659428/2/check/build-openstack-releasenotes/49d89a1/html/ : SUCCESS in 2m 25s\n- openstack-ansible-ovs-ubuntu-bionic http://logs.openstack.org/28/659428/2/check/openstack-ansible-ovs-ubuntu-bionic/bf13a82/ : SUCCESS in 41m 38s\n- openstack-ansible-ovn-ubuntu-bionic http://logs.openstack.org/28/659428/2/check/openstack-ansible-ovn-ubuntu-bionic/f9bd9de/ : SUCCESS in 42m 22s\n- openstack-ansible-ovn-centos-7 http://logs.openstack.org/28/659428/2/check/openstack-ansible-ovn-centos-7/3a2a423/ : FAILURE in 29m 43s (non-voting)\n- openstack-ansible-calico-ubuntu-bionic http://logs.openstack.org/28/659428/2/check/openstack-ansible-calico-ubuntu-bionic/ad0df5a/ : FAILURE in 39m 11s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic http://logs.openstack.org/28/659428/2/check/openstack-ansible-opendaylight-ubuntu-bionic/413961a/ : SUCCESS in 1h 01m 09s\n- openstack-ansible-ovs-nsh-ubuntu-bionic http://logs.openstack.org/28/659428/2/check/openstack-ansible-ovs-nsh-ubuntu-bionic/fb23860/ : SUCCESS in 56m 19s\n- openstack-ansible-odl-sfc-ubuntu-bionic http://logs.openstack.org/28/659428/2/check/openstack-ansible-odl-sfc-ubuntu-bionic/df7cafb/ : SUCCESS in 1h 01m 01s\n- openstack-ansible-ovs-sfc-ubuntu-bionic http://logs.openstack.org/28/659428/2/check/openstack-ansible-ovs-sfc-ubuntu-bionic/85731cb/ : SUCCESS in 33m 03s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic http://logs.openstack.org/28/659428/2/check/openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic/7597e39/ : SUCCESS in 58m 00s (non-voting)\n- openstack-ansible-ovs-fwaasv2-ubuntu-bionic http://logs.openstack.org/28/659428/2/check/openstack-ansible-ovs-fwaasv2-ubuntu-bionic/7605f02/ : SUCCESS in 32m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ff91d9794c77519752aa8b710de8b88b0023417f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-05-24 16:39:39.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e0bdb15113e0526f4db19fed921a845d45cd334d","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-05-24 19:00:50.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"fa934c910dd88b04b8e8b8afde9c9131c6665d9a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 19:01:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fb68220cf7b192fd2a0d8985b5d7f8e72a2c7a1f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 20:33:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"98f1d7cf3879c4ad729148e4a94958790f4fafe3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 20:33:22.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/28/659428/2/gate/openstack-ansible-linters/4a09d46/ : SUCCESS in 4m 54s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/28/659428/2/gate/openstack-ansible-deploy-aio_metal-debian-stable/154000e/ : SUCCESS in 1h 00m 41s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/28/659428/2/gate/openstack-ansible-deploy-aio_metal-centos-7/79c2c5b/ : SUCCESS in 1h 15m 06s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/28/659428/2/gate/openstack-ansible-deploy-aio_metal-opensuse-150/c37fa44/ : SUCCESS in 1h 23m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/28/659428/2/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/edc06bd/ : SUCCESS in 56m 53s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/28/659428/2/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/2ffffbd/ : SUCCESS in 51m 57s\n- openstack-tox-docs http://logs.openstack.org/28/659428/2/gate/openstack-tox-docs/409a5f9/html/ : SUCCESS in 2m 30s\n- build-openstack-releasenotes http://logs.openstack.org/28/659428/2/gate/build-openstack-releasenotes/3517c17/html/ : SUCCESS in 2m 45s\n- openstack-ansible-ovs-ubuntu-bionic http://logs.openstack.org/28/659428/2/gate/openstack-ansible-ovs-ubuntu-bionic/75560f7/ : SUCCESS in 39m 44s\n- openstack-ansible-ovn-ubuntu-bionic http://logs.openstack.org/28/659428/2/gate/openstack-ansible-ovn-ubuntu-bionic/90bb637/ : SUCCESS in 42m 29s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic http://logs.openstack.org/28/659428/2/gate/openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic/10825df/ : SUCCESS in 1h 02m 43s\n- openstack-ansible-opendaylight-ubuntu-bionic http://logs.openstack.org/28/659428/2/gate/openstack-ansible-opendaylight-ubuntu-bionic/290208f/ : SUCCESS in 1h 06m 01s\n- openstack-ansible-ovs-nsh-ubuntu-bionic http://logs.openstack.org/28/659428/2/gate/openstack-ansible-ovs-nsh-ubuntu-bionic/408342b/ : SUCCESS in 54m 10s\n- openstack-ansible-odl-sfc-ubuntu-bionic http://logs.openstack.org/28/659428/2/gate/openstack-ansible-odl-sfc-ubuntu-bionic/1725dac/ : SUCCESS in 1h 02m 03s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c23a94294deb6ad38cc42e7af0f2dd4863bbf2b6","revisions":{"809653234157cbf06e3fbcfcebd3d3ce7e4c0c1b":{"kind":"REWORK","_number":1,"created":"2019-05-16 03:39:12.000000000","uploader":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"ref":"refs/changes/28/659428/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/28/659428/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/28/659428/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/28/659428/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/28/659428/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/28/659428/1"}}},"commit":{"parents":[{"commit":"423a63c53c1671028c9dac7ad38ba3ac534272d6","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/423a63c53c1671028c9dac7ad38ba3ac534272d6"}]}],"author":{"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","date":"2019-05-16 03:37:48.000000000","tz":-300},"committer":{"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","date":"2019-05-16 03:37:48.000000000","tz":-300},"subject":"Update neutron_bgp_speaker_driver","message":"Update neutron_bgp_speaker_driver\n\nThe RyuBgpDriver driver is now replaced by OsKenBgpDriver\nof the neutron-dynamic-routing project\n\nChange-Id: I024777fa1d4d9880ea904a0f1d8a786b6adb2987\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/809653234157cbf06e3fbcfcebd3d3ce7e4c0c1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/809653234157cbf06e3fbcfcebd3d3ce7e4c0c1b"}]},"branch":"refs/heads/master"},"c23a94294deb6ad38cc42e7af0f2dd4863bbf2b6":{"kind":"REWORK","_number":2,"created":"2019-05-16 03:41:31.000000000","uploader":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"ref":"refs/changes/28/659428/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/28/659428/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/28/659428/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/28/659428/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/28/659428/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/28/659428/2"}}},"commit":{"parents":[{"commit":"423a63c53c1671028c9dac7ad38ba3ac534272d6","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/423a63c53c1671028c9dac7ad38ba3ac534272d6"}]}],"author":{"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","date":"2019-05-16 03:37:48.000000000","tz":-300},"committer":{"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","date":"2019-05-16 03:41:22.000000000","tz":-300},"subject":"Update neutron_bgp_speaker_driver","message":"Update neutron_bgp_speaker_driver\n\nThe RyuBgpDriver driver is now replaced by OsKenBgpDriver\nof the neutron-dynamic-routing project\n\nChange-Id: I024777fa1d4d9880ea904a0f1d8a786b6adb2987\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/c23a94294deb6ad38cc42e7af0f2dd4863bbf2b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/c23a94294deb6ad38cc42e7af0f2dd4863bbf2b6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
