)]}'
{"id":"openstack%2Fopenstack-ansible-os_neutron~659714","triplet_id":"openstack%2Fopenstack-ansible-os_neutron~stable%2Frocky~I91d3fef8d1b89f595b529c7c96b3d7249e5b7291","project":"openstack/openstack-ansible-os_neutron","branch":"stable/rocky","topic":"calico3","hashtags":[],"change_id":"I91d3fef8d1b89f595b529c7c96b3d7249e5b7291","subject":"Update Calico Felix to v3","status":"MERGED","created":"2019-05-17 03:55:47.000000000","updated":"2019-05-22 05:14:34.000000000","submitted":"2019-05-22 05:14:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659714-1558502074371-fb151f8d","meta_rev_id":"5b5418dbf2aa055f52490116e0a248f2f86bba40","_number":659714,"virtual_id_number":659714,"owner":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":2,"date":"2019-05-22 05:14:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":2,"date":"2019-05-22 04:18:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-20 16:49:52.000000000","_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":1,"date":"2019-05-22 04:18:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-20 16:49:52.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-05-22 04:18:55.000000000","updated_by":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"reviewer":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"state":"REVIEWER"},{"updated":"2019-05-22 05:14:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"46c0b66cb45a1a07825a4d93694933539966a0b2","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-17 03:55:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c74a1a9f397bf21fd871264b989dab6bed820c87","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 04:53:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/14/659714/1/check/openstack-ansible-linters/deaeda8/ : SUCCESS in 5m 13s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/14/659714/1/check/openstack-ansible-functional-centos-7/b5d07c2/ : SUCCESS in 48m 32s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/14/659714/1/check/openstack-ansible-functional-opensuse-423/172203d/ : FAILURE in 9m 42s (non-voting)\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/14/659714/1/check/openstack-ansible-functional-opensuse-150/e68d840/ : FAILURE in 10m 03s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/14/659714/1/check/openstack-ansible-functional-ubuntu-bionic/68a08cf/ : SUCCESS in 37m 04s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/14/659714/1/check/openstack-ansible-functional-ubuntu-xenial/4d5abd6/ : SUCCESS in 36m 51s\n- build-openstack-sphinx-docs http://logs.openstack.org/14/659714/1/check/build-openstack-sphinx-docs/722fb22/html/ : SUCCESS in 2m 43s\n- openstack-ansible-ovs-ubuntu-xenial http://logs.openstack.org/14/659714/1/check/openstack-ansible-ovs-ubuntu-xenial/4a02922/ : SUCCESS in 36m 48s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-xenial http://logs.openstack.org/14/659714/1/check/openstack-ansible-deploy-aio_metal_calico-ubuntu-xenial/425afed/ : FAILURE in 8m 35s\n- openstack-ansible-dragonflow-ubuntu-xenial http://logs.openstack.org/14/659714/1/check/openstack-ansible-dragonflow-ubuntu-xenial/69d615f/ : FAILURE in 40m 56s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-xenial http://logs.openstack.org/14/659714/1/check/openstack-ansible-opendaylight-ubuntu-xenial/aa8e3f4/ : SUCCESS in 50m 10s\n- openstack-ansible-ovs-nsh-ubuntu-xenial http://logs.openstack.org/14/659714/1/check/openstack-ansible-ovs-nsh-ubuntu-xenial/2442204/ : SUCCESS in 49m 13s\n- openstack-ansible-odl-sfc-ubuntu-xenial http://logs.openstack.org/14/659714/1/check/openstack-ansible-odl-sfc-ubuntu-xenial/fd9c98f/ : SUCCESS in 46m 02s\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-xenial http://logs.openstack.org/14/659714/1/check/openstack-ansible-opendaylight-bgpvpn-ubuntu-xenial/fefb389/ : SUCCESS in 52m 20s (non-voting)\n- openstack-ansible-functional-distro_install-ubuntu-bionic http://logs.openstack.org/14/659714/1/check/openstack-ansible-functional-distro_install-ubuntu-bionic/5b0a061/ : SUCCESS in 34m 54s\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/14/659714/1/check/openstack-ansible-functional-distro_install-centos-7/ca2d6dd/ : SUCCESS in 38m 30s (non-voting)\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/14/659714/1/check/openstack-ansible-functional-distro_install-opensuse-423/f394210/ : FAILURE in 9m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b0f5a1eab75312f7a2975484072b3aec7f2880b3","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-17 21:21:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"607d33fce53c20d9023bfcfa7e893adf4b7ac00e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 22:33:07.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/14/659714/1/check/openstack-ansible-linters/acfa27a/ : SUCCESS in 4m 41s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/14/659714/1/check/openstack-ansible-functional-centos-7/256faaf/ : SUCCESS in 46m 15s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/14/659714/1/check/openstack-ansible-functional-opensuse-423/435f20d/ : FAILURE in 10m 03s (non-voting)\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/14/659714/1/check/openstack-ansible-functional-opensuse-150/0fffe23/ : FAILURE in 9m 47s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/14/659714/1/check/openstack-ansible-functional-ubuntu-bionic/55af707/ : SUCCESS in 48m 28s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/14/659714/1/check/openstack-ansible-functional-ubuntu-xenial/1a3961d/ : SUCCESS in 35m 17s\n- build-openstack-sphinx-docs http://logs.openstack.org/14/659714/1/check/build-openstack-sphinx-docs/0c12a63/html/ : SUCCESS in 2m 55s\n- openstack-ansible-ovs-ubuntu-xenial http://logs.openstack.org/14/659714/1/check/openstack-ansible-ovs-ubuntu-xenial/2be6910/ : SUCCESS in 40m 13s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-xenial http://logs.openstack.org/14/659714/1/check/openstack-ansible-deploy-aio_metal_calico-ubuntu-xenial/a01f417/ : FAILURE in 1h 05m 52s\n- openstack-ansible-dragonflow-ubuntu-xenial http://logs.openstack.org/14/659714/1/check/openstack-ansible-dragonflow-ubuntu-xenial/1807514/ : FAILURE in 28m 48s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-xenial http://logs.openstack.org/14/659714/1/check/openstack-ansible-opendaylight-ubuntu-xenial/24e2850/ : SUCCESS in 56m 45s\n- openstack-ansible-ovs-nsh-ubuntu-xenial http://logs.openstack.org/14/659714/1/check/openstack-ansible-ovs-nsh-ubuntu-xenial/1a2d94d/ : SUCCESS in 54m 27s\n- openstack-ansible-odl-sfc-ubuntu-xenial http://logs.openstack.org/14/659714/1/check/openstack-ansible-odl-sfc-ubuntu-xenial/82e95c2/ : SUCCESS in 51m 45s\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-xenial http://logs.openstack.org/14/659714/1/check/openstack-ansible-opendaylight-bgpvpn-ubuntu-xenial/c128a3e/ : SUCCESS in 58m 17s (non-voting)\n- openstack-ansible-functional-distro_install-ubuntu-bionic http://logs.openstack.org/14/659714/1/check/openstack-ansible-functional-distro_install-ubuntu-bionic/8b0c0fe/ : SUCCESS in 41m 50s\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/14/659714/1/check/openstack-ansible-functional-distro_install-centos-7/a06d533/ : SUCCESS in 41m 26s (non-voting)\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/14/659714/1/check/openstack-ansible-functional-distro_install-opensuse-423/12adf9e/ : FAILURE in 9m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"56d2852c2009abd1ec9fdbb2ef94d85841db7226","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-18 05:06:14.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6f7c1ec3ab9358b1417e2036c07351ce36c65054","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-18 06:39:14.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/14/659714/1/check/openstack-ansible-linters/421ea8b/ : SUCCESS in 5m 30s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/14/659714/1/check/openstack-ansible-functional-centos-7/7e764c1/ : SUCCESS in 42m 20s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/14/659714/1/check/openstack-ansible-functional-opensuse-423/40bbf67/ : FAILURE in 12m 32s (non-voting)\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/14/659714/1/check/openstack-ansible-functional-opensuse-150/2e45a1d/ : FAILURE in 10m 04s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/14/659714/1/check/openstack-ansible-functional-ubuntu-bionic/b3b6c8f/ : SUCCESS in 43m 02s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/14/659714/1/check/openstack-ansible-functional-ubuntu-xenial/e887e7d/ : SUCCESS in 36m 21s\n- build-openstack-sphinx-docs http://logs.openstack.org/14/659714/1/check/build-openstack-sphinx-docs/90f8f32/html/ : SUCCESS in 3m 22s\n- openstack-ansible-ovs-ubuntu-xenial http://logs.openstack.org/14/659714/1/check/openstack-ansible-ovs-ubuntu-xenial/adeb8ba/ : SUCCESS in 41m 15s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-xenial http://logs.openstack.org/14/659714/1/check/openstack-ansible-deploy-aio_metal_calico-ubuntu-xenial/d3a6acb/ : FAILURE in 1h 04m 42s\n- openstack-ansible-dragonflow-ubuntu-xenial http://logs.openstack.org/14/659714/1/check/openstack-ansible-dragonflow-ubuntu-xenial/d1291c5/ : FAILURE in 31m 43s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-xenial http://logs.openstack.org/14/659714/1/check/openstack-ansible-opendaylight-ubuntu-xenial/f737fc2/ : SUCCESS in 54m 43s\n- openstack-ansible-ovs-nsh-ubuntu-xenial http://logs.openstack.org/14/659714/1/check/openstack-ansible-ovs-nsh-ubuntu-xenial/b0779be/ : SUCCESS in 53m 23s\n- openstack-ansible-odl-sfc-ubuntu-xenial http://logs.openstack.org/14/659714/1/check/openstack-ansible-odl-sfc-ubuntu-xenial/dd5a513/ : SUCCESS in 47m 13s\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-xenial http://logs.openstack.org/14/659714/1/check/openstack-ansible-opendaylight-bgpvpn-ubuntu-xenial/e1aad7a/ : SUCCESS in 45m 54s (non-voting)\n- openstack-ansible-functional-distro_install-ubuntu-bionic http://logs.openstack.org/14/659714/1/check/openstack-ansible-functional-distro_install-ubuntu-bionic/315c86b/ : SUCCESS in 47m 27s\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/14/659714/1/check/openstack-ansible-functional-distro_install-centos-7/a3f5009/ : SUCCESS in 47m 24s (non-voting)\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/14/659714/1/check/openstack-ansible-functional-distro_install-opensuse-423/01cab7a/ : FAILURE in 12m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8c3b39f299e6afce8e050ff6c5e8dde8c7f8074c","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-18 20:47:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2da8c16c6c60036e1e32d8ee1d220ae1dfa8edd8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-18 21:52:06.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/14/659714/1/check/openstack-ansible-linters/0710751/ : SUCCESS in 4m 16s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/14/659714/1/check/openstack-ansible-functional-centos-7/278dba2/ : SUCCESS in 43m 00s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/14/659714/1/check/openstack-ansible-functional-opensuse-423/bc04bfa/ : FAILURE in 9m 58s (non-voting)\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/14/659714/1/check/openstack-ansible-functional-opensuse-150/ef91e8b/ : FAILURE in 7m 40s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/14/659714/1/check/openstack-ansible-functional-ubuntu-bionic/610b79e/ : SUCCESS in 38m 13s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/14/659714/1/check/openstack-ansible-functional-ubuntu-xenial/e3436dd/ : SUCCESS in 33m 54s\n- build-openstack-sphinx-docs http://logs.openstack.org/14/659714/1/check/build-openstack-sphinx-docs/f2fcb38/html/ : SUCCESS in 2m 29s\n- openstack-ansible-ovs-ubuntu-xenial http://logs.openstack.org/14/659714/1/check/openstack-ansible-ovs-ubuntu-xenial/d8a7dae/ : SUCCESS in 46m 52s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-xenial http://logs.openstack.org/14/659714/1/check/openstack-ansible-deploy-aio_metal_calico-ubuntu-xenial/65ff2f3/ : FAILURE in 54m 51s\n- openstack-ansible-dragonflow-ubuntu-xenial http://logs.openstack.org/14/659714/1/check/openstack-ansible-dragonflow-ubuntu-xenial/2556ee7/ : FAILURE in 35m 46s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-xenial http://logs.openstack.org/14/659714/1/check/openstack-ansible-opendaylight-ubuntu-xenial/eaf0257/ : SUCCESS in 51m 55s\n- openstack-ansible-ovs-nsh-ubuntu-xenial http://logs.openstack.org/14/659714/1/check/openstack-ansible-ovs-nsh-ubuntu-xenial/814c55a/ : SUCCESS in 48m 28s\n- openstack-ansible-odl-sfc-ubuntu-xenial http://logs.openstack.org/14/659714/1/check/openstack-ansible-odl-sfc-ubuntu-xenial/ee810d0/ : SUCCESS in 45m 09s\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-xenial http://logs.openstack.org/14/659714/1/check/openstack-ansible-opendaylight-bgpvpn-ubuntu-xenial/9c65ae5/ : SUCCESS in 56m 46s (non-voting)\n- openstack-ansible-functional-distro_install-ubuntu-bionic http://logs.openstack.org/14/659714/1/check/openstack-ansible-functional-distro_install-ubuntu-bionic/7f791e0/ : SUCCESS in 38m 19s\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/14/659714/1/check/openstack-ansible-functional-distro_install-centos-7/4ec8c5d/ : SUCCESS in 44m 38s (non-voting)\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/14/659714/1/check/openstack-ansible-functional-distro_install-opensuse-423/5165749/ : FAILURE in 14m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4a762c0949b966b20d663a542fd9983a5996a3cf","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-19 02:35:59.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"4e8a426767ba98b54cb117386afffadbf2ddb079","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-19 02:36:15.000000000","message":"Patch Set 2: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"233fdcb02861e92dd90f7f56b455e081167f4ca8","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-19 03:40:11.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"5536ec75c31063f99078a17e9a6fbf0698178d6a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-19 04:08:22.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"ce5744644a4224563cd2a1af38e7a71250529896","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-19 04:10:08.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"50592b3eb8f3b5ef8fcc12c556a633a64a2ae7e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-19 05:19:12.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/14/659714/2/check/openstack-ansible-linters/ac60a11/ : SUCCESS in 6m 33s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/14/659714/2/check/openstack-ansible-functional-centos-7/2391d5a/ : SUCCESS in 46m 56s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/14/659714/2/check/openstack-ansible-functional-opensuse-423/4ffbb09/ : FAILURE in 12m 37s (non-voting)\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/14/659714/2/check/openstack-ansible-functional-opensuse-150/6c5f763/ : FAILURE in 8m 58s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/14/659714/2/check/openstack-ansible-functional-ubuntu-bionic/509de91/ : SUCCESS in 43m 57s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/14/659714/2/check/openstack-ansible-functional-ubuntu-xenial/638130b/ : SUCCESS in 51m 52s\n- build-openstack-sphinx-docs http://logs.openstack.org/14/659714/2/check/build-openstack-sphinx-docs/5cc0b83/html/ : SUCCESS in 3m 25s\n- openstack-ansible-ovs-ubuntu-xenial http://logs.openstack.org/14/659714/2/check/openstack-ansible-ovs-ubuntu-xenial/ba47460/ : SUCCESS in 52m 44s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-xenial http://logs.openstack.org/14/659714/2/check/openstack-ansible-deploy-aio_metal_calico-ubuntu-xenial/c5b2f5c/ : FAILURE in 1h 01m 10s\n- openstack-ansible-dragonflow-ubuntu-xenial http://logs.openstack.org/14/659714/2/check/openstack-ansible-dragonflow-ubuntu-xenial/fd489cb/ : FAILURE in 39m 46s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-xenial http://logs.openstack.org/14/659714/2/check/openstack-ansible-opendaylight-ubuntu-xenial/b9ba446/ : SUCCESS in 48m 25s\n- openstack-ansible-ovs-nsh-ubuntu-xenial http://logs.openstack.org/14/659714/2/check/openstack-ansible-ovs-nsh-ubuntu-xenial/08c37fd/ : SUCCESS in 58m 11s\n- openstack-ansible-odl-sfc-ubuntu-xenial http://logs.openstack.org/14/659714/2/check/openstack-ansible-odl-sfc-ubuntu-xenial/7082b24/ : SUCCESS in 46m 34s\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-xenial http://logs.openstack.org/14/659714/2/check/openstack-ansible-opendaylight-bgpvpn-ubuntu-xenial/e63ce74/ : SUCCESS in 1h 01m 00s (non-voting)\n- openstack-ansible-functional-distro_install-ubuntu-bionic http://logs.openstack.org/14/659714/2/check/openstack-ansible-functional-distro_install-ubuntu-bionic/dec318e/ : SUCCESS in 38m 54s\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/14/659714/2/check/openstack-ansible-functional-distro_install-centos-7/83fbb8f/ : SUCCESS in 37m 55s (non-voting)\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/14/659714/2/check/openstack-ansible-functional-distro_install-opensuse-423/262de87/ : FAILURE in 8m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"aaff0e1413fc507567d1bb8608047ae64de392b6","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-19 07:01:25.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"800fe712c78c5138bd18c6c13a0c8bbe4122c66b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-19 08:04:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/14/659714/3/check/openstack-ansible-linters/b5a638b/ : SUCCESS in 5m 48s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/14/659714/3/check/openstack-ansible-functional-centos-7/443c506/ : SUCCESS in 44m 47s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/14/659714/3/check/openstack-ansible-functional-opensuse-423/35041b5/ : FAILURE in 9m 17s (non-voting)\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/14/659714/3/check/openstack-ansible-functional-opensuse-150/100b5d0/ : FAILURE in 9m 44s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/14/659714/3/check/openstack-ansible-functional-ubuntu-bionic/26bac4c/ : SUCCESS in 34m 41s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/14/659714/3/check/openstack-ansible-functional-ubuntu-xenial/37bc3aa/ : SUCCESS in 34m 33s\n- build-openstack-sphinx-docs http://logs.openstack.org/14/659714/3/check/build-openstack-sphinx-docs/5871876/html/ : SUCCESS in 2m 40s\n- openstack-ansible-ovs-ubuntu-xenial http://logs.openstack.org/14/659714/3/check/openstack-ansible-ovs-ubuntu-xenial/df43ee3/ : SUCCESS in 34m 28s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-xenial http://logs.openstack.org/14/659714/3/check/openstack-ansible-deploy-aio_metal_calico-ubuntu-xenial/3c11bda/ : SUCCESS in 57m 10s\n- openstack-ansible-dragonflow-ubuntu-xenial http://logs.openstack.org/14/659714/3/check/openstack-ansible-dragonflow-ubuntu-xenial/2755eaf/ : FAILURE in 28m 10s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-xenial http://logs.openstack.org/14/659714/3/check/openstack-ansible-opendaylight-ubuntu-xenial/e8925e9/ : SUCCESS in 47m 47s\n- openstack-ansible-ovs-nsh-ubuntu-xenial http://logs.openstack.org/14/659714/3/check/openstack-ansible-ovs-nsh-ubuntu-xenial/1eb9d05/ : SUCCESS in 47m 33s\n- openstack-ansible-odl-sfc-ubuntu-xenial http://logs.openstack.org/14/659714/3/check/openstack-ansible-odl-sfc-ubuntu-xenial/5b73cdf/ : SUCCESS in 45m 43s\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-xenial http://logs.openstack.org/14/659714/3/check/openstack-ansible-opendaylight-bgpvpn-ubuntu-xenial/764b2bb/ : SUCCESS in 49m 55s (non-voting)\n- openstack-ansible-functional-distro_install-ubuntu-bionic http://logs.openstack.org/14/659714/3/check/openstack-ansible-functional-distro_install-ubuntu-bionic/decec94/ : SUCCESS in 40m 12s\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/14/659714/3/check/openstack-ansible-functional-distro_install-centos-7/258d0c7/ : SUCCESS in 39m 29s (non-voting)\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/14/659714/3/check/openstack-ansible-functional-distro_install-opensuse-423/8ea1264/ : FAILURE in 9m 17s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"36600fbbad46310ce9db65edb00ef5f189f04cc3","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-05-20 16:49:52.000000000","message":"Patch Set 3: Code-Review+2\n\npasses ci.","accounts_in_message":[],"_revision_number":3},{"id":"1bde81d1122b027be3d321e2d2b9b8e34dc5fe6e","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2019-05-22 04:18:55.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"b419aceb80bc53b925214d208f92cc59ed034a1a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 04:19:08.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e459c0d5c5622f4bc47a48b1e1eff83609b93e88","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 05:14:34.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/14/659714/3/gate/openstack-ansible-linters/1cd31bf/ : SUCCESS in 5m 21s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/14/659714/3/gate/openstack-ansible-functional-centos-7/c4a9d0c/ : SUCCESS in 42m 57s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/14/659714/3/gate/openstack-ansible-functional-opensuse-423/b407b90/ : FAILURE in 11m 00s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/14/659714/3/gate/openstack-ansible-functional-ubuntu-bionic/7ee31b4/ : SUCCESS in 41m 20s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/14/659714/3/gate/openstack-ansible-functional-ubuntu-xenial/c2d4220/ : SUCCESS in 37m 25s\n- build-openstack-sphinx-docs http://logs.openstack.org/14/659714/3/gate/build-openstack-sphinx-docs/e7fc796/html/ : SUCCESS in 2m 26s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-xenial http://logs.openstack.org/14/659714/3/gate/openstack-ansible-deploy-aio_metal_calico-ubuntu-xenial/909f066/ : SUCCESS in 54m 18s\n- openstack-ansible-ovs-ubuntu-xenial http://logs.openstack.org/14/659714/3/gate/openstack-ansible-ovs-ubuntu-xenial/900cf2f/ : SUCCESS in 35m 19s\n- openstack-ansible-opendaylight-ubuntu-xenial http://logs.openstack.org/14/659714/3/gate/openstack-ansible-opendaylight-ubuntu-xenial/4308af5/ : SUCCESS in 51m 07s\n- openstack-ansible-ovs-nsh-ubuntu-xenial http://logs.openstack.org/14/659714/3/gate/openstack-ansible-ovs-nsh-ubuntu-xenial/ddc6b1f/ : SUCCESS in 50m 31s\n- openstack-ansible-odl-sfc-ubuntu-xenial http://logs.openstack.org/14/659714/3/gate/openstack-ansible-odl-sfc-ubuntu-xenial/1f2aadd/ : SUCCESS in 50m 42s\n- openstack-ansible-functional-distro_install-ubuntu-bionic http://logs.openstack.org/14/659714/3/gate/openstack-ansible-functional-distro_install-ubuntu-bionic/9f72418/ : SUCCESS in 43m 58s\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/14/659714/3/gate/openstack-ansible-functional-distro_install-opensuse-423/b937143/ : FAILURE in 9m 26s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"04ecb95747de3361f49c433755d778c571684503","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 05:14:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f6e7d93d69b7e0359abb2aafb17da67fc60ff0a2","revisions":{"a4181a77c10c62dd7205b5452fbb4a63391c1e13":{"kind":"REWORK","_number":1,"created":"2019-05-17 03:55:47.000000000","uploader":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"ref":"refs/changes/14/659714/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/14/659714/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/14/659714/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/14/659714/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/14/659714/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/14/659714/1"}}},"commit":{"parents":[{"commit":"5a20db85d2d24d5ea8901feb0de8b892ec6b7c0c","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/5a20db85d2d24d5ea8901feb0de8b892ec6b7c0c"}]}],"author":{"name":"Logan V","email":"logan2211@gmail.com","date":"2018-11-08 03:50:13.000000000","tz":-360},"committer":{"name":"Logan V","email":"logan2211@gmail.com","date":"2019-05-17 03:55:40.000000000","tz":-300},"subject":"Update Calico Felix to v3","message":"Update Calico Felix to v3\n\nMoving to v3 requires etcd3, which uses a new L4 gateway setup and\ndeprecates v2 proxy. The role configuration is updated to use the\netcd3 L4 gateway.\n\nThis also includes several dependent fixes and moves the Calico\ntest to use the integrated repo\u0027s Calico scenario.\n\nCombined backport of:\nhttps://review.opendev.org/#/c/641483/\nhttps://review.opendev.org/#/c/645971/\nhttps://review.opendev.org/#/c/659710/\nhttps://review.opendev.org/#/c/659711/\n\nDepends-On: Iaf392898c072ac4b39d878bb27654406c5ab0d8b\nChange-Id: I91d3fef8d1b89f595b529c7c96b3d7249e5b7291\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/a4181a77c10c62dd7205b5452fbb4a63391c1e13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/a4181a77c10c62dd7205b5452fbb4a63391c1e13"}]},"branch":"refs/heads/stable/rocky"},"d741a379f7f5bc5c654506e9c9b8388cbc43d251":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-05-19 02:35:59.000000000","uploader":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"ref":"refs/changes/14/659714/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/14/659714/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/14/659714/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/14/659714/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/14/659714/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/14/659714/2"}}},"commit":{"parents":[{"commit":"5a20db85d2d24d5ea8901feb0de8b892ec6b7c0c","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/5a20db85d2d24d5ea8901feb0de8b892ec6b7c0c"}]}],"author":{"name":"Logan V","email":"logan2211@gmail.com","date":"2018-11-08 03:50:13.000000000","tz":-360},"committer":{"name":"Logan V","email":"logan2211@gmail.com","date":"2019-05-19 02:35:56.000000000","tz":0},"subject":"Update Calico Felix to v3","message":"Update Calico Felix to v3\n\nMoving to v3 requires etcd3, which uses a new L4 gateway setup and\ndeprecates v2 proxy. The role configuration is updated to use the\netcd3 L4 gateway.\n\nThis also includes several dependent fixes and moves the Calico\ntest to use the integrated repo\u0027s Calico scenario.\n\nCombined backport of:\nhttps://review.opendev.org/#/c/641483/\nhttps://review.opendev.org/#/c/645971/\nhttps://review.opendev.org/#/c/659710/\nhttps://review.opendev.org/#/c/659711/\n\nDepends-On: I746b9dbb090dd484a36e826f06beda1eba3f55f4\nDepends-On: Iaf392898c072ac4b39d878bb27654406c5ab0d8b\nChange-Id: I91d3fef8d1b89f595b529c7c96b3d7249e5b7291\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/d741a379f7f5bc5c654506e9c9b8388cbc43d251"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/d741a379f7f5bc5c654506e9c9b8388cbc43d251"}]},"branch":"refs/heads/stable/rocky"},"f6e7d93d69b7e0359abb2aafb17da67fc60ff0a2":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-05-19 07:01:25.000000000","uploader":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"ref":"refs/changes/14/659714/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/14/659714/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/14/659714/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/14/659714/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/14/659714/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/14/659714/3"}}},"commit":{"parents":[{"commit":"76e62caa931a27253f31c235ee27169d8911f775","subject":"Avoid looping systemd_service role include","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/76e62caa931a27253f31c235ee27169d8911f775"}]}],"author":{"name":"Logan V","email":"logan2211@gmail.com","date":"2018-11-08 03:50:13.000000000","tz":-360},"committer":{"name":"Logan V","email":"logan2211@gmail.com","date":"2019-05-19 07:01:20.000000000","tz":-300},"subject":"Update Calico Felix to v3","message":"Update Calico Felix to v3\n\nMoving to v3 requires etcd3, which uses a new L4 gateway setup and\ndeprecates v2 proxy. The role configuration is updated to use the\netcd3 L4 gateway.\n\nThis also includes several dependent fixes and moves the Calico\ntest to use the integrated repo\u0027s Calico scenario.\n\nCombined backport of:\nhttps://review.opendev.org/#/c/641483/\nhttps://review.opendev.org/#/c/645971/\nhttps://review.opendev.org/#/c/659710/\nhttps://review.opendev.org/#/c/659711/\n\nDepends-On: I746b9dbb090dd484a36e826f06beda1eba3f55f4\nDepends-On: Iaf392898c072ac4b39d878bb27654406c5ab0d8b\nChange-Id: I91d3fef8d1b89f595b529c7c96b3d7249e5b7291\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/f6e7d93d69b7e0359abb2aafb17da67fc60ff0a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/f6e7d93d69b7e0359abb2aafb17da67fc60ff0a2"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
