)]}'
{"id":"openstack%2Fopenstack-ansible-os_neutron~660270","triplet_id":"openstack%2Fopenstack-ansible-os_neutron~master~I069e43812acfe537f7c8bd48e1511ea2f18ceda1","project":"openstack/openstack-ansible-os_neutron","branch":"master","topic":"clarify-suse-support","hashtags":[],"change_id":"I069e43812acfe537f7c8bd48e1511ea2f18ceda1","subject":"Clarify supported SUSE versions","status":"MERGED","created":"2019-05-21 07:51:17.000000000","updated":"2019-05-22 15:06:22.000000000","submitted":"2019-05-22 15:06:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660270-1558537582352-fb151f8d","meta_rev_id":"491cbd66564704e28c361ec9cab313c490b95978","_number":660270,"virtual_id_number":660270,"owner":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"date":"2019-05-22 07:05:24.000000000","_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":2,"date":"2019-05-22 15:06:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2019-05-22 13:32:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-22 13:36:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-22 13:36:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-22 13:32:37.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-05-22 13:36:14.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-05-22 15:06:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7826fdc40f7d4fd38dc30a9fcc75f0a389075fff","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-05-21 07:51:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0b88599ad1da8236071af3bc55de0e181a273fa3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 09:43:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/70/660270/1/check/openstack-ansible-linters/12db5d7/ : SUCCESS in 5m 36s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/70/660270/1/check/openstack-ansible-deploy-aio_metal-debian-stable/567622c/ : SUCCESS in 1h 01m 46s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/70/660270/1/check/openstack-ansible-deploy-aio_metal-centos-7/458c043/ : SUCCESS in 1h 13m 21s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/70/660270/1/check/openstack-ansible-deploy-aio_metal-opensuse-150/4a799b8/ : SUCCESS in 1h 03m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/70/660270/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/a3d183d/ : SUCCESS in 1h 00m 02s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/70/660270/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/f369c0f/ : SUCCESS in 54m 15s\n- openstack-tox-docs http://logs.openstack.org/70/660270/1/check/openstack-tox-docs/fed8788/html/ : SUCCESS in 3m 23s\n- openstack-ansible-ovs-ubuntu-bionic http://logs.openstack.org/70/660270/1/check/openstack-ansible-ovs-ubuntu-bionic/11c954c/ : FAILURE in 16m 59s\n- openstack-ansible-ovn-ubuntu-bionic http://logs.openstack.org/70/660270/1/check/openstack-ansible-ovn-ubuntu-bionic/1928413/ : SUCCESS in 39m 45s\n- openstack-ansible-ovn-centos-7 http://logs.openstack.org/70/660270/1/check/openstack-ansible-ovn-centos-7/084b62a/ : FAILURE in 30m 27s (non-voting)\n- openstack-ansible-calico-ubuntu-bionic http://logs.openstack.org/70/660270/1/check/openstack-ansible-calico-ubuntu-bionic/8ebb7c3/ : FAILURE in 37m 27s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic http://logs.openstack.org/70/660270/1/check/openstack-ansible-opendaylight-ubuntu-bionic/d0cae8e/ : SUCCESS in 50m 37s\n- openstack-ansible-ovs-nsh-ubuntu-bionic http://logs.openstack.org/70/660270/1/check/openstack-ansible-ovs-nsh-ubuntu-bionic/115e931/ : SUCCESS in 55m 37s\n- openstack-ansible-odl-sfc-ubuntu-bionic http://logs.openstack.org/70/660270/1/check/openstack-ansible-odl-sfc-ubuntu-bionic/bea8b7f/ : SUCCESS in 55m 01s\n- openstack-ansible-ovs-sfc-ubuntu-bionic http://logs.openstack.org/70/660270/1/check/openstack-ansible-ovs-sfc-ubuntu-bionic/ea63378/ : SUCCESS in 37m 58s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic http://logs.openstack.org/70/660270/1/check/openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic/d44da0a/ : TIMED_OUT in 1h 33m 33s (non-voting)\n- openstack-ansible-ovs-fwaasv2-ubuntu-bionic http://logs.openstack.org/70/660270/1/check/openstack-ansible-ovs-fwaasv2-ubuntu-bionic/4a9add6/ : SUCCESS in 38m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"418a2526e916e9ac9b4f26825450f5f32efc38a8","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-05-22 07:05:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"412c8b9810277fa9a528087c00035714ca99a1a5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 08:52:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/70/660270/1/check/openstack-ansible-linters/af7a339/ : SUCCESS in 6m 06s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/70/660270/1/check/openstack-ansible-deploy-aio_metal-debian-stable/ab8ebd7/ : SUCCESS in 57m 59s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/70/660270/1/check/openstack-ansible-deploy-aio_metal-centos-7/65d00a9/ : SUCCESS in 1h 12m 33s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/70/660270/1/check/openstack-ansible-deploy-aio_metal-opensuse-150/4878722/ : SUCCESS in 1h 15m 01s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/70/660270/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/fb46e11/ : SUCCESS in 54m 09s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/70/660270/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/c5fdaa0/ : SUCCESS in 54m 28s\n- openstack-tox-docs http://logs.openstack.org/70/660270/1/check/openstack-tox-docs/a35a0b4/html/ : SUCCESS in 2m 35s\n- openstack-ansible-ovs-ubuntu-bionic http://logs.openstack.org/70/660270/1/check/openstack-ansible-ovs-ubuntu-bionic/e92273a/ : SUCCESS in 35m 40s\n- openstack-ansible-ovn-ubuntu-bionic http://logs.openstack.org/70/660270/1/check/openstack-ansible-ovn-ubuntu-bionic/e0ebf58/ : SUCCESS in 36m 33s\n- openstack-ansible-ovn-centos-7 http://logs.openstack.org/70/660270/1/check/openstack-ansible-ovn-centos-7/10d1b6d/ : FAILURE in 30m 31s (non-voting)\n- openstack-ansible-calico-ubuntu-bionic http://logs.openstack.org/70/660270/1/check/openstack-ansible-calico-ubuntu-bionic/7cd2b15/ : FAILURE in 38m 18s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic http://logs.openstack.org/70/660270/1/check/openstack-ansible-opendaylight-ubuntu-bionic/14737f0/ : SUCCESS in 54m 36s\n- openstack-ansible-ovs-nsh-ubuntu-bionic http://logs.openstack.org/70/660270/1/check/openstack-ansible-ovs-nsh-ubuntu-bionic/c6ea45e/ : SUCCESS in 49m 46s\n- openstack-ansible-odl-sfc-ubuntu-bionic http://logs.openstack.org/70/660270/1/check/openstack-ansible-odl-sfc-ubuntu-bionic/c81d86a/ : SUCCESS in 53m 26s\n- openstack-ansible-ovs-sfc-ubuntu-bionic http://logs.openstack.org/70/660270/1/check/openstack-ansible-ovs-sfc-ubuntu-bionic/d7ef8bc/ : SUCCESS in 36m 51s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic http://logs.openstack.org/70/660270/1/check/openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic/d39077a/ : SUCCESS in 1h 08m 36s (non-voting)\n- openstack-ansible-ovs-fwaasv2-ubuntu-bionic http://logs.openstack.org/70/660270/1/check/openstack-ansible-ovs-fwaasv2-ubuntu-bionic/a695aa4/ : SUCCESS in 42m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"66d36971b11f49213bd6deac92e55e313f6c16e1","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-05-22 13:32:37.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4f2ba7ec638b5d391523a163cad508f95bbce2a6","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-05-22 13:36:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cbdd6a4f336cbc4fb5138b3f80c17abc64358fbb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 13:36:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c596e648754da575ee13165963c6857b7524e4e9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 15:06:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"636c51f12d966fcf1f86be31b21dc72fcc6bcbf7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 15:06:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/70/660270/1/gate/openstack-ansible-linters/d916533/ : SUCCESS in 5m 39s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/70/660270/1/gate/openstack-ansible-deploy-aio_metal-debian-stable/50ece88/ : SUCCESS in 1h 03m 23s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/70/660270/1/gate/openstack-ansible-deploy-aio_metal-centos-7/f06393f/ : SUCCESS in 1h 17m 06s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/70/660270/1/gate/openstack-ansible-deploy-aio_metal-opensuse-150/e04b1ab/ : SUCCESS in 1h 03m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/70/660270/1/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/04021cf/ : SUCCESS in 59m 23s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/70/660270/1/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/53813e8/ : SUCCESS in 56m 21s\n- openstack-tox-docs http://logs.openstack.org/70/660270/1/gate/openstack-tox-docs/ccf92cb/html/ : SUCCESS in 3m 06s\n- openstack-ansible-ovs-ubuntu-bionic http://logs.openstack.org/70/660270/1/gate/openstack-ansible-ovs-ubuntu-bionic/89a5024/ : SUCCESS in 39m 38s\n- openstack-ansible-ovn-ubuntu-bionic http://logs.openstack.org/70/660270/1/gate/openstack-ansible-ovn-ubuntu-bionic/8b0979a/ : SUCCESS in 36m 41s\n- openstack-ansible-opendaylight-ubuntu-bionic http://logs.openstack.org/70/660270/1/gate/openstack-ansible-opendaylight-ubuntu-bionic/0f6148b/ : SUCCESS in 1h 07m 28s\n- openstack-ansible-ovs-nsh-ubuntu-bionic http://logs.openstack.org/70/660270/1/gate/openstack-ansible-ovs-nsh-ubuntu-bionic/1e447a9/ : SUCCESS in 1h 02m 04s\n- openstack-ansible-odl-sfc-ubuntu-bionic http://logs.openstack.org/70/660270/1/gate/openstack-ansible-odl-sfc-ubuntu-bionic/4989d0e/ : SUCCESS in 1h 03m 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a37d9a070b202917892dfa6e77a25c9bc175e597","revisions":{"a37d9a070b202917892dfa6e77a25c9bc175e597":{"kind":"REWORK","_number":1,"created":"2019-05-21 07:51:17.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/70/660270/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/70/660270/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/70/660270/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/70/660270/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/70/660270/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/70/660270/1"}}},"commit":{"parents":[{"commit":"9f619e3f059d2bf8db930ee23b8847456775708a","subject":"Remove LBaaS v2 plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/9f619e3f059d2bf8db930ee23b8847456775708a"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-21 07:50:25.000000000","tz":120},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-21 07:50:57.000000000","tz":120},"subject":"Clarify supported SUSE versions","message":"Clarify supported SUSE versions\n\nThis role only support openSUSE Leap 15, not 42.3, not tumbleweed.\nSaying all is confusing and a bad practice. This fixes it, while\nensuring the job is properly defined to test that assertion.\n\nChange-Id: I069e43812acfe537f7c8bd48e1511ea2f18ceda1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/a37d9a070b202917892dfa6e77a25c9bc175e597"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/a37d9a070b202917892dfa6e77a25c9bc175e597"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
